)]}'
{"id":"openstack%2Frequirements~664150","triplet_id":"openstack%2Frequirements~master~Id18255e061b71d4476256323fdc9c8e7858a2565","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Id18255e061b71d4476256323fdc9c8e7858a2565","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-06-09 06:17:24.000000000","updated":"2019-06-10 03:53:51.000000000","submitted":"2019-06-10 03:53:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664150-1560138831003-0eece10d","meta_rev_id":"39d17df93d97dc5e9e6d1a684f8869970964ab5c","_number":664150,"virtual_id_number":664150,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-10 03:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-09 20:47:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-09 20:47:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-09 20:47:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-10 03:53:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fcd07c3593e96ef362192de6235e3f826f9f7250","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-09 06:17:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"474bd5fa669fe72ccc2948d632c522125fbdeeac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 08:13:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/664150/1/check/openstack-tox-docs/e80fcb3/html/ : SUCCESS in 3m 22s\n- openstack-tox-pypy http://logs.openstack.org/50/664150/1/check/openstack-tox-pypy/907e2fe/ : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/50/664150/1/check/openstack-tox-pep8/f4a55b0/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/50/664150/1/check/openstack-tox-py27/a39ecb1/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/50/664150/1/check/openstack-tox-py36/4015587/ : SUCCESS in 3m 12s\n- openstack-tox-validate http://logs.openstack.org/50/664150/1/check/openstack-tox-validate/d73fdda/ : SUCCESS in 3m 19s\n- requirements-tox-babel http://logs.openstack.org/50/664150/1/check/requirements-tox-babel/5e90168/ : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/50/664150/1/check/requirements-tox-py27-check-uc/c28bbdc/ : FAILURE in 6m 51s\n- requirements-tox-py36-check-uc http://logs.openstack.org/50/664150/1/check/requirements-tox-py36-check-uc/76360db/ : FAILURE in 7m 23s\n- cross-cinder-py27 http://logs.openstack.org/50/664150/1/check/cross-cinder-py27/91d7756/ : SUCCESS in 9m 04s\n- cross-cinder-py36 http://logs.openstack.org/50/664150/1/check/cross-cinder-py36/8dca11e/ : SUCCESS in 8m 18s\n- cross-glance-py27 http://logs.openstack.org/50/664150/1/check/cross-glance-py27/00ec795/ : SUCCESS in 5m 33s\n- cross-glance-py36 http://logs.openstack.org/50/664150/1/check/cross-glance-py36/8e2c012/ : FAILURE in 5m 36s\n- cross-horizon-py27 http://logs.openstack.org/50/664150/1/check/cross-horizon-py27/4791658/ : SUCCESS in 8m 03s\n- cross-horizon-py36 http://logs.openstack.org/50/664150/1/check/cross-horizon-py36/06221d6/ : SUCCESS in 8m 14s\n- cross-horizon-npm http://logs.openstack.org/50/664150/1/check/cross-horizon-npm/5d4f1c0/ : SUCCESS in 5m 38s\n- cross-keystone-py27 http://logs.openstack.org/50/664150/1/check/cross-keystone-py27/73d8e0b/ : SUCCESS in 30m 43s\n- cross-keystone-py36 http://logs.openstack.org/50/664150/1/check/cross-keystone-py36/95a057e/ : SUCCESS in 32m 44s\n- cross-neutron-py27 http://logs.openstack.org/50/664150/1/check/cross-neutron-py27/4e9c72e/ : SUCCESS in 22m 04s\n- cross-neutron-py36 http://logs.openstack.org/50/664150/1/check/cross-neutron-py36/575831f/ : SUCCESS in 21m 32s\n- cross-nova-py27 http://logs.openstack.org/50/664150/1/check/cross-nova-py27/972ae1d/ : SUCCESS in 11m 21s\n- cross-nova-py36 http://logs.openstack.org/50/664150/1/check/cross-nova-py36/5017740/ : SUCCESS in 10m 42s\n- cross-nova-functional http://logs.openstack.org/50/664150/1/check/cross-nova-functional/ccb1109/ : SUCCESS in 23m 30s\n- cross-osc-tox-docs http://logs.openstack.org/50/664150/1/check/cross-osc-tox-docs/fcb8b23/html/ : SUCCESS in 5m 33s\n- cross-swift-py27 http://logs.openstack.org/50/664150/1/check/cross-swift-py27/5b3c9b8/ : SUCCESS in 7m 30s\n- cross-swift-py36 http://logs.openstack.org/50/664150/1/check/cross-swift-py36/ee41bee/ : SUCCESS in 8m 16s\n- requirements-integration http://logs.openstack.org/50/664150/1/check/requirements-integration/59b7dd6/ : SUCCESS in 55m 28s\n- neutron-grenade-multinode http://logs.openstack.org/50/664150/1/check/neutron-grenade-multinode/37f02f8/ : SUCCESS in 1h 07m 51s\n- tempest-full http://logs.openstack.org/50/664150/1/check/tempest-full/400fed1/ : SUCCESS in 1h 46m 13s\n- neutron-grenade http://logs.openstack.org/50/664150/1/check/neutron-grenade/2c4ffdb/ : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":1},{"id":"678738a5c27919b08f15d2cd07af1351d96684cd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-09 18:47:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef3326f76cb34b6383209bc4387c62969777b0d6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-09 18:49:15.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"543cb9d2c3f01826ef77c589f8a0e6383a37cb14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 20:38:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/664150/2/check/openstack-tox-docs/e0b2821/html/ : SUCCESS in 3m 11s\n- openstack-tox-pypy http://logs.openstack.org/50/664150/2/check/openstack-tox-pypy/6e8bfbe/ : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/50/664150/2/check/openstack-tox-pep8/5fa5439/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/50/664150/2/check/openstack-tox-py27/c63adf8/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/50/664150/2/check/openstack-tox-py36/86defb9/ : SUCCESS in 3m 35s\n- openstack-tox-validate http://logs.openstack.org/50/664150/2/check/openstack-tox-validate/1ac3f3b/ : SUCCESS in 3m 05s\n- requirements-tox-babel http://logs.openstack.org/50/664150/2/check/requirements-tox-babel/9c35441/ : SUCCESS in 3m 41s\n- requirements-tox-py27-check-uc http://logs.openstack.org/50/664150/2/check/requirements-tox-py27-check-uc/878aa7b/ : SUCCESS in 6m 25s\n- requirements-tox-py36-check-uc http://logs.openstack.org/50/664150/2/check/requirements-tox-py36-check-uc/c52d18f/ : SUCCESS in 7m 45s\n- cross-cinder-py27 http://logs.openstack.org/50/664150/2/check/cross-cinder-py27/2eb6028/ : SUCCESS in 7m 47s\n- cross-cinder-py36 http://logs.openstack.org/50/664150/2/check/cross-cinder-py36/262eea9/ : SUCCESS in 8m 39s\n- cross-glance-py27 http://logs.openstack.org/50/664150/2/check/cross-glance-py27/dff3850/ : SUCCESS in 4m 41s\n- cross-glance-py36 http://logs.openstack.org/50/664150/2/check/cross-glance-py36/2dbf33b/ : SUCCESS in 4m 28s\n- cross-horizon-py27 http://logs.openstack.org/50/664150/2/check/cross-horizon-py27/0a20d9f/ : SUCCESS in 9m 55s\n- cross-horizon-py36 http://logs.openstack.org/50/664150/2/check/cross-horizon-py36/f76b183/ : SUCCESS in 6m 52s\n- cross-horizon-npm http://logs.openstack.org/50/664150/2/check/cross-horizon-npm/f45f253/ : SUCCESS in 6m 32s\n- cross-keystone-py27 http://logs.openstack.org/50/664150/2/check/cross-keystone-py27/8f79750/ : SUCCESS in 18m 30s\n- cross-keystone-py36 http://logs.openstack.org/50/664150/2/check/cross-keystone-py36/7b28b0c/ : SUCCESS in 19m 55s\n- cross-neutron-py27 http://logs.openstack.org/50/664150/2/check/cross-neutron-py27/0183fa6/ : SUCCESS in 20m 13s\n- cross-neutron-py36 http://logs.openstack.org/50/664150/2/check/cross-neutron-py36/c20acd1/ : SUCCESS in 18m 17s\n- cross-nova-py27 http://logs.openstack.org/50/664150/2/check/cross-nova-py27/17cdce3/ : SUCCESS in 13m 38s\n- cross-nova-py36 http://logs.openstack.org/50/664150/2/check/cross-nova-py36/d283b2a/ : SUCCESS in 10m 59s\n- cross-nova-functional http://logs.openstack.org/50/664150/2/check/cross-nova-functional/77658c9/ : SUCCESS in 16m 47s\n- cross-osc-tox-docs http://logs.openstack.org/50/664150/2/check/cross-osc-tox-docs/38b6d91/html/ : SUCCESS in 5m 00s\n- cross-swift-py27 http://logs.openstack.org/50/664150/2/check/cross-swift-py27/a8e70bb/ : SUCCESS in 7m 21s\n- cross-swift-py36 http://logs.openstack.org/50/664150/2/check/cross-swift-py36/81faef3/ : SUCCESS in 7m 12s\n- requirements-integration http://logs.openstack.org/50/664150/2/check/requirements-integration/a646aff/ : SUCCESS in 48m 59s\n- neutron-grenade-multinode http://logs.openstack.org/50/664150/2/check/neutron-grenade-multinode/5ade496/ : SUCCESS in 1h 03m 54s\n- tempest-full http://logs.openstack.org/50/664150/2/check/tempest-full/ffc7445/ : FAILURE in 1h 24m 52s\n- neutron-grenade http://logs.openstack.org/50/664150/2/check/neutron-grenade/c547d29/ : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":2},{"id":"0b1386f3dc5c6456111dea773b88280e772dc896","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-09 20:47:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"491496c9bf3f53dc8584bfef17384c09e87e8bdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:35:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/664150/2/check/openstack-tox-docs/65f083f/html/ : SUCCESS in 3m 48s\n- openstack-tox-pypy http://logs.openstack.org/50/664150/2/check/openstack-tox-pypy/4e479a8/ : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/50/664150/2/check/openstack-tox-pep8/5034924/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/50/664150/2/check/openstack-tox-py27/f50b31c/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/50/664150/2/check/openstack-tox-py36/bfa7ca5/ : SUCCESS in 3m 55s\n- openstack-tox-validate http://logs.openstack.org/50/664150/2/check/openstack-tox-validate/03d5335/ : SUCCESS in 3m 44s\n- requirements-tox-babel http://logs.openstack.org/50/664150/2/check/requirements-tox-babel/725820c/ : SUCCESS in 2m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/50/664150/2/check/requirements-tox-py27-check-uc/ac877bf/ : SUCCESS in 8m 04s\n- requirements-tox-py36-check-uc http://logs.openstack.org/50/664150/2/check/requirements-tox-py36-check-uc/ac7f02a/ : SUCCESS in 7m 55s\n- cross-cinder-py27 http://logs.openstack.org/50/664150/2/check/cross-cinder-py27/138bc13/ : SUCCESS in 11m 07s\n- cross-cinder-py36 http://logs.openstack.org/50/664150/2/check/cross-cinder-py36/9e06d6d/ : SUCCESS in 8m 14s\n- cross-glance-py27 http://logs.openstack.org/50/664150/2/check/cross-glance-py27/e828de3/ : SUCCESS in 4m 16s\n- cross-glance-py36 http://logs.openstack.org/50/664150/2/check/cross-glance-py36/fc4d318/ : SUCCESS in 5m 03s\n- cross-horizon-py27 http://logs.openstack.org/50/664150/2/check/cross-horizon-py27/7c46b2c/ : SUCCESS in 11m 12s\n- cross-horizon-py36 http://logs.openstack.org/50/664150/2/check/cross-horizon-py36/bba03b9/ : SUCCESS in 7m 36s\n- cross-horizon-npm http://logs.openstack.org/50/664150/2/check/cross-horizon-npm/ababecc/ : SUCCESS in 5m 55s\n- cross-keystone-py27 http://logs.openstack.org/50/664150/2/check/cross-keystone-py27/02a43ee/ : SUCCESS in 21m 36s\n- cross-keystone-py36 http://logs.openstack.org/50/664150/2/check/cross-keystone-py36/402d8cc/ : SUCCESS in 19m 52s\n- cross-neutron-py27 http://logs.openstack.org/50/664150/2/check/cross-neutron-py27/fd87a5b/ : SUCCESS in 23m 34s\n- cross-neutron-py36 http://logs.openstack.org/50/664150/2/check/cross-neutron-py36/23b54fc/ : SUCCESS in 23m 15s\n- cross-nova-py27 http://logs.openstack.org/50/664150/2/check/cross-nova-py27/6237967/ : SUCCESS in 22m 38s\n- cross-nova-py36 http://logs.openstack.org/50/664150/2/check/cross-nova-py36/1017dd9/ : SUCCESS in 11m 21s\n- cross-nova-functional http://logs.openstack.org/50/664150/2/check/cross-nova-functional/79901a6/ : SUCCESS in 17m 37s\n- cross-osc-tox-docs http://logs.openstack.org/50/664150/2/check/cross-osc-tox-docs/f46d2e0/html/ : SUCCESS in 4m 56s\n- cross-swift-py27 http://logs.openstack.org/50/664150/2/check/cross-swift-py27/bedb764/ : SUCCESS in 8m 54s\n- cross-swift-py36 http://logs.openstack.org/50/664150/2/check/cross-swift-py36/4291186/ : SUCCESS in 15m 48s\n- requirements-integration http://logs.openstack.org/50/664150/2/check/requirements-integration/16fb4ce/ : SUCCESS in 42m 02s\n- neutron-grenade-multinode http://logs.openstack.org/50/664150/2/check/neutron-grenade-multinode/ad2f06d/ : SUCCESS in 1h 06m 30s\n- tempest-full http://logs.openstack.org/50/664150/2/check/tempest-full/5a59bb5/ : SUCCESS in 1h 42m 47s\n- neutron-grenade http://logs.openstack.org/50/664150/2/check/neutron-grenade/cec9d78/ : SUCCESS in 54m 07s","accounts_in_message":[],"_revision_number":2},{"id":"4111ef0b074050524ce7bb52ea39be6f93f408c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:35:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"92dd10a61e2d1973293ddd655876a2fbeeb1fab7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 00:16:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/50/664150/2/gate/openstack-tox-docs/81dc809/html/ : SUCCESS in 3m 06s\n- openstack-tox-pep8 http://logs.openstack.org/50/664150/2/gate/openstack-tox-pep8/e4111b7/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/50/664150/2/gate/openstack-tox-py27/259266d/ : SUCCESS in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/50/664150/2/gate/openstack-tox-py36/3d4775c/ : SUCCESS in 4m 29s\n- openstack-tox-validate http://logs.openstack.org/50/664150/2/gate/openstack-tox-validate/24a86d4/ : SUCCESS in 3m 13s\n- requirements-tox-babel http://logs.openstack.org/50/664150/2/gate/requirements-tox-babel/aaf2aca/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/50/664150/2/gate/requirements-tox-py27-check-uc/763da7d/ : SUCCESS in 7m 56s\n- requirements-tox-py36-check-uc http://logs.openstack.org/50/664150/2/gate/requirements-tox-py36-check-uc/3095b48/ : SUCCESS in 7m 39s\n- cross-cinder-py27 http://logs.openstack.org/50/664150/2/gate/cross-cinder-py27/3469857/ : SUCCESS in 9m 54s\n- cross-cinder-py36 http://logs.openstack.org/50/664150/2/gate/cross-cinder-py36/ab3fb67/ : SUCCESS in 7m 36s\n- cross-glance-py27 http://logs.openstack.org/50/664150/2/gate/cross-glance-py27/1e70f5d/ : SUCCESS in 5m 33s\n- cross-glance-py36 http://logs.openstack.org/50/664150/2/gate/cross-glance-py36/fb6e497/ : SUCCESS in 4m 23s\n- cross-horizon-py27 http://logs.openstack.org/50/664150/2/gate/cross-horizon-py27/81ef0d2/ : SUCCESS in 8m 46s\n- cross-horizon-py36 http://logs.openstack.org/50/664150/2/gate/cross-horizon-py36/306e08c/ : SUCCESS in 9m 12s\n- cross-horizon-npm http://logs.openstack.org/50/664150/2/gate/cross-horizon-npm/bc8a506/ : SUCCESS in 6m 41s\n- cross-keystone-py27 http://logs.openstack.org/50/664150/2/gate/cross-keystone-py27/5c1aa8a/ : SUCCESS in 22m 34s\n- cross-keystone-py36 http://logs.openstack.org/50/664150/2/gate/cross-keystone-py36/a1832bd/ : SUCCESS in 23m 24s\n- cross-neutron-py27 http://logs.openstack.org/50/664150/2/gate/cross-neutron-py27/b96b4f4/ : SUCCESS in 25m 19s\n- cross-neutron-py36 http://logs.openstack.org/50/664150/2/gate/cross-neutron-py36/c66c487/ : SUCCESS in 20m 53s\n- cross-nova-py27 http://logs.openstack.org/50/664150/2/gate/cross-nova-py27/255dcbb/ : SUCCESS in 11m 56s\n- cross-nova-py36 http://logs.openstack.org/50/664150/2/gate/cross-nova-py36/0bf1c87/ : SUCCESS in 12m 24s\n- cross-nova-functional http://logs.openstack.org/50/664150/2/gate/cross-nova-functional/7035a4d/ : SUCCESS in 17m 59s\n- cross-osc-tox-docs http://logs.openstack.org/50/664150/2/gate/cross-osc-tox-docs/c5ec574/html/ : SUCCESS in 5m 06s\n- cross-swift-py27 http://logs.openstack.org/50/664150/2/gate/cross-swift-py27/899096f/ : SUCCESS in 8m 19s\n- cross-swift-py36 http://logs.openstack.org/50/664150/2/gate/cross-swift-py36/391e329/ : SUCCESS in 8m 04s\n- requirements-integration http://logs.openstack.org/50/664150/2/gate/requirements-integration/c8caee9/ : SUCCESS in 55m 58s\n- neutron-grenade-multinode http://logs.openstack.org/50/664150/2/gate/neutron-grenade-multinode/f6ff89f/ : SUCCESS in 1h 10m 51s\n- tempest-full http://logs.openstack.org/50/664150/2/gate/tempest-full/576b44e/ : FAILURE in 1h 38m 58s\n- neutron-grenade http://logs.openstack.org/50/664150/2/gate/neutron-grenade/4cf7fac/ : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":2},{"id":"857ec63ba7711d5dc973c2cd1897525bb64c64ab","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-10 00:27:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d81b232f446cb7490851f4a1959ed8fbd5befe2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 01:56:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/664150/2/check/openstack-tox-docs/c6008aa/html/ : SUCCESS in 2m 52s\n- openstack-tox-pypy http://logs.openstack.org/50/664150/2/check/openstack-tox-pypy/7cb741c/ : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/50/664150/2/check/openstack-tox-pep8/e58581a/ : SUCCESS in 2m 52s\n- openstack-tox-py27 http://logs.openstack.org/50/664150/2/check/openstack-tox-py27/dbf62ba/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/50/664150/2/check/openstack-tox-py36/4691e2b/ : SUCCESS in 3m 10s\n- openstack-tox-validate http://logs.openstack.org/50/664150/2/check/openstack-tox-validate/2a19779/ : SUCCESS in 2m 59s\n- requirements-tox-babel http://logs.openstack.org/50/664150/2/check/requirements-tox-babel/ecaa68e/ : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/50/664150/2/check/requirements-tox-py27-check-uc/6f76ab6/ : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc http://logs.openstack.org/50/664150/2/check/requirements-tox-py36-check-uc/4a6a2cb/ : SUCCESS in 5m 44s\n- cross-cinder-py27 http://logs.openstack.org/50/664150/2/check/cross-cinder-py27/f6c4228/ : SUCCESS in 11m 29s\n- cross-cinder-py36 http://logs.openstack.org/50/664150/2/check/cross-cinder-py36/1d8e299/ : SUCCESS in 7m 58s\n- cross-glance-py27 http://logs.openstack.org/50/664150/2/check/cross-glance-py27/14b2e2e/ : SUCCESS in 4m 13s\n- cross-glance-py36 http://logs.openstack.org/50/664150/2/check/cross-glance-py36/04febb2/ : SUCCESS in 4m 22s\n- cross-horizon-py27 http://logs.openstack.org/50/664150/2/check/cross-horizon-py27/d1928e5/ : SUCCESS in 10m 26s\n- cross-horizon-py36 http://logs.openstack.org/50/664150/2/check/cross-horizon-py36/8dbfe9e/ : SUCCESS in 9m 08s\n- cross-horizon-npm http://logs.openstack.org/50/664150/2/check/cross-horizon-npm/9f8bacc/ : SUCCESS in 6m 28s\n- cross-keystone-py27 http://logs.openstack.org/50/664150/2/check/cross-keystone-py27/4e2f44c/ : SUCCESS in 20m 59s\n- cross-keystone-py36 http://logs.openstack.org/50/664150/2/check/cross-keystone-py36/b0444f7/ : SUCCESS in 22m 06s\n- cross-neutron-py27 http://logs.openstack.org/50/664150/2/check/cross-neutron-py27/a342fbc/ : SUCCESS in 23m 28s\n- cross-neutron-py36 http://logs.openstack.org/50/664150/2/check/cross-neutron-py36/d6b2460/ : SUCCESS in 27m 30s\n- cross-nova-py27 http://logs.openstack.org/50/664150/2/check/cross-nova-py27/64164e6/ : SUCCESS in 11m 50s\n- cross-nova-py36 http://logs.openstack.org/50/664150/2/check/cross-nova-py36/16be4bc/ : SUCCESS in 16m 19s\n- cross-nova-functional http://logs.openstack.org/50/664150/2/check/cross-nova-functional/55cf4c8/ : SUCCESS in 17m 58s\n- cross-osc-tox-docs http://logs.openstack.org/50/664150/2/check/cross-osc-tox-docs/81a20e2/html/ : SUCCESS in 5m 10s\n- cross-swift-py27 http://logs.openstack.org/50/664150/2/check/cross-swift-py27/94045ca/ : SUCCESS in 9m 03s\n- cross-swift-py36 http://logs.openstack.org/50/664150/2/check/cross-swift-py36/c2f6f14/ : SUCCESS in 7m 52s\n- requirements-integration http://logs.openstack.org/50/664150/2/check/requirements-integration/23915f4/ : SUCCESS in 55m 24s\n- neutron-grenade-multinode http://logs.openstack.org/50/664150/2/check/neutron-grenade-multinode/8ce65a4/ : SUCCESS in 1h 04m 53s\n- tempest-full http://logs.openstack.org/50/664150/2/check/tempest-full/27a6cf1/ : SUCCESS in 1h 28m 03s\n- neutron-grenade http://logs.openstack.org/50/664150/2/check/neutron-grenade/2f448c5/ : SUCCESS in 50m 24s","accounts_in_message":[],"_revision_number":2},{"id":"d5605a59034b769fbf1f02443f170da162be04d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 01:56:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d63d856ca524583fc928fab51e070915f625d02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 03:53:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/664150/2/gate/openstack-tox-docs/7410d97/html/ : SUCCESS in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/50/664150/2/gate/openstack-tox-pep8/3115af2/ : SUCCESS in 3m 53s\n- openstack-tox-py27 http://logs.openstack.org/50/664150/2/gate/openstack-tox-py27/ee6dce6/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/50/664150/2/gate/openstack-tox-py36/68591b8/ : SUCCESS in 3m 32s\n- openstack-tox-validate http://logs.openstack.org/50/664150/2/gate/openstack-tox-validate/b75ff4c/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/50/664150/2/gate/requirements-tox-babel/68fdd94/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/50/664150/2/gate/requirements-tox-py27-check-uc/75e16f3/ : SUCCESS in 6m 13s\n- requirements-tox-py36-check-uc http://logs.openstack.org/50/664150/2/gate/requirements-tox-py36-check-uc/b7d6f60/ : SUCCESS in 6m 32s\n- cross-cinder-py27 http://logs.openstack.org/50/664150/2/gate/cross-cinder-py27/ec5a048/ : SUCCESS in 9m 50s\n- cross-cinder-py36 http://logs.openstack.org/50/664150/2/gate/cross-cinder-py36/71cffcd/ : SUCCESS in 7m 28s\n- cross-glance-py27 http://logs.openstack.org/50/664150/2/gate/cross-glance-py27/21e388f/ : SUCCESS in 3m 50s\n- cross-glance-py36 http://logs.openstack.org/50/664150/2/gate/cross-glance-py36/9c1ebcd/ : SUCCESS in 4m 05s\n- cross-horizon-py27 http://logs.openstack.org/50/664150/2/gate/cross-horizon-py27/08c7b99/ : SUCCESS in 9m 08s\n- cross-horizon-py36 http://logs.openstack.org/50/664150/2/gate/cross-horizon-py36/8395cea/ : SUCCESS in 7m 29s\n- cross-horizon-npm http://logs.openstack.org/50/664150/2/gate/cross-horizon-npm/49cfa71/ : SUCCESS in 6m 28s\n- cross-keystone-py27 http://logs.openstack.org/50/664150/2/gate/cross-keystone-py27/7e8f564/ : SUCCESS in 20m 03s\n- cross-keystone-py36 http://logs.openstack.org/50/664150/2/gate/cross-keystone-py36/3dc97f9/ : SUCCESS in 21m 07s\n- cross-neutron-py27 http://logs.openstack.org/50/664150/2/gate/cross-neutron-py27/1a61ca6/ : SUCCESS in 24m 17s\n- cross-neutron-py36 http://logs.openstack.org/50/664150/2/gate/cross-neutron-py36/32c572c/ : SUCCESS in 19m 46s\n- cross-nova-py27 http://logs.openstack.org/50/664150/2/gate/cross-nova-py27/1bf66c1/ : SUCCESS in 13m 24s\n- cross-nova-py36 http://logs.openstack.org/50/664150/2/gate/cross-nova-py36/02dedbf/ : SUCCESS in 15m 59s\n- cross-nova-functional http://logs.openstack.org/50/664150/2/gate/cross-nova-functional/510e54b/ : SUCCESS in 17m 59s\n- cross-osc-tox-docs http://logs.openstack.org/50/664150/2/gate/cross-osc-tox-docs/34270f2/html/ : SUCCESS in 5m 12s\n- cross-swift-py27 http://logs.openstack.org/50/664150/2/gate/cross-swift-py27/554b0cb/ : SUCCESS in 8m 16s\n- cross-swift-py36 http://logs.openstack.org/50/664150/2/gate/cross-swift-py36/81bedab/ : SUCCESS in 7m 50s\n- requirements-integration http://logs.openstack.org/50/664150/2/gate/requirements-integration/f89bc7b/ : SUCCESS in 1h 00m 21s\n- neutron-grenade-multinode http://logs.openstack.org/50/664150/2/gate/neutron-grenade-multinode/ff97e0c/ : SUCCESS in 1h 08m 10s\n- tempest-full http://logs.openstack.org/50/664150/2/gate/tempest-full/83cec00/ : SUCCESS in 1h 51m 17s\n- neutron-grenade http://logs.openstack.org/50/664150/2/gate/neutron-grenade/fe607db/ : SUCCESS in 58m 25s","accounts_in_message":[],"_revision_number":2},{"id":"39d17df93d97dc5e9e6d1a684f8869970964ab5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 03:53:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98a5e8f0b47309e23edd429cb85704413876ce6d","revisions":{"95313494d3aa2f556be15cc311724d8cba8b0039":{"kind":"REWORK","_number":1,"created":"2019-06-09 06:17:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/664150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/664150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/664150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/664150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/664150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/664150/1"}}},"commit":{"parents":[{"commit":"4f3252cbd7c63fd1c60e9bd09748e39dc2d9f8fa","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f3252cbd7c63fd1c60e9bd09748e39dc2d9f8fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-09 06:17:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-09 06:17:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id18255e061b71d4476256323fdc9c8e7858a2565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95313494d3aa2f556be15cc311724d8cba8b0039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95313494d3aa2f556be15cc311724d8cba8b0039"}]},"branch":"refs/heads/master"},"98a5e8f0b47309e23edd429cb85704413876ce6d":{"kind":"REWORK","_number":2,"created":"2019-06-09 18:47:13.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/50/664150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/664150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/664150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/664150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/664150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/664150/2"}}},"commit":{"parents":[{"commit":"4f3252cbd7c63fd1c60e9bd09748e39dc2d9f8fa","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f3252cbd7c63fd1c60e9bd09748e39dc2d9f8fa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-09 06:17:13.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-06-09 18:45:16.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id18255e061b71d4476256323fdc9c8e7858a2565\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98a5e8f0b47309e23edd429cb85704413876ce6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98a5e8f0b47309e23edd429cb85704413876ce6d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
