)]}'
{"id":"openstack%2Frequirements~664381","triplet_id":"openstack%2Frequirements~master~Ia1698b32acc9b91ee108fd1fb150dd1f1f5e12bf","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia1698b32acc9b91ee108fd1fb150dd1f1f5e12bf","subject":"update constraint for openstacksdk to new release 0.31.0","status":"MERGED","created":"2019-06-10 15:47:18.000000000","updated":"2019-06-11 00:42:59.000000000","submitted":"2019-06-11 00:42:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664381-1560213779693-4a3a791a","meta_rev_id":"2c45c83d63676e88d7bd3d5eb36cb700d97b0ec5","_number":664381,"virtual_id_number":664381,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-06-11 00:42:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-10 22:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-10 16:11:13.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-06-10 16:08:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-10 22:39:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 16:08:48.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-06-10 16:11:13.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-06-10 22:39:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-11 00:42:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21cd3969f6a03fb8c070ca70cf391ed4755fa887","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06fdf33ffd5e31767a6a2cd967e280b206d47853","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-06-10 16:08:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9526c390f6c28bdf4c4987f5dc3a6b5fbe00ac1a","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-06-10 16:11:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8f0419a3f6199c5ff2d570e515bdf68aa5c325a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 22:19:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/664381/1/check/openstack-tox-docs/a3219db/html/ : SUCCESS in 3m 22s\n- openstack-tox-pypy http://logs.openstack.org/81/664381/1/check/openstack-tox-pypy/a7698d9/ : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/81/664381/1/check/openstack-tox-pep8/635ce2c/ : SUCCESS in 4m 12s\n- openstack-tox-py27 http://logs.openstack.org/81/664381/1/check/openstack-tox-py27/5514061/ : SUCCESS in 5m 13s\n- openstack-tox-py36 http://logs.openstack.org/81/664381/1/check/openstack-tox-py36/d065e3d/ : SUCCESS in 3m 19s\n- openstack-tox-validate http://logs.openstack.org/81/664381/1/check/openstack-tox-validate/3889a90/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/81/664381/1/check/requirements-tox-babel/56d35e5/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/81/664381/1/check/requirements-tox-py27-check-uc/6ae7f87/ : SUCCESS in 11m 06s\n- requirements-tox-py36-check-uc http://logs.openstack.org/81/664381/1/check/requirements-tox-py36-check-uc/995a96d/ : SUCCESS in 6m 44s\n- cross-cinder-py27 http://logs.openstack.org/81/664381/1/check/cross-cinder-py27/493fb3e/ : SUCCESS in 9m 05s\n- cross-cinder-py36 http://logs.openstack.org/81/664381/1/check/cross-cinder-py36/e58e4b9/ : SUCCESS in 7m 32s\n- cross-glance-py27 http://logs.openstack.org/81/664381/1/check/cross-glance-py27/d034dc7/ : SUCCESS in 4m 03s\n- cross-glance-py36 http://logs.openstack.org/81/664381/1/check/cross-glance-py36/bbd836c/ : SUCCESS in 4m 33s\n- cross-horizon-py27 http://logs.openstack.org/81/664381/1/check/cross-horizon-py27/a0b0d99/ : SUCCESS in 9m 18s\n- cross-horizon-py36 http://logs.openstack.org/81/664381/1/check/cross-horizon-py36/1c8d055/ : SUCCESS in 9m 58s\n- cross-horizon-npm http://logs.openstack.org/81/664381/1/check/cross-horizon-npm/e060d65/ : SUCCESS in 5m 47s\n- cross-keystone-py27 http://logs.openstack.org/81/664381/1/check/cross-keystone-py27/006666f/ : SUCCESS in 28m 33s\n- cross-keystone-py36 http://logs.openstack.org/81/664381/1/check/cross-keystone-py36/6e9cb53/ : SUCCESS in 26m 21s\n- cross-neutron-py27 http://logs.openstack.org/81/664381/1/check/cross-neutron-py27/03640a9/ : SUCCESS in 23m 17s\n- cross-neutron-py36 http://logs.openstack.org/81/664381/1/check/cross-neutron-py36/ec99707/ : SUCCESS in 22m 59s\n- cross-nova-py27 http://logs.openstack.org/81/664381/1/check/cross-nova-py27/09064fa/ : SUCCESS in 13m 36s\n- cross-nova-py36 http://logs.openstack.org/81/664381/1/check/cross-nova-py36/5cd1c5f/ : SUCCESS in 14m 40s\n- cross-nova-functional http://logs.openstack.org/81/664381/1/check/cross-nova-functional/dcec3cc/ : SUCCESS in 19m 53s\n- cross-osc-tox-docs http://logs.openstack.org/81/664381/1/check/cross-osc-tox-docs/c828047/html/ : SUCCESS in 6m 09s\n- cross-swift-py27 http://logs.openstack.org/81/664381/1/check/cross-swift-py27/630b5c8/ : SUCCESS in 8m 17s\n- cross-swift-py36 http://logs.openstack.org/81/664381/1/check/cross-swift-py36/c8c908e/ : SUCCESS in 8m 45s\n- requirements-integration http://logs.openstack.org/81/664381/1/check/requirements-integration/4e172cf/ : SUCCESS in 1h 04m 37s\n- neutron-grenade-multinode http://logs.openstack.org/81/664381/1/check/neutron-grenade-multinode/a1b68dd/ : SUCCESS in 1h 07m 06s\n- tempest-full http://logs.openstack.org/81/664381/1/check/tempest-full/1da8ffb/ : SUCCESS in 2h 02m 06s\n- neutron-grenade http://logs.openstack.org/81/664381/1/check/neutron-grenade/fa01d38/ : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":1},{"id":"288a90c578e580dd6021d6f9f0887a6e25ea3a28","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-10 22:39:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83803da4fd5a5487f48160715bccfe7e7154866b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 22:39:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0c8c6ea75c4af7ffe69166f43bdd98769306728","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 00:42:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/81/664381/1/gate/openstack-tox-docs/2b162d0/html/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/81/664381/1/gate/openstack-tox-pep8/c097d67/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/81/664381/1/gate/openstack-tox-py27/8b22603/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/81/664381/1/gate/openstack-tox-py36/b1c3cee/ : SUCCESS in 5m 22s\n- openstack-tox-validate http://logs.openstack.org/81/664381/1/gate/openstack-tox-validate/86f96f7/ : SUCCESS in 5m 00s\n- requirements-tox-babel http://logs.openstack.org/81/664381/1/gate/requirements-tox-babel/b874858/ : SUCCESS in 4m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/81/664381/1/gate/requirements-tox-py27-check-uc/5d8a592/ : SUCCESS in 9m 45s\n- requirements-tox-py36-check-uc http://logs.openstack.org/81/664381/1/gate/requirements-tox-py36-check-uc/0f202ad/ : SUCCESS in 7m 19s\n- cross-cinder-py27 http://logs.openstack.org/81/664381/1/gate/cross-cinder-py27/1016a89/ : SUCCESS in 8m 45s\n- cross-cinder-py36 http://logs.openstack.org/81/664381/1/gate/cross-cinder-py36/56fe062/ : SUCCESS in 8m 52s\n- cross-glance-py27 http://logs.openstack.org/81/664381/1/gate/cross-glance-py27/310b305/ : SUCCESS in 5m 09s\n- cross-glance-py36 http://logs.openstack.org/81/664381/1/gate/cross-glance-py36/76c2dd6/ : SUCCESS in 4m 12s\n- cross-horizon-py27 http://logs.openstack.org/81/664381/1/gate/cross-horizon-py27/7afc063/ : SUCCESS in 11m 25s\n- cross-horizon-py36 http://logs.openstack.org/81/664381/1/gate/cross-horizon-py36/56b9642/ : SUCCESS in 10m 07s\n- cross-horizon-npm http://logs.openstack.org/81/664381/1/gate/cross-horizon-npm/8bdd23b/ : SUCCESS in 6m 44s\n- cross-keystone-py27 http://logs.openstack.org/81/664381/1/gate/cross-keystone-py27/7bb919e/ : SUCCESS in 23m 45s\n- cross-keystone-py36 http://logs.openstack.org/81/664381/1/gate/cross-keystone-py36/d336fe4/ : SUCCESS in 24m 20s\n- cross-neutron-py27 http://logs.openstack.org/81/664381/1/gate/cross-neutron-py27/c50798c/ : SUCCESS in 21m 26s\n- cross-neutron-py36 http://logs.openstack.org/81/664381/1/gate/cross-neutron-py36/5ad39e7/ : SUCCESS in 24m 35s\n- cross-nova-py27 http://logs.openstack.org/81/664381/1/gate/cross-nova-py27/e4fbe9c/ : SUCCESS in 13m 33s\n- cross-nova-py36 http://logs.openstack.org/81/664381/1/gate/cross-nova-py36/aefef51/ : SUCCESS in 12m 52s\n- cross-nova-functional http://logs.openstack.org/81/664381/1/gate/cross-nova-functional/7ff2cb3/ : SUCCESS in 18m 13s\n- cross-osc-tox-docs http://logs.openstack.org/81/664381/1/gate/cross-osc-tox-docs/dbe7557/html/ : SUCCESS in 5m 51s\n- cross-swift-py27 http://logs.openstack.org/81/664381/1/gate/cross-swift-py27/56d241e/ : SUCCESS in 8m 43s\n- cross-swift-py36 http://logs.openstack.org/81/664381/1/gate/cross-swift-py36/8e44b92/ : SUCCESS in 9m 21s\n- requirements-integration http://logs.openstack.org/81/664381/1/gate/requirements-integration/35037fc/ : SUCCESS in 1h 10m 21s\n- neutron-grenade-multinode http://logs.openstack.org/81/664381/1/gate/neutron-grenade-multinode/f53fef8/ : SUCCESS in 1h 20m 17s\n- tempest-full http://logs.openstack.org/81/664381/1/gate/tempest-full/77d0746/ : SUCCESS in 1h 54m 57s\n- neutron-grenade http://logs.openstack.org/81/664381/1/gate/neutron-grenade/6d87d48/ : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":1},{"id":"31ebdf0ddb0c5809a7f994c55cb3263d7afa9f5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 00:42:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8460419afec69641402f9f02adc84c1b69dd6420","revisions":{"8460419afec69641402f9f02adc84c1b69dd6420":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:47:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/664381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/664381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/664381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/664381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/664381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/664381/1"}}},"commit":{"parents":[{"commit":"98a5e8f0b47309e23edd429cb85704413876ce6d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98a5e8f0b47309e23edd429cb85704413876ce6d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:47:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:47:17.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.31.0","message":"update constraint for openstacksdk to new release 0.31.0\n\nChange-Id: Ia1698b32acc9b91ee108fd1fb150dd1f1f5e12bf\nmeta:version: 0.31.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Change-Id: I0c5bb1a0b193b15e8551544c82b37e83b4ab474c\nmeta:release:Code-Review+1: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8460419afec69641402f9f02adc84c1b69dd6420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8460419afec69641402f9f02adc84c1b69dd6420"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
