)]}'
{"id":"openstack%2Frequirements~664384","triplet_id":"openstack%2Frequirements~master~I7a9fa4871cedc7d26324ec7b48ef98ca31945fbb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7a9fa4871cedc7d26324ec7b48ef98ca31945fbb","subject":"update constraint for tooz to new release 1.65.0","status":"MERGED","created":"2019-06-10 15:53:13.000000000","updated":"2019-06-14 10:06:43.000000000","submitted":"2019-06-14 10:06:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664384-1560506803828-b1f8f8f4","meta_rev_id":"edfda1c45f6c450b13931e000cea28b0e1879f77","_number":664384,"virtual_id_number":664384,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-06-14 10:06:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-12 15:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-13 11:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-06-13 11:06:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 15:34:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-13 11:06:56.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-06-14 10:06:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6ee810e2c98aa45df7c3e7cebfa2dc6c175e662","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:53:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b711572a0ddf7f543c7203dae3e99d590aa82cdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 23:20:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/1/check/openstack-tox-docs/ac39d28/html/ : SUCCESS in 3m 16s\n- openstack-tox-pypy http://logs.openstack.org/84/664384/1/check/openstack-tox-pypy/522895c/ : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/1/check/openstack-tox-pep8/521a672/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/1/check/openstack-tox-py27/0eead3d/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/1/check/openstack-tox-py36/5e22338/ : SUCCESS in 3m 56s\n- openstack-tox-validate http://logs.openstack.org/84/664384/1/check/openstack-tox-validate/5fcbb55/ : SUCCESS in 5m 09s\n- requirements-tox-babel http://logs.openstack.org/84/664384/1/check/requirements-tox-babel/2bc50e2/ : SUCCESS in 4m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/1/check/requirements-tox-py27-check-uc/7ce8b82/ : FAILURE in 6m 27s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/1/check/requirements-tox-py36-check-uc/8efda0d/ : FAILURE in 7m 46s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/1/check/cross-cinder-py27/d2fc2e8/ : SUCCESS in 9m 57s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/1/check/cross-cinder-py36/c356f65/ : SUCCESS in 9m 36s\n- cross-glance-py27 http://logs.openstack.org/84/664384/1/check/cross-glance-py27/5553a8c/ : SUCCESS in 4m 39s\n- cross-glance-py36 http://logs.openstack.org/84/664384/1/check/cross-glance-py36/d211469/ : SUCCESS in 4m 33s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/1/check/cross-horizon-py27/67c249e/ : SUCCESS in 10m 34s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/1/check/cross-horizon-py36/aff957a/ : SUCCESS in 9m 30s\n- cross-horizon-npm http://logs.openstack.org/84/664384/1/check/cross-horizon-npm/b2efde6/ : SUCCESS in 5m 57s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/1/check/cross-keystone-py27/dae4c61/ : SUCCESS in 22m 23s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/1/check/cross-keystone-py36/be8d92b/ : SUCCESS in 24m 28s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/1/check/cross-neutron-py27/a22cf2b/ : SUCCESS in 21m 56s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/1/check/cross-neutron-py36/070ef9f/ : SUCCESS in 23m 05s\n- cross-nova-py27 http://logs.openstack.org/84/664384/1/check/cross-nova-py27/d116bc5/ : SUCCESS in 18m 49s\n- cross-nova-py36 http://logs.openstack.org/84/664384/1/check/cross-nova-py36/23d8a75/ : SUCCESS in 12m 29s\n- cross-nova-functional http://logs.openstack.org/84/664384/1/check/cross-nova-functional/3ee678b/ : SUCCESS in 18m 28s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/1/check/cross-osc-tox-docs/2cf8a41/html/ : SUCCESS in 5m 42s\n- cross-swift-py27 http://logs.openstack.org/84/664384/1/check/cross-swift-py27/4d6b562/ : SUCCESS in 8m 19s\n- cross-swift-py36 http://logs.openstack.org/84/664384/1/check/cross-swift-py36/89da2f4/ : SUCCESS in 8m 41s\n- requirements-integration http://logs.openstack.org/84/664384/1/check/requirements-integration/400ab01/ : SUCCESS in 45m 42s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/1/check/neutron-grenade-multinode/b1a3aff/ : SUCCESS in 1h 08m 53s\n- tempest-full http://logs.openstack.org/84/664384/1/check/tempest-full/3c1ca23/ : SUCCESS in 1h 48m 12s\n- neutron-grenade http://logs.openstack.org/84/664384/1/check/neutron-grenade/d9869d9/ : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"606dd8aa29f47c199fb81d2a2b6761c7130e8659","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-11 00:21:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"831869008642d41ae1d13dd9dc0cefe4968f17be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 06:53:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/2/check/openstack-tox-docs/31d3231/html/ : SUCCESS in 3m 04s\n- openstack-tox-pypy http://logs.openstack.org/84/664384/2/check/openstack-tox-pypy/df39a2f/ : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/2/check/openstack-tox-pep8/d065c3a/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/2/check/openstack-tox-py27/0808c9c/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/2/check/openstack-tox-py36/26bc416/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/84/664384/2/check/openstack-tox-validate/0efaac9/ : SUCCESS in 3m 33s\n- requirements-tox-babel http://logs.openstack.org/84/664384/2/check/requirements-tox-babel/8076a29/ : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py27-check-uc/4d500c9/ : SUCCESS in 7m 55s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py36-check-uc/3132d87/ : SUCCESS in 6m 50s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/2/check/cross-cinder-py27/eccdba1/ : SUCCESS in 10m 51s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/2/check/cross-cinder-py36/8a457b1/ : SUCCESS in 9m 10s\n- cross-glance-py27 http://logs.openstack.org/84/664384/2/check/cross-glance-py27/952c050/ : SUCCESS in 4m 14s\n- cross-glance-py36 http://logs.openstack.org/84/664384/2/check/cross-glance-py36/426cd40/ : SUCCESS in 4m 07s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/2/check/cross-horizon-py27/889528f/ : SUCCESS in 10m 35s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/2/check/cross-horizon-py36/0a377fc/ : SUCCESS in 9m 16s\n- cross-horizon-npm http://logs.openstack.org/84/664384/2/check/cross-horizon-npm/d7b89e4/ : SUCCESS in 6m 16s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/2/check/cross-keystone-py27/9db8bd6/ : SUCCESS in 20m 08s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/2/check/cross-keystone-py36/592fc10/ : SUCCESS in 23m 20s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/2/check/cross-neutron-py27/496ac52/ : SUCCESS in 22m 53s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/2/check/cross-neutron-py36/da374e1/ : SUCCESS in 22m 36s\n- cross-nova-py27 http://logs.openstack.org/84/664384/2/check/cross-nova-py27/8927fae/ : FAILURE in 11m 20s\n- cross-nova-py36 http://logs.openstack.org/84/664384/2/check/cross-nova-py36/642609c/ : SUCCESS in 13m 31s\n- cross-nova-functional http://logs.openstack.org/84/664384/2/check/cross-nova-functional/fabf730/ : SUCCESS in 20m 12s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/2/check/cross-osc-tox-docs/466cb71/html/ : SUCCESS in 5m 20s\n- cross-swift-py27 http://logs.openstack.org/84/664384/2/check/cross-swift-py27/7e4d2c8/ : SUCCESS in 8m 00s\n- cross-swift-py36 http://logs.openstack.org/84/664384/2/check/cross-swift-py36/f55eebf/ : SUCCESS in 9m 06s\n- requirements-integration http://logs.openstack.org/84/664384/2/check/requirements-integration/9397085/ : SUCCESS in 51m 49s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/2/check/neutron-grenade-multinode/b426fd0/ : SUCCESS in 1h 20m 45s\n- tempest-full http://logs.openstack.org/84/664384/2/check/tempest-full/9eb3b06/ : SUCCESS in 1h 55m 38s\n- neutron-grenade http://logs.openstack.org/84/664384/2/check/neutron-grenade/1cbf187/ : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":2},{"id":"24105bcdc6521aafe8b4a31cf1c00559e3bbd9bd","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-11 08:24:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d77f546b96adc88b151418a1a1ca785fec8133bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:35:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/2/check/openstack-tox-docs/0158b49/html/ : SUCCESS in 3m 35s\n- openstack-tox-pypy http://logs.openstack.org/84/664384/2/check/openstack-tox-pypy/16d6169/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/2/check/openstack-tox-pep8/2eda6e6/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/2/check/openstack-tox-py27/763c804/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/2/check/openstack-tox-py36/498721b/ : SUCCESS in 3m 08s\n- openstack-tox-validate http://logs.openstack.org/84/664384/2/check/openstack-tox-validate/89e3be7/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/84/664384/2/check/requirements-tox-babel/e591708/ : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py27-check-uc/cbc5e6f/ : SUCCESS in 8m 25s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py36-check-uc/561fa72/ : SUCCESS in 7m 53s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/2/check/cross-cinder-py27/afb7219/ : SUCCESS in 7m 57s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/2/check/cross-cinder-py36/132cb09/ : SUCCESS in 10m 26s\n- cross-glance-py27 http://logs.openstack.org/84/664384/2/check/cross-glance-py27/71908ae/ : SUCCESS in 4m 59s\n- cross-glance-py36 http://logs.openstack.org/84/664384/2/check/cross-glance-py36/36f0520/ : SUCCESS in 4m 54s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/2/check/cross-horizon-py27/a84420c/ : SUCCESS in 9m 06s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/2/check/cross-horizon-py36/c318127/ : SUCCESS in 7m 55s\n- cross-horizon-npm http://logs.openstack.org/84/664384/2/check/cross-horizon-npm/910907f/ : SUCCESS in 5m 53s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/2/check/cross-keystone-py27/cea98d2/ : SUCCESS in 20m 26s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/2/check/cross-keystone-py36/164a151/ : SUCCESS in 24m 56s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/2/check/cross-neutron-py27/c8fedcd/ : SUCCESS in 23m 28s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/2/check/cross-neutron-py36/149c69c/ : SUCCESS in 24m 57s\n- cross-nova-py27 http://logs.openstack.org/84/664384/2/check/cross-nova-py27/e59bf2f/ : FAILURE in 12m 45s\n- cross-nova-py36 http://logs.openstack.org/84/664384/2/check/cross-nova-py36/2e6cf2a/ : SUCCESS in 20m 17s\n- cross-nova-functional http://logs.openstack.org/84/664384/2/check/cross-nova-functional/a79035e/ : SUCCESS in 22m 07s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/2/check/cross-osc-tox-docs/4d22eb8/html/ : SUCCESS in 7m 12s\n- cross-swift-py27 http://logs.openstack.org/84/664384/2/check/cross-swift-py27/3bec1d4/ : SUCCESS in 8m 30s\n- cross-swift-py36 http://logs.openstack.org/84/664384/2/check/cross-swift-py36/0ed4b0e/ : SUCCESS in 9m 10s\n- requirements-integration http://logs.openstack.org/84/664384/2/check/requirements-integration/ec5d081/ : SUCCESS in 44m 58s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/2/check/neutron-grenade-multinode/4407769/ : SUCCESS in 1h 07m 25s\n- tempest-full http://logs.openstack.org/84/664384/2/check/tempest-full/3a97c08/ : SUCCESS in 1h 46m 47s\n- neutron-grenade http://logs.openstack.org/84/664384/2/check/neutron-grenade/b5fe148/ : FAILURE in 41m 34s","accounts_in_message":[],"_revision_number":2},{"id":"aae8e5d97fb88a3be6ff8092cfd8be59a5f9db34","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-11 14:20:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f125b998aeba9752b9a32e854de1e95b905eaac5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:12:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/2/check/openstack-tox-docs/dc01897/html/ : SUCCESS in 3m 16s\n- openstack-tox-pypy http://logs.openstack.org/84/664384/2/check/openstack-tox-pypy/c44773f/ : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/2/check/openstack-tox-pep8/bb4fabc/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/2/check/openstack-tox-py27/e085f87/ : RETRY_LIMIT in 2m 24s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/2/check/openstack-tox-py36/fc43ea7/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/84/664384/2/check/openstack-tox-validate/d5b081c/ : SUCCESS in 4m 50s\n- requirements-tox-babel http://logs.openstack.org/84/664384/2/check/requirements-tox-babel/db8db15/ : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py27-check-uc/5377271/ : SUCCESS in 7m 04s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py36-check-uc/5076446/ : SUCCESS in 6m 55s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/2/check/cross-cinder-py27/f9ff9e3/ : SUCCESS in 8m 56s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/2/check/cross-cinder-py36/f968c49/ : SUCCESS in 7m 47s\n- cross-glance-py27 http://logs.openstack.org/84/664384/2/check/cross-glance-py27/4f2f0ca/ : SUCCESS in 4m 40s\n- cross-glance-py36 http://logs.openstack.org/84/664384/2/check/cross-glance-py36/36636ac/ : SUCCESS in 4m 27s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/2/check/cross-horizon-py27/38a57b2/ : SUCCESS in 10m 29s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/2/check/cross-horizon-py36/dda22e4/ : SUCCESS in 9m 22s\n- cross-horizon-npm http://logs.openstack.org/84/664384/2/check/cross-horizon-npm/22a04c0/ : SUCCESS in 6m 50s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/2/check/cross-keystone-py27/ea4c288/ : SUCCESS in 20m 50s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/2/check/cross-keystone-py36/4e4ffb9/ : SUCCESS in 28m 25s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/2/check/cross-neutron-py27/e3b12fe/ : SUCCESS in 28m 42s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/2/check/cross-neutron-py36/95cfa9e/ : SUCCESS in 28m 15s\n- cross-nova-py27 http://logs.openstack.org/84/664384/2/check/cross-nova-py27/bfe4e0b/ : FAILURE in 11m 48s\n- cross-nova-py36 http://logs.openstack.org/84/664384/2/check/cross-nova-py36/fb08c16/ : SUCCESS in 12m 47s\n- cross-nova-functional http://logs.openstack.org/84/664384/2/check/cross-nova-functional/41427b3/ : SUCCESS in 21m 02s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/2/check/cross-osc-tox-docs/50f0d61/html/ : SUCCESS in 5m 04s\n- cross-swift-py27 http://logs.openstack.org/84/664384/2/check/cross-swift-py27/d2fd295/ : SUCCESS in 7m 49s\n- cross-swift-py36 http://logs.openstack.org/84/664384/2/check/cross-swift-py36/18210c9/ : SUCCESS in 7m 48s\n- requirements-integration http://logs.openstack.org/84/664384/2/check/requirements-integration/e556280/ : SUCCESS in 1h 11m 57s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/2/check/neutron-grenade-multinode/c2f7d88/ : SUCCESS in 1h 08m 24s\n- tempest-full http://logs.openstack.org/84/664384/2/check/tempest-full/933593e/ : SUCCESS in 1h 44m 37s\n- neutron-grenade http://logs.openstack.org/84/664384/2/check/neutron-grenade/cde3eb5/ : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":2},{"id":"142d098c3e9fd06116aaf0614282fd3d0b38ec36","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-12 15:34:41.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4b31421bce0012580da3c3ed3c2b25b305a879d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 00:27:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/2/check/openstack-tox-docs/9864fc3/html/ : SUCCESS in 3m 05s\n- openstack-tox-pypy http://logs.openstack.org/84/664384/2/check/openstack-tox-pypy/ae2f725/ : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/2/check/openstack-tox-pep8/fb3b2d4/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/2/check/openstack-tox-py27/d096fc5/ : SUCCESS in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/2/check/openstack-tox-py36/65d0236/ : SUCCESS in 3m 22s\n- openstack-tox-validate http://logs.openstack.org/84/664384/2/check/openstack-tox-validate/754feab/ : SUCCESS in 4m 44s\n- requirements-tox-babel http://logs.openstack.org/84/664384/2/check/requirements-tox-babel/431a83f/ : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py27-check-uc/ecf9c15/ : SUCCESS in 7m 59s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py36-check-uc/3f3dd83/ : SUCCESS in 7m 18s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/2/check/cross-cinder-py27/a73463c/ : FAILURE in 8m 55s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/2/check/cross-cinder-py36/8f6b94a/ : SUCCESS in 7m 59s\n- cross-glance-py27 http://logs.openstack.org/84/664384/2/check/cross-glance-py27/4f77e4f/ : SUCCESS in 4m 39s\n- cross-glance-py36 http://logs.openstack.org/84/664384/2/check/cross-glance-py36/fbcda35/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/2/check/cross-horizon-py27/7e94095/ : SUCCESS in 10m 08s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/2/check/cross-horizon-py36/d76ebe7/ : SUCCESS in 8m 04s\n- cross-horizon-npm http://logs.openstack.org/84/664384/2/check/cross-horizon-npm/4ded0b6/ : SUCCESS in 6m 23s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/2/check/cross-keystone-py27/7d76f39/ : SUCCESS in 20m 35s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/2/check/cross-keystone-py36/19519fd/ : SUCCESS in 26m 18s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/2/check/cross-neutron-py27/0eb2001/ : SUCCESS in 22m 45s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/2/check/cross-neutron-py36/0198d3c/ : SUCCESS in 21m 01s\n- cross-nova-py27 http://logs.openstack.org/84/664384/2/check/cross-nova-py27/26a2462/ : SUCCESS in 13m 07s\n- cross-nova-py36 http://logs.openstack.org/84/664384/2/check/cross-nova-py36/1fba944/ : SUCCESS in 17m 24s\n- cross-nova-functional http://logs.openstack.org/84/664384/2/check/cross-nova-functional/f758d1d/ : SUCCESS in 19m 29s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/2/check/cross-osc-tox-docs/994e309/html/ : SUCCESS in 5m 43s\n- cross-swift-py27 http://logs.openstack.org/84/664384/2/check/cross-swift-py27/eae0a9c/ : SUCCESS in 7m 52s\n- cross-swift-py36 http://logs.openstack.org/84/664384/2/check/cross-swift-py36/db05696/ : SUCCESS in 8m 31s\n- requirements-integration http://logs.openstack.org/84/664384/2/check/requirements-integration/98d7a72/ : SUCCESS in 1h 03m 23s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/2/check/neutron-grenade-multinode/82e3ccf/ : FAILURE in 37m 06s\n- tempest-full http://logs.openstack.org/84/664384/2/check/tempest-full/7b60f1e/ : SUCCESS in 1h 41m 41s\n- neutron-grenade http://logs.openstack.org/84/664384/2/check/neutron-grenade/5726edb/ : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":2},{"id":"019b091323a6bad7335b5c2e6b6897b19e9ac607","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-13 00:30:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d3e59b4382525acfb2a61f865ddc1021d144f972","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 10:07:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/2/check/openstack-tox-docs/0262cb7/html/ : SUCCESS in 3m 57s\n- openstack-tox-pypy http://logs.openstack.org/84/664384/2/check/openstack-tox-pypy/21b9030/ : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/2/check/openstack-tox-pep8/ace8a1c/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/2/check/openstack-tox-py27/2a09b0a/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/2/check/openstack-tox-py36/b23bbb8/ : SUCCESS in 3m 17s\n- openstack-tox-validate http://logs.openstack.org/84/664384/2/check/openstack-tox-validate/45cfe40/ : SUCCESS in 3m 35s\n- requirements-tox-babel http://logs.openstack.org/84/664384/2/check/requirements-tox-babel/7453eb6/ : SUCCESS in 3m 37s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py27-check-uc/dcc4700/ : SUCCESS in 7m 20s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py36-check-uc/5a77151/ : SUCCESS in 8m 14s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/2/check/cross-cinder-py27/5233c97/ : SUCCESS in 8m 31s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/2/check/cross-cinder-py36/217c49f/ : SUCCESS in 8m 04s\n- cross-glance-py27 http://logs.openstack.org/84/664384/2/check/cross-glance-py27/c74f52f/ : SUCCESS in 4m 15s\n- cross-glance-py36 http://logs.openstack.org/84/664384/2/check/cross-glance-py36/0aaad43/ : SUCCESS in 4m 43s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/2/check/cross-horizon-py27/5352fb0/ : SUCCESS in 10m 52s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/2/check/cross-horizon-py36/cc29364/ : SUCCESS in 8m 34s\n- cross-horizon-npm http://logs.openstack.org/84/664384/2/check/cross-horizon-npm/0aef4a1/ : SUCCESS in 6m 16s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/2/check/cross-keystone-py27/4703ff5/ : SUCCESS in 22m 24s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/2/check/cross-keystone-py36/4863649/ : SUCCESS in 23m 14s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/2/check/cross-neutron-py27/0a26c9c/ : SUCCESS in 22m 46s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/2/check/cross-neutron-py36/209a571/ : SUCCESS in 21m 53s\n- cross-nova-py27 http://logs.openstack.org/84/664384/2/check/cross-nova-py27/0748bab/ : SUCCESS in 12m 37s\n- cross-nova-py36 http://logs.openstack.org/84/664384/2/check/cross-nova-py36/ab9cd08/ : SUCCESS in 10m 48s\n- cross-nova-functional http://logs.openstack.org/84/664384/2/check/cross-nova-functional/0c4785a/ : SUCCESS in 18m 40s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/2/check/cross-osc-tox-docs/4f709e4/html/ : SUCCESS in 5m 48s\n- cross-swift-py27 http://logs.openstack.org/84/664384/2/check/cross-swift-py27/40950bc/ : SUCCESS in 8m 18s\n- cross-swift-py36 http://logs.openstack.org/84/664384/2/check/cross-swift-py36/4ee0ec6/ : SUCCESS in 7m 57s\n- requirements-integration http://logs.openstack.org/84/664384/2/check/requirements-integration/1534cb4/ : SUCCESS in 55m 55s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/2/check/neutron-grenade-multinode/9917f60/ : SUCCESS in 1h 08m 32s\n- tempest-full http://logs.openstack.org/84/664384/2/check/tempest-full/1747b9d/ : SUCCESS in 1h 52m 53s\n- neutron-grenade http://logs.openstack.org/84/664384/2/check/neutron-grenade/aaaea08/ : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":2},{"id":"162b9a17c05856e7b26e94f3388eb00d8c1f7432","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-13 11:06:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"839ae513cfccda44417e89cde0396100879f0baf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 11:07:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"89cf997bd801f4d2b61730c8f158d147feea6ff1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 12:27:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/2/gate/openstack-tox-docs/38c2993/html/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/2/gate/openstack-tox-pep8/9d42066/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/2/gate/openstack-tox-py27/bd8a150/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/2/gate/openstack-tox-py36/9b75edb/ : SUCCESS in 3m 44s\n- openstack-tox-validate http://logs.openstack.org/84/664384/2/gate/openstack-tox-validate/116dd2c/ : SUCCESS in 3m 04s\n- requirements-tox-babel http://logs.openstack.org/84/664384/2/gate/requirements-tox-babel/6e2f530/ : SUCCESS in 4m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/2/gate/requirements-tox-py27-check-uc/99d8cdd/ : SUCCESS in 7m 22s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/2/gate/requirements-tox-py36-check-uc/782774e/ : SUCCESS in 6m 32s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/2/gate/cross-cinder-py27/484c92c/ : SUCCESS in 10m 09s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/2/gate/cross-cinder-py36/60d3635/ : SUCCESS in 7m 54s\n- cross-glance-py27 http://logs.openstack.org/84/664384/2/gate/cross-glance-py27/7dce0e3/ : SUCCESS in 4m 55s\n- cross-glance-py36 http://logs.openstack.org/84/664384/2/gate/cross-glance-py36/a72144d/ : SUCCESS in 4m 35s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/2/gate/cross-horizon-py27/080d836/ : SUCCESS in 10m 30s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/2/gate/cross-horizon-py36/bd1ed9e/ : SUCCESS in 9m 41s\n- cross-horizon-npm http://logs.openstack.org/84/664384/2/gate/cross-horizon-npm/56c1075/ : SUCCESS in 6m 09s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/2/gate/cross-keystone-py27/3faed26/ : SUCCESS in 21m 01s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/2/gate/cross-keystone-py36/0f6b61a/ : SUCCESS in 23m 34s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/2/gate/cross-neutron-py27/d86a2a3/ : SUCCESS in 24m 41s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/2/gate/cross-neutron-py36/4a3d80e/ : SUCCESS in 23m 48s\n- cross-nova-py27 http://logs.openstack.org/84/664384/2/gate/cross-nova-py27/e6af084/ : SUCCESS in 13m 06s\n- cross-nova-py36 http://logs.openstack.org/84/664384/2/gate/cross-nova-py36/7612898/ : SUCCESS in 12m 42s\n- cross-nova-functional http://logs.openstack.org/84/664384/2/gate/cross-nova-functional/8cf7891/ : SUCCESS in 18m 04s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/2/gate/cross-osc-tox-docs/9a0895f/html/ : SUCCESS in 5m 35s\n- cross-swift-py27 http://logs.openstack.org/84/664384/2/gate/cross-swift-py27/2f6aa21/ : SUCCESS in 8m 25s\n- cross-swift-py36 http://logs.openstack.org/84/664384/2/gate/cross-swift-py36/f4cbdad/ : SUCCESS in 7m 59s\n- requirements-integration http://logs.openstack.org/84/664384/2/gate/requirements-integration/d23a5cf/ : SUCCESS in 56m 21s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/2/gate/neutron-grenade-multinode/0733e3f/ : SUCCESS in 1h 05m 08s\n- tempest-full http://logs.openstack.org/84/664384/2/gate/tempest-full/027484b/ : FAILURE in 17m 59s\n- neutron-grenade http://logs.openstack.org/84/664384/2/gate/neutron-grenade/cb9815e/ : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":2},{"id":"621bd85edf1d9b8e0cf7bdaefcb089029356c620","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-13 13:49:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"35803f6f0408daf523a3763a367ea6efe2ad1ab6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 00:11:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/2/check/openstack-tox-docs/9273cf1/html/ : SUCCESS in 2m 58s\n- openstack-tox-pypy http://logs.openstack.org/84/664384/2/check/openstack-tox-pypy/dfaf12a/ : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/2/check/openstack-tox-pep8/a269c3c/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/2/check/openstack-tox-py27/318e3ce/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/2/check/openstack-tox-py36/b50ea74/ : SUCCESS in 3m 11s\n- openstack-tox-validate http://logs.openstack.org/84/664384/2/check/openstack-tox-validate/32559bb/ : SUCCESS in 2m 54s\n- requirements-tox-babel http://logs.openstack.org/84/664384/2/check/requirements-tox-babel/b2ffdc0/ : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py27-check-uc/101e07b/ : SUCCESS in 8m 01s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py36-check-uc/4bfbadc/ : SUCCESS in 7m 27s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/2/check/cross-cinder-py27/211ed8e/ : SUCCESS in 9m 25s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/2/check/cross-cinder-py36/be5e53f/ : SUCCESS in 10m 05s\n- cross-glance-py27 http://logs.openstack.org/84/664384/2/check/cross-glance-py27/c845c9e/ : SUCCESS in 5m 07s\n- cross-glance-py36 http://logs.openstack.org/84/664384/2/check/cross-glance-py36/f177c64/ : SUCCESS in 4m 46s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/2/check/cross-horizon-py27/cd08d9d/ : SUCCESS in 10m 42s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/2/check/cross-horizon-py36/777e479/ : SUCCESS in 8m 59s\n- cross-horizon-npm http://logs.openstack.org/84/664384/2/check/cross-horizon-npm/cfc4272/ : SUCCESS in 6m 36s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/2/check/cross-keystone-py27/14a0aa4/ : SUCCESS in 23m 40s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/2/check/cross-keystone-py36/d38dde6/ : SUCCESS in 24m 26s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/2/check/cross-neutron-py27/806269d/ : SUCCESS in 23m 52s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/2/check/cross-neutron-py36/4f128d3/ : SUCCESS in 23m 09s\n- cross-nova-py27 http://logs.openstack.org/84/664384/2/check/cross-nova-py27/b0a359f/ : SUCCESS in 20m 55s\n- cross-nova-py36 http://logs.openstack.org/84/664384/2/check/cross-nova-py36/637f9a1/ : SUCCESS in 12m 29s\n- cross-nova-functional http://logs.openstack.org/84/664384/2/check/cross-nova-functional/0feed0c/ : SUCCESS in 19m 08s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/2/check/cross-osc-tox-docs/07c4ae2/html/ : SUCCESS in 5m 15s\n- cross-swift-py27 http://logs.openstack.org/84/664384/2/check/cross-swift-py27/d9c6eb1/ : SUCCESS in 11m 09s\n- cross-swift-py36 http://logs.openstack.org/84/664384/2/check/cross-swift-py36/1476e51/ : SUCCESS in 11m 31s\n- requirements-integration http://logs.openstack.org/84/664384/2/check/requirements-integration/6c2a639/ : SUCCESS in 1h 01m 55s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/2/check/neutron-grenade-multinode/e60597a/ : FAILURE in 24m 46s\n- tempest-full http://logs.openstack.org/84/664384/2/check/tempest-full/fc711db/ : SUCCESS in 2h 02m 16s\n- neutron-grenade http://logs.openstack.org/84/664384/2/check/neutron-grenade/5831915/ : SUCCESS in 56m 11s","accounts_in_message":[],"_revision_number":2},{"id":"1f2162a5de941e8f2996dd9eb50574c9a20c641a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-14 00:19:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4f3299c581a2aef36d972326cb8e0cdb1d4d237d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 07:14:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/2/check/openstack-tox-docs/0a9ac64/html/ : SUCCESS in 4m 00s\n- openstack-tox-pypy http://logs.openstack.org/84/664384/2/check/openstack-tox-pypy/fd532f5/ : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/2/check/openstack-tox-pep8/a4794c9/ : SUCCESS in 3m 21s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/2/check/openstack-tox-py27/1c3f1f5/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/2/check/openstack-tox-py36/2a3683d/ : SUCCESS in 3m 28s\n- openstack-tox-validate http://logs.openstack.org/84/664384/2/check/openstack-tox-validate/3764a7a/ : SUCCESS in 3m 16s\n- requirements-tox-babel http://logs.openstack.org/84/664384/2/check/requirements-tox-babel/853ce54/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py27-check-uc/dab447b/ : SUCCESS in 7m 03s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/2/check/requirements-tox-py36-check-uc/b29f9cf/ : SUCCESS in 7m 23s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/2/check/cross-cinder-py27/2985c6d/ : SUCCESS in 8m 51s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/2/check/cross-cinder-py36/ada10cd/ : SUCCESS in 8m 54s\n- cross-glance-py27 http://logs.openstack.org/84/664384/2/check/cross-glance-py27/abd84fe/ : SUCCESS in 5m 45s\n- cross-glance-py36 http://logs.openstack.org/84/664384/2/check/cross-glance-py36/e0fbf5c/ : SUCCESS in 5m 56s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/2/check/cross-horizon-py27/8a16d61/ : SUCCESS in 11m 48s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/2/check/cross-horizon-py36/29354c0/ : SUCCESS in 9m 26s\n- cross-horizon-npm http://logs.openstack.org/84/664384/2/check/cross-horizon-npm/adbc650/ : SUCCESS in 6m 42s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/2/check/cross-keystone-py27/b370a99/ : SUCCESS in 21m 38s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/2/check/cross-keystone-py36/2a6ea3e/ : SUCCESS in 21m 51s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/2/check/cross-neutron-py27/3f2a8b7/ : SUCCESS in 28m 36s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/2/check/cross-neutron-py36/51a9a69/ : SUCCESS in 29m 54s\n- cross-nova-py27 http://logs.openstack.org/84/664384/2/check/cross-nova-py27/743abe4/ : SUCCESS in 11m 57s\n- cross-nova-py36 http://logs.openstack.org/84/664384/2/check/cross-nova-py36/5c16186/ : SUCCESS in 13m 49s\n- cross-nova-functional http://logs.openstack.org/84/664384/2/check/cross-nova-functional/d9599c5/ : SUCCESS in 20m 51s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/2/check/cross-osc-tox-docs/b63594b/html/ : SUCCESS in 5m 53s\n- cross-swift-py27 http://logs.openstack.org/84/664384/2/check/cross-swift-py27/1d73502/ : SUCCESS in 8m 41s\n- cross-swift-py36 http://logs.openstack.org/84/664384/2/check/cross-swift-py36/febab9d/ : SUCCESS in 8m 14s\n- requirements-integration http://logs.openstack.org/84/664384/2/check/requirements-integration/fe2fc20/ : SUCCESS in 58m 58s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/2/check/neutron-grenade-multinode/1250c37/ : SUCCESS in 1h 09m 48s\n- tempest-full http://logs.openstack.org/84/664384/2/check/tempest-full/3a4e76c/ : SUCCESS in 1h 47m 07s\n- neutron-grenade http://logs.openstack.org/84/664384/2/check/neutron-grenade/39e347c/ : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8274cc931ec0d44d65a82d6b4fa6e918f2f59cef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 07:14:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f0aff4522ac148528538508f35f7b3a2a88f4787","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 10:06:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ccaa695cdaa9bbefa627a1d9cc8a2d4ac32214dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 10:06:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/84/664384/2/gate/openstack-tox-docs/03adc79/html/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/84/664384/2/gate/openstack-tox-pep8/453e0a6/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/84/664384/2/gate/openstack-tox-py27/0b52411/ : SUCCESS in 3m 47s\n- openstack-tox-py36 http://logs.openstack.org/84/664384/2/gate/openstack-tox-py36/51d3249/ : SUCCESS in 3m 57s\n- openstack-tox-validate http://logs.openstack.org/84/664384/2/gate/openstack-tox-validate/38c891d/ : SUCCESS in 3m 35s\n- requirements-tox-babel http://logs.openstack.org/84/664384/2/gate/requirements-tox-babel/a1cdecd/ : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/84/664384/2/gate/requirements-tox-py27-check-uc/d4a8859/ : SUCCESS in 8m 21s\n- requirements-tox-py36-check-uc http://logs.openstack.org/84/664384/2/gate/requirements-tox-py36-check-uc/40bbe37/ : SUCCESS in 6m 28s\n- cross-cinder-py27 http://logs.openstack.org/84/664384/2/gate/cross-cinder-py27/8ae3f23/ : SUCCESS in 8m 37s\n- cross-cinder-py36 http://logs.openstack.org/84/664384/2/gate/cross-cinder-py36/0acb194/ : SUCCESS in 9m 42s\n- cross-glance-py27 http://logs.openstack.org/84/664384/2/gate/cross-glance-py27/d1bf26b/ : SUCCESS in 4m 08s\n- cross-glance-py36 http://logs.openstack.org/84/664384/2/gate/cross-glance-py36/aabc3da/ : SUCCESS in 4m 32s\n- cross-horizon-py27 http://logs.openstack.org/84/664384/2/gate/cross-horizon-py27/85f9423/ : SUCCESS in 10m 12s\n- cross-horizon-py36 http://logs.openstack.org/84/664384/2/gate/cross-horizon-py36/771a906/ : SUCCESS in 8m 54s\n- cross-horizon-npm http://logs.openstack.org/84/664384/2/gate/cross-horizon-npm/e4cf505/ : SUCCESS in 6m 36s\n- cross-keystone-py27 http://logs.openstack.org/84/664384/2/gate/cross-keystone-py27/6a4c88a/ : SUCCESS in 25m 28s\n- cross-keystone-py36 http://logs.openstack.org/84/664384/2/gate/cross-keystone-py36/17e9376/ : SUCCESS in 24m 12s\n- cross-neutron-py27 http://logs.openstack.org/84/664384/2/gate/cross-neutron-py27/c796833/ : SUCCESS in 22m 17s\n- cross-neutron-py36 http://logs.openstack.org/84/664384/2/gate/cross-neutron-py36/d81fb43/ : SUCCESS in 21m 54s\n- cross-nova-py27 http://logs.openstack.org/84/664384/2/gate/cross-nova-py27/922acb9/ : SUCCESS in 15m 28s\n- cross-nova-py36 http://logs.openstack.org/84/664384/2/gate/cross-nova-py36/f7e86e8/ : SUCCESS in 10m 36s\n- cross-nova-functional http://logs.openstack.org/84/664384/2/gate/cross-nova-functional/22cc870/ : SUCCESS in 18m 55s\n- cross-osc-tox-docs http://logs.openstack.org/84/664384/2/gate/cross-osc-tox-docs/023f20e/html/ : SUCCESS in 4m 43s\n- cross-swift-py27 http://logs.openstack.org/84/664384/2/gate/cross-swift-py27/241f26d/ : SUCCESS in 8m 02s\n- cross-swift-py36 http://logs.openstack.org/84/664384/2/gate/cross-swift-py36/bda9542/ : SUCCESS in 8m 34s\n- requirements-integration http://logs.openstack.org/84/664384/2/gate/requirements-integration/aa14c7a/ : SUCCESS in 55m 17s\n- neutron-grenade-multinode http://logs.openstack.org/84/664384/2/gate/neutron-grenade-multinode/06814e9/ : SUCCESS in 1h 13m 51s\n- tempest-full http://logs.openstack.org/84/664384/2/gate/tempest-full/ec92508/ : SUCCESS in 1h 45m 22s\n- neutron-grenade http://logs.openstack.org/84/664384/2/gate/neutron-grenade/ffc9061/ : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57814fff7d519e1637ff73e6cd4c8bf05325e6e3","revisions":{"bcf19353e68dfffaaf400b0e1243be03cf273128":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:53:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/664384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/664384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/664384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/664384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/664384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/664384/1"}}},"commit":{"parents":[{"commit":"98a5e8f0b47309e23edd429cb85704413876ce6d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98a5e8f0b47309e23edd429cb85704413876ce6d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:53:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:53:08.000000000","tz":0},"subject":"update constraint for tooz to new release 1.65.0","message":"update constraint for tooz to new release 1.65.0\n\nChange-Id: I7a9fa4871cedc7d26324ec7b48ef98ca31945fbb\nmeta:version: 1.65.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Matthew Thode \u003cmthode@mthode.org\u003e\nmeta:release:Commit: Matthew Thode \u003cmthode@mthode.org\u003e\nmeta:release:Change-Id: Id5de482ec90bf0ae5b1365c7905d776c886ef4d6\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bcf19353e68dfffaaf400b0e1243be03cf273128"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bcf19353e68dfffaaf400b0e1243be03cf273128"}]},"branch":"refs/heads/master"},"57814fff7d519e1637ff73e6cd4c8bf05325e6e3":{"kind":"REWORK","_number":2,"created":"2019-06-11 00:21:54.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/84/664384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/664384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/664384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/664384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/664384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/664384/2"}}},"commit":{"parents":[{"commit":"98a5e8f0b47309e23edd429cb85704413876ce6d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98a5e8f0b47309e23edd429cb85704413876ce6d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:53:08.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-06-11 00:21:13.000000000","tz":-300},"subject":"update constraint for tooz to new release 1.65.0","message":"update constraint for tooz to new release 1.65.0\n\nAlso need to update grpcio, how was this gated in the first place?\n\nChange-Id: I7a9fa4871cedc7d26324ec7b48ef98ca31945fbb\nmeta:version: 1.65.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Matthew Thode \u003cmthode@mthode.org\u003e\nmeta:release:Commit: Matthew Thode \u003cmthode@mthode.org\u003e\nmeta:release:Change-Id: Id5de482ec90bf0ae5b1365c7905d776c886ef4d6\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57814fff7d519e1637ff73e6cd4c8bf05325e6e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57814fff7d519e1637ff73e6cd4c8bf05325e6e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
