)]}'
{"id":"openstack%2Frequirements~664385","triplet_id":"openstack%2Frequirements~master~I88da0ac7a8fc0d91f22310c5908c5786a63e6ae3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I88da0ac7a8fc0d91f22310c5908c5786a63e6ae3","subject":"update constraint for octavia-lib to new release 1.2.0","status":"MERGED","created":"2019-06-10 15:54:02.000000000","updated":"2019-06-13 05:56:30.000000000","submitted":"2019-06-13 05:56:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664385-1560405390138-216c3fa0","meta_rev_id":"7cadd249c54a794a893ae9ca3c004a620cc415b5","_number":664385,"virtual_id_number":664385,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2019-06-12 01:20:04.000000000","_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-06-13 05:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-11 00:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-11 08:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-11 00:20:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 00:20:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-11 08:37:07.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-06-12 01:20:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-06-13 05:56:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"336f9c62ef928fb068e2e347b04d7963f1d5ab2f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:54:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c26e9a0c2c2345437712df5fcbb1ead6d36204e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 23:01:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/85/664385/1/check/openstack-tox-docs/25b0905/html/ : SUCCESS in 3m 34s\n- openstack-tox-pypy http://logs.openstack.org/85/664385/1/check/openstack-tox-pypy/31beec0/ : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/664385/1/check/openstack-tox-pep8/d10b0b4/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/85/664385/1/check/openstack-tox-py27/4e353b3/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/85/664385/1/check/openstack-tox-py36/690cd09/ : SUCCESS in 3m 20s\n- openstack-tox-validate http://logs.openstack.org/85/664385/1/check/openstack-tox-validate/882a482/ : SUCCESS in 3m 41s\n- requirements-tox-babel http://logs.openstack.org/85/664385/1/check/requirements-tox-babel/f8aa6dc/ : SUCCESS in 3m 46s\n- requirements-tox-py27-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py27-check-uc/42cd3cf/ : SUCCESS in 5m 46s\n- requirements-tox-py36-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py36-check-uc/194b884/ : SUCCESS in 8m 25s\n- cross-cinder-py27 http://logs.openstack.org/85/664385/1/check/cross-cinder-py27/e2561ca/ : SUCCESS in 10m 17s\n- cross-cinder-py36 http://logs.openstack.org/85/664385/1/check/cross-cinder-py36/b931121/ : SUCCESS in 8m 52s\n- cross-glance-py27 http://logs.openstack.org/85/664385/1/check/cross-glance-py27/9fc3e95/ : SUCCESS in 4m 28s\n- cross-glance-py36 http://logs.openstack.org/85/664385/1/check/cross-glance-py36/2d58fd1/ : SUCCESS in 4m 48s\n- cross-horizon-py27 http://logs.openstack.org/85/664385/1/check/cross-horizon-py27/1aaceba/ : SUCCESS in 11m 06s\n- cross-horizon-py36 http://logs.openstack.org/85/664385/1/check/cross-horizon-py36/1a9fdc6/ : SUCCESS in 11m 57s\n- cross-horizon-npm http://logs.openstack.org/85/664385/1/check/cross-horizon-npm/1ff77cd/ : SUCCESS in 7m 15s\n- cross-keystone-py27 http://logs.openstack.org/85/664385/1/check/cross-keystone-py27/bd5e78d/ : SUCCESS in 21m 50s\n- cross-keystone-py36 http://logs.openstack.org/85/664385/1/check/cross-keystone-py36/6e9707a/ : SUCCESS in 34m 12s\n- cross-neutron-py27 http://logs.openstack.org/85/664385/1/check/cross-neutron-py27/94d6dc2/ : SUCCESS in 25m 31s\n- cross-neutron-py36 http://logs.openstack.org/85/664385/1/check/cross-neutron-py36/ceff5b3/ : SUCCESS in 27m 14s\n- cross-nova-py27 http://logs.openstack.org/85/664385/1/check/cross-nova-py27/cc6d433/ : SUCCESS in 12m 44s\n- cross-nova-py36 http://logs.openstack.org/85/664385/1/check/cross-nova-py36/81b0237/ : SUCCESS in 12m 07s\n- cross-nova-functional http://logs.openstack.org/85/664385/1/check/cross-nova-functional/fdef493/ : SUCCESS in 20m 25s\n- cross-osc-tox-docs http://logs.openstack.org/85/664385/1/check/cross-osc-tox-docs/b28585c/html/ : SUCCESS in 5m 23s\n- cross-swift-py27 http://logs.openstack.org/85/664385/1/check/cross-swift-py27/e68dd57/ : SUCCESS in 8m 15s\n- cross-swift-py36 http://logs.openstack.org/85/664385/1/check/cross-swift-py36/0c53814/ : SUCCESS in 8m 14s\n- requirements-integration http://logs.openstack.org/85/664385/1/check/requirements-integration/b1d6e3a/ : SUCCESS in 58m 14s\n- neutron-grenade-multinode http://logs.openstack.org/85/664385/1/check/neutron-grenade-multinode/27604a1/ : SUCCESS in 1h 26m 18s\n- tempest-full http://logs.openstack.org/85/664385/1/check/tempest-full/efd2132/ : SUCCESS in 1h 31m 31s\n- neutron-grenade http://logs.openstack.org/85/664385/1/check/neutron-grenade/ff03a54/ : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":1},{"id":"87c990608062b4140ee46fcfb04d2bed3e2c0e44","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-11 00:20:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2970fcacdfd04496211462e7552962bef49d98dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 00:21:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7dd9963a736a4938dfdd090ddecf4c7713394fd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 02:49:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/85/664385/1/gate/openstack-tox-docs/2de42c2/html/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/85/664385/1/gate/openstack-tox-pep8/801ef22/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/85/664385/1/gate/openstack-tox-py27/d686843/ : SUCCESS in 3m 39s\n- openstack-tox-py36 http://logs.openstack.org/85/664385/1/gate/openstack-tox-py36/b8a4a59/ : SUCCESS in 4m 15s\n- openstack-tox-validate http://logs.openstack.org/85/664385/1/gate/openstack-tox-validate/6dd84d9/ : SUCCESS in 4m 02s\n- requirements-tox-babel http://logs.openstack.org/85/664385/1/gate/requirements-tox-babel/77028e7/ : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/85/664385/1/gate/requirements-tox-py27-check-uc/4e406d4/ : SUCCESS in 9m 24s\n- requirements-tox-py36-check-uc http://logs.openstack.org/85/664385/1/gate/requirements-tox-py36-check-uc/bdcf0b2/ : SUCCESS in 6m 06s\n- cross-cinder-py27 http://logs.openstack.org/85/664385/1/gate/cross-cinder-py27/18ac707/ : SUCCESS in 15m 06s\n- cross-cinder-py36 http://logs.openstack.org/85/664385/1/gate/cross-cinder-py36/59b2774/ : SUCCESS in 10m 14s\n- cross-glance-py27 http://logs.openstack.org/85/664385/1/gate/cross-glance-py27/3cd658d/ : SUCCESS in 4m 21s\n- cross-glance-py36 http://logs.openstack.org/85/664385/1/gate/cross-glance-py36/57b8626/ : SUCCESS in 5m 26s\n- cross-horizon-py27 http://logs.openstack.org/85/664385/1/gate/cross-horizon-py27/998f83f/ : SUCCESS in 11m 14s\n- cross-horizon-py36 http://logs.openstack.org/85/664385/1/gate/cross-horizon-py36/118c0f6/ : SUCCESS in 9m 20s\n- cross-horizon-npm http://logs.openstack.org/85/664385/1/gate/cross-horizon-npm/5c9b38b/ : SUCCESS in 6m 38s\n- cross-keystone-py27 http://logs.openstack.org/85/664385/1/gate/cross-keystone-py27/cfb7b96/ : SUCCESS in 22m 12s\n- cross-keystone-py36 http://logs.openstack.org/85/664385/1/gate/cross-keystone-py36/0577908/ : SUCCESS in 22m 28s\n- cross-neutron-py27 http://logs.openstack.org/85/664385/1/gate/cross-neutron-py27/973717d/ : SUCCESS in 21m 09s\n- cross-neutron-py36 http://logs.openstack.org/85/664385/1/gate/cross-neutron-py36/ba1d0ef/ : SUCCESS in 26m 34s\n- cross-nova-py27 http://logs.openstack.org/85/664385/1/gate/cross-nova-py27/31b91c5/ : FAILURE in 16m 13s\n- cross-nova-py36 http://logs.openstack.org/85/664385/1/gate/cross-nova-py36/c681b47/ : SUCCESS in 11m 36s\n- cross-nova-functional http://logs.openstack.org/85/664385/1/gate/cross-nova-functional/36ff919/ : SUCCESS in 18m 37s\n- cross-osc-tox-docs http://logs.openstack.org/85/664385/1/gate/cross-osc-tox-docs/3c6ef44/html/ : SUCCESS in 7m 08s\n- cross-swift-py27 http://logs.openstack.org/85/664385/1/gate/cross-swift-py27/bdac44e/ : SUCCESS in 9m 30s\n- cross-swift-py36 http://logs.openstack.org/85/664385/1/gate/cross-swift-py36/fd2e274/ : SUCCESS in 8m 16s\n- requirements-integration http://logs.openstack.org/85/664385/1/gate/requirements-integration/d9bfa83/ : SUCCESS in 1h 02m 19s\n- neutron-grenade-multinode http://logs.openstack.org/85/664385/1/gate/neutron-grenade-multinode/6545421/ : SUCCESS in 1h 07m 30s\n- tempest-full http://logs.openstack.org/85/664385/1/gate/tempest-full/1727885/ : FAILURE in 8m 12s\n- neutron-grenade http://logs.openstack.org/85/664385/1/gate/neutron-grenade/e66522f/ : SUCCESS in 48m 21s","accounts_in_message":[],"_revision_number":1},{"id":"93a6289a11d42d7cd12347ae1cbbb2fc7e0f79c8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-11 02:54:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"beb9256572f6f47338f55f15cd2ef2949e8ecfa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 07:29:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/85/664385/1/check/openstack-tox-docs/84533d2/html/ : SUCCESS in 3m 17s\n- openstack-tox-pypy http://logs.openstack.org/85/664385/1/check/openstack-tox-pypy/b73e8d3/ : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/664385/1/check/openstack-tox-pep8/e6163ab/ : SUCCESS in 3m 13s\n- openstack-tox-py27 http://logs.openstack.org/85/664385/1/check/openstack-tox-py27/4234da9/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/85/664385/1/check/openstack-tox-py36/1fabd7f/ : SUCCESS in 4m 35s\n- openstack-tox-validate http://logs.openstack.org/85/664385/1/check/openstack-tox-validate/adbe19b/ : SUCCESS in 3m 23s\n- requirements-tox-babel http://logs.openstack.org/85/664385/1/check/requirements-tox-babel/e7eae87/ : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py27-check-uc/cd58d53/ : SUCCESS in 7m 03s\n- requirements-tox-py36-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py36-check-uc/6f19dfd/ : SUCCESS in 6m 37s\n- cross-cinder-py27 http://logs.openstack.org/85/664385/1/check/cross-cinder-py27/3f486b5/ : SUCCESS in 11m 53s\n- cross-cinder-py36 http://logs.openstack.org/85/664385/1/check/cross-cinder-py36/e17278d/ : SUCCESS in 10m 50s\n- cross-glance-py27 http://logs.openstack.org/85/664385/1/check/cross-glance-py27/11789f1/ : SUCCESS in 5m 43s\n- cross-glance-py36 http://logs.openstack.org/85/664385/1/check/cross-glance-py36/e642677/ : SUCCESS in 4m 24s\n- cross-horizon-py27 http://logs.openstack.org/85/664385/1/check/cross-horizon-py27/c5e214d/ : SUCCESS in 10m 34s\n- cross-horizon-py36 http://logs.openstack.org/85/664385/1/check/cross-horizon-py36/3d4dae2/ : SUCCESS in 9m 11s\n- cross-horizon-npm http://logs.openstack.org/85/664385/1/check/cross-horizon-npm/28e2e06/ : SUCCESS in 6m 17s\n- cross-keystone-py27 http://logs.openstack.org/85/664385/1/check/cross-keystone-py27/1e839c7/ : SUCCESS in 21m 51s\n- cross-keystone-py36 http://logs.openstack.org/85/664385/1/check/cross-keystone-py36/1893100/ : SUCCESS in 28m 01s\n- cross-neutron-py27 http://logs.openstack.org/85/664385/1/check/cross-neutron-py27/31348b0/ : SUCCESS in 24m 19s\n- cross-neutron-py36 http://logs.openstack.org/85/664385/1/check/cross-neutron-py36/aace574/ : SUCCESS in 21m 46s\n- cross-nova-py27 http://logs.openstack.org/85/664385/1/check/cross-nova-py27/59a47bd/ : FAILURE in 11m 20s\n- cross-nova-py36 http://logs.openstack.org/85/664385/1/check/cross-nova-py36/cb3da78/ : SUCCESS in 10m 51s\n- cross-nova-functional http://logs.openstack.org/85/664385/1/check/cross-nova-functional/308e5d8/ : SUCCESS in 17m 41s\n- cross-osc-tox-docs http://logs.openstack.org/85/664385/1/check/cross-osc-tox-docs/7de48e3/html/ : SUCCESS in 5m 18s\n- cross-swift-py27 http://logs.openstack.org/85/664385/1/check/cross-swift-py27/2b4ab7e/ : SUCCESS in 7m 38s\n- cross-swift-py36 http://logs.openstack.org/85/664385/1/check/cross-swift-py36/33ffb2f/ : SUCCESS in 11m 37s\n- requirements-integration http://logs.openstack.org/85/664385/1/check/requirements-integration/557405b/ : SUCCESS in 56m 04s\n- neutron-grenade-multinode http://logs.openstack.org/85/664385/1/check/neutron-grenade-multinode/b7a1076/ : SUCCESS in 1h 34m 17s\n- tempest-full http://logs.openstack.org/85/664385/1/check/tempest-full/4c7aa51/ : SUCCESS in 1h 51m 43s\n- neutron-grenade http://logs.openstack.org/85/664385/1/check/neutron-grenade/ad608ef/ : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b6186805aacfab854695bf7432e306b55fecd468","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-11 08:37:07.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fe7080b3e7d09bf6b74af198a6eb1e8df66fdd63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 12:43:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/85/664385/1/check/openstack-tox-docs/ccbb534/html/ : SUCCESS in 3m 21s\n- openstack-tox-pypy http://logs.openstack.org/85/664385/1/check/openstack-tox-pypy/fa412ce/ : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/664385/1/check/openstack-tox-pep8/f591136/ : SUCCESS in 3m 10s\n- openstack-tox-py27 http://logs.openstack.org/85/664385/1/check/openstack-tox-py27/ab4b43b/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/85/664385/1/check/openstack-tox-py36/c6ec054/ : SUCCESS in 3m 42s\n- openstack-tox-validate http://logs.openstack.org/85/664385/1/check/openstack-tox-validate/d57a8f2/ : SUCCESS in 3m 19s\n- requirements-tox-babel http://logs.openstack.org/85/664385/1/check/requirements-tox-babel/c89deeb/ : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py27-check-uc/df4fa16/ : SUCCESS in 7m 16s\n- requirements-tox-py36-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py36-check-uc/4de2961/ : SUCCESS in 7m 10s\n- cross-cinder-py27 http://logs.openstack.org/85/664385/1/check/cross-cinder-py27/5e4b50f/ : SUCCESS in 10m 15s\n- cross-cinder-py36 http://logs.openstack.org/85/664385/1/check/cross-cinder-py36/28e36a7/ : SUCCESS in 9m 54s\n- cross-glance-py27 http://logs.openstack.org/85/664385/1/check/cross-glance-py27/f047389/ : SUCCESS in 4m 28s\n- cross-glance-py36 http://logs.openstack.org/85/664385/1/check/cross-glance-py36/a215e5e/ : SUCCESS in 4m 14s\n- cross-horizon-py27 http://logs.openstack.org/85/664385/1/check/cross-horizon-py27/7edc1b5/ : SUCCESS in 11m 45s\n- cross-horizon-py36 http://logs.openstack.org/85/664385/1/check/cross-horizon-py36/bfcd729/ : SUCCESS in 9m 00s\n- cross-horizon-npm http://logs.openstack.org/85/664385/1/check/cross-horizon-npm/c692478/ : SUCCESS in 8m 16s\n- cross-keystone-py27 http://logs.openstack.org/85/664385/1/check/cross-keystone-py27/3529b58/ : SUCCESS in 24m 03s\n- cross-keystone-py36 http://logs.openstack.org/85/664385/1/check/cross-keystone-py36/9be2564/ : SUCCESS in 24m 05s\n- cross-neutron-py27 http://logs.openstack.org/85/664385/1/check/cross-neutron-py27/96fecb9/ : SUCCESS in 20m 57s\n- cross-neutron-py36 http://logs.openstack.org/85/664385/1/check/cross-neutron-py36/874366b/ : SUCCESS in 22m 16s\n- cross-nova-py27 http://logs.openstack.org/85/664385/1/check/cross-nova-py27/a18497b/ : FAILURE in 14m 31s\n- cross-nova-py36 http://logs.openstack.org/85/664385/1/check/cross-nova-py36/8842a62/ : SUCCESS in 11m 47s\n- cross-nova-functional http://logs.openstack.org/85/664385/1/check/cross-nova-functional/6e62dc5/ : SUCCESS in 17m 23s\n- cross-osc-tox-docs http://logs.openstack.org/85/664385/1/check/cross-osc-tox-docs/dddc6b8/html/ : SUCCESS in 5m 34s\n- cross-swift-py27 http://logs.openstack.org/85/664385/1/check/cross-swift-py27/fdc86a1/ : SUCCESS in 7m 51s\n- cross-swift-py36 http://logs.openstack.org/85/664385/1/check/cross-swift-py36/7781b7b/ : SUCCESS in 9m 27s\n- requirements-integration http://logs.openstack.org/85/664385/1/check/requirements-integration/be4cd38/ : SUCCESS in 55m 52s\n- neutron-grenade-multinode http://logs.openstack.org/85/664385/1/check/neutron-grenade-multinode/b8b95dd/ : SUCCESS in 1h 01m 01s\n- tempest-full http://logs.openstack.org/85/664385/1/check/tempest-full/dcf4b1d/ : SUCCESS in 1h 40m 52s\n- neutron-grenade http://logs.openstack.org/85/664385/1/check/neutron-grenade/5e53378/ : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":1},{"id":"840fa511badf84ac7126e70511ce18d73b17f100","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-11 14:21:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d813a50c624c18784c0e2c7fc8e2ac31b498836","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:00:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/85/664385/1/check/openstack-tox-docs/2806b5a/html/ : SUCCESS in 3m 30s\n- openstack-tox-pypy http://logs.openstack.org/85/664385/1/check/openstack-tox-pypy/e14b19c/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/664385/1/check/openstack-tox-pep8/1a8eda9/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/85/664385/1/check/openstack-tox-py27/2a6e915/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/85/664385/1/check/openstack-tox-py36/340bbba/ : SUCCESS in 3m 23s\n- openstack-tox-validate http://logs.openstack.org/85/664385/1/check/openstack-tox-validate/b1653ee/ : SUCCESS in 3m 08s\n- requirements-tox-babel http://logs.openstack.org/85/664385/1/check/requirements-tox-babel/5deec8a/ : SUCCESS in 3m 32s\n- requirements-tox-py27-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py27-check-uc/740003c/ : SUCCESS in 7m 40s\n- requirements-tox-py36-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py36-check-uc/4fa5329/ : SUCCESS in 6m 50s\n- cross-cinder-py27 http://logs.openstack.org/85/664385/1/check/cross-cinder-py27/80e15a2/ : SUCCESS in 8m 30s\n- cross-cinder-py36 http://logs.openstack.org/85/664385/1/check/cross-cinder-py36/82bec83/ : SUCCESS in 10m 44s\n- cross-glance-py27 http://logs.openstack.org/85/664385/1/check/cross-glance-py27/fdb394d/ : SUCCESS in 5m 32s\n- cross-glance-py36 http://logs.openstack.org/85/664385/1/check/cross-glance-py36/c58cd45/ : SUCCESS in 4m 17s\n- cross-horizon-py27 http://logs.openstack.org/85/664385/1/check/cross-horizon-py27/de73141/ : SUCCESS in 10m 19s\n- cross-horizon-py36 http://logs.openstack.org/85/664385/1/check/cross-horizon-py36/9ec50ec/ : SUCCESS in 8m 48s\n- cross-horizon-npm http://logs.openstack.org/85/664385/1/check/cross-horizon-npm/1375b89/ : SUCCESS in 8m 51s\n- cross-keystone-py27 http://logs.openstack.org/85/664385/1/check/cross-keystone-py27/feacdcb/ : SUCCESS in 25m 32s\n- cross-keystone-py36 http://logs.openstack.org/85/664385/1/check/cross-keystone-py36/8975ad6/ : SUCCESS in 22m 52s\n- cross-neutron-py27 http://logs.openstack.org/85/664385/1/check/cross-neutron-py27/af1c861/ : SUCCESS in 29m 13s\n- cross-neutron-py36 http://logs.openstack.org/85/664385/1/check/cross-neutron-py36/c02c31f/ : RETRY_LIMIT in 2m 04s\n- cross-nova-py27 http://logs.openstack.org/85/664385/1/check/cross-nova-py27/8b9bd66/ : RETRY_LIMIT in 1m 27s\n- cross-nova-py36 http://logs.openstack.org/85/664385/1/check/cross-nova-py36/ab67f41/ : SUCCESS in 23m 22s\n- cross-nova-functional http://logs.openstack.org/85/664385/1/check/cross-nova-functional/cd86b09/ : SUCCESS in 20m 47s\n- cross-osc-tox-docs http://logs.openstack.org/85/664385/1/check/cross-osc-tox-docs/1cd7f16/html/ : SUCCESS in 5m 28s\n- cross-swift-py27 http://logs.openstack.org/85/664385/1/check/cross-swift-py27/a55f151/ : SUCCESS in 7m 53s\n- cross-swift-py36 http://logs.openstack.org/85/664385/1/check/cross-swift-py36/527fbb1/ : SUCCESS in 8m 31s\n- requirements-integration http://logs.openstack.org/85/664385/1/check/requirements-integration/bfccf02/ : SUCCESS in 53m 49s\n- neutron-grenade-multinode http://logs.openstack.org/85/664385/1/check/neutron-grenade-multinode/208084e/ : SUCCESS in 1h 16m 42s\n- tempest-full http://logs.openstack.org/85/664385/1/check/tempest-full/ed88a0c/ : SUCCESS in 1h 45m 14s\n- neutron-grenade http://logs.openstack.org/85/664385/1/check/neutron-grenade/2037d9a/ : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"632929c4b4a7d1fafa621ecbb0fbdcee55f0a85e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 01:20:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2712d9a2e1c883d42bdfe3041eb7e736e9b4d460","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 12:21:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/85/664385/1/check/openstack-tox-docs/61fcecf/html/ : SUCCESS in 3m 44s\n- openstack-tox-pypy http://logs.openstack.org/85/664385/1/check/openstack-tox-pypy/891cd87/ : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/664385/1/check/openstack-tox-pep8/28298bd/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/85/664385/1/check/openstack-tox-py27/2549134/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/85/664385/1/check/openstack-tox-py36/07801ec/ : SUCCESS in 4m 38s\n- openstack-tox-validate http://logs.openstack.org/85/664385/1/check/openstack-tox-validate/2a6a1a4/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/85/664385/1/check/requirements-tox-babel/f5a5d69/ : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py27-check-uc/9030661/ : SUCCESS in 6m 24s\n- requirements-tox-py36-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py36-check-uc/864d6e1/ : SUCCESS in 7m 16s\n- cross-cinder-py27 http://logs.openstack.org/85/664385/1/check/cross-cinder-py27/64e982e/ : SUCCESS in 8m 29s\n- cross-cinder-py36 http://logs.openstack.org/85/664385/1/check/cross-cinder-py36/018da9e/ : SUCCESS in 8m 25s\n- cross-glance-py27 http://logs.openstack.org/85/664385/1/check/cross-glance-py27/0dabd0e/ : SUCCESS in 4m 27s\n- cross-glance-py36 http://logs.openstack.org/85/664385/1/check/cross-glance-py36/73c8355/ : SUCCESS in 6m 18s\n- cross-horizon-py27 http://logs.openstack.org/85/664385/1/check/cross-horizon-py27/006f6bf/ : SUCCESS in 10m 33s\n- cross-horizon-py36 http://logs.openstack.org/85/664385/1/check/cross-horizon-py36/68fba74/ : SUCCESS in 8m 58s\n- cross-horizon-npm http://logs.openstack.org/85/664385/1/check/cross-horizon-npm/87f9112/ : SUCCESS in 8m 07s\n- cross-keystone-py27 http://logs.openstack.org/85/664385/1/check/cross-keystone-py27/decd2df/ : SUCCESS in 20m 54s\n- cross-keystone-py36 http://logs.openstack.org/85/664385/1/check/cross-keystone-py36/9bdc245/ : SUCCESS in 24m 34s\n- cross-neutron-py27 http://logs.openstack.org/85/664385/1/check/cross-neutron-py27/465029e/ : SUCCESS in 25m 59s\n- cross-neutron-py36 http://logs.openstack.org/85/664385/1/check/cross-neutron-py36/67068bd/ : SUCCESS in 25m 59s\n- cross-nova-py27 http://logs.openstack.org/85/664385/1/check/cross-nova-py27/c4d77d3/ : SUCCESS in 13m 14s\n- cross-nova-py36 http://logs.openstack.org/85/664385/1/check/cross-nova-py36/38e6866/ : SUCCESS in 11m 42s\n- cross-nova-functional http://logs.openstack.org/85/664385/1/check/cross-nova-functional/0980ff0/ : SUCCESS in 17m 58s\n- cross-osc-tox-docs http://logs.openstack.org/85/664385/1/check/cross-osc-tox-docs/5b19083/html/ : SUCCESS in 5m 05s\n- cross-swift-py27 http://logs.openstack.org/85/664385/1/check/cross-swift-py27/d0ca07f/ : SUCCESS in 7m 20s\n- cross-swift-py36 http://logs.openstack.org/85/664385/1/check/cross-swift-py36/7cdafc9/ : SUCCESS in 8m 22s\n- requirements-integration http://logs.openstack.org/85/664385/1/check/requirements-integration/ab68e9e/ : SUCCESS in 1h 04m 35s\n- neutron-grenade-multinode http://logs.openstack.org/85/664385/1/check/neutron-grenade-multinode/ba7be12/ : SUCCESS in 1h 07m 55s\n- tempest-full http://logs.openstack.org/85/664385/1/check/tempest-full/7ae97b4/ : SUCCESS in 1h 46m 02s\n- neutron-grenade http://logs.openstack.org/85/664385/1/check/neutron-grenade/ea7f9a4/ : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4f8fefde3cdfb5199621d9e8d6d3cd27fb8cb859","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 12:21:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7f6caad6f5b55c0b2ca869df8e4c29cd1e6b6f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 16:14:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/85/664385/1/gate/openstack-tox-docs/9a7b32e/html/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/85/664385/1/gate/openstack-tox-pep8/77d1ab8/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/85/664385/1/gate/openstack-tox-py27/fdfa68e/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/85/664385/1/gate/openstack-tox-py36/cb7b511/ : SUCCESS in 3m 16s\n- openstack-tox-validate http://logs.openstack.org/85/664385/1/gate/openstack-tox-validate/4863aa1/ : SUCCESS in 3m 25s\n- requirements-tox-babel http://logs.openstack.org/85/664385/1/gate/requirements-tox-babel/77bdc5f/ : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc http://logs.openstack.org/85/664385/1/gate/requirements-tox-py27-check-uc/d8eb120/ : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/85/664385/1/gate/requirements-tox-py36-check-uc/5d67fa4/ : SUCCESS in 7m 07s\n- cross-cinder-py27 http://logs.openstack.org/85/664385/1/gate/cross-cinder-py27/fdf0e06/ : SUCCESS in 8m 42s\n- cross-cinder-py36 http://logs.openstack.org/85/664385/1/gate/cross-cinder-py36/0f26fc2/ : SUCCESS in 9m 28s\n- cross-glance-py27 http://logs.openstack.org/85/664385/1/gate/cross-glance-py27/937b2b7/ : SUCCESS in 4m 19s\n- cross-glance-py36 http://logs.openstack.org/85/664385/1/gate/cross-glance-py36/6da0a8b/ : SUCCESS in 4m 12s\n- cross-horizon-py27 http://logs.openstack.org/85/664385/1/gate/cross-horizon-py27/0d424c3/ : SUCCESS in 9m 44s\n- cross-horizon-py36 http://logs.openstack.org/85/664385/1/gate/cross-horizon-py36/71cd23e/ : SUCCESS in 8m 52s\n- cross-horizon-npm http://logs.openstack.org/85/664385/1/gate/cross-horizon-npm/e8e95a0/ : SUCCESS in 6m 19s\n- cross-keystone-py27 http://logs.openstack.org/85/664385/1/gate/cross-keystone-py27/f3b0ee2/ : SUCCESS in 20m 05s\n- cross-keystone-py36 http://logs.openstack.org/85/664385/1/gate/cross-keystone-py36/952f019/ : SUCCESS in 30m 02s\n- cross-neutron-py27 http://logs.openstack.org/85/664385/1/gate/cross-neutron-py27/263a871/ : SUCCESS in 31m 31s\n- cross-neutron-py36 http://logs.openstack.org/85/664385/1/gate/cross-neutron-py36/2319c85/ : SUCCESS in 22m 27s\n- cross-nova-py27 http://logs.openstack.org/85/664385/1/gate/cross-nova-py27/e503975/ : SUCCESS in 11m 56s\n- cross-nova-py36 http://logs.openstack.org/85/664385/1/gate/cross-nova-py36/21b9181/ : SUCCESS in 13m 10s\n- cross-nova-functional http://logs.openstack.org/85/664385/1/gate/cross-nova-functional/3c0a4f0/ : SUCCESS in 17m 19s\n- cross-osc-tox-docs http://logs.openstack.org/85/664385/1/gate/cross-osc-tox-docs/e1effb1/html/ : SUCCESS in 6m 42s\n- cross-swift-py27 http://logs.openstack.org/85/664385/1/gate/cross-swift-py27/ddb85a9/ : SUCCESS in 20m 53s\n- cross-swift-py36 http://logs.openstack.org/85/664385/1/gate/cross-swift-py36/efa3d27/ : SUCCESS in 8m 17s\n- requirements-integration http://logs.openstack.org/85/664385/1/gate/requirements-integration/130fcb5/ : SUCCESS in 53m 46s\n- neutron-grenade-multinode http://logs.openstack.org/85/664385/1/gate/neutron-grenade-multinode/d92ef6e/ : FAILURE in 14m 00s\n- tempest-full http://logs.openstack.org/85/664385/1/gate/tempest-full/bf8546d/ : SUCCESS in 1h 50m 47s\n- neutron-grenade http://logs.openstack.org/85/664385/1/gate/neutron-grenade/a5ecbd8/ : SUCCESS in 1h 16m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7de765633c4b2a93abd06da256c9d007cf0ce51e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-12 16:21:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c7a9932252910ef33c411f508662e3a893db9ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 01:11:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/85/664385/1/check/openstack-tox-docs/1a9471f/html/ : SUCCESS in 3m 02s\n- openstack-tox-pypy http://logs.openstack.org/85/664385/1/check/openstack-tox-pypy/c9f819a/ : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/664385/1/check/openstack-tox-pep8/833f100/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/85/664385/1/check/openstack-tox-py27/b1911ca/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/85/664385/1/check/openstack-tox-py36/84abaf5/ : SUCCESS in 4m 49s\n- openstack-tox-validate http://logs.openstack.org/85/664385/1/check/openstack-tox-validate/af7ffa3/ : SUCCESS in 3m 17s\n- requirements-tox-babel http://logs.openstack.org/85/664385/1/check/requirements-tox-babel/61791f6/ : SUCCESS in 4m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py27-check-uc/86f98b2/ : SUCCESS in 9m 15s\n- requirements-tox-py36-check-uc http://logs.openstack.org/85/664385/1/check/requirements-tox-py36-check-uc/e6577d3/ : SUCCESS in 8m 27s\n- cross-cinder-py27 http://logs.openstack.org/85/664385/1/check/cross-cinder-py27/5197591/ : SUCCESS in 9m 01s\n- cross-cinder-py36 http://logs.openstack.org/85/664385/1/check/cross-cinder-py36/ea17179/ : SUCCESS in 10m 07s\n- cross-glance-py27 http://logs.openstack.org/85/664385/1/check/cross-glance-py27/dc9725c/ : SUCCESS in 6m 27s\n- cross-glance-py36 http://logs.openstack.org/85/664385/1/check/cross-glance-py36/4d928f9/ : SUCCESS in 4m 06s\n- cross-horizon-py27 http://logs.openstack.org/85/664385/1/check/cross-horizon-py27/b6b80c8/ : SUCCESS in 10m 38s\n- cross-horizon-py36 http://logs.openstack.org/85/664385/1/check/cross-horizon-py36/c03a592/ : SUCCESS in 8m 48s\n- cross-horizon-npm http://logs.openstack.org/85/664385/1/check/cross-horizon-npm/acfda8d/ : SUCCESS in 5m 39s\n- cross-keystone-py27 http://logs.openstack.org/85/664385/1/check/cross-keystone-py27/bc0b636/ : SUCCESS in 32m 00s\n- cross-keystone-py36 http://logs.openstack.org/85/664385/1/check/cross-keystone-py36/91ca7f7/ : SUCCESS in 24m 14s\n- cross-neutron-py27 http://logs.openstack.org/85/664385/1/check/cross-neutron-py27/f6e85ee/ : SUCCESS in 22m 16s\n- cross-neutron-py36 http://logs.openstack.org/85/664385/1/check/cross-neutron-py36/15e5796/ : SUCCESS in 19m 43s\n- cross-nova-py27 http://logs.openstack.org/85/664385/1/check/cross-nova-py27/c453d3d/ : SUCCESS in 12m 22s\n- cross-nova-py36 http://logs.openstack.org/85/664385/1/check/cross-nova-py36/75aef84/ : SUCCESS in 13m 23s\n- cross-nova-functional http://logs.openstack.org/85/664385/1/check/cross-nova-functional/0c3cc86/ : SUCCESS in 20m 36s\n- cross-osc-tox-docs http://logs.openstack.org/85/664385/1/check/cross-osc-tox-docs/a14572f/html/ : SUCCESS in 5m 39s\n- cross-swift-py27 http://logs.openstack.org/85/664385/1/check/cross-swift-py27/80d11e3/ : SUCCESS in 7m 38s\n- cross-swift-py36 http://logs.openstack.org/85/664385/1/check/cross-swift-py36/cc23abd/ : SUCCESS in 8m 36s\n- requirements-integration http://logs.openstack.org/85/664385/1/check/requirements-integration/602ffdf/ : SUCCESS in 1h 06m 34s\n- neutron-grenade-multinode http://logs.openstack.org/85/664385/1/check/neutron-grenade-multinode/ebb4fb0/ : SUCCESS in 1h 14m 39s\n- tempest-full http://logs.openstack.org/85/664385/1/check/tempest-full/8f18120/ : SUCCESS in 1h 44m 06s\n- neutron-grenade http://logs.openstack.org/85/664385/1/check/neutron-grenade/cc72b3e/ : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0b00ea563576fd2d3f1a8359cae81726bf0d0720","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 01:12:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b70994381992ead2006a608bcb5edef375890898","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 05:56:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/85/664385/1/gate/openstack-tox-docs/d9f952f/html/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/85/664385/1/gate/openstack-tox-pep8/fda3c8c/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/85/664385/1/gate/openstack-tox-py27/9f9d9b5/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/85/664385/1/gate/openstack-tox-py36/cad121e/ : SUCCESS in 3m 42s\n- openstack-tox-validate http://logs.openstack.org/85/664385/1/gate/openstack-tox-validate/761adf2/ : SUCCESS in 4m 35s\n- requirements-tox-babel http://logs.openstack.org/85/664385/1/gate/requirements-tox-babel/2cae7ed/ : SUCCESS in 3m 39s\n- requirements-tox-py27-check-uc http://logs.openstack.org/85/664385/1/gate/requirements-tox-py27-check-uc/a8cde22/ : SUCCESS in 7m 47s\n- requirements-tox-py36-check-uc http://logs.openstack.org/85/664385/1/gate/requirements-tox-py36-check-uc/3ab728d/ : SUCCESS in 7m 25s\n- cross-cinder-py27 http://logs.openstack.org/85/664385/1/gate/cross-cinder-py27/602c016/ : SUCCESS in 9m 46s\n- cross-cinder-py36 http://logs.openstack.org/85/664385/1/gate/cross-cinder-py36/7f14d2c/ : SUCCESS in 8m 11s\n- cross-glance-py27 http://logs.openstack.org/85/664385/1/gate/cross-glance-py27/dff1491/ : SUCCESS in 4m 21s\n- cross-glance-py36 http://logs.openstack.org/85/664385/1/gate/cross-glance-py36/f3021a9/ : SUCCESS in 6m 30s\n- cross-horizon-py27 http://logs.openstack.org/85/664385/1/gate/cross-horizon-py27/6daaf6f/ : SUCCESS in 11m 30s\n- cross-horizon-py36 http://logs.openstack.org/85/664385/1/gate/cross-horizon-py36/f18009c/ : SUCCESS in 10m 05s\n- cross-horizon-npm http://logs.openstack.org/85/664385/1/gate/cross-horizon-npm/421b1e4/ : SUCCESS in 7m 01s\n- cross-keystone-py27 http://logs.openstack.org/85/664385/1/gate/cross-keystone-py27/81db727/ : SUCCESS in 21m 01s\n- cross-keystone-py36 http://logs.openstack.org/85/664385/1/gate/cross-keystone-py36/b7ba33f/ : SUCCESS in 25m 09s\n- cross-neutron-py27 http://logs.openstack.org/85/664385/1/gate/cross-neutron-py27/b160869/ : SUCCESS in 24m 19s\n- cross-neutron-py36 http://logs.openstack.org/85/664385/1/gate/cross-neutron-py36/66cb866/ : SUCCESS in 24m 22s\n- cross-nova-py27 http://logs.openstack.org/85/664385/1/gate/cross-nova-py27/bbdc3b5/ : SUCCESS in 10m 51s\n- cross-nova-py36 http://logs.openstack.org/85/664385/1/gate/cross-nova-py36/c894ac7/ : SUCCESS in 11m 00s\n- cross-nova-functional http://logs.openstack.org/85/664385/1/gate/cross-nova-functional/3ecea8e/ : SUCCESS in 19m 13s\n- cross-osc-tox-docs http://logs.openstack.org/85/664385/1/gate/cross-osc-tox-docs/0f20f3e/html/ : SUCCESS in 5m 41s\n- cross-swift-py27 http://logs.openstack.org/85/664385/1/gate/cross-swift-py27/bb52628/ : SUCCESS in 8m 44s\n- cross-swift-py36 http://logs.openstack.org/85/664385/1/gate/cross-swift-py36/c2e3e3d/ : SUCCESS in 8m 10s\n- requirements-integration http://logs.openstack.org/85/664385/1/gate/requirements-integration/017fe25/ : SUCCESS in 53m 47s\n- neutron-grenade-multinode http://logs.openstack.org/85/664385/1/gate/neutron-grenade-multinode/6eaef60/ : SUCCESS in 1h 05m 14s\n- tempest-full http://logs.openstack.org/85/664385/1/gate/tempest-full/5cfc5e6/ : SUCCESS in 1h 46m 10s\n- neutron-grenade http://logs.openstack.org/85/664385/1/gate/neutron-grenade/c3eaeca/ : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":1},{"id":"006bc0b3bcfeb7e0d067a8f6c0016c745e6dcb2c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 05:56:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f360fa1ce8bb58bc7fb60957990e1e19fa82d78","revisions":{"9f360fa1ce8bb58bc7fb60957990e1e19fa82d78":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:54:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/664385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/664385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/664385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/664385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/664385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/664385/1"}}},"commit":{"parents":[{"commit":"98a5e8f0b47309e23edd429cb85704413876ce6d","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98a5e8f0b47309e23edd429cb85704413876ce6d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:54:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:54:00.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 1.2.0","message":"update constraint for octavia-lib to new release 1.2.0\n\nChange-Id: I88da0ac7a8fc0d91f22310c5908c5786a63e6ae3\nmeta:version: 1.2.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Adam Harwell \u003cflux.adam@gmail.com\u003e\nmeta:release:Commit: Adam Harwell \u003cflux.adam@gmail.com\u003e\nmeta:release:Change-Id: I65d8a000f58c094874230b282a4b764d178f2ef0\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f360fa1ce8bb58bc7fb60957990e1e19fa82d78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f360fa1ce8bb58bc7fb60957990e1e19fa82d78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
