)]}'
{"id":"openstack%2Frequirements~665109","triplet_id":"openstack%2Frequirements~stable%2Fstein~I67aac41d32ca9c503c1cc1e59832f06c4cd3ccf4","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I67aac41d32ca9c503c1cc1e59832f06c4cd3ccf4","subject":"update constraint for python-novaclient to new release 13.0.1","status":"MERGED","created":"2019-06-13 10:56:55.000000000","updated":"2019-06-15 12:09:19.000000000","submitted":"2019-06-15 12:09:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665109-1560600559581-9b2c7f29","meta_rev_id":"1ec45e7b37106eb8968e4ca17751e4a66a46d563","_number":665109,"virtual_id_number":665109,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-15 12:09:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-15 00:27:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-15 00:27:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-15 00:27:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-15 12:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0ef25cb59a6ab60133e175aa002958024046241","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-13 10:56:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"309cbb6e7b576e7af19341e3c00035eaf61c4a45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 22:53:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/09/665109/1/check/openstack-tox-pypy/b278703/ : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/09/665109/1/check/openstack-tox-pep8/3585ac2/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/09/665109/1/check/openstack-tox-py27/ce16ec4/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/09/665109/1/check/openstack-tox-py35/66fe054/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/09/665109/1/check/openstack-tox-py36/d6b8e5e/ : SUCCESS in 3m 03s\n- openstack-tox-validate http://logs.openstack.org/09/665109/1/check/openstack-tox-validate/46738e7/ : SUCCESS in 3m 30s\n- requirements-tox-babel http://logs.openstack.org/09/665109/1/check/requirements-tox-babel/5cf8ce4/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py27-check-uc/7336d43/ : SUCCESS in 7m 13s\n- requirements-tox-py35-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py35-check-uc/4df5c7a/ : SUCCESS in 8m 51s\n- requirements-tox-py36-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py36-check-uc/c85752f/ : SUCCESS in 6m 26s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/09/665109/1/check/cross-cinder-py27/59dd312/ : SUCCESS in 9m 42s\n- cross-cinder-py35 http://logs.openstack.org/09/665109/1/check/cross-cinder-py35/ce0beee/ : SUCCESS in 9m 25s\n- cross-glance-py27 http://logs.openstack.org/09/665109/1/check/cross-glance-py27/933566f/ : SUCCESS in 4m 43s\n- cross-glance-py35 http://logs.openstack.org/09/665109/1/check/cross-glance-py35/8ae86c0/ : SUCCESS in 4m 26s\n- cross-horizon-py27 http://logs.openstack.org/09/665109/1/check/cross-horizon-py27/a74b0dd/ : SUCCESS in 10m 59s\n- cross-horizon-py35 http://logs.openstack.org/09/665109/1/check/cross-horizon-py35/334ab8e/ : SUCCESS in 10m 34s\n- cross-horizon-npm http://logs.openstack.org/09/665109/1/check/cross-horizon-npm/a556565/ : SUCCESS in 8m 31s\n- cross-keystone-py27 http://logs.openstack.org/09/665109/1/check/cross-keystone-py27/a90d4cf/ : SUCCESS in 23m 30s\n- cross-keystone-py35 http://logs.openstack.org/09/665109/1/check/cross-keystone-py35/d0f4e2a/ : SUCCESS in 21m 53s\n- cross-neutron-py27 http://logs.openstack.org/09/665109/1/check/cross-neutron-py27/0536ddd/ : SUCCESS in 30m 06s\n- cross-neutron-py35 http://logs.openstack.org/09/665109/1/check/cross-neutron-py35/1bf3d60/ : SUCCESS in 25m 01s\n- cross-nova-py27 http://logs.openstack.org/09/665109/1/check/cross-nova-py27/161523a/ : FAILURE in 12m 07s\n- cross-nova-py35 http://logs.openstack.org/09/665109/1/check/cross-nova-py35/868194f/ : SUCCESS in 12m 17s\n- cross-nova-functional http://logs.openstack.org/09/665109/1/check/cross-nova-functional/54fd295/ : SUCCESS in 18m 40s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/09/665109/1/check/cross-osc-build-sphinx-docs/559f4e1/html/ : SUCCESS in 5m 59s\n- cross-swift-py27 http://logs.openstack.org/09/665109/1/check/cross-swift-py27/bc055fc/ : SUCCESS in 8m 21s\n- cross-swift-py35 http://logs.openstack.org/09/665109/1/check/cross-swift-py35/68f32ae/ : SUCCESS in 8m 24s\n- requirements-integration http://logs.openstack.org/09/665109/1/check/requirements-integration/d0b34f8/ : SUCCESS in 54m 04s\n- neutron-grenade-multinode http://logs.openstack.org/09/665109/1/check/neutron-grenade-multinode/741a8fd/ : SUCCESS in 1h 02m 56s\n- tempest-full http://logs.openstack.org/09/665109/1/check/tempest-full/10ddd90/ : SUCCESS in 1h 32m 36s\n- neutron-grenade http://logs.openstack.org/09/665109/1/check/neutron-grenade/c81bcf1/ : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bbf6882dd9edea710c96232be3d750b947c76c8b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-14 02:35:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"033cd0172d7c75535c059af7195680171c9ede89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 09:43:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/09/665109/1/check/openstack-tox-pypy/30dd7bf/ : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/09/665109/1/check/openstack-tox-pep8/d0083a4/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/09/665109/1/check/openstack-tox-py27/f654450/ : SUCCESS in 3m 23s\n- openstack-tox-py35 http://logs.openstack.org/09/665109/1/check/openstack-tox-py35/820a665/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/09/665109/1/check/openstack-tox-py36/277216b/ : SUCCESS in 3m 44s\n- openstack-tox-validate http://logs.openstack.org/09/665109/1/check/openstack-tox-validate/9fd8c66/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/09/665109/1/check/requirements-tox-babel/34a65f9/ : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py27-check-uc/3953ebd/ : SUCCESS in 6m 09s\n- requirements-tox-py35-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py35-check-uc/87f73fe/ : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py36-check-uc/c24b944/ : SUCCESS in 6m 54s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/09/665109/1/check/cross-cinder-py27/1630dee/ : SUCCESS in 8m 30s\n- cross-cinder-py35 http://logs.openstack.org/09/665109/1/check/cross-cinder-py35/71556ed/ : SUCCESS in 8m 57s\n- cross-glance-py27 http://logs.openstack.org/09/665109/1/check/cross-glance-py27/334f4ef/ : SUCCESS in 4m 21s\n- cross-glance-py35 http://logs.openstack.org/09/665109/1/check/cross-glance-py35/07873be/ : SUCCESS in 4m 17s\n- cross-horizon-py27 http://logs.openstack.org/09/665109/1/check/cross-horizon-py27/481d217/ : SUCCESS in 9m 46s\n- cross-horizon-py35 http://logs.openstack.org/09/665109/1/check/cross-horizon-py35/393be43/ : SUCCESS in 9m 11s\n- cross-horizon-npm http://logs.openstack.org/09/665109/1/check/cross-horizon-npm/8401e63/ : SUCCESS in 8m 50s\n- cross-keystone-py27 http://logs.openstack.org/09/665109/1/check/cross-keystone-py27/51b3f07/ : SUCCESS in 19m 19s\n- cross-keystone-py35 http://logs.openstack.org/09/665109/1/check/cross-keystone-py35/b2797e2/ : SUCCESS in 27m 50s\n- cross-neutron-py27 http://logs.openstack.org/09/665109/1/check/cross-neutron-py27/d17b6cb/ : SUCCESS in 23m 44s\n- cross-neutron-py35 http://logs.openstack.org/09/665109/1/check/cross-neutron-py35/1ffe0d4/ : SUCCESS in 22m 16s\n- cross-nova-py27 http://logs.openstack.org/09/665109/1/check/cross-nova-py27/a2cd09d/ : FAILURE in 13m 29s\n- cross-nova-py35 http://logs.openstack.org/09/665109/1/check/cross-nova-py35/f3c8d73/ : SUCCESS in 12m 47s\n- cross-nova-functional http://logs.openstack.org/09/665109/1/check/cross-nova-functional/6c96a9c/ : SUCCESS in 19m 30s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/09/665109/1/check/cross-osc-build-sphinx-docs/b42aadb/html/ : SUCCESS in 5m 53s\n- cross-swift-py27 http://logs.openstack.org/09/665109/1/check/cross-swift-py27/ba21e9a/ : SUCCESS in 7m 20s\n- cross-swift-py35 http://logs.openstack.org/09/665109/1/check/cross-swift-py35/d6b97d5/ : SUCCESS in 7m 23s\n- requirements-integration http://logs.openstack.org/09/665109/1/check/requirements-integration/06343b1/ : SUCCESS in 56m 20s\n- neutron-grenade-multinode http://logs.openstack.org/09/665109/1/check/neutron-grenade-multinode/4a18e9e/ : SUCCESS in 1h 09m 16s\n- tempest-full http://logs.openstack.org/09/665109/1/check/tempest-full/757f399/ : SUCCESS in 1h 55m 25s\n- neutron-grenade http://logs.openstack.org/09/665109/1/check/neutron-grenade/7eab838/ : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0ed35bccf79178b3a1b002e088fc50b3c429869f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-14 14:56:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75a276a1d1142ee28e129fb3b3144e151e8caf5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 00:14:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/09/665109/1/check/openstack-tox-pypy/cb9beaf/ : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/09/665109/1/check/openstack-tox-pep8/847b34a/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/09/665109/1/check/openstack-tox-py27/09e95b4/ : SUCCESS in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/09/665109/1/check/openstack-tox-py35/407764d/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/09/665109/1/check/openstack-tox-py36/528766c/ : SUCCESS in 3m 29s\n- openstack-tox-validate http://logs.openstack.org/09/665109/1/check/openstack-tox-validate/a41886f/ : SUCCESS in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/09/665109/1/check/requirements-tox-babel/a138fa7/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py27-check-uc/894706f/ : SUCCESS in 6m 38s\n- requirements-tox-py35-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py35-check-uc/42dbcea/ : SUCCESS in 7m 48s\n- requirements-tox-py36-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py36-check-uc/dca8c90/ : SUCCESS in 7m 51s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/09/665109/1/check/cross-cinder-py27/c1f2cb3/ : SUCCESS in 8m 38s\n- cross-cinder-py35 http://logs.openstack.org/09/665109/1/check/cross-cinder-py35/4cb7d18/ : SUCCESS in 10m 49s\n- cross-glance-py27 http://logs.openstack.org/09/665109/1/check/cross-glance-py27/24b7c89/ : SUCCESS in 4m 16s\n- cross-glance-py35 http://logs.openstack.org/09/665109/1/check/cross-glance-py35/86eca60/ : SUCCESS in 4m 21s\n- cross-horizon-py27 http://logs.openstack.org/09/665109/1/check/cross-horizon-py27/f304577/ : SUCCESS in 10m 52s\n- cross-horizon-py35 http://logs.openstack.org/09/665109/1/check/cross-horizon-py35/4d4daf4/ : SUCCESS in 9m 59s\n- cross-horizon-npm http://logs.openstack.org/09/665109/1/check/cross-horizon-npm/d938421/ : SUCCESS in 6m 01s\n- cross-keystone-py27 http://logs.openstack.org/09/665109/1/check/cross-keystone-py27/4ba9675/ : SUCCESS in 21m 44s\n- cross-keystone-py35 http://logs.openstack.org/09/665109/1/check/cross-keystone-py35/6e5fa39/ : SUCCESS in 29m 45s\n- cross-neutron-py27 http://logs.openstack.org/09/665109/1/check/cross-neutron-py27/91ccd7b/ : SUCCESS in 28m 17s\n- cross-neutron-py35 http://logs.openstack.org/09/665109/1/check/cross-neutron-py35/13fb81f/ : SUCCESS in 24m 29s\n- cross-nova-py27 http://logs.openstack.org/09/665109/1/check/cross-nova-py27/e1f55c1/ : SUCCESS in 14m 31s\n- cross-nova-py35 http://logs.openstack.org/09/665109/1/check/cross-nova-py35/40b7a62/ : SUCCESS in 14m 18s\n- cross-nova-functional http://logs.openstack.org/09/665109/1/check/cross-nova-functional/a0741ce/ : SUCCESS in 18m 44s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/09/665109/1/check/cross-osc-build-sphinx-docs/b1bbe31/html/ : SUCCESS in 6m 19s\n- cross-swift-py27 http://logs.openstack.org/09/665109/1/check/cross-swift-py27/bc6ce06/ : SUCCESS in 9m 56s\n- cross-swift-py35 http://logs.openstack.org/09/665109/1/check/cross-swift-py35/ac43cd7/ : SUCCESS in 7m 55s\n- requirements-integration http://logs.openstack.org/09/665109/1/check/requirements-integration/f6ec218/ : SUCCESS in 56m 05s\n- neutron-grenade-multinode http://logs.openstack.org/09/665109/1/check/neutron-grenade-multinode/2958c3c/ : SUCCESS in 1h 04m 45s\n- tempest-full http://logs.openstack.org/09/665109/1/check/tempest-full/b8520d9/ : SUCCESS in 1h 45m 13s\n- neutron-grenade http://logs.openstack.org/09/665109/1/check/neutron-grenade/9e7a293/ : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bc796cb761fa8f4322bd2c461879c18021da0891","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-15 00:27:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6573bf05c73eebaac5693cb2c970904c12378457","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 00:27:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3f4a5db9130ffa74d68543f054393a61394076b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 02:19:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/09/665109/1/gate/openstack-tox-pep8/568dc3a/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/09/665109/1/gate/openstack-tox-py27/0f22667/ : SUCCESS in 3m 33s\n- openstack-tox-py35 http://logs.openstack.org/09/665109/1/gate/openstack-tox-py35/c8fcd45/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/09/665109/1/gate/openstack-tox-py36/c08cd59/ : SUCCESS in 3m 22s\n- openstack-tox-validate http://logs.openstack.org/09/665109/1/gate/openstack-tox-validate/1479392/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/09/665109/1/gate/requirements-tox-babel/fd87247/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/665109/1/gate/requirements-tox-py27-check-uc/9b9c2e0/ : SUCCESS in 6m 46s\n- requirements-tox-py35-check-uc http://logs.openstack.org/09/665109/1/gate/requirements-tox-py35-check-uc/dfd7fec/ : SUCCESS in 7m 09s\n- cross-cinder-py27 http://logs.openstack.org/09/665109/1/gate/cross-cinder-py27/4cf4d90/ : SUCCESS in 10m 31s\n- cross-cinder-py35 http://logs.openstack.org/09/665109/1/gate/cross-cinder-py35/78b04e8/ : SUCCESS in 9m 22s\n- cross-glance-py27 http://logs.openstack.org/09/665109/1/gate/cross-glance-py27/941403b/ : SUCCESS in 4m 30s\n- cross-glance-py35 http://logs.openstack.org/09/665109/1/gate/cross-glance-py35/f1dfd5b/ : SUCCESS in 6m 25s\n- cross-horizon-py27 http://logs.openstack.org/09/665109/1/gate/cross-horizon-py27/6297e1f/ : SUCCESS in 10m 38s\n- cross-horizon-py35 http://logs.openstack.org/09/665109/1/gate/cross-horizon-py35/0553297/ : SUCCESS in 10m 03s\n- cross-horizon-npm http://logs.openstack.org/09/665109/1/gate/cross-horizon-npm/164abc7/ : SUCCESS in 6m 32s\n- cross-keystone-py27 http://logs.openstack.org/09/665109/1/gate/cross-keystone-py27/528d419/ : SUCCESS in 20m 48s\n- cross-keystone-py35 http://logs.openstack.org/09/665109/1/gate/cross-keystone-py35/7eca66d/ : SUCCESS in 26m 31s\n- cross-neutron-py27 http://logs.openstack.org/09/665109/1/gate/cross-neutron-py27/46d25e2/ : SUCCESS in 21m 45s\n- cross-neutron-py35 http://logs.openstack.org/09/665109/1/gate/cross-neutron-py35/768f512/ : SUCCESS in 25m 41s\n- cross-nova-py27 http://logs.openstack.org/09/665109/1/gate/cross-nova-py27/47daff8/ : SUCCESS in 13m 33s\n- cross-nova-py35 http://logs.openstack.org/09/665109/1/gate/cross-nova-py35/a1993ac/ : SUCCESS in 13m 42s\n- cross-nova-functional http://logs.openstack.org/09/665109/1/gate/cross-nova-functional/b205c15/ : FAILURE in 34m 28s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/09/665109/1/gate/cross-osc-build-sphinx-docs/a9ac859/html/ : SUCCESS in 7m 34s\n- cross-swift-py27 http://logs.openstack.org/09/665109/1/gate/cross-swift-py27/75cf8c7/ : SUCCESS in 7m 56s\n- cross-swift-py35 http://logs.openstack.org/09/665109/1/gate/cross-swift-py35/e36a20f/ : SUCCESS in 7m 33s\n- requirements-integration http://logs.openstack.org/09/665109/1/gate/requirements-integration/69daa9b/ : SUCCESS in 52m 30s\n- neutron-grenade-multinode http://logs.openstack.org/09/665109/1/gate/neutron-grenade-multinode/f21d349/ : SUCCESS in 1h 06m 11s\n- tempest-full http://logs.openstack.org/09/665109/1/gate/tempest-full/42a0668/ : SUCCESS in 1h 46m 02s\n- neutron-grenade http://logs.openstack.org/09/665109/1/gate/neutron-grenade/95e1688/ : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b6c4dfdc60ea8a05d77e54984bb0e835ab96b594","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-15 02:22:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e81e5aa769fb2786d5b8a1664c48b1b1a4c120e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 09:31:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/09/665109/1/check/openstack-tox-pypy/7cb03ea/ : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/09/665109/1/check/openstack-tox-pep8/9c07323/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/09/665109/1/check/openstack-tox-py27/bd633ea/ : SUCCESS in 3m 19s\n- openstack-tox-py35 http://logs.openstack.org/09/665109/1/check/openstack-tox-py35/af11cdb/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/09/665109/1/check/openstack-tox-py36/8beafd0/ : SUCCESS in 3m 28s\n- openstack-tox-validate http://logs.openstack.org/09/665109/1/check/openstack-tox-validate/13a2316/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/09/665109/1/check/requirements-tox-babel/c8090b9/ : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py27-check-uc/e039a25/ : SUCCESS in 6m 40s\n- requirements-tox-py35-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py35-check-uc/d8bc78b/ : SUCCESS in 9m 16s\n- requirements-tox-py36-check-uc http://logs.openstack.org/09/665109/1/check/requirements-tox-py36-check-uc/37f8b1a/ : SUCCESS in 7m 04s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/09/665109/1/check/cross-cinder-py27/8a9a3fb/ : SUCCESS in 9m 20s\n- cross-cinder-py35 http://logs.openstack.org/09/665109/1/check/cross-cinder-py35/b6ef3a0/ : SUCCESS in 11m 15s\n- cross-glance-py27 http://logs.openstack.org/09/665109/1/check/cross-glance-py27/3508f85/ : SUCCESS in 6m 08s\n- cross-glance-py35 http://logs.openstack.org/09/665109/1/check/cross-glance-py35/1ea7019/ : SUCCESS in 4m 29s\n- cross-horizon-py27 http://logs.openstack.org/09/665109/1/check/cross-horizon-py27/dced883/ : SUCCESS in 11m 46s\n- cross-horizon-py35 http://logs.openstack.org/09/665109/1/check/cross-horizon-py35/955a60f/ : SUCCESS in 10m 16s\n- cross-horizon-npm http://logs.openstack.org/09/665109/1/check/cross-horizon-npm/0a09d94/ : SUCCESS in 6m 30s\n- cross-keystone-py27 http://logs.openstack.org/09/665109/1/check/cross-keystone-py27/ffa12fc/ : SUCCESS in 22m 50s\n- cross-keystone-py35 http://logs.openstack.org/09/665109/1/check/cross-keystone-py35/0fa3026/ : SUCCESS in 25m 40s\n- cross-neutron-py27 http://logs.openstack.org/09/665109/1/check/cross-neutron-py27/7745399/ : SUCCESS in 21m 53s\n- cross-neutron-py35 http://logs.openstack.org/09/665109/1/check/cross-neutron-py35/7f763f0/ : SUCCESS in 27m 25s\n- cross-nova-py27 http://logs.openstack.org/09/665109/1/check/cross-nova-py27/850e36e/ : SUCCESS in 13m 24s\n- cross-nova-py35 http://logs.openstack.org/09/665109/1/check/cross-nova-py35/a705e7b/ : SUCCESS in 14m 51s\n- cross-nova-functional http://logs.openstack.org/09/665109/1/check/cross-nova-functional/474c4b5/ : SUCCESS in 18m 07s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/09/665109/1/check/cross-osc-build-sphinx-docs/9f95a05/html/ : SUCCESS in 5m 31s\n- cross-swift-py27 http://logs.openstack.org/09/665109/1/check/cross-swift-py27/8aa4a5e/ : SUCCESS in 8m 25s\n- cross-swift-py35 http://logs.openstack.org/09/665109/1/check/cross-swift-py35/984d54d/ : SUCCESS in 11m 43s\n- requirements-integration http://logs.openstack.org/09/665109/1/check/requirements-integration/f87ab06/ : SUCCESS in 48m 49s\n- neutron-grenade-multinode http://logs.openstack.org/09/665109/1/check/neutron-grenade-multinode/7200b7d/ : SUCCESS in 1h 09m 18s\n- tempest-full http://logs.openstack.org/09/665109/1/check/tempest-full/c031f54/ : SUCCESS in 1h 44m 31s\n- neutron-grenade http://logs.openstack.org/09/665109/1/check/neutron-grenade/db5157b/ : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3f000a00a485e8a909ad526c213a6ee9585b8932","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 09:31:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d3f59bd8b27b68777835ae476f2d3e32db6b846","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 12:09:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/09/665109/1/gate/openstack-tox-pep8/3516096/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/09/665109/1/gate/openstack-tox-py27/a7870e8/ : SUCCESS in 3m 20s\n- openstack-tox-py35 http://logs.openstack.org/09/665109/1/gate/openstack-tox-py35/cc51df6/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/09/665109/1/gate/openstack-tox-py36/4cc3274/ : SUCCESS in 3m 37s\n- openstack-tox-validate http://logs.openstack.org/09/665109/1/gate/openstack-tox-validate/fb44522/ : SUCCESS in 2m 58s\n- requirements-tox-babel http://logs.openstack.org/09/665109/1/gate/requirements-tox-babel/c2227d3/ : SUCCESS in 3m 29s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/665109/1/gate/requirements-tox-py27-check-uc/5d446d0/ : SUCCESS in 6m 49s\n- requirements-tox-py35-check-uc http://logs.openstack.org/09/665109/1/gate/requirements-tox-py35-check-uc/ed6592e/ : SUCCESS in 7m 15s\n- cross-cinder-py27 http://logs.openstack.org/09/665109/1/gate/cross-cinder-py27/12037d1/ : SUCCESS in 8m 39s\n- cross-cinder-py35 http://logs.openstack.org/09/665109/1/gate/cross-cinder-py35/98eeed2/ : SUCCESS in 8m 46s\n- cross-glance-py27 http://logs.openstack.org/09/665109/1/gate/cross-glance-py27/37ac1e0/ : SUCCESS in 4m 25s\n- cross-glance-py35 http://logs.openstack.org/09/665109/1/gate/cross-glance-py35/9f2d4fe/ : SUCCESS in 4m 29s\n- cross-horizon-py27 http://logs.openstack.org/09/665109/1/gate/cross-horizon-py27/c347976/ : SUCCESS in 10m 30s\n- cross-horizon-py35 http://logs.openstack.org/09/665109/1/gate/cross-horizon-py35/3335192/ : SUCCESS in 9m 54s\n- cross-horizon-npm http://logs.openstack.org/09/665109/1/gate/cross-horizon-npm/b8713b2/ : SUCCESS in 6m 44s\n- cross-keystone-py27 http://logs.openstack.org/09/665109/1/gate/cross-keystone-py27/7f8e559/ : SUCCESS in 21m 58s\n- cross-keystone-py35 http://logs.openstack.org/09/665109/1/gate/cross-keystone-py35/defb6f9/ : SUCCESS in 23m 47s\n- cross-neutron-py27 http://logs.openstack.org/09/665109/1/gate/cross-neutron-py27/1561537/ : SUCCESS in 25m 12s\n- cross-neutron-py35 http://logs.openstack.org/09/665109/1/gate/cross-neutron-py35/8aeed23/ : SUCCESS in 23m 41s\n- cross-nova-py27 http://logs.openstack.org/09/665109/1/gate/cross-nova-py27/24e8f9f/ : SUCCESS in 11m 56s\n- cross-nova-py35 http://logs.openstack.org/09/665109/1/gate/cross-nova-py35/e027f19/ : SUCCESS in 12m 39s\n- cross-nova-functional http://logs.openstack.org/09/665109/1/gate/cross-nova-functional/f8830d4/ : SUCCESS in 19m 18s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/09/665109/1/gate/cross-osc-build-sphinx-docs/8e3e1f0/html/ : SUCCESS in 5m 30s\n- cross-swift-py27 http://logs.openstack.org/09/665109/1/gate/cross-swift-py27/da57418/ : SUCCESS in 7m 44s\n- cross-swift-py35 http://logs.openstack.org/09/665109/1/gate/cross-swift-py35/07d99e6/ : SUCCESS in 7m 35s\n- requirements-integration http://logs.openstack.org/09/665109/1/gate/requirements-integration/daf09be/ : SUCCESS in 52m 43s\n- neutron-grenade-multinode http://logs.openstack.org/09/665109/1/gate/neutron-grenade-multinode/6289dc1/ : SUCCESS in 1h 07m 02s\n- tempest-full http://logs.openstack.org/09/665109/1/gate/tempest-full/09557a1/ : SUCCESS in 1h 32m 43s\n- neutron-grenade http://logs.openstack.org/09/665109/1/gate/neutron-grenade/601117b/ : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f62a15118e3a5c29c79f3146837ad94241754456","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 12:09:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7549d9ac7c6cf8447a2bf415bc9692742dad6b32","revisions":{"7549d9ac7c6cf8447a2bf415bc9692742dad6b32":{"kind":"REWORK","_number":1,"created":"2019-06-13 10:56:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/665109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/665109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/665109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/665109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/665109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/665109/1"}}},"commit":{"parents":[{"commit":"0bcef059e5b64ec843937f5a5d85f1d1cd248803","subject":"update constraint for ironic-lib to new release 2.16.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0bcef059e5b64ec843937f5a5d85f1d1cd248803"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-13 10:56:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-13 10:56:52.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 13.0.1","message":"update constraint for python-novaclient to new release 13.0.1\n\nChange-Id: I67aac41d32ca9c503c1cc1e59832f06c4cd3ccf4\nmeta:version: 13.0.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta:release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta:release:Change-Id: I1fd7fd693a401e7a370ac5743981a185511d5cd1\nmeta:release:Code-Review+1: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7549d9ac7c6cf8447a2bf415bc9692742dad6b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7549d9ac7c6cf8447a2bf415bc9692742dad6b32"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
