)]}'
{"id":"openstack%2Frequirements~665174","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ie1aa0887f1e73ca2ca763c4e85ca102d0d9f80e2","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Ie1aa0887f1e73ca2ca763c4e85ca102d0d9f80e2","subject":"update constraint for ovsdbapp to new release 0.12.4","status":"MERGED","created":"2019-06-13 16:42:27.000000000","updated":"2019-06-14 20:29:17.000000000","submitted":"2019-06-14 20:29:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665174-1560544157531-e4049d2a","meta_rev_id":"e364b41bbace29c27ae64ee11d4ef823bc9cb510","_number":665174,"virtual_id_number":665174,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-06-14 20:29:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-14 14:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-14 13:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-06-14 14:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-14 14:55:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 13:23:03.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-06-14 14:22:57.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-06-14 14:55:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-14 20:29:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7deee91061259a12c8f1a56a93fff6f528d680f6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-13 16:42:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04a4b527ef238dfbd683de49eb7084c0d43e7293","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 05:21:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/74/665174/1/check/openstack-tox-pypy/c2a0ee2/ : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/74/665174/1/check/openstack-tox-pep8/f25e8a6/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/74/665174/1/check/openstack-tox-py27/dc91c91/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/74/665174/1/check/openstack-tox-py35/a52ff4c/ : SUCCESS in 3m 12s\n- openstack-tox-validate http://logs.openstack.org/74/665174/1/check/openstack-tox-validate/638f3e8/ : SUCCESS in 3m 48s\n- requirements-tox-babel http://logs.openstack.org/74/665174/1/check/requirements-tox-babel/d51b9c9/ : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/74/665174/1/check/requirements-tox-py27-check-uc/1df9276/ : SUCCESS in 7m 15s\n- requirements-tox-py35-check-uc http://logs.openstack.org/74/665174/1/check/requirements-tox-py35-check-uc/9a2175c/ : SUCCESS in 9m 34s\n- requirements-tox-py36-check-uc http://logs.openstack.org/74/665174/1/check/requirements-tox-py36-check-uc/7288358/ : SUCCESS in 7m 11s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/74/665174/1/check/cross-cinder-py27/cc88e90/ : SUCCESS in 8m 50s\n- cross-cinder-py35 http://logs.openstack.org/74/665174/1/check/cross-cinder-py35/2299220/ : SUCCESS in 8m 31s\n- cross-glance-py27 http://logs.openstack.org/74/665174/1/check/cross-glance-py27/5c1b267/ : SUCCESS in 5m 10s\n- cross-glance-py35 http://logs.openstack.org/74/665174/1/check/cross-glance-py35/5a79cd3/ : SUCCESS in 5m 01s\n- cross-horizon-py27 http://logs.openstack.org/74/665174/1/check/cross-horizon-py27/88e9d0d/ : SUCCESS in 10m 38s\n- cross-horizon-py35 http://logs.openstack.org/74/665174/1/check/cross-horizon-py35/0b2748c/ : SUCCESS in 8m 59s\n- cross-keystone-py27 http://logs.openstack.org/74/665174/1/check/cross-keystone-py27/8ab34ca/ : SUCCESS in 20m 25s\n- cross-keystone-py35 http://logs.openstack.org/74/665174/1/check/cross-keystone-py35/72c34b0/ : SUCCESS in 11m 40s\n- cross-neutron-py27 http://logs.openstack.org/74/665174/1/check/cross-neutron-py27/9ff0876/ : SUCCESS in 26m 28s\n- cross-neutron-py35 http://logs.openstack.org/74/665174/1/check/cross-neutron-py35/ed237bf/ : SUCCESS in 26m 27s\n- cross-nova-py27 http://logs.openstack.org/74/665174/1/check/cross-nova-py27/4a5e46f/ : SUCCESS in 12m 25s\n- cross-nova-py35 http://logs.openstack.org/74/665174/1/check/cross-nova-py35/73707f7/ : SUCCESS in 13m 42s\n- cross-nova-functional http://logs.openstack.org/74/665174/1/check/cross-nova-functional/ed402f8/ : SUCCESS in 19m 11s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/74/665174/1/check/cross-osc-build-sphinx-docs/b6f2e68/html/ : SUCCESS in 6m 38s\n- cross-swift-py27 http://logs.openstack.org/74/665174/1/check/cross-swift-py27/7cf64aa/ : SUCCESS in 7m 54s\n- cross-swift-py35 http://logs.openstack.org/74/665174/1/check/cross-swift-py35/cfe678d/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/74/665174/1/check/openstack-tox-py36/c2f16c1/ : SUCCESS in 3m 36s\n- requirements-integration http://logs.openstack.org/74/665174/1/check/requirements-integration/4706e07/ : SUCCESS in 58m 25s\n- neutron-grenade-multinode http://logs.openstack.org/74/665174/1/check/neutron-grenade-multinode/af20681/ : SUCCESS in 1h 19m 32s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/74/665174/1/check/legacy-tempest-dsvm-neutron-full/ee0b62b/ : SUCCESS in 1h 35m 43s\n- neutron-grenade http://logs.openstack.org/74/665174/1/check/neutron-grenade/654841d/ : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f85d266d3f8196fbebc4be02e6a9c477839ff62e","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-14 13:23:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"356656571b332b5ac5261597da99e76f5c4149e9","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:22:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"99c883741ce5634dc90f52d3550c0a74ac66d78c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-14 14:55:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d035e4cc05b6fab9c19036849949b23d5c036b3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 14:55:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"606c9fdf401e2d29cd8b9cb48c7916bb150d0a0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:29:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"65e1a17b8107f8faafcafc483b48947ccbcad46e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:29:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/74/665174/1/gate/openstack-tox-pep8/a1ed672/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/74/665174/1/gate/openstack-tox-py27/3c41686/ : SUCCESS in 3m 49s\n- openstack-tox-py35 http://logs.openstack.org/74/665174/1/gate/openstack-tox-py35/f8c04e3/ : SUCCESS in 3m 41s\n- openstack-tox-validate http://logs.openstack.org/74/665174/1/gate/openstack-tox-validate/60cedb4/ : SUCCESS in 3m 38s\n- requirements-tox-babel http://logs.openstack.org/74/665174/1/gate/requirements-tox-babel/ef4993b/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/74/665174/1/gate/requirements-tox-py27-check-uc/002dc3d/ : SUCCESS in 12m 24s\n- requirements-tox-py35-check-uc http://logs.openstack.org/74/665174/1/gate/requirements-tox-py35-check-uc/d066b84/ : SUCCESS in 7m 20s\n- cross-cinder-py27 http://logs.openstack.org/74/665174/1/gate/cross-cinder-py27/d8bd565/ : SUCCESS in 10m 01s\n- cross-cinder-py35 http://logs.openstack.org/74/665174/1/gate/cross-cinder-py35/645f25a/ : SUCCESS in 8m 19s\n- cross-glance-py27 http://logs.openstack.org/74/665174/1/gate/cross-glance-py27/c99c4c9/ : SUCCESS in 4m 39s\n- cross-glance-py35 http://logs.openstack.org/74/665174/1/gate/cross-glance-py35/6e16b5d/ : SUCCESS in 4m 44s\n- cross-horizon-py27 http://logs.openstack.org/74/665174/1/gate/cross-horizon-py27/aefa22b/ : SUCCESS in 8m 50s\n- cross-horizon-py35 http://logs.openstack.org/74/665174/1/gate/cross-horizon-py35/f7fbdc3/ : SUCCESS in 10m 43s\n- cross-keystone-py27 http://logs.openstack.org/74/665174/1/gate/cross-keystone-py27/ec06ab2/ : SUCCESS in 22m 58s\n- cross-keystone-py35 http://logs.openstack.org/74/665174/1/gate/cross-keystone-py35/0f3f95a/ : SUCCESS in 22m 29s\n- cross-neutron-py27 http://logs.openstack.org/74/665174/1/gate/cross-neutron-py27/b7e92b3/ : SUCCESS in 28m 01s\n- cross-neutron-py35 http://logs.openstack.org/74/665174/1/gate/cross-neutron-py35/b55570b/ : SUCCESS in 29m 59s\n- cross-nova-py27 http://logs.openstack.org/74/665174/1/gate/cross-nova-py27/27dfc75/ : SUCCESS in 16m 48s\n- cross-nova-py35 http://logs.openstack.org/74/665174/1/gate/cross-nova-py35/a26fc3b/ : SUCCESS in 13m 12s\n- cross-nova-functional http://logs.openstack.org/74/665174/1/gate/cross-nova-functional/14c63d0/ : SUCCESS in 16m 15s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/74/665174/1/gate/cross-osc-build-sphinx-docs/6378911/html/ : SUCCESS in 6m 24s\n- cross-swift-py27 http://logs.openstack.org/74/665174/1/gate/cross-swift-py27/9ac666e/ : SUCCESS in 10m 38s\n- cross-swift-py35 http://logs.openstack.org/74/665174/1/gate/cross-swift-py35/fef2fed/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/74/665174/1/gate/openstack-tox-py36/21da395/ : SUCCESS in 3m 21s\n- requirements-integration http://logs.openstack.org/74/665174/1/gate/requirements-integration/118f33e/ : SUCCESS in 53m 49s\n- neutron-grenade-multinode http://logs.openstack.org/74/665174/1/gate/neutron-grenade-multinode/c0ac658/ : SUCCESS in 1h 07m 10s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/74/665174/1/gate/legacy-tempest-dsvm-neutron-full/2234aa7/ : SUCCESS in 1h 32m 13s\n- neutron-grenade http://logs.openstack.org/74/665174/1/gate/neutron-grenade/b59af36/ : SUCCESS in 55m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14579e115c6250ca45e8afa936374630089f2338","revisions":{"14579e115c6250ca45e8afa936374630089f2338":{"kind":"REWORK","_number":1,"created":"2019-06-13 16:42:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/665174/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/665174/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/665174/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/665174/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/665174/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/665174/1"}}},"commit":{"parents":[{"commit":"2b2ed555ac701aabad0b143feb4a0b08c5921ba2","subject":"Merge \"update constraint for ironic-lib to new release 2.14.2\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b2ed555ac701aabad0b143feb4a0b08c5921ba2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-13 16:42:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-13 16:42:26.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 0.12.4","message":"update constraint for ovsdbapp to new release 0.12.4\n\nChange-Id: Ie1aa0887f1e73ca2ca763c4e85ca102d0d9f80e2\nmeta:version: 0.12.4\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nmeta:release:Commit: Jakub Libosvar \u003clibosvar@redhat.com\u003e\nmeta:release:Change-Id: Ida5e0f2ee367ec0dd37160f8a7a5955ff59a910d\nmeta:release:Code-Review+1: Terry Wilson \u003ctwilson@redhat.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14579e115c6250ca45e8afa936374630089f2338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14579e115c6250ca45e8afa936374630089f2338"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
