)]}'
{"id":"openstack%2Frequirements~665229","triplet_id":"openstack%2Frequirements~master~I18c09f26569703faddfe0ea51f774eddbe32a96c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I18c09f26569703faddfe0ea51f774eddbe32a96c","subject":"update constraint for ovsdbapp to new release 0.16.0","status":"MERGED","created":"2019-06-13 19:55:30.000000000","updated":"2019-06-14 20:55:12.000000000","submitted":"2019-06-14 20:55:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665229-1560545712865-b1f8f8f4","meta_rev_id":"247d4d67c542d6f6a10f773623c5e866aeb45186","_number":665229,"virtual_id_number":665229,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-06-14 20:55:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-14 14:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-14 13:22:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-06-14 14:22:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-14 14:55:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 13:22:34.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-06-14 14:22:48.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-06-14 14:55:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-14 20:55:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ddc8a6f2b4d504631d7de5bed3d7bc318d734a0a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-13 19:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"381f7f2fcbdfa9b1f65cac75017302fdb6443bea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 06:28:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/29/665229/1/check/openstack-tox-docs/55d45c0/html/ : SUCCESS in 3m 09s\n- openstack-tox-pypy http://logs.openstack.org/29/665229/1/check/openstack-tox-pypy/be9db57/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/29/665229/1/check/openstack-tox-pep8/7c72a47/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/29/665229/1/check/openstack-tox-py27/e6892a4/ : SUCCESS in 3m 25s\n- openstack-tox-py36 http://logs.openstack.org/29/665229/1/check/openstack-tox-py36/f699adf/ : SUCCESS in 3m 16s\n- openstack-tox-validate http://logs.openstack.org/29/665229/1/check/openstack-tox-validate/11d1d02/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/29/665229/1/check/requirements-tox-babel/be27af8/ : SUCCESS in 3m 25s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/665229/1/check/requirements-tox-py27-check-uc/a8164ab/ : SUCCESS in 6m 28s\n- requirements-tox-py36-check-uc http://logs.openstack.org/29/665229/1/check/requirements-tox-py36-check-uc/9e09c77/ : SUCCESS in 5m 58s\n- cross-cinder-py27 http://logs.openstack.org/29/665229/1/check/cross-cinder-py27/2307a6b/ : SUCCESS in 7m 49s\n- cross-cinder-py36 http://logs.openstack.org/29/665229/1/check/cross-cinder-py36/3ddc726/ : SUCCESS in 8m 59s\n- cross-glance-py27 http://logs.openstack.org/29/665229/1/check/cross-glance-py27/11b45eb/ : SUCCESS in 4m 10s\n- cross-glance-py36 http://logs.openstack.org/29/665229/1/check/cross-glance-py36/df8bd2a/ : SUCCESS in 4m 26s\n- cross-horizon-py27 http://logs.openstack.org/29/665229/1/check/cross-horizon-py27/f2f0503/ : SUCCESS in 10m 37s\n- cross-horizon-py36 http://logs.openstack.org/29/665229/1/check/cross-horizon-py36/2554fd1/ : SUCCESS in 9m 32s\n- cross-horizon-npm http://logs.openstack.org/29/665229/1/check/cross-horizon-npm/c640166/ : SUCCESS in 6m 12s\n- cross-keystone-py27 http://logs.openstack.org/29/665229/1/check/cross-keystone-py27/aec33cd/ : SUCCESS in 23m 22s\n- cross-keystone-py36 http://logs.openstack.org/29/665229/1/check/cross-keystone-py36/9a1f142/ : SUCCESS in 25m 17s\n- cross-neutron-py27 http://logs.openstack.org/29/665229/1/check/cross-neutron-py27/9a7198f/ : SUCCESS in 20m 12s\n- cross-neutron-py36 http://logs.openstack.org/29/665229/1/check/cross-neutron-py36/016724e/ : SUCCESS in 24m 29s\n- cross-nova-py27 http://logs.openstack.org/29/665229/1/check/cross-nova-py27/9ce040f/ : SUCCESS in 11m 25s\n- cross-nova-py36 http://logs.openstack.org/29/665229/1/check/cross-nova-py36/61847d7/ : SUCCESS in 11m 15s\n- cross-nova-functional http://logs.openstack.org/29/665229/1/check/cross-nova-functional/c2f7a4d/ : SUCCESS in 19m 41s\n- cross-osc-tox-docs http://logs.openstack.org/29/665229/1/check/cross-osc-tox-docs/8a055d8/html/ : SUCCESS in 5m 10s\n- cross-swift-py27 http://logs.openstack.org/29/665229/1/check/cross-swift-py27/0d82d17/ : SUCCESS in 7m 49s\n- cross-swift-py36 http://logs.openstack.org/29/665229/1/check/cross-swift-py36/e4d924d/ : SUCCESS in 8m 36s\n- requirements-integration http://logs.openstack.org/29/665229/1/check/requirements-integration/9b42c81/ : SUCCESS in 1h 07m 53s\n- neutron-grenade-multinode http://logs.openstack.org/29/665229/1/check/neutron-grenade-multinode/b7b217a/ : SUCCESS in 1h 17m 27s\n- tempest-full http://logs.openstack.org/29/665229/1/check/tempest-full/71918cf/ : SUCCESS in 1h 59m 04s\n- neutron-grenade http://logs.openstack.org/29/665229/1/check/neutron-grenade/6be7e11/ : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5ecadd12f79f826fe029e9aaf3f86054ecbbf621","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-14 13:22:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f3272aba99d5266eb83460939503b6589b5499d","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:22:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d274b0aa670ef582f581a640d8447302ef135bb2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-14 14:55:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"658e682dfb59e6254d821603d25dad2ebfd73d5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 14:55:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b955ab771bdafa3e5916b3c4350641d0ecc6f9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:55:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fb037b0d8008968ed2f047aee6c647e6c777b3e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:55:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/29/665229/1/gate/openstack-tox-docs/7760ef7/html/ : SUCCESS in 4m 28s\n- openstack-tox-pep8 http://logs.openstack.org/29/665229/1/gate/openstack-tox-pep8/18f7f3d/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/29/665229/1/gate/openstack-tox-py27/f01d9d6/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/29/665229/1/gate/openstack-tox-py36/0149fb4/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/29/665229/1/gate/openstack-tox-validate/777b168/ : SUCCESS in 4m 29s\n- requirements-tox-babel http://logs.openstack.org/29/665229/1/gate/requirements-tox-babel/c0fac28/ : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/29/665229/1/gate/requirements-tox-py27-check-uc/3239479/ : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc http://logs.openstack.org/29/665229/1/gate/requirements-tox-py36-check-uc/52e6666/ : SUCCESS in 6m 51s\n- cross-cinder-py27 http://logs.openstack.org/29/665229/1/gate/cross-cinder-py27/5447cc8/ : SUCCESS in 9m 20s\n- cross-cinder-py36 http://logs.openstack.org/29/665229/1/gate/cross-cinder-py36/33e273d/ : SUCCESS in 7m 39s\n- cross-glance-py27 http://logs.openstack.org/29/665229/1/gate/cross-glance-py27/99cf116/ : SUCCESS in 4m 39s\n- cross-glance-py36 http://logs.openstack.org/29/665229/1/gate/cross-glance-py36/d42521f/ : SUCCESS in 4m 12s\n- cross-horizon-py27 http://logs.openstack.org/29/665229/1/gate/cross-horizon-py27/ef57e60/ : SUCCESS in 9m 23s\n- cross-horizon-py36 http://logs.openstack.org/29/665229/1/gate/cross-horizon-py36/6b3e7e3/ : SUCCESS in 9m 13s\n- cross-horizon-npm http://logs.openstack.org/29/665229/1/gate/cross-horizon-npm/37ce555/ : SUCCESS in 7m 31s\n- cross-keystone-py27 http://logs.openstack.org/29/665229/1/gate/cross-keystone-py27/4df632c/ : SUCCESS in 21m 58s\n- cross-keystone-py36 http://logs.openstack.org/29/665229/1/gate/cross-keystone-py36/f91bc29/ : SUCCESS in 24m 43s\n- cross-neutron-py27 http://logs.openstack.org/29/665229/1/gate/cross-neutron-py27/45c599f/ : SUCCESS in 22m 57s\n- cross-neutron-py36 http://logs.openstack.org/29/665229/1/gate/cross-neutron-py36/785c278/ : SUCCESS in 24m 09s\n- cross-nova-py27 http://logs.openstack.org/29/665229/1/gate/cross-nova-py27/9e8311b/ : SUCCESS in 11m 56s\n- cross-nova-py36 http://logs.openstack.org/29/665229/1/gate/cross-nova-py36/e084413/ : SUCCESS in 14m 13s\n- cross-nova-functional http://logs.openstack.org/29/665229/1/gate/cross-nova-functional/7a9591c/ : SUCCESS in 18m 53s\n- cross-osc-tox-docs http://logs.openstack.org/29/665229/1/gate/cross-osc-tox-docs/cd938a0/html/ : SUCCESS in 5m 22s\n- cross-swift-py27 http://logs.openstack.org/29/665229/1/gate/cross-swift-py27/f161fd2/ : SUCCESS in 7m 17s\n- cross-swift-py36 http://logs.openstack.org/29/665229/1/gate/cross-swift-py36/7ec1a49/ : SUCCESS in 8m 39s\n- requirements-integration http://logs.openstack.org/29/665229/1/gate/requirements-integration/6f6a3a4/ : SUCCESS in 59m 06s\n- neutron-grenade-multinode http://logs.openstack.org/29/665229/1/gate/neutron-grenade-multinode/78fb768/ : SUCCESS in 1h 11m 57s\n- tempest-full http://logs.openstack.org/29/665229/1/gate/tempest-full/2d37f5b/ : SUCCESS in 1h 46m 25s\n- neutron-grenade http://logs.openstack.org/29/665229/1/gate/neutron-grenade/3516365/ : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00cdb9ccd49928fcbc5ed5247ac017688d7fc16f","revisions":{"00cdb9ccd49928fcbc5ed5247ac017688d7fc16f":{"kind":"REWORK","_number":1,"created":"2019-06-13 19:55:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/665229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/665229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/665229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/665229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/665229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/665229/1"}}},"commit":{"parents":[{"commit":"67d9f8452bfec111f99e8e655a7372c414d0e6b2","subject":"Merge \"update constraint for octavia-lib to new release 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67d9f8452bfec111f99e8e655a7372c414d0e6b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-13 19:55:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-13 19:55:24.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 0.16.0","message":"update constraint for ovsdbapp to new release 0.16.0\n\nChange-Id: I18c09f26569703faddfe0ea51f774eddbe32a96c\nmeta:version: 0.16.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: yes\nmeta:release:Author: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Commit: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Change-Id: I4f5d4bd34bc5f761693c771109581ffd96953d0d\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Miguel Lavalle \u003cmiguel.lavalle@verizonmedia.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00cdb9ccd49928fcbc5ed5247ac017688d7fc16f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00cdb9ccd49928fcbc5ed5247ac017688d7fc16f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
