)]}'
{"id":"openstack%2Frequirements~665273","triplet_id":"openstack%2Frequirements~master~I81d8f13de62b9575248c2ff3b8bb6d0bed19c8c5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I81d8f13de62b9575248c2ff3b8bb6d0bed19c8c5","subject":"update constraint for metalsmith to new release 0.13.0","status":"MERGED","created":"2019-06-14 02:03:23.000000000","updated":"2019-06-15 20:34:29.000000000","submitted":"2019-06-15 20:34:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665273-1560630869523-018dc456","meta_rev_id":"0dcd69b46f21e652a653560e45955cbee99b93ee","_number":665273,"virtual_id_number":665273,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-15 20:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-15 18:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-15 18:49:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-15 18:49:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-15 20:34:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1aca7080d1d0a98454df30c7d49c53885a77dd58","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-14 02:03:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed3bd293007dfbbf30db6781a4785c35a3eb5972","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 08:01:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/73/665273/1/check/openstack-tox-docs/f980037/html/ : SUCCESS in 3m 36s\n- openstack-tox-pypy http://logs.openstack.org/73/665273/1/check/openstack-tox-pypy/018c050/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/665273/1/check/openstack-tox-pep8/b1f3db4/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/73/665273/1/check/openstack-tox-py27/d16d5a7/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/73/665273/1/check/openstack-tox-py36/5da3f68/ : SUCCESS in 3m 14s\n- openstack-tox-validate http://logs.openstack.org/73/665273/1/check/openstack-tox-validate/e01daef/ : SUCCESS in 4m 14s\n- requirements-tox-babel http://logs.openstack.org/73/665273/1/check/requirements-tox-babel/768eb2d/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/665273/1/check/requirements-tox-py27-check-uc/509fa15/ : SUCCESS in 6m 23s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/665273/1/check/requirements-tox-py36-check-uc/eef2210/ : SUCCESS in 7m 34s\n- cross-cinder-py27 http://logs.openstack.org/73/665273/1/check/cross-cinder-py27/92bcd3d/ : SUCCESS in 11m 17s\n- cross-cinder-py36 http://logs.openstack.org/73/665273/1/check/cross-cinder-py36/49964c8/ : SUCCESS in 8m 02s\n- cross-glance-py27 http://logs.openstack.org/73/665273/1/check/cross-glance-py27/d580579/ : SUCCESS in 4m 25s\n- cross-glance-py36 http://logs.openstack.org/73/665273/1/check/cross-glance-py36/13a90cd/ : SUCCESS in 4m 29s\n- cross-horizon-py27 http://logs.openstack.org/73/665273/1/check/cross-horizon-py27/049ceff/ : SUCCESS in 10m 55s\n- cross-horizon-py36 http://logs.openstack.org/73/665273/1/check/cross-horizon-py36/7cce40d/ : SUCCESS in 7m 54s\n- cross-horizon-npm http://logs.openstack.org/73/665273/1/check/cross-horizon-npm/f3ed6b1/ : SUCCESS in 6m 27s\n- cross-keystone-py27 http://logs.openstack.org/73/665273/1/check/cross-keystone-py27/a720607/ : SUCCESS in 20m 18s\n- cross-keystone-py36 http://logs.openstack.org/73/665273/1/check/cross-keystone-py36/c242b0a/ : SUCCESS in 21m 10s\n- cross-neutron-py27 http://logs.openstack.org/73/665273/1/check/cross-neutron-py27/f0908fd/ : SUCCESS in 22m 40s\n- cross-neutron-py36 http://logs.openstack.org/73/665273/1/check/cross-neutron-py36/01e054b/ : SUCCESS in 22m 23s\n- cross-nova-py27 http://logs.openstack.org/73/665273/1/check/cross-nova-py27/b538974/ : SUCCESS in 11m 54s\n- cross-nova-py36 http://logs.openstack.org/73/665273/1/check/cross-nova-py36/d9a1ead/ : SUCCESS in 13m 46s\n- cross-nova-functional http://logs.openstack.org/73/665273/1/check/cross-nova-functional/52dd6fd/ : SUCCESS in 18m 42s\n- cross-osc-tox-docs http://logs.openstack.org/73/665273/1/check/cross-osc-tox-docs/76a6934/html/ : SUCCESS in 5m 36s\n- cross-swift-py27 http://logs.openstack.org/73/665273/1/check/cross-swift-py27/ac35eba/ : SUCCESS in 7m 56s\n- cross-swift-py36 http://logs.openstack.org/73/665273/1/check/cross-swift-py36/9b82708/ : SUCCESS in 8m 12s\n- requirements-integration http://logs.openstack.org/73/665273/1/check/requirements-integration/a6fb57f/ : SUCCESS in 1h 03m 48s\n- neutron-grenade-multinode http://logs.openstack.org/73/665273/1/check/neutron-grenade-multinode/17a388c/ : SUCCESS in 1h 12m 54s\n- tempest-full http://logs.openstack.org/73/665273/1/check/tempest-full/a5ab3ab/ : FAILURE in 1h 50m 17s\n- neutron-grenade http://logs.openstack.org/73/665273/1/check/neutron-grenade/f4287f0/ : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"adb930b7efa85f78c0bacd9f45894a4499835ebd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-14 14:56:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"296cf7e48f0a98c040e2ad71e714edc520ac62d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 23:50:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/73/665273/1/check/openstack-tox-docs/761d74d/html/ : SUCCESS in 3m 05s\n- openstack-tox-pypy http://logs.openstack.org/73/665273/1/check/openstack-tox-pypy/e4353f2/ : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/665273/1/check/openstack-tox-pep8/618c7d1/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/73/665273/1/check/openstack-tox-py27/ccb8753/ : SUCCESS in 3m 23s\n- openstack-tox-py36 http://logs.openstack.org/73/665273/1/check/openstack-tox-py36/f1a3f4b/ : SUCCESS in 3m 21s\n- openstack-tox-validate http://logs.openstack.org/73/665273/1/check/openstack-tox-validate/ac7f882/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/73/665273/1/check/requirements-tox-babel/6a8bf3d/ : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/665273/1/check/requirements-tox-py27-check-uc/6fdf22e/ : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/665273/1/check/requirements-tox-py36-check-uc/af6957c/ : SUCCESS in 6m 39s\n- cross-cinder-py27 http://logs.openstack.org/73/665273/1/check/cross-cinder-py27/d4d6cbb/ : SUCCESS in 8m 39s\n- cross-cinder-py36 http://logs.openstack.org/73/665273/1/check/cross-cinder-py36/6dba68d/ : SUCCESS in 9m 56s\n- cross-glance-py27 http://logs.openstack.org/73/665273/1/check/cross-glance-py27/a70622e/ : SUCCESS in 4m 29s\n- cross-glance-py36 http://logs.openstack.org/73/665273/1/check/cross-glance-py36/ed5539e/ : SUCCESS in 4m 54s\n- cross-horizon-py27 http://logs.openstack.org/73/665273/1/check/cross-horizon-py27/ac7c00c/ : SUCCESS in 10m 17s\n- cross-horizon-py36 http://logs.openstack.org/73/665273/1/check/cross-horizon-py36/cd34fa2/ : SUCCESS in 9m 56s\n- cross-horizon-npm http://logs.openstack.org/73/665273/1/check/cross-horizon-npm/83a4a58/ : SUCCESS in 6m 25s\n- cross-keystone-py27 http://logs.openstack.org/73/665273/1/check/cross-keystone-py27/29a9a15/ : SUCCESS in 22m 10s\n- cross-keystone-py36 http://logs.openstack.org/73/665273/1/check/cross-keystone-py36/d150104/ : SUCCESS in 26m 07s\n- cross-neutron-py27 http://logs.openstack.org/73/665273/1/check/cross-neutron-py27/9e353dc/ : SUCCESS in 28m 14s\n- cross-neutron-py36 http://logs.openstack.org/73/665273/1/check/cross-neutron-py36/4d3d94b/ : SUCCESS in 27m 46s\n- cross-nova-py27 http://logs.openstack.org/73/665273/1/check/cross-nova-py27/a3053cc/ : SUCCESS in 12m 44s\n- cross-nova-py36 http://logs.openstack.org/73/665273/1/check/cross-nova-py36/804e3ea/ : SUCCESS in 13m 49s\n- cross-nova-functional http://logs.openstack.org/73/665273/1/check/cross-nova-functional/180c2e0/ : SUCCESS in 17m 59s\n- cross-osc-tox-docs http://logs.openstack.org/73/665273/1/check/cross-osc-tox-docs/ac76a32/html/ : SUCCESS in 7m 05s\n- cross-swift-py27 http://logs.openstack.org/73/665273/1/check/cross-swift-py27/c86389e/ : SUCCESS in 7m 19s\n- cross-swift-py36 http://logs.openstack.org/73/665273/1/check/cross-swift-py36/46cafed/ : SUCCESS in 9m 07s\n- requirements-integration http://logs.openstack.org/73/665273/1/check/requirements-integration/cd22c5d/ : SUCCESS in 55m 24s\n- neutron-grenade-multinode http://logs.openstack.org/73/665273/1/check/neutron-grenade-multinode/4364e00/ : SUCCESS in 1h 15m 16s\n- tempest-full http://logs.openstack.org/73/665273/1/check/tempest-full/a38101d/ : FAILURE in 1h 40m 22s\n- neutron-grenade http://logs.openstack.org/73/665273/1/check/neutron-grenade/7af5c45/ : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9a072ba7ee5b3a28d2118555b2f800d60ab033e1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-15 17:07:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e755e7d30fd416ac71f7687a142b590784a9b43d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 18:45:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/665273/1/check/openstack-tox-docs/73c9328/html/ : SUCCESS in 3m 02s\n- openstack-tox-pypy http://logs.openstack.org/73/665273/1/check/openstack-tox-pypy/af1cdde/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/665273/1/check/openstack-tox-pep8/5f8ed95/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/73/665273/1/check/openstack-tox-py27/3640884/ : SUCCESS in 3m 08s\n- openstack-tox-py36 http://logs.openstack.org/73/665273/1/check/openstack-tox-py36/945170b/ : SUCCESS in 4m 03s\n- openstack-tox-validate http://logs.openstack.org/73/665273/1/check/openstack-tox-validate/3f2ae5e/ : SUCCESS in 3m 35s\n- requirements-tox-babel http://logs.openstack.org/73/665273/1/check/requirements-tox-babel/d8682d3/ : SUCCESS in 4m 33s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/665273/1/check/requirements-tox-py27-check-uc/644c5db/ : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/665273/1/check/requirements-tox-py36-check-uc/bb20388/ : SUCCESS in 6m 32s\n- cross-cinder-py27 http://logs.openstack.org/73/665273/1/check/cross-cinder-py27/e76f763/ : SUCCESS in 11m 05s\n- cross-cinder-py36 http://logs.openstack.org/73/665273/1/check/cross-cinder-py36/9983cd7/ : SUCCESS in 12m 32s\n- cross-glance-py27 http://logs.openstack.org/73/665273/1/check/cross-glance-py27/7cfe4d1/ : SUCCESS in 4m 24s\n- cross-glance-py36 http://logs.openstack.org/73/665273/1/check/cross-glance-py36/a5a6fc3/ : SUCCESS in 5m 21s\n- cross-horizon-py27 http://logs.openstack.org/73/665273/1/check/cross-horizon-py27/7adc3d1/ : SUCCESS in 8m 52s\n- cross-horizon-py36 http://logs.openstack.org/73/665273/1/check/cross-horizon-py36/85f2277/ : SUCCESS in 7m 54s\n- cross-horizon-npm http://logs.openstack.org/73/665273/1/check/cross-horizon-npm/05f988a/ : SUCCESS in 6m 19s\n- cross-keystone-py27 http://logs.openstack.org/73/665273/1/check/cross-keystone-py27/f76308a/ : SUCCESS in 22m 49s\n- cross-keystone-py36 http://logs.openstack.org/73/665273/1/check/cross-keystone-py36/95bd271/ : SUCCESS in 23m 48s\n- cross-neutron-py27 http://logs.openstack.org/73/665273/1/check/cross-neutron-py27/d8dc753/ : SUCCESS in 21m 01s\n- cross-neutron-py36 http://logs.openstack.org/73/665273/1/check/cross-neutron-py36/30c85a6/ : SUCCESS in 20m 30s\n- cross-nova-py27 http://logs.openstack.org/73/665273/1/check/cross-nova-py27/d0595a5/ : SUCCESS in 10m 46s\n- cross-nova-py36 http://logs.openstack.org/73/665273/1/check/cross-nova-py36/3a4967c/ : SUCCESS in 18m 50s\n- cross-nova-functional http://logs.openstack.org/73/665273/1/check/cross-nova-functional/17cee24/ : SUCCESS in 16m 18s\n- cross-osc-tox-docs http://logs.openstack.org/73/665273/1/check/cross-osc-tox-docs/9ab9a23/html/ : SUCCESS in 5m 06s\n- cross-swift-py27 http://logs.openstack.org/73/665273/1/check/cross-swift-py27/485c75b/ : SUCCESS in 12m 01s\n- cross-swift-py36 http://logs.openstack.org/73/665273/1/check/cross-swift-py36/0019d9f/ : SUCCESS in 8m 43s\n- requirements-integration http://logs.openstack.org/73/665273/1/check/requirements-integration/1a01fb5/ : SUCCESS in 51m 25s\n- neutron-grenade-multinode http://logs.openstack.org/73/665273/1/check/neutron-grenade-multinode/da614c5/ : SUCCESS in 1h 06m 14s\n- tempest-full http://logs.openstack.org/73/665273/1/check/tempest-full/90d1a71/ : SUCCESS in 1h 36m 09s\n- neutron-grenade http://logs.openstack.org/73/665273/1/check/neutron-grenade/c22f276/ : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":1},{"id":"30e674e310c36ae5c8298b1ab6f95da26433dda1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-15 18:49:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ec2a73a0cf90e9621bb856deb85faf42351a097","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 18:49:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f923a203f63ec97439f49b26dd54f8a1c70c5951","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 20:34:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"88456ed440045ef64e8451c99bde96d22024b020","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 20:34:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/665273/1/gate/openstack-tox-docs/95d323a/html/ : SUCCESS in 5m 51s\n- openstack-tox-pep8 http://logs.openstack.org/73/665273/1/gate/openstack-tox-pep8/13a1d74/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/73/665273/1/gate/openstack-tox-py27/c0a02b9/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/73/665273/1/gate/openstack-tox-py36/88ba3f9/ : SUCCESS in 5m 01s\n- openstack-tox-validate http://logs.openstack.org/73/665273/1/gate/openstack-tox-validate/5f8ef83/ : SUCCESS in 3m 04s\n- requirements-tox-babel http://logs.openstack.org/73/665273/1/gate/requirements-tox-babel/b990384/ : SUCCESS in 4m 41s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/665273/1/gate/requirements-tox-py27-check-uc/51d7ed8/ : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/665273/1/gate/requirements-tox-py36-check-uc/7b8cc5a/ : SUCCESS in 9m 33s\n- cross-cinder-py27 http://logs.openstack.org/73/665273/1/gate/cross-cinder-py27/81d3750/ : SUCCESS in 9m 28s\n- cross-cinder-py36 http://logs.openstack.org/73/665273/1/gate/cross-cinder-py36/188a104/ : SUCCESS in 8m 44s\n- cross-glance-py27 http://logs.openstack.org/73/665273/1/gate/cross-glance-py27/6f464cc/ : SUCCESS in 4m 33s\n- cross-glance-py36 http://logs.openstack.org/73/665273/1/gate/cross-glance-py36/7fb857c/ : SUCCESS in 4m 08s\n- cross-horizon-py27 http://logs.openstack.org/73/665273/1/gate/cross-horizon-py27/6144348/ : SUCCESS in 8m 56s\n- cross-horizon-py36 http://logs.openstack.org/73/665273/1/gate/cross-horizon-py36/728fbbe/ : SUCCESS in 8m 47s\n- cross-horizon-npm http://logs.openstack.org/73/665273/1/gate/cross-horizon-npm/e53eea6/ : SUCCESS in 8m 19s\n- cross-keystone-py27 http://logs.openstack.org/73/665273/1/gate/cross-keystone-py27/53f23d4/ : SUCCESS in 22m 07s\n- cross-keystone-py36 http://logs.openstack.org/73/665273/1/gate/cross-keystone-py36/8e9da3f/ : SUCCESS in 25m 56s\n- cross-neutron-py27 http://logs.openstack.org/73/665273/1/gate/cross-neutron-py27/9321bf9/ : SUCCESS in 22m 49s\n- cross-neutron-py36 http://logs.openstack.org/73/665273/1/gate/cross-neutron-py36/97ca716/ : SUCCESS in 22m 29s\n- cross-nova-py27 http://logs.openstack.org/73/665273/1/gate/cross-nova-py27/83efa56/ : SUCCESS in 10m 50s\n- cross-nova-py36 http://logs.openstack.org/73/665273/1/gate/cross-nova-py36/4c7d9c7/ : SUCCESS in 11m 23s\n- cross-nova-functional http://logs.openstack.org/73/665273/1/gate/cross-nova-functional/45649a4/ : SUCCESS in 18m 19s\n- cross-osc-tox-docs http://logs.openstack.org/73/665273/1/gate/cross-osc-tox-docs/a76effc/html/ : SUCCESS in 6m 21s\n- cross-swift-py27 http://logs.openstack.org/73/665273/1/gate/cross-swift-py27/aeeab4d/ : SUCCESS in 7m 49s\n- cross-swift-py36 http://logs.openstack.org/73/665273/1/gate/cross-swift-py36/279f4c0/ : SUCCESS in 10m 20s\n- requirements-integration http://logs.openstack.org/73/665273/1/gate/requirements-integration/4510e38/ : SUCCESS in 52m 08s\n- neutron-grenade-multinode http://logs.openstack.org/73/665273/1/gate/neutron-grenade-multinode/092ea75/ : SUCCESS in 1h 10m 41s\n- tempest-full http://logs.openstack.org/73/665273/1/gate/tempest-full/eaa8103/ : SUCCESS in 1h 39m 36s\n- neutron-grenade http://logs.openstack.org/73/665273/1/gate/neutron-grenade/79a6a3f/ : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f23317e65d3411b7637ce06d230d6063e3edc21e","revisions":{"f23317e65d3411b7637ce06d230d6063e3edc21e":{"kind":"REWORK","_number":1,"created":"2019-06-14 02:03:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/665273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/665273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/665273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/665273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/665273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/665273/1"}}},"commit":{"parents":[{"commit":"67d9f8452bfec111f99e8e655a7372c414d0e6b2","subject":"Merge \"update constraint for octavia-lib to new release 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67d9f8452bfec111f99e8e655a7372c414d0e6b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-14 02:03:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-14 02:03:21.000000000","tz":0},"subject":"update constraint for metalsmith to new release 0.13.0","message":"update constraint for metalsmith to new release 0.13.0\n\nChange-Id: I81d8f13de62b9575248c2ff3b8bb6d0bed19c8c5\nmeta:version: 0.13.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Change-Id: Id4b661f36725aac0347a66bc284eb734eaeacd36\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Code-Review+1: Vieri \u003c15050873171@163.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f23317e65d3411b7637ce06d230d6063e3edc21e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f23317e65d3411b7637ce06d230d6063e3edc21e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
