)]}'
{"id":"openstack%2Frequirements~665291","triplet_id":"openstack%2Frequirements~master~Ia9d8c77926a41570f58ad513c9285a4768030621","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia9d8c77926a41570f58ad513c9285a4768030621","subject":"update constraint for keystonemiddleware to new release 6.1.0","status":"MERGED","created":"2019-06-14 04:06:16.000000000","updated":"2019-06-14 20:55:14.000000000","submitted":"2019-06-14 20:55:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665291-1560545714158-0738595d","meta_rev_id":"2c299282bdc38312b593f1a70475bf7ec7f6d1fa","_number":665291,"virtual_id_number":665291,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":2,"date":"2019-06-14 20:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-14 14:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-14 13:21:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-06-14 14:23:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-14 14:55:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 13:21:56.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-06-14 14:23:22.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-06-14 14:55:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-14 20:55:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0baf7f0ab7242443a876acbbb461c310c5b2b455","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-14 04:06:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06debbe18694e0a2519f12e82583fca73c5a965d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 10:39:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/665291/1/check/openstack-tox-docs/29e6ce7/html/ : SUCCESS in 3m 18s\n- openstack-tox-pypy http://logs.openstack.org/91/665291/1/check/openstack-tox-pypy/eb40b60/ : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/91/665291/1/check/openstack-tox-pep8/f095598/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/91/665291/1/check/openstack-tox-py27/799fa40/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/91/665291/1/check/openstack-tox-py36/af758a9/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/91/665291/1/check/openstack-tox-validate/8ad918c/ : SUCCESS in 3m 28s\n- requirements-tox-babel http://logs.openstack.org/91/665291/1/check/requirements-tox-babel/a2d321a/ : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/665291/1/check/requirements-tox-py27-check-uc/8cba7dc/ : SUCCESS in 7m 49s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/665291/1/check/requirements-tox-py36-check-uc/819d99c/ : SUCCESS in 7m 56s\n- cross-cinder-py27 http://logs.openstack.org/91/665291/1/check/cross-cinder-py27/1b273df/ : SUCCESS in 11m 53s\n- cross-cinder-py36 http://logs.openstack.org/91/665291/1/check/cross-cinder-py36/c571433/ : SUCCESS in 8m 09s\n- cross-glance-py27 http://logs.openstack.org/91/665291/1/check/cross-glance-py27/a70a292/ : SUCCESS in 4m 40s\n- cross-glance-py36 http://logs.openstack.org/91/665291/1/check/cross-glance-py36/e7d2e46/ : SUCCESS in 4m 49s\n- cross-horizon-py27 http://logs.openstack.org/91/665291/1/check/cross-horizon-py27/c975c6c/ : SUCCESS in 10m 13s\n- cross-horizon-py36 http://logs.openstack.org/91/665291/1/check/cross-horizon-py36/3044068/ : SUCCESS in 9m 33s\n- cross-horizon-npm http://logs.openstack.org/91/665291/1/check/cross-horizon-npm/f059f9d/ : SUCCESS in 6m 41s\n- cross-keystone-py27 http://logs.openstack.org/91/665291/1/check/cross-keystone-py27/5b72c52/ : SUCCESS in 22m 13s\n- cross-keystone-py36 http://logs.openstack.org/91/665291/1/check/cross-keystone-py36/bc67d53/ : SUCCESS in 24m 28s\n- cross-neutron-py27 http://logs.openstack.org/91/665291/1/check/cross-neutron-py27/3b732af/ : SUCCESS in 25m 34s\n- cross-neutron-py36 http://logs.openstack.org/91/665291/1/check/cross-neutron-py36/c313529/ : SUCCESS in 20m 45s\n- cross-nova-py27 http://logs.openstack.org/91/665291/1/check/cross-nova-py27/20f691e/ : SUCCESS in 12m 39s\n- cross-nova-py36 http://logs.openstack.org/91/665291/1/check/cross-nova-py36/120292c/ : SUCCESS in 11m 27s\n- cross-nova-functional http://logs.openstack.org/91/665291/1/check/cross-nova-functional/88eef66/ : SUCCESS in 19m 35s\n- cross-osc-tox-docs http://logs.openstack.org/91/665291/1/check/cross-osc-tox-docs/8a48bec/html/ : SUCCESS in 5m 38s\n- cross-swift-py27 http://logs.openstack.org/91/665291/1/check/cross-swift-py27/73f4f73/ : SUCCESS in 7m 37s\n- cross-swift-py36 http://logs.openstack.org/91/665291/1/check/cross-swift-py36/3191ac4/ : SUCCESS in 9m 50s\n- requirements-integration http://logs.openstack.org/91/665291/1/check/requirements-integration/315f00c/ : SUCCESS in 54m 04s\n- neutron-grenade-multinode http://logs.openstack.org/91/665291/1/check/neutron-grenade-multinode/8b43bd3/ : SUCCESS in 1h 07m 10s\n- tempest-full http://logs.openstack.org/91/665291/1/check/tempest-full/80efd9b/ : SUCCESS in 1h 43m 54s\n- neutron-grenade http://logs.openstack.org/91/665291/1/check/neutron-grenade/40584e6/ : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c15a790c2030c9d8abee7d1c19436d0a2de38661","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-14 13:21:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f5df6df85530798ed5cea0a1d812639a8a05f3b","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:23:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ca38d51b75df9f1d15bc08569dd376765f2fd89a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-14 14:55:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba20b9dfb8a3ab7c57cef1eff643c3fea2803869","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 14:55:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e57e08f1a277e7e135fa8f683a3601ed037888a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:55:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/91/665291/1/gate/openstack-tox-docs/88d2ee9/html/ : SUCCESS in 2m 59s\n- openstack-tox-pep8 http://logs.openstack.org/91/665291/1/gate/openstack-tox-pep8/93bdc73/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/91/665291/1/gate/openstack-tox-py27/e2c9f09/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/91/665291/1/gate/openstack-tox-py36/32e560c/ : SUCCESS in 3m 27s\n- openstack-tox-validate http://logs.openstack.org/91/665291/1/gate/openstack-tox-validate/6b69132/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/91/665291/1/gate/requirements-tox-babel/3885708/ : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc http://logs.openstack.org/91/665291/1/gate/requirements-tox-py27-check-uc/2f83b3d/ : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc http://logs.openstack.org/91/665291/1/gate/requirements-tox-py36-check-uc/677d3ed/ : SUCCESS in 6m 22s\n- cross-cinder-py27 http://logs.openstack.org/91/665291/1/gate/cross-cinder-py27/570da7c/ : SUCCESS in 9m 18s\n- cross-cinder-py36 http://logs.openstack.org/91/665291/1/gate/cross-cinder-py36/2cf70ca/ : SUCCESS in 8m 10s\n- cross-glance-py27 http://logs.openstack.org/91/665291/1/gate/cross-glance-py27/09e73d8/ : SUCCESS in 4m 32s\n- cross-glance-py36 http://logs.openstack.org/91/665291/1/gate/cross-glance-py36/1c10631/ : SUCCESS in 4m 18s\n- cross-horizon-py27 http://logs.openstack.org/91/665291/1/gate/cross-horizon-py27/eea9211/ : SUCCESS in 10m 23s\n- cross-horizon-py36 http://logs.openstack.org/91/665291/1/gate/cross-horizon-py36/702f668/ : SUCCESS in 8m 29s\n- cross-horizon-npm http://logs.openstack.org/91/665291/1/gate/cross-horizon-npm/3cccb00/ : SUCCESS in 6m 04s\n- cross-keystone-py27 http://logs.openstack.org/91/665291/1/gate/cross-keystone-py27/2721f0a/ : SUCCESS in 21m 15s\n- cross-keystone-py36 http://logs.openstack.org/91/665291/1/gate/cross-keystone-py36/312fc09/ : SUCCESS in 28m 01s\n- cross-neutron-py27 http://logs.openstack.org/91/665291/1/gate/cross-neutron-py27/22faa65/ : SUCCESS in 24m 37s\n- cross-neutron-py36 http://logs.openstack.org/91/665291/1/gate/cross-neutron-py36/063117b/ : SUCCESS in 19m 23s\n- cross-nova-py27 http://logs.openstack.org/91/665291/1/gate/cross-nova-py27/efdccb5/ : SUCCESS in 11m 24s\n- cross-nova-py36 http://logs.openstack.org/91/665291/1/gate/cross-nova-py36/cba2e4b/ : SUCCESS in 11m 50s\n- cross-nova-functional http://logs.openstack.org/91/665291/1/gate/cross-nova-functional/33fee7f/ : SUCCESS in 18m 20s\n- cross-osc-tox-docs http://logs.openstack.org/91/665291/1/gate/cross-osc-tox-docs/358a269/html/ : SUCCESS in 5m 05s\n- cross-swift-py27 http://logs.openstack.org/91/665291/1/gate/cross-swift-py27/40ce5da/ : SUCCESS in 12m 41s\n- cross-swift-py36 http://logs.openstack.org/91/665291/1/gate/cross-swift-py36/b44ff07/ : SUCCESS in 8m 52s\n- requirements-integration http://logs.openstack.org/91/665291/1/gate/requirements-integration/5194431/ : SUCCESS in 51m 24s\n- neutron-grenade-multinode http://logs.openstack.org/91/665291/1/gate/neutron-grenade-multinode/fd0985b/ : SUCCESS in 1h 09m 06s\n- tempest-full http://logs.openstack.org/91/665291/1/gate/tempest-full/702d38d/ : SUCCESS in 1h 42m 47s\n- neutron-grenade http://logs.openstack.org/91/665291/1/gate/neutron-grenade/8957a9b/ : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0fb9a12c24256a08f4e1ea30010c82be203bb6c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:55:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58beb385f6d144427195de25cc715872dbf5cb39","revisions":{"58beb385f6d144427195de25cc715872dbf5cb39":{"kind":"REWORK","_number":1,"created":"2019-06-14 04:06:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/91/665291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/665291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/665291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/665291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/665291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/665291/1"}}},"commit":{"parents":[{"commit":"67d9f8452bfec111f99e8e655a7372c414d0e6b2","subject":"Merge \"update constraint for octavia-lib to new release 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67d9f8452bfec111f99e8e655a7372c414d0e6b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-14 04:06:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-14 04:06:14.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 6.1.0","message":"update constraint for keystonemiddleware to new release 6.1.0\n\nChange-Id: Ia9d8c77926a41570f58ad513c9285a4768030621\nmeta:version: 6.1.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Jens Harbott \u003cj.harbott@x-ion.de\u003e\nmeta:release:Commit: Jens Harbott \u003cj.harbott@x-ion.de\u003e\nmeta:release:Change-Id: I862c5695d6a1fbf14b53c35af02b8a946f6a47d8\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58beb385f6d144427195de25cc715872dbf5cb39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/58beb385f6d144427195de25cc715872dbf5cb39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
