)]}'
{"id":"openstack%2Frequirements~665294","triplet_id":"openstack%2Frequirements~master~Idfabcb67adad4183b886bdc69bd82272bc1c6081","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idfabcb67adad4183b886bdc69bd82272bc1c6081","subject":"update constraint for oslo.config to new release 6.10.0","status":"MERGED","created":"2019-06-14 04:15:42.000000000","updated":"2019-06-15 12:09:17.000000000","submitted":"2019-06-15 12:09:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665294-1560600557320-eb0806c3","meta_rev_id":"e0d55953a70ce9822a822228eab9dede2d71fd2f","_number":665294,"virtual_id_number":665294,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-06-14 13:26:11.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-06-15 12:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-15 00:27:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-15 00:27:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 13:26:11.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-06-15 00:27:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-15 12:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"236a67e28185c3afb4601b030573a2d1a2aa7ecb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-14 04:15:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ec8ab637998ac8fc1b930d92627c16f062071aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 11:22:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/94/665294/1/check/openstack-tox-docs/76d58f0/html/ : SUCCESS in 3m 02s\n- openstack-tox-pypy http://logs.openstack.org/94/665294/1/check/openstack-tox-pypy/f6d6dce/ : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/94/665294/1/check/openstack-tox-pep8/00d942a/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/94/665294/1/check/openstack-tox-py27/346a53a/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/94/665294/1/check/openstack-tox-py36/8ffedc3/ : SUCCESS in 2m 51s\n- openstack-tox-validate http://logs.openstack.org/94/665294/1/check/openstack-tox-validate/1fff8e2/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/94/665294/1/check/requirements-tox-babel/607a9f6/ : SUCCESS in 2m 53s\n- requirements-tox-py27-check-uc http://logs.openstack.org/94/665294/1/check/requirements-tox-py27-check-uc/672dbc3/ : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc http://logs.openstack.org/94/665294/1/check/requirements-tox-py36-check-uc/2d3b542/ : SUCCESS in 6m 04s\n- cross-cinder-py27 http://logs.openstack.org/94/665294/1/check/cross-cinder-py27/e511749/ : SUCCESS in 7m 39s\n- cross-cinder-py36 http://logs.openstack.org/94/665294/1/check/cross-cinder-py36/e6e0f0a/ : SUCCESS in 8m 57s\n- cross-glance-py27 http://logs.openstack.org/94/665294/1/check/cross-glance-py27/e2c8461/ : SUCCESS in 4m 49s\n- cross-glance-py36 http://logs.openstack.org/94/665294/1/check/cross-glance-py36/c14e676/ : SUCCESS in 4m 25s\n- cross-horizon-py27 http://logs.openstack.org/94/665294/1/check/cross-horizon-py27/0051f2c/ : SUCCESS in 10m 45s\n- cross-horizon-py36 http://logs.openstack.org/94/665294/1/check/cross-horizon-py36/5c85da4/ : SUCCESS in 9m 54s\n- cross-horizon-npm http://logs.openstack.org/94/665294/1/check/cross-horizon-npm/efb4f25/ : SUCCESS in 6m 16s\n- cross-keystone-py27 http://logs.openstack.org/94/665294/1/check/cross-keystone-py27/1b427c8/ : SUCCESS in 23m 38s\n- cross-keystone-py36 http://logs.openstack.org/94/665294/1/check/cross-keystone-py36/629d30d/ : SUCCESS in 22m 57s\n- cross-neutron-py27 http://logs.openstack.org/94/665294/1/check/cross-neutron-py27/96197fc/ : SUCCESS in 24m 32s\n- cross-neutron-py36 http://logs.openstack.org/94/665294/1/check/cross-neutron-py36/d49f0b5/ : SUCCESS in 22m 34s\n- cross-nova-py27 http://logs.openstack.org/94/665294/1/check/cross-nova-py27/8949657/ : FAILURE in 38m 33s\n- cross-nova-py36 http://logs.openstack.org/94/665294/1/check/cross-nova-py36/6a6b5fd/ : SUCCESS in 13m 21s\n- cross-nova-functional http://logs.openstack.org/94/665294/1/check/cross-nova-functional/82e525c/ : SUCCESS in 17m 14s\n- cross-osc-tox-docs http://logs.openstack.org/94/665294/1/check/cross-osc-tox-docs/decde2f/html/ : SUCCESS in 8m 05s\n- cross-swift-py27 http://logs.openstack.org/94/665294/1/check/cross-swift-py27/44d0f81/ : SUCCESS in 7m 54s\n- cross-swift-py36 http://logs.openstack.org/94/665294/1/check/cross-swift-py36/451c3d7/ : SUCCESS in 8m 31s\n- requirements-integration http://logs.openstack.org/94/665294/1/check/requirements-integration/841f675/ : SUCCESS in 57m 06s\n- neutron-grenade-multinode http://logs.openstack.org/94/665294/1/check/neutron-grenade-multinode/8fae25e/ : SUCCESS in 1h 07m 16s\n- tempest-full http://logs.openstack.org/94/665294/1/check/tempest-full/3b4e808/ : SUCCESS in 2h 04m 54s\n- neutron-grenade http://logs.openstack.org/94/665294/1/check/neutron-grenade/9cffdaf/ : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3e323678a55014a52f5dc2d638a4c555ac165b31","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-14 13:26:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"533b4c92d50a792b5d85417b1a08cb3eab8e124d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-14 14:56:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"256f865f565629f8aed32ab54164cbf19a6ab933","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 22:52:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/94/665294/1/check/openstack-tox-docs/18b66f1/html/ : SUCCESS in 3m 15s\n- openstack-tox-pypy http://logs.openstack.org/94/665294/1/check/openstack-tox-pypy/de8e241/ : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/94/665294/1/check/openstack-tox-pep8/8f38a16/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/94/665294/1/check/openstack-tox-py27/86ff85d/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/94/665294/1/check/openstack-tox-py36/e92e763/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/94/665294/1/check/openstack-tox-validate/4fadf9e/ : SUCCESS in 2m 54s\n- requirements-tox-babel http://logs.openstack.org/94/665294/1/check/requirements-tox-babel/8abdf9e/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/94/665294/1/check/requirements-tox-py27-check-uc/bfd306f/ : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc http://logs.openstack.org/94/665294/1/check/requirements-tox-py36-check-uc/6d0052e/ : SUCCESS in 6m 09s\n- cross-cinder-py27 http://logs.openstack.org/94/665294/1/check/cross-cinder-py27/e122612/ : SUCCESS in 12m 53s\n- cross-cinder-py36 http://logs.openstack.org/94/665294/1/check/cross-cinder-py36/814e279/ : FAILURE in 14m 29s\n- cross-glance-py27 http://logs.openstack.org/94/665294/1/check/cross-glance-py27/512578f/ : SUCCESS in 5m 32s\n- cross-glance-py36 http://logs.openstack.org/94/665294/1/check/cross-glance-py36/5c33872/ : SUCCESS in 5m 55s\n- cross-horizon-py27 http://logs.openstack.org/94/665294/1/check/cross-horizon-py27/b513801/ : SUCCESS in 11m 18s\n- cross-horizon-py36 http://logs.openstack.org/94/665294/1/check/cross-horizon-py36/f2222eb/ : SUCCESS in 9m 13s\n- cross-horizon-npm http://logs.openstack.org/94/665294/1/check/cross-horizon-npm/dd91517/ : SUCCESS in 6m 03s\n- cross-keystone-py27 http://logs.openstack.org/94/665294/1/check/cross-keystone-py27/e211d92/ : SUCCESS in 20m 47s\n- cross-keystone-py36 http://logs.openstack.org/94/665294/1/check/cross-keystone-py36/9060167/ : SUCCESS in 26m 05s\n- cross-neutron-py27 http://logs.openstack.org/94/665294/1/check/cross-neutron-py27/bdbfcbc/ : SUCCESS in 27m 54s\n- cross-neutron-py36 http://logs.openstack.org/94/665294/1/check/cross-neutron-py36/5eec402/ : SUCCESS in 23m 04s\n- cross-nova-py27 http://logs.openstack.org/94/665294/1/check/cross-nova-py27/f2cd800/ : SUCCESS in 15m 15s\n- cross-nova-py36 http://logs.openstack.org/94/665294/1/check/cross-nova-py36/7ddd396/ : SUCCESS in 11m 35s\n- cross-nova-functional http://logs.openstack.org/94/665294/1/check/cross-nova-functional/222e574/ : SUCCESS in 18m 13s\n- cross-osc-tox-docs http://logs.openstack.org/94/665294/1/check/cross-osc-tox-docs/eefbbf6/html/ : SUCCESS in 5m 53s\n- cross-swift-py27 http://logs.openstack.org/94/665294/1/check/cross-swift-py27/1ffda84/ : SUCCESS in 8m 06s\n- cross-swift-py36 http://logs.openstack.org/94/665294/1/check/cross-swift-py36/a8e7939/ : SUCCESS in 8m 07s\n- requirements-integration http://logs.openstack.org/94/665294/1/check/requirements-integration/65cdf75/ : SUCCESS in 1h 03m 52s\n- neutron-grenade-multinode http://logs.openstack.org/94/665294/1/check/neutron-grenade-multinode/09a0a5a/ : SUCCESS in 1h 09m 37s\n- tempest-full http://logs.openstack.org/94/665294/1/check/tempest-full/eedad02/ : SUCCESS in 1h 53m 11s\n- neutron-grenade http://logs.openstack.org/94/665294/1/check/neutron-grenade/5dcad25/ : SUCCESS in 1h 06m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b406e7c6336b9af0d378f51b669fe1649fa413ea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-15 00:27:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50afcc3afedf2086f5280137e5b83ba7c12f9d68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 08:33:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/94/665294/1/check/openstack-tox-docs/0168e38/html/ : SUCCESS in 2m 41s\n- openstack-tox-pypy http://logs.openstack.org/94/665294/1/check/openstack-tox-pypy/118b6e4/ : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/94/665294/1/check/openstack-tox-pep8/73fb0e5/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/94/665294/1/check/openstack-tox-py27/0c3f90b/ : SUCCESS in 3m 05s\n- openstack-tox-py36 http://logs.openstack.org/94/665294/1/check/openstack-tox-py36/0bd5f00/ : SUCCESS in 3m 06s\n- openstack-tox-validate http://logs.openstack.org/94/665294/1/check/openstack-tox-validate/144b34d/ : SUCCESS in 2m 50s\n- requirements-tox-babel http://logs.openstack.org/94/665294/1/check/requirements-tox-babel/8350d57/ : SUCCESS in 3m 15s\n- requirements-tox-py27-check-uc http://logs.openstack.org/94/665294/1/check/requirements-tox-py27-check-uc/f2f6009/ : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc http://logs.openstack.org/94/665294/1/check/requirements-tox-py36-check-uc/14bdfa0/ : SUCCESS in 6m 47s\n- cross-cinder-py27 http://logs.openstack.org/94/665294/1/check/cross-cinder-py27/be361b6/ : SUCCESS in 8m 33s\n- cross-cinder-py36 http://logs.openstack.org/94/665294/1/check/cross-cinder-py36/c7ff601/ : SUCCESS in 8m 32s\n- cross-glance-py27 http://logs.openstack.org/94/665294/1/check/cross-glance-py27/b1c0a47/ : SUCCESS in 4m 32s\n- cross-glance-py36 http://logs.openstack.org/94/665294/1/check/cross-glance-py36/0b11ee0/ : SUCCESS in 4m 40s\n- cross-horizon-py27 http://logs.openstack.org/94/665294/1/check/cross-horizon-py27/3eb1b1c/ : SUCCESS in 9m 18s\n- cross-horizon-py36 http://logs.openstack.org/94/665294/1/check/cross-horizon-py36/9d6fe14/ : SUCCESS in 9m 04s\n- cross-horizon-npm http://logs.openstack.org/94/665294/1/check/cross-horizon-npm/e09ed73/ : SUCCESS in 6m 00s\n- cross-keystone-py27 http://logs.openstack.org/94/665294/1/check/cross-keystone-py27/a55e954/ : SUCCESS in 21m 07s\n- cross-keystone-py36 http://logs.openstack.org/94/665294/1/check/cross-keystone-py36/39610e1/ : SUCCESS in 26m 29s\n- cross-neutron-py27 http://logs.openstack.org/94/665294/1/check/cross-neutron-py27/b1d9791/ : SUCCESS in 22m 23s\n- cross-neutron-py36 http://logs.openstack.org/94/665294/1/check/cross-neutron-py36/f366379/ : SUCCESS in 22m 27s\n- cross-nova-py27 http://logs.openstack.org/94/665294/1/check/cross-nova-py27/31c8cc6/ : SUCCESS in 13m 02s\n- cross-nova-py36 http://logs.openstack.org/94/665294/1/check/cross-nova-py36/525a3c4/ : SUCCESS in 11m 27s\n- cross-nova-functional http://logs.openstack.org/94/665294/1/check/cross-nova-functional/4418bb2/ : SUCCESS in 19m 25s\n- cross-osc-tox-docs http://logs.openstack.org/94/665294/1/check/cross-osc-tox-docs/53854c0/html/ : SUCCESS in 5m 44s\n- cross-swift-py27 http://logs.openstack.org/94/665294/1/check/cross-swift-py27/6000a30/ : SUCCESS in 9m 38s\n- cross-swift-py36 http://logs.openstack.org/94/665294/1/check/cross-swift-py36/55f6987/ : SUCCESS in 11m 20s\n- requirements-integration http://logs.openstack.org/94/665294/1/check/requirements-integration/48de639/ : SUCCESS in 56m 12s\n- neutron-grenade-multinode http://logs.openstack.org/94/665294/1/check/neutron-grenade-multinode/bbbb589/ : SUCCESS in 1h 10m 44s\n- tempest-full http://logs.openstack.org/94/665294/1/check/tempest-full/8bb0c29/ : SUCCESS in 1h 45m 52s\n- neutron-grenade http://logs.openstack.org/94/665294/1/check/neutron-grenade/1c16a89/ : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f82178ebb57e9ab46dffee9514205b8659f982d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 08:33:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb3e614e4808d2ba132bbdc277fae6036e5ca520","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 12:09:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"242f5b49293261d23e23200396e695402f7132eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 12:09:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/94/665294/1/gate/openstack-tox-docs/bb822cb/html/ : SUCCESS in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/94/665294/1/gate/openstack-tox-pep8/74761c9/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/94/665294/1/gate/openstack-tox-py27/dae62ba/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/94/665294/1/gate/openstack-tox-py36/3f67031/ : SUCCESS in 3m 42s\n- openstack-tox-validate http://logs.openstack.org/94/665294/1/gate/openstack-tox-validate/a48c16d/ : SUCCESS in 3m 25s\n- requirements-tox-babel http://logs.openstack.org/94/665294/1/gate/requirements-tox-babel/ceb91db/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/94/665294/1/gate/requirements-tox-py27-check-uc/fae287c/ : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc http://logs.openstack.org/94/665294/1/gate/requirements-tox-py36-check-uc/8f78618/ : SUCCESS in 6m 24s\n- cross-cinder-py27 http://logs.openstack.org/94/665294/1/gate/cross-cinder-py27/ceeb326/ : SUCCESS in 9m 42s\n- cross-cinder-py36 http://logs.openstack.org/94/665294/1/gate/cross-cinder-py36/737ca23/ : SUCCESS in 9m 25s\n- cross-glance-py27 http://logs.openstack.org/94/665294/1/gate/cross-glance-py27/8ee5ee6/ : SUCCESS in 4m 37s\n- cross-glance-py36 http://logs.openstack.org/94/665294/1/gate/cross-glance-py36/07ab9a4/ : SUCCESS in 4m 32s\n- cross-horizon-py27 http://logs.openstack.org/94/665294/1/gate/cross-horizon-py27/5d1521c/ : SUCCESS in 10m 33s\n- cross-horizon-py36 http://logs.openstack.org/94/665294/1/gate/cross-horizon-py36/80662bc/ : SUCCESS in 7m 15s\n- cross-horizon-npm http://logs.openstack.org/94/665294/1/gate/cross-horizon-npm/104720e/ : SUCCESS in 6m 01s\n- cross-keystone-py27 http://logs.openstack.org/94/665294/1/gate/cross-keystone-py27/e5e1bf4/ : SUCCESS in 21m 35s\n- cross-keystone-py36 http://logs.openstack.org/94/665294/1/gate/cross-keystone-py36/15c3109/ : SUCCESS in 23m 59s\n- cross-neutron-py27 http://logs.openstack.org/94/665294/1/gate/cross-neutron-py27/4b02b09/ : SUCCESS in 23m 22s\n- cross-neutron-py36 http://logs.openstack.org/94/665294/1/gate/cross-neutron-py36/3d7d428/ : SUCCESS in 21m 51s\n- cross-nova-py27 http://logs.openstack.org/94/665294/1/gate/cross-nova-py27/44f4a5b/ : SUCCESS in 14m 42s\n- cross-nova-py36 http://logs.openstack.org/94/665294/1/gate/cross-nova-py36/6008d4a/ : SUCCESS in 15m 52s\n- cross-nova-functional http://logs.openstack.org/94/665294/1/gate/cross-nova-functional/008b67d/ : SUCCESS in 20m 27s\n- cross-osc-tox-docs http://logs.openstack.org/94/665294/1/gate/cross-osc-tox-docs/7839fce/html/ : SUCCESS in 6m 03s\n- cross-swift-py27 http://logs.openstack.org/94/665294/1/gate/cross-swift-py27/789b7a3/ : SUCCESS in 9m 04s\n- cross-swift-py36 http://logs.openstack.org/94/665294/1/gate/cross-swift-py36/f6568b5/ : SUCCESS in 10m 11s\n- requirements-integration http://logs.openstack.org/94/665294/1/gate/requirements-integration/9d91b27/ : SUCCESS in 51m 48s\n- neutron-grenade-multinode http://logs.openstack.org/94/665294/1/gate/neutron-grenade-multinode/3f0a3c3/ : SUCCESS in 1h 06m 53s\n- tempest-full http://logs.openstack.org/94/665294/1/gate/tempest-full/a828890/ : SUCCESS in 1h 50m 34s\n- neutron-grenade http://logs.openstack.org/94/665294/1/gate/neutron-grenade/034935a/ : SUCCESS in 44m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec3b7034972e45890d313f8b723efbeef1195bc2","revisions":{"ec3b7034972e45890d313f8b723efbeef1195bc2":{"kind":"REWORK","_number":1,"created":"2019-06-14 04:15:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/94/665294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/94/665294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/665294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/665294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/94/665294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/94/665294/1"}}},"commit":{"parents":[{"commit":"67d9f8452bfec111f99e8e655a7372c414d0e6b2","subject":"Merge \"update constraint for octavia-lib to new release 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67d9f8452bfec111f99e8e655a7372c414d0e6b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-14 04:15:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-14 04:15:40.000000000","tz":0},"subject":"update constraint for oslo.config to new release 6.10.0","message":"update constraint for oslo.config to new release 6.10.0\n\nChange-Id: Idfabcb67adad4183b886bdc69bd82272bc1c6081\nmeta:version: 6.10.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ic36a8a77f83f5513bd95781e445231cfb55edd3a\nmeta:release:Code-Review+1: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec3b7034972e45890d313f8b723efbeef1195bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec3b7034972e45890d313f8b723efbeef1195bc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
