)]}'
{"id":"openstack%2Frequirements~665296","triplet_id":"openstack%2Frequirements~master~Id33da669015332d7236d381eb9b37969474770b1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id33da669015332d7236d381eb9b37969474770b1","subject":"update constraint for pbr to new release 5.3.0","status":"MERGED","created":"2019-06-14 04:20:07.000000000","updated":"2019-06-14 20:57:28.000000000","submitted":"2019-06-14 20:57:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665296-1560545848082-1e9fde82","meta_rev_id":"6e1a1b861a33bbffe62f71e508dfe92333950c5e","_number":665296,"virtual_id_number":665296,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-14 20:57:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-14 14:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-14 14:55:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-14 14:55:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-14 20:57:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6b5503a5ca89a1cca8ac5f54d49bc6336e7e390","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-14 04:20:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d09a024bc9ddbee83a269d1d5a97c886863b00c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 10:49:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/665296/1/check/openstack-tox-docs/2b81791/html/ : SUCCESS in 4m 53s\n- openstack-tox-pypy http://logs.openstack.org/96/665296/1/check/openstack-tox-pypy/970698a/ : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/96/665296/1/check/openstack-tox-pep8/2f51e28/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/96/665296/1/check/openstack-tox-py27/9cf78f6/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/96/665296/1/check/openstack-tox-py36/49c0c84/ : SUCCESS in 3m 24s\n- openstack-tox-validate http://logs.openstack.org/96/665296/1/check/openstack-tox-validate/79c2d38/ : SUCCESS in 3m 32s\n- requirements-tox-babel http://logs.openstack.org/96/665296/1/check/requirements-tox-babel/fab285b/ : SUCCESS in 3m 08s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/665296/1/check/requirements-tox-py27-check-uc/84f9299/ : SUCCESS in 6m 03s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/665296/1/check/requirements-tox-py36-check-uc/72379d4/ : SUCCESS in 8m 10s\n- cross-cinder-py27 http://logs.openstack.org/96/665296/1/check/cross-cinder-py27/325d9ca/ : SUCCESS in 8m 52s\n- cross-cinder-py36 http://logs.openstack.org/96/665296/1/check/cross-cinder-py36/6228634/ : SUCCESS in 8m 55s\n- cross-glance-py27 http://logs.openstack.org/96/665296/1/check/cross-glance-py27/6948d93/ : SUCCESS in 4m 27s\n- cross-glance-py36 http://logs.openstack.org/96/665296/1/check/cross-glance-py36/f197dd2/ : SUCCESS in 4m 49s\n- cross-horizon-py27 http://logs.openstack.org/96/665296/1/check/cross-horizon-py27/e1e02ea/ : SUCCESS in 9m 41s\n- cross-horizon-py36 http://logs.openstack.org/96/665296/1/check/cross-horizon-py36/625cf97/ : SUCCESS in 9m 31s\n- cross-horizon-npm http://logs.openstack.org/96/665296/1/check/cross-horizon-npm/b144179/ : SUCCESS in 6m 50s\n- cross-keystone-py27 http://logs.openstack.org/96/665296/1/check/cross-keystone-py27/bba823a/ : SUCCESS in 24m 05s\n- cross-keystone-py36 http://logs.openstack.org/96/665296/1/check/cross-keystone-py36/7783a8d/ : SUCCESS in 29m 24s\n- cross-neutron-py27 http://logs.openstack.org/96/665296/1/check/cross-neutron-py27/6758e64/ : SUCCESS in 30m 51s\n- cross-neutron-py36 http://logs.openstack.org/96/665296/1/check/cross-neutron-py36/c16b853/ : SUCCESS in 22m 30s\n- cross-nova-py27 http://logs.openstack.org/96/665296/1/check/cross-nova-py27/098b59e/ : SUCCESS in 11m 26s\n- cross-nova-py36 http://logs.openstack.org/96/665296/1/check/cross-nova-py36/d2f0f48/ : SUCCESS in 11m 23s\n- cross-nova-functional http://logs.openstack.org/96/665296/1/check/cross-nova-functional/af996f6/ : SUCCESS in 21m 15s\n- cross-osc-tox-docs http://logs.openstack.org/96/665296/1/check/cross-osc-tox-docs/7920818/html/ : SUCCESS in 5m 13s\n- cross-swift-py27 http://logs.openstack.org/96/665296/1/check/cross-swift-py27/45387b4/ : SUCCESS in 24m 22s\n- cross-swift-py36 http://logs.openstack.org/96/665296/1/check/cross-swift-py36/55ff79e/ : SUCCESS in 8m 30s\n- requirements-integration http://logs.openstack.org/96/665296/1/check/requirements-integration/bf50ea7/ : SUCCESS in 56m 28s\n- neutron-grenade-multinode http://logs.openstack.org/96/665296/1/check/neutron-grenade-multinode/6e408ac/ : SUCCESS in 1h 06m 59s\n- tempest-full http://logs.openstack.org/96/665296/1/check/tempest-full/f4baca7/ : SUCCESS in 1h 50m 41s\n- neutron-grenade http://logs.openstack.org/96/665296/1/check/neutron-grenade/d7264d9/ : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b8c8bb7859e364879e6154cea9d481537e7f36ea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-14 14:55:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3f37bfd6c501cc8896074e6a8a5c5647980e3f2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 14:55:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4865ed5284f6deab9acb043dc6bab3420530f0fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:57:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/665296/1/gate/openstack-tox-docs/42c8991/html/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/96/665296/1/gate/openstack-tox-pep8/b6da728/ : SUCCESS in 2m 45s\n- openstack-tox-py27 http://logs.openstack.org/96/665296/1/gate/openstack-tox-py27/a628be3/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/96/665296/1/gate/openstack-tox-py36/43c5d8f/ : SUCCESS in 3m 04s\n- openstack-tox-validate http://logs.openstack.org/96/665296/1/gate/openstack-tox-validate/5e425ca/ : SUCCESS in 3m 00s\n- requirements-tox-babel http://logs.openstack.org/96/665296/1/gate/requirements-tox-babel/1201361/ : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/96/665296/1/gate/requirements-tox-py27-check-uc/a293ed3/ : SUCCESS in 9m 36s\n- requirements-tox-py36-check-uc http://logs.openstack.org/96/665296/1/gate/requirements-tox-py36-check-uc/d731188/ : SUCCESS in 6m 36s\n- cross-cinder-py27 http://logs.openstack.org/96/665296/1/gate/cross-cinder-py27/fcdd71f/ : SUCCESS in 8m 30s\n- cross-cinder-py36 http://logs.openstack.org/96/665296/1/gate/cross-cinder-py36/6789344/ : SUCCESS in 9m 00s\n- cross-glance-py27 http://logs.openstack.org/96/665296/1/gate/cross-glance-py27/0e05ed3/ : SUCCESS in 5m 51s\n- cross-glance-py36 http://logs.openstack.org/96/665296/1/gate/cross-glance-py36/39a0947/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/96/665296/1/gate/cross-horizon-py27/d4b5867/ : SUCCESS in 10m 27s\n- cross-horizon-py36 http://logs.openstack.org/96/665296/1/gate/cross-horizon-py36/8c91715/ : SUCCESS in 9m 30s\n- cross-horizon-npm http://logs.openstack.org/96/665296/1/gate/cross-horizon-npm/2117464/ : SUCCESS in 6m 14s\n- cross-keystone-py27 http://logs.openstack.org/96/665296/1/gate/cross-keystone-py27/fb34e75/ : SUCCESS in 22m 42s\n- cross-keystone-py36 http://logs.openstack.org/96/665296/1/gate/cross-keystone-py36/3a60439/ : SUCCESS in 27m 27s\n- cross-neutron-py27 http://logs.openstack.org/96/665296/1/gate/cross-neutron-py27/817cac0/ : SUCCESS in 22m 57s\n- cross-neutron-py36 http://logs.openstack.org/96/665296/1/gate/cross-neutron-py36/2ed948c/ : SUCCESS in 22m 36s\n- cross-nova-py27 http://logs.openstack.org/96/665296/1/gate/cross-nova-py27/2d565dd/ : SUCCESS in 15m 09s\n- cross-nova-py36 http://logs.openstack.org/96/665296/1/gate/cross-nova-py36/13ddd6e/ : SUCCESS in 11m 14s\n- cross-nova-functional http://logs.openstack.org/96/665296/1/gate/cross-nova-functional/3e08797/ : SUCCESS in 20m 52s\n- cross-osc-tox-docs http://logs.openstack.org/96/665296/1/gate/cross-osc-tox-docs/665a7e6/html/ : SUCCESS in 5m 16s\n- cross-swift-py27 http://logs.openstack.org/96/665296/1/gate/cross-swift-py27/b39a7b0/ : SUCCESS in 8m 17s\n- cross-swift-py36 http://logs.openstack.org/96/665296/1/gate/cross-swift-py36/0df0b42/ : SUCCESS in 8m 18s\n- requirements-integration http://logs.openstack.org/96/665296/1/gate/requirements-integration/f7bd1ac/ : SUCCESS in 52m 34s\n- neutron-grenade-multinode http://logs.openstack.org/96/665296/1/gate/neutron-grenade-multinode/dd5d20d/ : SUCCESS in 1h 15m 03s\n- tempest-full http://logs.openstack.org/96/665296/1/gate/tempest-full/bc39988/ : SUCCESS in 1h 45m 07s\n- neutron-grenade http://logs.openstack.org/96/665296/1/gate/neutron-grenade/7d7c40e/ : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2aa4728ca9b29fa11bf9a24e8ef903be4d16c684","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:57:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00625a6ddb519fb682099298bb5bdb937c5eb6f0","revisions":{"00625a6ddb519fb682099298bb5bdb937c5eb6f0":{"kind":"REWORK","_number":1,"created":"2019-06-14 04:20:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/665296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/665296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/665296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/665296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/665296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/665296/1"}}},"commit":{"parents":[{"commit":"67d9f8452bfec111f99e8e655a7372c414d0e6b2","subject":"Merge \"update constraint for octavia-lib to new release 1.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/67d9f8452bfec111f99e8e655a7372c414d0e6b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-14 04:20:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-14 04:20:03.000000000","tz":0},"subject":"update constraint for pbr to new release 5.3.0","message":"update constraint for pbr to new release 5.3.0\n\nChange-Id: Id33da669015332d7236d381eb9b37969474770b1\nmeta:version: 5.3.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: Ic36a8a77f83f5513bd95781e445231cfb55edd3a\nmeta:release:Code-Review+1: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00625a6ddb519fb682099298bb5bdb937c5eb6f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00625a6ddb519fb682099298bb5bdb937c5eb6f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
