)]}'
{"id":"openstack%2Frequirements~665563","triplet_id":"openstack%2Frequirements~master~I9c28ee745e6d169eb8e2f15b7192b6460e11fd13","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I9c28ee745e6d169eb8e2f15b7192b6460e11fd13","subject":"update constraint for osc-lib to new release 1.13.0","status":"MERGED","created":"2019-06-17 00:50:09.000000000","updated":"2019-06-17 18:34:55.000000000","submitted":"2019-06-17 18:34:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665563-1560796495200-d473cd44","meta_rev_id":"fd319a62666781109b1458d36dbc8e8f0da5044b","_number":665563,"virtual_id_number":665563,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2019-06-17 09:08:10.000000000","_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":2,"date":"2019-06-17 18:34:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-17 15:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-17 15:55:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 09:08:10.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-06-17 15:55:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-17 18:34:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c774c722edec3ef9606a53c690d149ce7b813f7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-17 00:50:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34c12009f3de3b1674213bbf30d8e4d45b0e5a65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 02:25:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/63/665563/1/check/openstack-tox-docs/49c7c31/html/ : SUCCESS in 4m 20s\n- openstack-tox-pypy http://logs.openstack.org/63/665563/1/check/openstack-tox-pypy/9b98dff/ : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/63/665563/1/check/openstack-tox-pep8/73f23d9/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/63/665563/1/check/openstack-tox-py27/48b744d/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/63/665563/1/check/openstack-tox-py36/feb04e1/ : SUCCESS in 3m 03s\n- openstack-tox-validate http://logs.openstack.org/63/665563/1/check/openstack-tox-validate/e5c505c/ : SUCCESS in 2m 42s\n- requirements-tox-babel http://logs.openstack.org/63/665563/1/check/requirements-tox-babel/0bbb88c/ : SUCCESS in 3m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/63/665563/1/check/requirements-tox-py27-check-uc/630e07a/ : SUCCESS in 6m 42s\n- requirements-tox-py36-check-uc http://logs.openstack.org/63/665563/1/check/requirements-tox-py36-check-uc/4e0ab26/ : SUCCESS in 6m 41s\n- cross-cinder-py27 http://logs.openstack.org/63/665563/1/check/cross-cinder-py27/b879568/ : SUCCESS in 9m 13s\n- cross-cinder-py36 http://logs.openstack.org/63/665563/1/check/cross-cinder-py36/a4ffe2f/ : SUCCESS in 8m 41s\n- cross-glance-py27 http://logs.openstack.org/63/665563/1/check/cross-glance-py27/6b1e5ac/ : SUCCESS in 4m 24s\n- cross-glance-py36 http://logs.openstack.org/63/665563/1/check/cross-glance-py36/6ce84d2/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/63/665563/1/check/cross-horizon-py27/b7734a4/ : SUCCESS in 10m 48s\n- cross-horizon-py36 http://logs.openstack.org/63/665563/1/check/cross-horizon-py36/3c8c4f0/ : SUCCESS in 8m 43s\n- cross-horizon-npm http://logs.openstack.org/63/665563/1/check/cross-horizon-npm/907a14e/ : SUCCESS in 6m 14s\n- cross-keystone-py27 http://logs.openstack.org/63/665563/1/check/cross-keystone-py27/d24c721/ : SUCCESS in 26m 48s\n- cross-keystone-py36 http://logs.openstack.org/63/665563/1/check/cross-keystone-py36/3e204f1/ : SUCCESS in 23m 57s\n- cross-neutron-py27 http://logs.openstack.org/63/665563/1/check/cross-neutron-py27/61cd171/ : SUCCESS in 20m 48s\n- cross-neutron-py36 http://logs.openstack.org/63/665563/1/check/cross-neutron-py36/90e5343/ : FAILURE in 3m 50s\n- cross-nova-py27 http://logs.openstack.org/63/665563/1/check/cross-nova-py27/81cc443/ : FAILURE in 4m 51s\n- cross-nova-py36 http://logs.openstack.org/63/665563/1/check/cross-nova-py36/1a7978d/ : SUCCESS in 11m 38s\n- cross-nova-functional http://logs.openstack.org/63/665563/1/check/cross-nova-functional/fc6a9a7/ : FAILURE in 4m 40s\n- cross-osc-tox-docs http://logs.openstack.org/63/665563/1/check/cross-osc-tox-docs/836508c/ : POST_FAILURE in 4m 58s\n- cross-swift-py27 http://logs.openstack.org/63/665563/1/check/cross-swift-py27/39e6616/ : SUCCESS in 8m 58s\n- cross-swift-py36 http://logs.openstack.org/63/665563/1/check/cross-swift-py36/62d2a8b/ : FAILURE in 3m 37s\n- requirements-integration http://logs.openstack.org/63/665563/1/check/requirements-integration/8308f9f/ : SUCCESS in 1h 01m 12s\n- neutron-grenade-multinode http://logs.openstack.org/63/665563/1/check/neutron-grenade-multinode/78f6e4a/ : SUCCESS in 55m 45s\n- tempest-full http://logs.openstack.org/63/665563/1/check/tempest-full/4f76465/ : SUCCESS in 1h 34m 32s\n- neutron-grenade http://logs.openstack.org/63/665563/1/check/neutron-grenade/01d28d2/ : SUCCESS in 47m 26s","accounts_in_message":[],"_revision_number":1},{"id":"99470339f35c36161dc2f5e8fe157819de3cace7","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-17 09:08:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbcd72397895c2f81a59bd4ba228c1011657ec02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 10:46:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/665563/1/check/openstack-tox-docs/9a58ecf/html/ : SUCCESS in 3m 08s\n- openstack-tox-pypy http://logs.openstack.org/63/665563/1/check/openstack-tox-pypy/4278456/ : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/63/665563/1/check/openstack-tox-pep8/799c97d/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/63/665563/1/check/openstack-tox-py27/5a0f273/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/63/665563/1/check/openstack-tox-py36/a142d82/ : SUCCESS in 3m 35s\n- openstack-tox-validate http://logs.openstack.org/63/665563/1/check/openstack-tox-validate/018236a/ : SUCCESS in 2m 48s\n- requirements-tox-babel http://logs.openstack.org/63/665563/1/check/requirements-tox-babel/8548bee/ : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/63/665563/1/check/requirements-tox-py27-check-uc/aa28671/ : SUCCESS in 6m 38s\n- requirements-tox-py36-check-uc http://logs.openstack.org/63/665563/1/check/requirements-tox-py36-check-uc/b98c76b/ : SUCCESS in 7m 16s\n- cross-cinder-py27 http://logs.openstack.org/63/665563/1/check/cross-cinder-py27/03603ec/ : SUCCESS in 8m 15s\n- cross-cinder-py36 http://logs.openstack.org/63/665563/1/check/cross-cinder-py36/11f1a43/ : SUCCESS in 7m 21s\n- cross-glance-py27 http://logs.openstack.org/63/665563/1/check/cross-glance-py27/8344566/ : SUCCESS in 4m 13s\n- cross-glance-py36 http://logs.openstack.org/63/665563/1/check/cross-glance-py36/dcfb612/ : SUCCESS in 4m 26s\n- cross-horizon-py27 http://logs.openstack.org/63/665563/1/check/cross-horizon-py27/3636dde/ : SUCCESS in 12m 16s\n- cross-horizon-py36 http://logs.openstack.org/63/665563/1/check/cross-horizon-py36/a9a6961/ : SUCCESS in 9m 08s\n- cross-horizon-npm http://logs.openstack.org/63/665563/1/check/cross-horizon-npm/1f110bb/ : SUCCESS in 6m 08s\n- cross-keystone-py27 http://logs.openstack.org/63/665563/1/check/cross-keystone-py27/e102fe1/ : SUCCESS in 21m 30s\n- cross-keystone-py36 http://logs.openstack.org/63/665563/1/check/cross-keystone-py36/1961c1a/ : SUCCESS in 29m 12s\n- cross-neutron-py27 http://logs.openstack.org/63/665563/1/check/cross-neutron-py27/2f78f06/ : SUCCESS in 19m 32s\n- cross-neutron-py36 http://logs.openstack.org/63/665563/1/check/cross-neutron-py36/8a99a92/ : SUCCESS in 22m 07s\n- cross-nova-py27 http://logs.openstack.org/63/665563/1/check/cross-nova-py27/cf7c7b3/ : SUCCESS in 12m 12s\n- cross-nova-py36 http://logs.openstack.org/63/665563/1/check/cross-nova-py36/0e14c96/ : SUCCESS in 10m 22s\n- cross-nova-functional http://logs.openstack.org/63/665563/1/check/cross-nova-functional/2dce050/ : SUCCESS in 18m 17s\n- cross-osc-tox-docs http://logs.openstack.org/63/665563/1/check/cross-osc-tox-docs/631432e/html/ : SUCCESS in 5m 27s\n- cross-swift-py27 http://logs.openstack.org/63/665563/1/check/cross-swift-py27/ef45dab/ : SUCCESS in 7m 02s\n- cross-swift-py36 http://logs.openstack.org/63/665563/1/check/cross-swift-py36/92fa8c0/ : SUCCESS in 8m 07s\n- requirements-integration http://logs.openstack.org/63/665563/1/check/requirements-integration/f1fd648/ : SUCCESS in 58m 18s\n- neutron-grenade-multinode http://logs.openstack.org/63/665563/1/check/neutron-grenade-multinode/9dbe3a5/ : SUCCESS in 1h 09m 52s\n- tempest-full http://logs.openstack.org/63/665563/1/check/tempest-full/2b7495e/ : SUCCESS in 1h 33m 19s\n- neutron-grenade http://logs.openstack.org/63/665563/1/check/neutron-grenade/cf3a5e2/ : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f40a12a0e5c51fa9dc64742b490bf77c32c5bde1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-17 15:55:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6e5d15b6b969f12a5ab1ec915dd309c830e3cc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 15:55:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2918d743a59b759984a1806a530fd8936cf44aa7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:34:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/665563/1/gate/openstack-tox-docs/3f7d8a9/html/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/63/665563/1/gate/openstack-tox-pep8/1964316/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/63/665563/1/gate/openstack-tox-py27/69c2a6f/ : SUCCESS in 4m 07s\n- openstack-tox-py36 http://logs.openstack.org/63/665563/1/gate/openstack-tox-py36/17e6fdf/ : SUCCESS in 7m 26s\n- openstack-tox-validate http://logs.openstack.org/63/665563/1/gate/openstack-tox-validate/86747ae/ : SUCCESS in 4m 25s\n- requirements-tox-babel http://logs.openstack.org/63/665563/1/gate/requirements-tox-babel/bc8dc94/ : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc http://logs.openstack.org/63/665563/1/gate/requirements-tox-py27-check-uc/f43bd64/ : SUCCESS in 6m 44s\n- requirements-tox-py36-check-uc http://logs.openstack.org/63/665563/1/gate/requirements-tox-py36-check-uc/67d629a/ : SUCCESS in 6m 55s\n- cross-cinder-py27 http://logs.openstack.org/63/665563/1/gate/cross-cinder-py27/b44149a/ : SUCCESS in 8m 35s\n- cross-cinder-py36 http://logs.openstack.org/63/665563/1/gate/cross-cinder-py36/8cfa870/ : SUCCESS in 8m 36s\n- cross-glance-py27 http://logs.openstack.org/63/665563/1/gate/cross-glance-py27/372ce37/ : SUCCESS in 7m 17s\n- cross-glance-py36 http://logs.openstack.org/63/665563/1/gate/cross-glance-py36/6d39820/ : SUCCESS in 5m 00s\n- cross-horizon-py27 http://logs.openstack.org/63/665563/1/gate/cross-horizon-py27/489ba49/ : SUCCESS in 11m 06s\n- cross-horizon-py36 http://logs.openstack.org/63/665563/1/gate/cross-horizon-py36/4647cd3/ : SUCCESS in 9m 07s\n- cross-horizon-npm http://logs.openstack.org/63/665563/1/gate/cross-horizon-npm/fb10438/ : SUCCESS in 6m 43s\n- cross-keystone-py27 http://logs.openstack.org/63/665563/1/gate/cross-keystone-py27/abb736e/ : SUCCESS in 24m 41s\n- cross-keystone-py36 http://logs.openstack.org/63/665563/1/gate/cross-keystone-py36/31343be/ : SUCCESS in 24m 51s\n- cross-neutron-py27 http://logs.openstack.org/63/665563/1/gate/cross-neutron-py27/b39c1b5/ : SUCCESS in 26m 04s\n- cross-neutron-py36 http://logs.openstack.org/63/665563/1/gate/cross-neutron-py36/1da2064/ : SUCCESS in 22m 43s\n- cross-nova-py27 http://logs.openstack.org/63/665563/1/gate/cross-nova-py27/53fd89c/ : SUCCESS in 12m 14s\n- cross-nova-py36 http://logs.openstack.org/63/665563/1/gate/cross-nova-py36/5deadd7/ : SUCCESS in 13m 40s\n- cross-nova-functional http://logs.openstack.org/63/665563/1/gate/cross-nova-functional/245f877/ : SUCCESS in 22m 57s\n- cross-osc-tox-docs http://logs.openstack.org/63/665563/1/gate/cross-osc-tox-docs/dadacc6/html/ : SUCCESS in 8m 29s\n- cross-swift-py27 http://logs.openstack.org/63/665563/1/gate/cross-swift-py27/6f7beef/ : SUCCESS in 18m 34s\n- cross-swift-py36 http://logs.openstack.org/63/665563/1/gate/cross-swift-py36/b1d44c7/ : SUCCESS in 10m 28s\n- requirements-integration http://logs.openstack.org/63/665563/1/gate/requirements-integration/d04e380/ : SUCCESS in 55m 05s\n- neutron-grenade-multinode http://logs.openstack.org/63/665563/1/gate/neutron-grenade-multinode/619f18b/ : SUCCESS in 1h 06m 01s\n- tempest-full http://logs.openstack.org/63/665563/1/gate/tempest-full/2487240/ : SUCCESS in 1h 59m 26s\n- neutron-grenade http://logs.openstack.org/63/665563/1/gate/neutron-grenade/75ee884/ : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4412275d83a8c301b11882917800fb474781c820","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 18:34:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2ab20d3584b99302fc19646972d945235cd74a8","revisions":{"b2ab20d3584b99302fc19646972d945235cd74a8":{"kind":"REWORK","_number":1,"created":"2019-06-17 00:50:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/665563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/665563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/665563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/665563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/665563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/665563/1"}}},"commit":{"parents":[{"commit":"2e402cc4ce15c3a69bbdf1143a8135855fbdbfbd","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e402cc4ce15c3a69bbdf1143a8135855fbdbfbd"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-17 00:50:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-17 00:50:08.000000000","tz":0},"subject":"update constraint for osc-lib to new release 1.13.0","message":"update constraint for osc-lib to new release 1.13.0\n\nChange-Id: I9c28ee745e6d169eb8e2f15b7192b6460e11fd13\nmeta:version: 1.13.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Commit: Dean Troyer \u003cdtroyer@gmail.com\u003e\nmeta:release:Change-Id: I13eae9b480499cb8432ddab5ce9cfbb40c3530d3\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\nmeta:release:Workflow+1: Tony Breeds \u003ctony@bakeyournoodle.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2ab20d3584b99302fc19646972d945235cd74a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2ab20d3584b99302fc19646972d945235cd74a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
