)]}'
{"id":"openstack%2Frequirements~666206","triplet_id":"openstack%2Frequirements~master~I836a6218ff5fccf4aab74b37a5a682601a388106","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I836a6218ff5fccf4aab74b37a5a682601a388106","subject":"Use UTF-8 to decode subprocess output on py3","status":"MERGED","created":"2019-06-19 03:39:58.000000000","updated":"2019-06-24 12:14:52.000000000","submitted":"2019-06-24 12:14:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666206-1561378492938-bfd37fb7","meta_rev_id":"b6bbb6b6b67a565eb8967acfe0ecb9cae02fbd20","_number":666206,"virtual_id_number":666206,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2019-06-24 12:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-19 22:43:48.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-19 12:01:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-06-24 01:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-19 20:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-06-19 20:33:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 12:01:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-19 20:33:20.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-06-24 01:54:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-24 12:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e135d3d9440ac79573e235515509cfe491142bfd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-19 03:39:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25eb97eb965173b5a4ccfd37512c16c927c051f4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-19 03:40:40.000000000","message":"Patch Set 1:\n\nSee http://logs.openstack.org/58/665758/2/check/requirements-check/1bb2ae1/job-output.txt.gz#_2019-06-19_02_14_01_844364 for an example of the bad in the gate.","accounts_in_message":[],"_revision_number":1},{"id":"f89637740ba105a4511b91cb2c1c9c6a8dd3de42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 10:20:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/666206/1/check/openstack-tox-docs/540b400/html/ : SUCCESS in 4m 26s\n- openstack-tox-pypy http://logs.openstack.org/06/666206/1/check/openstack-tox-pypy/721a1f0/ : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/06/666206/1/check/openstack-tox-pep8/ccf8710/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/06/666206/1/check/openstack-tox-py27/2eb4faf/ : SUCCESS in 3m 28s\n- openstack-tox-py36 http://logs.openstack.org/06/666206/1/check/openstack-tox-py36/e96519b/ : SUCCESS in 4m 09s\n- openstack-tox-validate http://logs.openstack.org/06/666206/1/check/openstack-tox-validate/2f7a0e8/ : SUCCESS in 3m 08s\n- requirements-integration http://logs.openstack.org/06/666206/1/check/requirements-integration/d550326/ : SUCCESS in 58m 00s\n- neutron-grenade-multinode http://logs.openstack.org/06/666206/1/check/neutron-grenade-multinode/a4fb4ea/ : SUCCESS in 1h 09m 37s\n- tempest-full http://logs.openstack.org/06/666206/1/check/tempest-full/4e878e2/ : SUCCESS in 1h 42m 44s\n- neutron-grenade http://logs.openstack.org/06/666206/1/check/neutron-grenade/32a9121/ : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d83d15b6c1e1b87fbf5eb0978d40e9022484e1a6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-19 12:01:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"42495c6b2950196a2936822336c97ef02e7a7826","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-19 20:33:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Tim!\n\nI don\u0027t understand why we haven\u0027t seen this before.\n\n#confused :(","accounts_in_message":[],"_revision_number":1},{"id":"08212144cf5f2cf758f9f8267e38bf324d6bb791","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 20:33:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c5916637108423313a4ecbf1243bdcf7b899eea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 22:29:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/06/666206/1/gate/openstack-tox-docs/1f23a7d/html/ : SUCCESS in 2m 55s\n- openstack-tox-pep8 http://logs.openstack.org/06/666206/1/gate/openstack-tox-pep8/0ef4944/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/06/666206/1/gate/openstack-tox-py27/72ca1d4/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/06/666206/1/gate/openstack-tox-py36/155ce52/ : SUCCESS in 3m 27s\n- openstack-tox-validate http://logs.openstack.org/06/666206/1/gate/openstack-tox-validate/9201ce8/ : SUCCESS in 3m 19s\n- requirements-integration http://logs.openstack.org/06/666206/1/gate/requirements-integration/7387874/ : SUCCESS in 1h 01m 07s\n- neutron-grenade-multinode http://logs.openstack.org/06/666206/1/gate/neutron-grenade-multinode/0b30964/ : FAILURE in 16m 14s\n- tempest-full http://logs.openstack.org/06/666206/1/gate/tempest-full/bd09a02/ : SUCCESS in 1h 49m 27s\n- neutron-grenade http://logs.openstack.org/06/666206/1/gate/neutron-grenade/d3cdeaf/ : SUCCESS in 54m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e355043467e96c4c71c60cefcf95938f3ef6815e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-06-19 22:43:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"df8711692ae0a4b41be96d22774bf1996c4ee065","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 03:30:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/06/666206/1/check/openstack-tox-docs/1522588/html/ : SUCCESS in 3m 00s\n- openstack-tox-pypy http://logs.openstack.org/06/666206/1/check/openstack-tox-pypy/f7bcd5c/ : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/06/666206/1/check/openstack-tox-pep8/60bb54a/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/06/666206/1/check/openstack-tox-py27/ef00c7c/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/06/666206/1/check/openstack-tox-py36/4d35be4/ : SUCCESS in 2m 56s\n- openstack-tox-validate http://logs.openstack.org/06/666206/1/check/openstack-tox-validate/7a35b9b/ : SUCCESS in 3m 26s\n- requirements-integration http://logs.openstack.org/06/666206/1/check/requirements-integration/c44be11/ : SUCCESS in 49m 36s\n- neutron-grenade-multinode http://logs.openstack.org/06/666206/1/check/neutron-grenade-multinode/c1d2cdb/ : FAILURE in 12m 25s\n- tempest-full http://logs.openstack.org/06/666206/1/check/tempest-full/85d4c76/ : SUCCESS in 1h 31m 15s\n- neutron-grenade http://logs.openstack.org/06/666206/1/check/neutron-grenade/6eb4797/ : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":1},{"id":"327890c7075682ff0534af56e2a2f5709580eb07","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-20 13:17:25.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5ce3c9186288e076dc27b965aab82327dc08ec6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-23 06:40:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"442704d253b39582584128b3482a7ace77646465","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-24 01:54:04.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9771db0f78b983ac70ef5a4fc9ca3a021dc4e00e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 03:46:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/06/666206/1/check/openstack-tox-docs/1594c24/html/ : SUCCESS in 3m 58s\n- openstack-tox-pypy http://logs.openstack.org/06/666206/1/check/openstack-tox-pypy/b7c84bc/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/06/666206/1/check/openstack-tox-pep8/55148d0/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/06/666206/1/check/openstack-tox-py27/6fb7dcf/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/06/666206/1/check/openstack-tox-py36/712d4fe/ : SUCCESS in 4m 12s\n- openstack-tox-validate http://logs.openstack.org/06/666206/1/check/openstack-tox-validate/c5a5960/ : SUCCESS in 2m 58s\n- requirements-integration http://logs.openstack.org/06/666206/1/check/requirements-integration/b608d3f/ : SUCCESS in 50m 38s\n- neutron-grenade-multinode http://logs.openstack.org/06/666206/1/check/neutron-grenade-multinode/6445978/ : SUCCESS in 1h 06m 15s\n- tempest-full http://logs.openstack.org/06/666206/1/check/tempest-full/e8e3fae/ : FAILURE in 1h 47m 23s\n- neutron-grenade http://logs.openstack.org/06/666206/1/check/neutron-grenade/b291b15/ : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0ffa09af7610ca0b4387f03e4133d10e73e508ed","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-24 03:53:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"728a1c1a1e13a0aadf0e8b0d5ebbd21eab9786f7","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-24 04:53:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21747b91b9beccee09b66c1bc10d1e24c60f6189","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2019-06-24 04:53:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d70a86e3244ca529d19eb56587370cd1129c6bfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 05:56:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/666206/1/check/openstack-tox-docs/ed73bb3/html/ : SUCCESS in 3m 55s\n- openstack-tox-pypy http://logs.openstack.org/06/666206/1/check/openstack-tox-pypy/bdda978/ : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/06/666206/1/check/openstack-tox-pep8/c499d80/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/06/666206/1/check/openstack-tox-py27/9ddd7d9/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/06/666206/1/check/openstack-tox-py36/c00037c/ : SUCCESS in 3m 49s\n- openstack-tox-validate http://logs.openstack.org/06/666206/1/check/openstack-tox-validate/e504291/ : SUCCESS in 3m 34s\n- requirements-integration http://logs.openstack.org/06/666206/1/check/requirements-integration/0af3a39/ : SUCCESS in 51m 40s\n- neutron-grenade-multinode http://logs.openstack.org/06/666206/1/check/neutron-grenade-multinode/83975dc/ : SUCCESS in 1h 02m 22s\n- tempest-full http://logs.openstack.org/06/666206/1/check/tempest-full/a5cccc9/ : SUCCESS in 1h 50m 16s\n- neutron-grenade http://logs.openstack.org/06/666206/1/check/neutron-grenade/f57d468/ : SUCCESS in 51m 34s","accounts_in_message":[],"_revision_number":1},{"id":"09a654b3b066e7f6bbbe818d8584328c23405895","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 05:56:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e1dab3b09c9e2f67cdcdc95f9db51328ec5db67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 08:09:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/06/666206/1/gate/openstack-tox-docs/0ee70f4/html/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/06/666206/1/gate/openstack-tox-pep8/7f79aac/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/06/666206/1/gate/openstack-tox-py27/f3eeeda/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/06/666206/1/gate/openstack-tox-py36/220e248/ : SUCCESS in 4m 39s\n- openstack-tox-validate http://logs.openstack.org/06/666206/1/gate/openstack-tox-validate/15c2669/ : SUCCESS in 4m 12s\n- requirements-integration http://logs.openstack.org/06/666206/1/gate/requirements-integration/e13ba65/ : SUCCESS in 1h 06m 17s\n- neutron-grenade-multinode http://logs.openstack.org/06/666206/1/gate/neutron-grenade-multinode/74ad07e/ : SUCCESS in 1h 18m 27s\n- tempest-full http://logs.openstack.org/06/666206/1/gate/tempest-full/a6879b8/ : TIMED_OUT in 2h 07m 09s\n- neutron-grenade http://logs.openstack.org/06/666206/1/gate/neutron-grenade/1f3eed0/ : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":1},{"id":"81d43405e25fa7a73f9eaaf4ce8faf715ba2394c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-24 08:10:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f9bd6bfc2ac41c7291cdd1c907cd52765547d594","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:09:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/666206/1/check/openstack-tox-docs/a08a96a/html/ : SUCCESS in 4m 03s\n- openstack-tox-pypy http://logs.openstack.org/06/666206/1/check/openstack-tox-pypy/329448d/ : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/06/666206/1/check/openstack-tox-pep8/12851bd/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/06/666206/1/check/openstack-tox-py27/cd7e995/ : SUCCESS in 4m 35s\n- openstack-tox-py36 http://logs.openstack.org/06/666206/1/check/openstack-tox-py36/7dc04ed/ : SUCCESS in 4m 49s\n- openstack-tox-validate http://logs.openstack.org/06/666206/1/check/openstack-tox-validate/cdde8b9/ : SUCCESS in 4m 16s\n- requirements-integration http://logs.openstack.org/06/666206/1/check/requirements-integration/c21c809/ : SUCCESS in 59m 18s\n- neutron-grenade-multinode http://logs.openstack.org/06/666206/1/check/neutron-grenade-multinode/7700a5d/ : SUCCESS in 1h 14m 07s\n- tempest-full http://logs.openstack.org/06/666206/1/check/tempest-full/22666db/ : SUCCESS in 1h 44m 27s\n- neutron-grenade http://logs.openstack.org/06/666206/1/check/neutron-grenade/ec851dc/ : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9680dfbe35b4f1897a3db906bf6858d1477dd85f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:09:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a183addb8458fe0f10151ef0344bdd2419d74952","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 12:14:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/06/666206/1/gate/openstack-tox-docs/3ec6b3a/html/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/06/666206/1/gate/openstack-tox-pep8/dd97006/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/06/666206/1/gate/openstack-tox-py27/f27455c/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/06/666206/1/gate/openstack-tox-py36/cf3563c/ : SUCCESS in 4m 14s\n- openstack-tox-validate http://logs.openstack.org/06/666206/1/gate/openstack-tox-validate/c7e4b42/ : SUCCESS in 4m 07s\n- requirements-integration http://logs.openstack.org/06/666206/1/gate/requirements-integration/4e567d6/ : SUCCESS in 1h 18m 03s\n- neutron-grenade-multinode http://logs.openstack.org/06/666206/1/gate/neutron-grenade-multinode/8139033/ : SUCCESS in 1h 21m 25s\n- tempest-full http://logs.openstack.org/06/666206/1/gate/tempest-full/32bba7f/ : SUCCESS in 2h 00m 45s\n- neutron-grenade http://logs.openstack.org/06/666206/1/gate/neutron-grenade/f1a493f/ : SUCCESS in 1h 04m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b6bbb6b6b67a565eb8967acfe0ecb9cae02fbd20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 12:14:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f094c78968ad746d1785cbcc20acb00c4c6acf21","revisions":{"f094c78968ad746d1785cbcc20acb00c4c6acf21":{"kind":"REWORK","_number":1,"created":"2019-06-19 03:39:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/666206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/666206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/666206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/666206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/666206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/666206/1"}}},"commit":{"parents":[{"commit":"9ebc858033d7998a167941f1b7b060de60d17244","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ebc858033d7998a167941f1b7b060de60d17244"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-19 03:34:38.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-06-19 03:39:57.000000000","tz":-420},"subject":"Use UTF-8 to decode subprocess output on py3","message":"Use UTF-8 to decode subprocess output on py3\n\nOtherwise, you get output like\n\n    Branch: master\n    Source: src/opendev.org/openstack/swift\n    Requirements: /home/zuul/src/opendev.org/openstack/requirements\n    git log -n 1 --format\u003d%H\n    Patch under test: b\u0027f683ca84318da31492f214f978a55bdc1be70c2c\u0027\n    git --git-dir /home/zuul/src/opendev.org/openstack/requirements/.git rev-parse HEAD\n    requirements git sha: b\u00279ebc858033d7998a167941f1b7b060de60d17244\u0027\n    virtualenv /tmp/tmp9l32_z5b/venv\n    /tmp/tmp9l32_z5b/venv/bin/pip install /home/zuul/src/opendev.org/openstack/requirements\n    Checking b\u0027f683ca84318da31492f214f978a55bdc1be70c2c\u0027\n\n(Note the bytes getting interpolated in the \"Patch under test\",\n\"requirements git sha\", and \"Checking\" lines.)\n\nChange-Id: I836a6218ff5fccf4aab74b37a5a682601a388106\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f094c78968ad746d1785cbcc20acb00c4c6acf21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f094c78968ad746d1785cbcc20acb00c4c6acf21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
