)]}'
{"id":"openstack%2Frequirements~666228","triplet_id":"openstack%2Frequirements~master~I80f3bf30f3e3bd012a0f9b666fbb09b7694bf066","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I80f3bf30f3e3bd012a0f9b666fbb09b7694bf066","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-06-19 06:14:05.000000000","updated":"2019-06-22 04:24:30.000000000","submitted":"2019-06-22 04:24:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666228-1561177470589-90acf6b2","meta_rev_id":"fbea0f0bf486d259a3072ec6d24c67498e07814e","_number":666228,"virtual_id_number":666228,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-22 04:24:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-20 14:47:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-20 14:47:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 14:47:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-22 04:24:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35f12ad3c0b805f52fd1dc22a3a3f88b76fff5bf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-19 06:14:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7425c8c012b458dfafae5aab6d282a337cde1349","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 13:07:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666228/1/check/openstack-tox-docs/a65f86d/html/ : SUCCESS in 3m 03s\n- openstack-tox-pypy http://logs.openstack.org/28/666228/1/check/openstack-tox-pypy/1db9c7d/ : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/28/666228/1/check/openstack-tox-pep8/0cb6e90/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/28/666228/1/check/openstack-tox-py27/6bd9170/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/28/666228/1/check/openstack-tox-py36/a8dc276/ : SUCCESS in 3m 30s\n- openstack-tox-validate http://logs.openstack.org/28/666228/1/check/openstack-tox-validate/e82d6bf/ : SUCCESS in 3m 04s\n- requirements-tox-babel http://logs.openstack.org/28/666228/1/check/requirements-tox-babel/811c079/ : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/666228/1/check/requirements-tox-py27-check-uc/c048656/ : FAILURE in 8m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/666228/1/check/requirements-tox-py36-check-uc/37a2214/ : FAILURE in 6m 57s\n- cross-cinder-py27 http://logs.openstack.org/28/666228/1/check/cross-cinder-py27/f62deff/ : SUCCESS in 10m 23s\n- cross-cinder-py36 http://logs.openstack.org/28/666228/1/check/cross-cinder-py36/4dc74eb/ : SUCCESS in 10m 57s\n- cross-glance-py27 http://logs.openstack.org/28/666228/1/check/cross-glance-py27/0658853/ : SUCCESS in 4m 39s\n- cross-glance-py36 http://logs.openstack.org/28/666228/1/check/cross-glance-py36/5132fe9/ : SUCCESS in 6m 10s\n- cross-horizon-py27 http://logs.openstack.org/28/666228/1/check/cross-horizon-py27/a7a4628/ : SUCCESS in 9m 14s\n- cross-horizon-py36 http://logs.openstack.org/28/666228/1/check/cross-horizon-py36/b63eaae/ : SUCCESS in 8m 09s\n- cross-horizon-npm http://logs.openstack.org/28/666228/1/check/cross-horizon-npm/4d30271/ : SUCCESS in 6m 32s\n- cross-keystone-py27 http://logs.openstack.org/28/666228/1/check/cross-keystone-py27/16420ca/ : SUCCESS in 23m 08s\n- cross-keystone-py36 http://logs.openstack.org/28/666228/1/check/cross-keystone-py36/6e15b46/ : SUCCESS in 27m 59s\n- cross-neutron-py27 http://logs.openstack.org/28/666228/1/check/cross-neutron-py27/18823be/ : SUCCESS in 25m 51s\n- cross-neutron-py36 http://logs.openstack.org/28/666228/1/check/cross-neutron-py36/f312279/ : SUCCESS in 23m 48s\n- cross-nova-py27 http://logs.openstack.org/28/666228/1/check/cross-nova-py27/478f6f5/ : SUCCESS in 15m 55s\n- cross-nova-py36 http://logs.openstack.org/28/666228/1/check/cross-nova-py36/97aca2a/ : SUCCESS in 12m 26s\n- cross-nova-functional http://logs.openstack.org/28/666228/1/check/cross-nova-functional/b6baa2d/ : SUCCESS in 19m 59s\n- cross-osc-tox-docs http://logs.openstack.org/28/666228/1/check/cross-osc-tox-docs/12924af/html/ : SUCCESS in 7m 21s\n- cross-swift-py27 http://logs.openstack.org/28/666228/1/check/cross-swift-py27/d6f6395/ : SUCCESS in 8m 19s\n- cross-swift-py36 http://logs.openstack.org/28/666228/1/check/cross-swift-py36/1c4bc22/ : SUCCESS in 7m 54s\n- requirements-integration http://logs.openstack.org/28/666228/1/check/requirements-integration/2672e3b/ : SUCCESS in 56m 48s\n- neutron-grenade-multinode http://logs.openstack.org/28/666228/1/check/neutron-grenade-multinode/d57f0dc/ : SUCCESS in 1h 14m 39s\n- tempest-full http://logs.openstack.org/28/666228/1/check/tempest-full/037e072/ : SUCCESS in 1h 53m 56s\n- neutron-grenade http://logs.openstack.org/28/666228/1/check/neutron-grenade/b44d2a1/ : SUCCESS in 1h 02m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3f5a3ea2075cde0bf7dfe8e49a6ab4e4f5855ed0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-19 14:19:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2bd6b2832dab4793db4be9fa7b95c449d8f7c464","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-19 14:19:15.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"d85654935cc43e73407ba89cd6f4af49c53b3253","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 20:53:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666228/2/check/openstack-tox-docs/54a3003/html/ : SUCCESS in 2m 39s\n- openstack-tox-pypy http://logs.openstack.org/28/666228/2/check/openstack-tox-pypy/56a2f68/ : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/28/666228/2/check/openstack-tox-pep8/30f63f8/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/28/666228/2/check/openstack-tox-py27/f82dd9c/ : SUCCESS in 3m 00s\n- openstack-tox-py36 http://logs.openstack.org/28/666228/2/check/openstack-tox-py36/0adb005/ : SUCCESS in 5m 27s\n- openstack-tox-validate http://logs.openstack.org/28/666228/2/check/openstack-tox-validate/6781876/ : SUCCESS in 3m 15s\n- requirements-tox-babel http://logs.openstack.org/28/666228/2/check/requirements-tox-babel/1b9f2c0/ : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py27-check-uc/3e89872/ : SUCCESS in 7m 58s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py36-check-uc/5dc04d7/ : SUCCESS in 8m 08s\n- cross-cinder-py27 http://logs.openstack.org/28/666228/2/check/cross-cinder-py27/c5feaad/ : SUCCESS in 8m 59s\n- cross-cinder-py36 http://logs.openstack.org/28/666228/2/check/cross-cinder-py36/830c9c3/ : SUCCESS in 8m 43s\n- cross-glance-py27 http://logs.openstack.org/28/666228/2/check/cross-glance-py27/4091076/ : SUCCESS in 6m 17s\n- cross-glance-py36 http://logs.openstack.org/28/666228/2/check/cross-glance-py36/9560c0b/ : SUCCESS in 5m 34s\n- cross-horizon-py27 http://logs.openstack.org/28/666228/2/check/cross-horizon-py27/946c766/ : SUCCESS in 10m 34s\n- cross-horizon-py36 http://logs.openstack.org/28/666228/2/check/cross-horizon-py36/07cdd2a/ : SUCCESS in 9m 52s\n- cross-horizon-npm http://logs.openstack.org/28/666228/2/check/cross-horizon-npm/618f072/ : SUCCESS in 6m 31s\n- cross-keystone-py27 http://logs.openstack.org/28/666228/2/check/cross-keystone-py27/1a09534/ : SUCCESS in 25m 26s\n- cross-keystone-py36 http://logs.openstack.org/28/666228/2/check/cross-keystone-py36/6500d3b/ : SUCCESS in 28m 13s\n- cross-neutron-py27 http://logs.openstack.org/28/666228/2/check/cross-neutron-py27/8584d30/ : SUCCESS in 25m 32s\n- cross-neutron-py36 http://logs.openstack.org/28/666228/2/check/cross-neutron-py36/5b600b9/ : SUCCESS in 23m 37s\n- cross-nova-py27 http://logs.openstack.org/28/666228/2/check/cross-nova-py27/77edb1f/ : SUCCESS in 13m 26s\n- cross-nova-py36 http://logs.openstack.org/28/666228/2/check/cross-nova-py36/fd5cca0/ : SUCCESS in 12m 19s\n- cross-nova-functional http://logs.openstack.org/28/666228/2/check/cross-nova-functional/3c9588b/ : SUCCESS in 19m 30s\n- cross-osc-tox-docs http://logs.openstack.org/28/666228/2/check/cross-osc-tox-docs/83743ec/html/ : SUCCESS in 5m 55s\n- cross-swift-py27 http://logs.openstack.org/28/666228/2/check/cross-swift-py27/2099211/ : SUCCESS in 7m 45s\n- cross-swift-py36 http://logs.openstack.org/28/666228/2/check/cross-swift-py36/93dfafa/ : SUCCESS in 7m 57s\n- requirements-integration http://logs.openstack.org/28/666228/2/check/requirements-integration/4505098/ : SUCCESS in 55m 50s\n- neutron-grenade-multinode http://logs.openstack.org/28/666228/2/check/neutron-grenade-multinode/27eaa0b/ : FAILURE in 15m 04s\n- tempest-full http://logs.openstack.org/28/666228/2/check/tempest-full/c3d2a7b/ : SUCCESS in 1h 46m 00s\n- neutron-grenade http://logs.openstack.org/28/666228/2/check/neutron-grenade/1c3d1ae/ : SUCCESS in 56m 43s","accounts_in_message":[],"_revision_number":2},{"id":"c020ac967991ab5adb4ea1dd2a153d8c614642e7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-19 23:36:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"90b6c3757fc429b0c53160b51893827d5d093ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 04:22:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666228/2/check/openstack-tox-docs/bc9a3e6/html/ : SUCCESS in 4m 28s\n- openstack-tox-pypy http://logs.openstack.org/28/666228/2/check/openstack-tox-pypy/5791df5/ : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/28/666228/2/check/openstack-tox-pep8/d739f92/ : SUCCESS in 2m 48s\n- openstack-tox-py27 http://logs.openstack.org/28/666228/2/check/openstack-tox-py27/a1d5c3b/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/28/666228/2/check/openstack-tox-py36/2d229f7/ : SUCCESS in 11m 50s\n- openstack-tox-validate http://logs.openstack.org/28/666228/2/check/openstack-tox-validate/bbd4889/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/28/666228/2/check/requirements-tox-babel/eed0cc5/ : SUCCESS in 4m 29s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py27-check-uc/60b7220/ : SUCCESS in 7m 19s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py36-check-uc/f99f27e/ : SUCCESS in 29m 45s\n- cross-cinder-py27 http://logs.openstack.org/28/666228/2/check/cross-cinder-py27/01d60c3/ : SUCCESS in 8m 56s\n- cross-cinder-py36 http://logs.openstack.org/28/666228/2/check/cross-cinder-py36/0127cac/ : SUCCESS in 8m 20s\n- cross-glance-py27 http://logs.openstack.org/28/666228/2/check/cross-glance-py27/beb1eda/ : SUCCESS in 13m 50s\n- cross-glance-py36 http://logs.openstack.org/28/666228/2/check/cross-glance-py36/207f15c/ : SUCCESS in 5m 11s\n- cross-horizon-py27 http://logs.openstack.org/28/666228/2/check/cross-horizon-py27/d4ad774/ : SUCCESS in 9m 19s\n- cross-horizon-py36 http://logs.openstack.org/28/666228/2/check/cross-horizon-py36/00d5004/ : SUCCESS in 8m 31s\n- cross-horizon-npm http://logs.openstack.org/28/666228/2/check/cross-horizon-npm/a89f664/ : SUCCESS in 5m 43s\n- cross-keystone-py27 http://logs.openstack.org/28/666228/2/check/cross-keystone-py27/fe025f3/ : SUCCESS in 24m 29s\n- cross-keystone-py36 http://logs.openstack.org/28/666228/2/check/cross-keystone-py36/0bc60f3/ : FAILURE in 6m 14s\n- cross-neutron-py27 http://logs.openstack.org/28/666228/2/check/cross-neutron-py27/0eed950/ : SUCCESS in 22m 14s\n- cross-neutron-py36 http://logs.openstack.org/28/666228/2/check/cross-neutron-py36/cfef8ed/ : TIMED_OUT in 41m 01s\n- cross-nova-py27 http://logs.openstack.org/28/666228/2/check/cross-nova-py27/c4e7e99/ : SUCCESS in 12m 58s\n- cross-nova-py36 http://logs.openstack.org/28/666228/2/check/cross-nova-py36/2394442/ : SUCCESS in 21m 00s\n- cross-nova-functional http://logs.openstack.org/28/666228/2/check/cross-nova-functional/220c17f/ : SUCCESS in 18m 03s\n- cross-osc-tox-docs http://logs.openstack.org/28/666228/2/check/cross-osc-tox-docs/d4460e8/html/ : SUCCESS in 6m 24s\n- cross-swift-py27 http://logs.openstack.org/28/666228/2/check/cross-swift-py27/c8a8fe4/ : SUCCESS in 10m 01s\n- cross-swift-py36 http://logs.openstack.org/28/666228/2/check/cross-swift-py36/3b8f710/ : SUCCESS in 14m 26s\n- requirements-integration http://logs.openstack.org/28/666228/2/check/requirements-integration/d5f1f3c/ : SUCCESS in 57m 01s\n- neutron-grenade-multinode http://logs.openstack.org/28/666228/2/check/neutron-grenade-multinode/9cca446/ : FAILURE in 32m 30s\n- tempest-full http://logs.openstack.org/28/666228/2/check/tempest-full/d96011c/ : SUCCESS in 2h 03m 47s\n- neutron-grenade http://logs.openstack.org/28/666228/2/check/neutron-grenade/c309369/ : FAILURE in 25m 54s","accounts_in_message":[],"_revision_number":2},{"id":"c60ce817cbfcc04507eddcd4cfeb56509029a61b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-20 05:00:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"22308ad5d137152b978c30f6eae7d82a5ca8b855","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 07:12:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666228/2/check/openstack-tox-docs/070a4a5/html/ : SUCCESS in 3m 59s\n- openstack-tox-pypy http://logs.openstack.org/28/666228/2/check/openstack-tox-pypy/8a7ab98/ : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/28/666228/2/check/openstack-tox-pep8/979ff79/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/28/666228/2/check/openstack-tox-py27/e7fb4b7/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/28/666228/2/check/openstack-tox-py36/bfa6911/ : SUCCESS in 3m 13s\n- openstack-tox-validate http://logs.openstack.org/28/666228/2/check/openstack-tox-validate/a550b96/ : SUCCESS in 4m 22s\n- requirements-tox-babel http://logs.openstack.org/28/666228/2/check/requirements-tox-babel/6ca32ae/ : SUCCESS in 2m 45s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py27-check-uc/38e56e1/ : SUCCESS in 8m 14s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py36-check-uc/8bb90fb/ : SUCCESS in 7m 55s\n- cross-cinder-py27 http://logs.openstack.org/28/666228/2/check/cross-cinder-py27/7686053/ : SUCCESS in 9m 39s\n- cross-cinder-py36 http://logs.openstack.org/28/666228/2/check/cross-cinder-py36/d90c873/ : SUCCESS in 8m 20s\n- cross-glance-py27 http://logs.openstack.org/28/666228/2/check/cross-glance-py27/1d80411/ : SUCCESS in 5m 20s\n- cross-glance-py36 http://logs.openstack.org/28/666228/2/check/cross-glance-py36/be3d542/ : SUCCESS in 5m 03s\n- cross-horizon-py27 http://logs.openstack.org/28/666228/2/check/cross-horizon-py27/cbb4bc2/ : SUCCESS in 10m 01s\n- cross-horizon-py36 http://logs.openstack.org/28/666228/2/check/cross-horizon-py36/7adc953/ : SUCCESS in 8m 28s\n- cross-horizon-npm http://logs.openstack.org/28/666228/2/check/cross-horizon-npm/8ca5806/ : SUCCESS in 8m 10s\n- cross-keystone-py27 http://logs.openstack.org/28/666228/2/check/cross-keystone-py27/7ad3c41/ : SUCCESS in 23m 21s\n- cross-keystone-py36 http://logs.openstack.org/28/666228/2/check/cross-keystone-py36/75561aa/ : SUCCESS in 22m 40s\n- cross-neutron-py27 http://logs.openstack.org/28/666228/2/check/cross-neutron-py27/8c7db56/ : SUCCESS in 23m 06s\n- cross-neutron-py36 http://logs.openstack.org/28/666228/2/check/cross-neutron-py36/8f899e3/ : SUCCESS in 20m 53s\n- cross-nova-py27 http://logs.openstack.org/28/666228/2/check/cross-nova-py27/2f3713a/ : SUCCESS in 21m 42s\n- cross-nova-py36 http://logs.openstack.org/28/666228/2/check/cross-nova-py36/445ac44/ : SUCCESS in 10m 34s\n- cross-nova-functional http://logs.openstack.org/28/666228/2/check/cross-nova-functional/2d2ded7/ : SUCCESS in 18m 48s\n- cross-osc-tox-docs http://logs.openstack.org/28/666228/2/check/cross-osc-tox-docs/f29610c/html/ : SUCCESS in 10m 17s\n- cross-swift-py27 http://logs.openstack.org/28/666228/2/check/cross-swift-py27/4d90da7/ : SUCCESS in 7m 20s\n- cross-swift-py36 http://logs.openstack.org/28/666228/2/check/cross-swift-py36/0433dab/ : SUCCESS in 8m 26s\n- requirements-integration http://logs.openstack.org/28/666228/2/check/requirements-integration/c4b84a2/ : SUCCESS in 1h 03m 10s\n- neutron-grenade-multinode http://logs.openstack.org/28/666228/2/check/neutron-grenade-multinode/f39b44b/ : SUCCESS in 1h 09m 14s\n- tempest-full http://logs.openstack.org/28/666228/2/check/tempest-full/12bf547/ : TIMED_OUT in 2h 06m 32s\n- neutron-grenade http://logs.openstack.org/28/666228/2/check/neutron-grenade/3ebb53a/ : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1cb87cfc94a27dcf359f8d4861a8693f25eb5416","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-20 14:47:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"24ae9035823dc02da41a775db85da03eb4ced84c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:17:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/666228/2/check/openstack-tox-docs/710cc5a/html/ : SUCCESS in 2m 40s\n- openstack-tox-pypy http://logs.openstack.org/28/666228/2/check/openstack-tox-pypy/a27965a/ : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/28/666228/2/check/openstack-tox-pep8/9cd032b/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/28/666228/2/check/openstack-tox-py27/2f534ae/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/28/666228/2/check/openstack-tox-py36/c483c6f/ : SUCCESS in 3m 29s\n- openstack-tox-validate http://logs.openstack.org/28/666228/2/check/openstack-tox-validate/d6ff469/ : SUCCESS in 3m 25s\n- requirements-tox-babel http://logs.openstack.org/28/666228/2/check/requirements-tox-babel/d412403/ : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py27-check-uc/9cbe6ab/ : SUCCESS in 8m 01s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py36-check-uc/02acc37/ : SUCCESS in 6m 48s\n- cross-cinder-py27 http://logs.openstack.org/28/666228/2/check/cross-cinder-py27/010f00d/ : SUCCESS in 9m 47s\n- cross-cinder-py36 http://logs.openstack.org/28/666228/2/check/cross-cinder-py36/0ee9d9d/ : SUCCESS in 9m 16s\n- cross-glance-py27 http://logs.openstack.org/28/666228/2/check/cross-glance-py27/bc62de3/ : SUCCESS in 4m 38s\n- cross-glance-py36 http://logs.openstack.org/28/666228/2/check/cross-glance-py36/0745afa/ : SUCCESS in 4m 30s\n- cross-horizon-py27 http://logs.openstack.org/28/666228/2/check/cross-horizon-py27/b127d23/ : SUCCESS in 10m 52s\n- cross-horizon-py36 http://logs.openstack.org/28/666228/2/check/cross-horizon-py36/ad8495b/ : SUCCESS in 9m 04s\n- cross-horizon-npm http://logs.openstack.org/28/666228/2/check/cross-horizon-npm/8431105/ : SUCCESS in 7m 59s\n- cross-keystone-py27 http://logs.openstack.org/28/666228/2/check/cross-keystone-py27/214accc/ : SUCCESS in 23m 03s\n- cross-keystone-py36 http://logs.openstack.org/28/666228/2/check/cross-keystone-py36/15e4aaf/ : SUCCESS in 23m 28s\n- cross-neutron-py27 http://logs.openstack.org/28/666228/2/check/cross-neutron-py27/c04d215/ : SUCCESS in 22m 36s\n- cross-neutron-py36 http://logs.openstack.org/28/666228/2/check/cross-neutron-py36/eae071f/ : SUCCESS in 28m 54s\n- cross-nova-py27 http://logs.openstack.org/28/666228/2/check/cross-nova-py27/8bbd980/ : SUCCESS in 12m 43s\n- cross-nova-py36 http://logs.openstack.org/28/666228/2/check/cross-nova-py36/9abda8c/ : SUCCESS in 22m 26s\n- cross-nova-functional http://logs.openstack.org/28/666228/2/check/cross-nova-functional/e8027aa/ : SUCCESS in 28m 37s\n- cross-osc-tox-docs http://logs.openstack.org/28/666228/2/check/cross-osc-tox-docs/38dfa4a/html/ : SUCCESS in 5m 51s\n- cross-swift-py27 http://logs.openstack.org/28/666228/2/check/cross-swift-py27/f68c519/ : SUCCESS in 8m 39s\n- cross-swift-py36 http://logs.openstack.org/28/666228/2/check/cross-swift-py36/a4b615b/ : SUCCESS in 14m 44s\n- requirements-integration http://logs.openstack.org/28/666228/2/check/requirements-integration/e81a5a0/ : SUCCESS in 52m 44s\n- neutron-grenade-multinode http://logs.openstack.org/28/666228/2/check/neutron-grenade-multinode/d434d87/ : SUCCESS in 1h 22m 37s\n- tempest-full http://logs.openstack.org/28/666228/2/check/tempest-full/00585d2/ : SUCCESS in 1h 39m 01s\n- neutron-grenade http://logs.openstack.org/28/666228/2/check/neutron-grenade/5ce0d73/ : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d39d1d79ed99e143e27e08e41b8fba611cbd35c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:17:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"56c340a6d4e3c894b1d5bc0989d6da52d0a6372d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 23:14:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666228/2/gate/openstack-tox-docs/d497d26/html/ : SUCCESS in 7m 25s\n- openstack-tox-pep8 http://logs.openstack.org/28/666228/2/gate/openstack-tox-pep8/a8f474c/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/28/666228/2/gate/openstack-tox-py27/f8f81a8/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/28/666228/2/gate/openstack-tox-py36/601d40d/ : SUCCESS in 3m 34s\n- openstack-tox-validate http://logs.openstack.org/28/666228/2/gate/openstack-tox-validate/347c1a0/ : SUCCESS in 3m 02s\n- requirements-tox-babel http://logs.openstack.org/28/666228/2/gate/requirements-tox-babel/4886ad3/ : SUCCESS in 7m 44s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/666228/2/gate/requirements-tox-py27-check-uc/659a862/ : SUCCESS in 7m 04s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/666228/2/gate/requirements-tox-py36-check-uc/9a5e560/ : SUCCESS in 6m 48s\n- cross-cinder-py27 http://logs.openstack.org/28/666228/2/gate/cross-cinder-py27/9c7dba6/ : SUCCESS in 14m 38s\n- cross-cinder-py36 http://logs.openstack.org/28/666228/2/gate/cross-cinder-py36/01a0a39/ : SUCCESS in 9m 10s\n- cross-glance-py27 http://logs.openstack.org/28/666228/2/gate/cross-glance-py27/a0f4d3b/ : SUCCESS in 4m 36s\n- cross-glance-py36 http://logs.openstack.org/28/666228/2/gate/cross-glance-py36/441bffa/ : SUCCESS in 4m 48s\n- cross-horizon-py27 http://logs.openstack.org/28/666228/2/gate/cross-horizon-py27/9e08f19/ : SUCCESS in 10m 25s\n- cross-horizon-py36 http://logs.openstack.org/28/666228/2/gate/cross-horizon-py36/247ccb8/ : SUCCESS in 17m 13s\n- cross-horizon-npm http://logs.openstack.org/28/666228/2/gate/cross-horizon-npm/77462df/ : SUCCESS in 5m 46s\n- cross-keystone-py27 http://logs.openstack.org/28/666228/2/gate/cross-keystone-py27/379dea2/ : SUCCESS in 28m 09s\n- cross-keystone-py36 http://logs.openstack.org/28/666228/2/gate/cross-keystone-py36/46e6f8e/ : SUCCESS in 22m 27s\n- cross-neutron-py27 http://logs.openstack.org/28/666228/2/gate/cross-neutron-py27/467fe42/ : SUCCESS in 26m 54s\n- cross-neutron-py36 http://logs.openstack.org/28/666228/2/gate/cross-neutron-py36/1fb447d/ : SUCCESS in 24m 36s\n- cross-nova-py27 http://logs.openstack.org/28/666228/2/gate/cross-nova-py27/b6802a4/ : SUCCESS in 12m 12s\n- cross-nova-py36 http://logs.openstack.org/28/666228/2/gate/cross-nova-py36/9c85aef/ : SUCCESS in 11m 34s\n- cross-nova-functional http://logs.openstack.org/28/666228/2/gate/cross-nova-functional/c7c8434/ : SUCCESS in 16m 21s\n- cross-osc-tox-docs http://logs.openstack.org/28/666228/2/gate/cross-osc-tox-docs/e30a94c/html/ : SUCCESS in 6m 53s\n- cross-swift-py27 http://logs.openstack.org/28/666228/2/gate/cross-swift-py27/eef5f10/ : SUCCESS in 29m 18s\n- cross-swift-py36 http://logs.openstack.org/28/666228/2/gate/cross-swift-py36/0f8bf09/ : SUCCESS in 29m 48s\n- requirements-integration http://logs.openstack.org/28/666228/2/gate/requirements-integration/cc31e73/ : FAILURE in 20m 06s\n- neutron-grenade-multinode http://logs.openstack.org/28/666228/2/gate/neutron-grenade-multinode/e2a7403/ : FAILURE in 20m 53s\n- tempest-full http://logs.openstack.org/28/666228/2/gate/tempest-full/9c9885f/ : SUCCESS in 1h 48m 44s\n- neutron-grenade http://logs.openstack.org/28/666228/2/gate/neutron-grenade/6888fb5/ : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":2},{"id":"11f31f5502b004a718ff45e6f158ddab6d5f9e30","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-21 14:09:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cfb0dc708fb8cbb8b78e7dca67860fe1beb6101b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 19:52:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/666228/2/check/openstack-tox-docs/eec6394/html/ : SUCCESS in 3m 09s\n- openstack-tox-pypy http://logs.openstack.org/28/666228/2/check/openstack-tox-pypy/8751951/ : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/28/666228/2/check/openstack-tox-pep8/59049ff/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/28/666228/2/check/openstack-tox-py27/b45bcba/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/28/666228/2/check/openstack-tox-py36/7d4a5ad/ : SUCCESS in 3m 15s\n- openstack-tox-validate http://logs.openstack.org/28/666228/2/check/openstack-tox-validate/a956285/ : SUCCESS in 2m 51s\n- requirements-tox-babel http://logs.openstack.org/28/666228/2/check/requirements-tox-babel/968fb8b/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py27-check-uc/1d78eae/ : SUCCESS in 6m 34s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py36-check-uc/f726574/ : SUCCESS in 8m 13s\n- cross-cinder-py27 http://logs.openstack.org/28/666228/2/check/cross-cinder-py27/3c3c2ef/ : SUCCESS in 8m 49s\n- cross-cinder-py36 http://logs.openstack.org/28/666228/2/check/cross-cinder-py36/01c0dc7/ : SUCCESS in 9m 39s\n- cross-glance-py27 http://logs.openstack.org/28/666228/2/check/cross-glance-py27/f497010/ : SUCCESS in 4m 33s\n- cross-glance-py36 http://logs.openstack.org/28/666228/2/check/cross-glance-py36/c4a5b5f/ : SUCCESS in 4m 47s\n- cross-horizon-py27 http://logs.openstack.org/28/666228/2/check/cross-horizon-py27/69b1b4c/ : SUCCESS in 11m 00s\n- cross-horizon-py36 http://logs.openstack.org/28/666228/2/check/cross-horizon-py36/e97511f/ : SUCCESS in 9m 12s\n- cross-horizon-npm http://logs.openstack.org/28/666228/2/check/cross-horizon-npm/f670cf1/ : SUCCESS in 6m 01s\n- cross-keystone-py27 http://logs.openstack.org/28/666228/2/check/cross-keystone-py27/ddf6bad/ : SUCCESS in 21m 34s\n- cross-keystone-py36 http://logs.openstack.org/28/666228/2/check/cross-keystone-py36/13e4290/ : SUCCESS in 27m 38s\n- cross-neutron-py27 http://logs.openstack.org/28/666228/2/check/cross-neutron-py27/4c5afff/ : SUCCESS in 22m 48s\n- cross-neutron-py36 http://logs.openstack.org/28/666228/2/check/cross-neutron-py36/0de7e73/ : SUCCESS in 23m 51s\n- cross-nova-py27 http://logs.openstack.org/28/666228/2/check/cross-nova-py27/f53dda1/ : SUCCESS in 12m 21s\n- cross-nova-py36 http://logs.openstack.org/28/666228/2/check/cross-nova-py36/4bd125e/ : SUCCESS in 13m 35s\n- cross-nova-functional http://logs.openstack.org/28/666228/2/check/cross-nova-functional/a279f19/ : SUCCESS in 19m 47s\n- cross-osc-tox-docs http://logs.openstack.org/28/666228/2/check/cross-osc-tox-docs/3ec0ac4/html/ : SUCCESS in 5m 28s\n- cross-swift-py27 http://logs.openstack.org/28/666228/2/check/cross-swift-py27/2244343/ : SUCCESS in 8m 37s\n- cross-swift-py36 http://logs.openstack.org/28/666228/2/check/cross-swift-py36/42b49c9/ : SUCCESS in 8m 35s\n- requirements-integration http://logs.openstack.org/28/666228/2/check/requirements-integration/394fdb3/ : SUCCESS in 1h 04m 15s\n- neutron-grenade-multinode http://logs.openstack.org/28/666228/2/check/neutron-grenade-multinode/dbd296e/ : SUCCESS in 1h 04m 36s\n- tempest-full http://logs.openstack.org/28/666228/2/check/tempest-full/e2f6725/ : FAILURE in 2h 00m 06s\n- neutron-grenade http://logs.openstack.org/28/666228/2/check/neutron-grenade/44e0ee2/ : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d5ea1c68fa018647b1ef64673ef03a5a827c433a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-21 20:33:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1c441a308cfb4eed1f1fb482333c012d6aee44be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 01:10:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/666228/2/check/openstack-tox-docs/904b716/html/ : SUCCESS in 2m 58s\n- openstack-tox-pypy http://logs.openstack.org/28/666228/2/check/openstack-tox-pypy/38d0510/ : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/28/666228/2/check/openstack-tox-pep8/b16998a/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/28/666228/2/check/openstack-tox-py27/0afab22/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/28/666228/2/check/openstack-tox-py36/ff08075/ : SUCCESS in 3m 30s\n- openstack-tox-validate http://logs.openstack.org/28/666228/2/check/openstack-tox-validate/11538d3/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/28/666228/2/check/requirements-tox-babel/ab994e2/ : SUCCESS in 6m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py27-check-uc/7dfffa5/ : SUCCESS in 6m 04s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/666228/2/check/requirements-tox-py36-check-uc/89922be/ : SUCCESS in 8m 51s\n- cross-cinder-py27 http://logs.openstack.org/28/666228/2/check/cross-cinder-py27/e0c59cd/ : SUCCESS in 8m 45s\n- cross-cinder-py36 http://logs.openstack.org/28/666228/2/check/cross-cinder-py36/6c2252d/ : SUCCESS in 8m 11s\n- cross-glance-py27 http://logs.openstack.org/28/666228/2/check/cross-glance-py27/6669e61/ : SUCCESS in 5m 38s\n- cross-glance-py36 http://logs.openstack.org/28/666228/2/check/cross-glance-py36/ab95344/ : SUCCESS in 5m 37s\n- cross-horizon-py27 http://logs.openstack.org/28/666228/2/check/cross-horizon-py27/2e555eb/ : SUCCESS in 10m 27s\n- cross-horizon-py36 http://logs.openstack.org/28/666228/2/check/cross-horizon-py36/6fabd95/ : SUCCESS in 8m 19s\n- cross-horizon-npm http://logs.openstack.org/28/666228/2/check/cross-horizon-npm/a02459f/ : SUCCESS in 6m 23s\n- cross-keystone-py27 http://logs.openstack.org/28/666228/2/check/cross-keystone-py27/2bfc5bb/ : SUCCESS in 21m 50s\n- cross-keystone-py36 http://logs.openstack.org/28/666228/2/check/cross-keystone-py36/fa697b4/ : SUCCESS in 25m 39s\n- cross-neutron-py27 http://logs.openstack.org/28/666228/2/check/cross-neutron-py27/e137bc4/ : SUCCESS in 23m 40s\n- cross-neutron-py36 http://logs.openstack.org/28/666228/2/check/cross-neutron-py36/d896ce2/ : SUCCESS in 23m 40s\n- cross-nova-py27 http://logs.openstack.org/28/666228/2/check/cross-nova-py27/83e4fd6/ : SUCCESS in 15m 10s\n- cross-nova-py36 http://logs.openstack.org/28/666228/2/check/cross-nova-py36/54a5e75/ : SUCCESS in 11m 41s\n- cross-nova-functional http://logs.openstack.org/28/666228/2/check/cross-nova-functional/c3a6af9/ : SUCCESS in 20m 19s\n- cross-osc-tox-docs http://logs.openstack.org/28/666228/2/check/cross-osc-tox-docs/02409d1/html/ : SUCCESS in 7m 03s\n- cross-swift-py27 http://logs.openstack.org/28/666228/2/check/cross-swift-py27/c3b8c7f/ : SUCCESS in 8m 15s\n- cross-swift-py36 http://logs.openstack.org/28/666228/2/check/cross-swift-py36/965f3ad/ : SUCCESS in 11m 49s\n- requirements-integration http://logs.openstack.org/28/666228/2/check/requirements-integration/21f3077/ : SUCCESS in 50m 37s\n- neutron-grenade-multinode http://logs.openstack.org/28/666228/2/check/neutron-grenade-multinode/99ea278/ : SUCCESS in 1h 04m 21s\n- tempest-full http://logs.openstack.org/28/666228/2/check/tempest-full/9a70fc5/ : SUCCESS in 1h 42m 55s\n- neutron-grenade http://logs.openstack.org/28/666228/2/check/neutron-grenade/8039928/ : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b2c76f75133d6dff85d49b8aff9b9742fd6f22fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 01:11:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ee35f1baca623c8ef31a5fd5f80f8a994ff2293f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 04:24:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fbea0f0bf486d259a3072ec6d24c67498e07814e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 04:24:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/666228/2/gate/openstack-tox-docs/31973e8/html/ : SUCCESS in 2m 41s\n- openstack-tox-pep8 http://logs.openstack.org/28/666228/2/gate/openstack-tox-pep8/e54b2ea/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/28/666228/2/gate/openstack-tox-py27/09d9a33/ : SUCCESS in 2m 56s\n- openstack-tox-py36 http://logs.openstack.org/28/666228/2/gate/openstack-tox-py36/b506d8f/ : SUCCESS in 2m 55s\n- openstack-tox-validate http://logs.openstack.org/28/666228/2/gate/openstack-tox-validate/4addc5f/ : SUCCESS in 3m 45s\n- requirements-tox-babel http://logs.openstack.org/28/666228/2/gate/requirements-tox-babel/e50a8e2/ : SUCCESS in 2m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/666228/2/gate/requirements-tox-py27-check-uc/7d75c3b/ : SUCCESS in 7m 10s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/666228/2/gate/requirements-tox-py36-check-uc/50dc8f2/ : SUCCESS in 5m 52s\n- cross-cinder-py27 http://logs.openstack.org/28/666228/2/gate/cross-cinder-py27/2916236/ : SUCCESS in 8m 01s\n- cross-cinder-py36 http://logs.openstack.org/28/666228/2/gate/cross-cinder-py36/e2e0804/ : SUCCESS in 8m 14s\n- cross-glance-py27 http://logs.openstack.org/28/666228/2/gate/cross-glance-py27/4b82413/ : SUCCESS in 4m 07s\n- cross-glance-py36 http://logs.openstack.org/28/666228/2/gate/cross-glance-py36/022b845/ : SUCCESS in 4m 51s\n- cross-horizon-py27 http://logs.openstack.org/28/666228/2/gate/cross-horizon-py27/0d95159/ : SUCCESS in 10m 17s\n- cross-horizon-py36 http://logs.openstack.org/28/666228/2/gate/cross-horizon-py36/045721c/ : SUCCESS in 7m 59s\n- cross-horizon-npm http://logs.openstack.org/28/666228/2/gate/cross-horizon-npm/54ce8bb/ : SUCCESS in 7m 49s\n- cross-keystone-py27 http://logs.openstack.org/28/666228/2/gate/cross-keystone-py27/dd2c611/ : SUCCESS in 25m 59s\n- cross-keystone-py36 http://logs.openstack.org/28/666228/2/gate/cross-keystone-py36/6b81c9b/ : SUCCESS in 22m 21s\n- cross-neutron-py27 http://logs.openstack.org/28/666228/2/gate/cross-neutron-py27/0e57861/ : SUCCESS in 30m 13s\n- cross-neutron-py36 http://logs.openstack.org/28/666228/2/gate/cross-neutron-py36/ea84711/ : SUCCESS in 18m 14s\n- cross-nova-py27 http://logs.openstack.org/28/666228/2/gate/cross-nova-py27/8d52359/ : SUCCESS in 17m 20s\n- cross-nova-py36 http://logs.openstack.org/28/666228/2/gate/cross-nova-py36/06b1a56/ : SUCCESS in 10m 19s\n- cross-nova-functional http://logs.openstack.org/28/666228/2/gate/cross-nova-functional/8b19e0e/ : SUCCESS in 17m 07s\n- cross-osc-tox-docs http://logs.openstack.org/28/666228/2/gate/cross-osc-tox-docs/162f7d7/html/ : SUCCESS in 6m 01s\n- cross-swift-py27 http://logs.openstack.org/28/666228/2/gate/cross-swift-py27/bc1b610/ : SUCCESS in 10m 01s\n- cross-swift-py36 http://logs.openstack.org/28/666228/2/gate/cross-swift-py36/9e2f310/ : SUCCESS in 7m 59s\n- requirements-integration http://logs.openstack.org/28/666228/2/gate/requirements-integration/0f92685/ : SUCCESS in 48m 45s\n- neutron-grenade-multinode http://logs.openstack.org/28/666228/2/gate/neutron-grenade-multinode/f40a409/ : SUCCESS in 1h 11m 51s\n- tempest-full http://logs.openstack.org/28/666228/2/gate/tempest-full/a0e9b2e/ : SUCCESS in 1h 46m 12s\n- neutron-grenade http://logs.openstack.org/28/666228/2/gate/neutron-grenade/aad452f/ : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9273441710632834b85737cda573b86694efab9","revisions":{"0f71a6980772a11ef33a297b98ac0e70df4214b1":{"kind":"REWORK","_number":1,"created":"2019-06-19 06:14:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/666228/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/666228/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/666228/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/666228/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/666228/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/666228/1"}}},"commit":{"parents":[{"commit":"9ebc858033d7998a167941f1b7b060de60d17244","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ebc858033d7998a167941f1b7b060de60d17244"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-19 06:13:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-19 06:13:59.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I80f3bf30f3e3bd012a0f9b666fbb09b7694bf066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f71a6980772a11ef33a297b98ac0e70df4214b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0f71a6980772a11ef33a297b98ac0e70df4214b1"}]},"branch":"refs/heads/master"},"f9273441710632834b85737cda573b86694efab9":{"kind":"REWORK","_number":2,"created":"2019-06-19 14:19:12.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/28/666228/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/666228/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/666228/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/666228/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/666228/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/666228/2"}}},"commit":{"parents":[{"commit":"9ebc858033d7998a167941f1b7b060de60d17244","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ebc858033d7998a167941f1b7b060de60d17244"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-19 06:13:59.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-06-19 14:18:57.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I80f3bf30f3e3bd012a0f9b666fbb09b7694bf066\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9273441710632834b85737cda573b86694efab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9273441710632834b85737cda573b86694efab9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
