)]}'
{"id":"openstack%2Frequirements~666844","triplet_id":"openstack%2Frequirements~stable%2Frocky~Ib0caaa2e1c26204f9fddb1027092f04064780371","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Ib0caaa2e1c26204f9fddb1027092f04064780371","subject":"update constraint for os-net-config to new release 9.4.0","status":"MERGED","created":"2019-06-21 14:05:10.000000000","updated":"2019-06-23 01:28:44.000000000","submitted":"2019-06-23 01:28:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666844-1561253324559-a6cfe23d","meta_rev_id":"38825b953d511eb9bfef7e7f471b811c429cf51a","_number":666844,"virtual_id_number":666844,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-23 01:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-21 19:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-22 06:07:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-22 06:07:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-23 01:28:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d769e095c70c823cf3a0f3ddad274c7f2682637","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-21 14:05:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"059ae48fd01b8ae1b90996b03aa87a7945147315","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 19:14:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/44/666844/1/check/openstack-tox-pypy/a597d36/ : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/44/666844/1/check/openstack-tox-pep8/20e5745/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/44/666844/1/check/openstack-tox-py27/c6748dd/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/44/666844/1/check/openstack-tox-py35/a0f0797/ : SUCCESS in 3m 18s\n- openstack-tox-validate http://logs.openstack.org/44/666844/1/check/openstack-tox-validate/e1565c4/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/44/666844/1/check/requirements-tox-babel/1855cd7/ : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/44/666844/1/check/requirements-tox-py27-check-uc/3604f93/ : SUCCESS in 6m 45s\n- requirements-tox-py35-check-uc http://logs.openstack.org/44/666844/1/check/requirements-tox-py35-check-uc/f18b376/ : SUCCESS in 6m 24s\n- requirements-tox-py36-check-uc http://logs.openstack.org/44/666844/1/check/requirements-tox-py36-check-uc/4bc878c/ : SUCCESS in 6m 15s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/44/666844/1/check/cross-cinder-py27/2ddd002/ : SUCCESS in 9m 05s\n- cross-cinder-py35 http://logs.openstack.org/44/666844/1/check/cross-cinder-py35/c41c8b8/ : SUCCESS in 10m 15s\n- cross-glance-py27 http://logs.openstack.org/44/666844/1/check/cross-glance-py27/aa87ee2/ : SUCCESS in 4m 44s\n- cross-glance-py35 http://logs.openstack.org/44/666844/1/check/cross-glance-py35/9368524/ : SUCCESS in 4m 30s\n- cross-horizon-py27 http://logs.openstack.org/44/666844/1/check/cross-horizon-py27/e032297/ : SUCCESS in 10m 07s\n- cross-horizon-py35 http://logs.openstack.org/44/666844/1/check/cross-horizon-py35/85dbc1b/ : SUCCESS in 9m 13s\n- cross-keystone-py27 http://logs.openstack.org/44/666844/1/check/cross-keystone-py27/c9ccfe4/ : SUCCESS in 12m 52s\n- cross-keystone-py35 http://logs.openstack.org/44/666844/1/check/cross-keystone-py35/5455e23/ : SUCCESS in 14m 40s\n- cross-neutron-py27 http://logs.openstack.org/44/666844/1/check/cross-neutron-py27/39f8da9/ : SUCCESS in 25m 52s\n- cross-neutron-py35 http://logs.openstack.org/44/666844/1/check/cross-neutron-py35/6fc849b/ : SUCCESS in 30m 42s\n- cross-nova-py27 http://logs.openstack.org/44/666844/1/check/cross-nova-py27/e65c201/ : SUCCESS in 12m 34s\n- cross-nova-py35 http://logs.openstack.org/44/666844/1/check/cross-nova-py35/550c445/ : SUCCESS in 13m 04s\n- cross-nova-functional http://logs.openstack.org/44/666844/1/check/cross-nova-functional/9e6b4d7/ : SUCCESS in 19m 00s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/44/666844/1/check/cross-osc-build-sphinx-docs/d914929/html/ : SUCCESS in 5m 45s\n- cross-swift-py27 http://logs.openstack.org/44/666844/1/check/cross-swift-py27/a692143/ : SUCCESS in 8m 19s\n- cross-swift-py35 http://logs.openstack.org/44/666844/1/check/cross-swift-py35/283276b/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/44/666844/1/check/openstack-tox-py36/a598e6f/ : SUCCESS in 3m 34s\n- requirements-integration http://logs.openstack.org/44/666844/1/check/requirements-integration/f1f356f/ : SUCCESS in 59m 49s\n- neutron-grenade-multinode http://logs.openstack.org/44/666844/1/check/neutron-grenade-multinode/7a2d6c8/ : SUCCESS in 1h 03m 44s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/44/666844/1/check/legacy-tempest-dsvm-neutron-full/b750042/ : SUCCESS in 1h 16m 25s\n- neutron-grenade http://logs.openstack.org/44/666844/1/check/neutron-grenade/b5bf28a/ : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":1},{"id":"22a830a3a9b00f057b0f03aed75d785fb36e3856","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-21 19:16:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"abe8b33c29883d738b4eceb0b336fe79463f73fc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 06:07:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9edb605fc253576eb7616cff5a7458110bd40ca0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 06:07:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7de3b71c4655a5d0bbd1ec38b34a4d19db141ce0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 09:32:02.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/666844/1/gate/openstack-tox-pep8/a35484c/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/44/666844/1/gate/openstack-tox-py27/8a52955/ : SUCCESS in 4m 44s\n- openstack-tox-py35 http://logs.openstack.org/44/666844/1/gate/openstack-tox-py35/8b58d49/ : SUCCESS in 3m 56s\n- openstack-tox-validate http://logs.openstack.org/44/666844/1/gate/openstack-tox-validate/00f2683/ : SUCCESS in 4m 13s\n- requirements-tox-babel http://logs.openstack.org/44/666844/1/gate/requirements-tox-babel/80decb2/ : SUCCESS in 10m 54s\n- requirements-tox-py27-check-uc http://logs.openstack.org/44/666844/1/gate/requirements-tox-py27-check-uc/58f86a3/ : SUCCESS in 8m 57s\n- requirements-tox-py35-check-uc http://logs.openstack.org/44/666844/1/gate/requirements-tox-py35-check-uc/b368c7f/ : SUCCESS in 7m 11s\n- cross-cinder-py27 http://logs.openstack.org/44/666844/1/gate/cross-cinder-py27/1ad08be/ : SUCCESS in 9m 10s\n- cross-cinder-py35 http://logs.openstack.org/44/666844/1/gate/cross-cinder-py35/2d7a18f/ : SUCCESS in 11m 08s\n- cross-glance-py27 http://logs.openstack.org/44/666844/1/gate/cross-glance-py27/709d4a4/ : SUCCESS in 4m 40s\n- cross-glance-py35 http://logs.openstack.org/44/666844/1/gate/cross-glance-py35/3c1429b/ : SUCCESS in 5m 35s\n- cross-horizon-py27 http://logs.openstack.org/44/666844/1/gate/cross-horizon-py27/9bcc6c4/ : SUCCESS in 8m 57s\n- cross-horizon-py35 http://logs.openstack.org/44/666844/1/gate/cross-horizon-py35/bc7f5a5/ : SUCCESS in 9m 06s\n- cross-keystone-py27 http://logs.openstack.org/44/666844/1/gate/cross-keystone-py27/d396233/ : SUCCESS in 11m 19s\n- cross-keystone-py35 http://logs.openstack.org/44/666844/1/gate/cross-keystone-py35/aa3fa29/ : SUCCESS in 14m 23s\n- cross-neutron-py27 http://logs.openstack.org/44/666844/1/gate/cross-neutron-py27/29a536c/ : SUCCESS in 23m 10s\n- cross-neutron-py35 http://logs.openstack.org/44/666844/1/gate/cross-neutron-py35/124c0e8/ : SUCCESS in 24m 18s\n- cross-nova-py27 http://logs.openstack.org/44/666844/1/gate/cross-nova-py27/e44f888/ : SUCCESS in 13m 41s\n- cross-nova-py35 http://logs.openstack.org/44/666844/1/gate/cross-nova-py35/67af9ca/ : SUCCESS in 14m 50s\n- cross-nova-functional http://logs.openstack.org/44/666844/1/gate/cross-nova-functional/950dec3/ : SUCCESS in 17m 26s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/44/666844/1/gate/cross-osc-build-sphinx-docs/96f5aa7/html/ : SUCCESS in 5m 20s\n- cross-swift-py27 http://logs.openstack.org/44/666844/1/gate/cross-swift-py27/0ecadf2/ : SUCCESS in 8m 01s\n- cross-swift-py35 http://logs.openstack.org/44/666844/1/gate/cross-swift-py35/e31f517/ : SUCCESS in 6m 03s\n- openstack-tox-py36 http://logs.openstack.org/44/666844/1/gate/openstack-tox-py36/d20cb94/ : SUCCESS in 3m 33s\n- requirements-integration http://logs.openstack.org/44/666844/1/gate/requirements-integration/08120cb/ : SUCCESS in 51m 21s\n- neutron-grenade-multinode http://logs.openstack.org/44/666844/1/gate/neutron-grenade-multinode/ba83eea/ : SUCCESS in 1h 02m 06s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/44/666844/1/gate/legacy-tempest-dsvm-neutron-full/e5ad7fb/ : FAILURE in 5m 46s\n- neutron-grenade http://logs.openstack.org/44/666844/1/gate/neutron-grenade/5cd0ffa/ : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":1},{"id":"cddaca8247411e9a54ef2056851e6dea73bad95a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 17:14:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b9252017c9c9204146457dc57dd97b8a30b36193","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 19:10:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/44/666844/1/check/openstack-tox-pypy/93740f4/ : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/44/666844/1/check/openstack-tox-pep8/cf6cce7/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/44/666844/1/check/openstack-tox-py27/7661f6b/ : SUCCESS in 4m 19s\n- openstack-tox-py35 http://logs.openstack.org/44/666844/1/check/openstack-tox-py35/adbc8ed/ : SUCCESS in 3m 01s\n- openstack-tox-validate http://logs.openstack.org/44/666844/1/check/openstack-tox-validate/a8a1dc0/ : SUCCESS in 4m 09s\n- requirements-tox-babel http://logs.openstack.org/44/666844/1/check/requirements-tox-babel/085fe9e/ : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc http://logs.openstack.org/44/666844/1/check/requirements-tox-py27-check-uc/9eb9c91/ : SUCCESS in 7m 32s\n- requirements-tox-py35-check-uc http://logs.openstack.org/44/666844/1/check/requirements-tox-py35-check-uc/8245844/ : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc http://logs.openstack.org/44/666844/1/check/requirements-tox-py36-check-uc/a1914eb/ : SUCCESS in 6m 31s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/44/666844/1/check/cross-cinder-py27/5cb51c3/ : SUCCESS in 9m 23s\n- cross-cinder-py35 http://logs.openstack.org/44/666844/1/check/cross-cinder-py35/672fb58/ : SUCCESS in 10m 59s\n- cross-glance-py27 http://logs.openstack.org/44/666844/1/check/cross-glance-py27/ca714d2/ : SUCCESS in 4m 02s\n- cross-glance-py35 http://logs.openstack.org/44/666844/1/check/cross-glance-py35/03f4abf/ : SUCCESS in 4m 34s\n- cross-horizon-py27 http://logs.openstack.org/44/666844/1/check/cross-horizon-py27/296a72d/ : SUCCESS in 7m 51s\n- cross-horizon-py35 http://logs.openstack.org/44/666844/1/check/cross-horizon-py35/f5e6826/ : SUCCESS in 9m 59s\n- cross-keystone-py27 http://logs.openstack.org/44/666844/1/check/cross-keystone-py27/bf9e3d0/ : SUCCESS in 11m 53s\n- cross-keystone-py35 http://logs.openstack.org/44/666844/1/check/cross-keystone-py35/b4c90de/ : SUCCESS in 13m 18s\n- cross-neutron-py27 http://logs.openstack.org/44/666844/1/check/cross-neutron-py27/0dd8d08/ : SUCCESS in 20m 41s\n- cross-neutron-py35 http://logs.openstack.org/44/666844/1/check/cross-neutron-py35/f055b73/ : SUCCESS in 25m 11s\n- cross-nova-py27 http://logs.openstack.org/44/666844/1/check/cross-nova-py27/cf6c74e/ : SUCCESS in 13m 17s\n- cross-nova-py35 http://logs.openstack.org/44/666844/1/check/cross-nova-py35/7f3c6d4/ : SUCCESS in 12m 14s\n- cross-nova-functional http://logs.openstack.org/44/666844/1/check/cross-nova-functional/ed3d5f7/ : SUCCESS in 15m 38s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/44/666844/1/check/cross-osc-build-sphinx-docs/5e1c36a/html/ : SUCCESS in 6m 18s\n- cross-swift-py27 http://logs.openstack.org/44/666844/1/check/cross-swift-py27/779b0ec/ : SUCCESS in 7m 15s\n- cross-swift-py35 http://logs.openstack.org/44/666844/1/check/cross-swift-py35/5510beb/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/44/666844/1/check/openstack-tox-py36/db2b668/ : SUCCESS in 3m 30s\n- requirements-integration http://logs.openstack.org/44/666844/1/check/requirements-integration/0ec1ee2/ : SUCCESS in 52m 28s\n- neutron-grenade-multinode http://logs.openstack.org/44/666844/1/check/neutron-grenade-multinode/e00b033/ : SUCCESS in 1h 01m 33s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/44/666844/1/check/legacy-tempest-dsvm-neutron-full/c0e2296/ : SUCCESS in 1h 36m 47s\n- neutron-grenade http://logs.openstack.org/44/666844/1/check/neutron-grenade/07e7f2b/ : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d653990e894750b1006765cbbf20d0bdbd22f3ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 19:10:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b00ce9544d4403f4b33b02870d486f7839d04b65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 20:56:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/44/666844/1/gate/openstack-tox-pep8/ef799cf/ : SUCCESS in 2m 44s\n- openstack-tox-py27 http://logs.openstack.org/44/666844/1/gate/openstack-tox-py27/db22cc2/ : SUCCESS in 3m 30s\n- openstack-tox-py35 http://logs.openstack.org/44/666844/1/gate/openstack-tox-py35/eb525e9/ : SUCCESS in 2m 55s\n- openstack-tox-validate http://logs.openstack.org/44/666844/1/gate/openstack-tox-validate/dfb5fe5/ : SUCCESS in 2m 47s\n- requirements-tox-babel http://logs.openstack.org/44/666844/1/gate/requirements-tox-babel/3c80602/ : SUCCESS in 2m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/44/666844/1/gate/requirements-tox-py27-check-uc/5beeb65/ : SUCCESS in 5m 52s\n- requirements-tox-py35-check-uc http://logs.openstack.org/44/666844/1/gate/requirements-tox-py35-check-uc/bfb1b20/ : SUCCESS in 5m 57s\n- cross-cinder-py27 http://logs.openstack.org/44/666844/1/gate/cross-cinder-py27/68089d2/ : SUCCESS in 9m 05s\n- cross-cinder-py35 http://logs.openstack.org/44/666844/1/gate/cross-cinder-py35/b1d2499/ : SUCCESS in 9m 10s\n- cross-glance-py27 http://logs.openstack.org/44/666844/1/gate/cross-glance-py27/637d66b/ : SUCCESS in 3m 57s\n- cross-glance-py35 http://logs.openstack.org/44/666844/1/gate/cross-glance-py35/0e8555d/ : SUCCESS in 4m 39s\n- cross-horizon-py27 http://logs.openstack.org/44/666844/1/gate/cross-horizon-py27/aa6920a/ : SUCCESS in 9m 51s\n- cross-horizon-py35 http://logs.openstack.org/44/666844/1/gate/cross-horizon-py35/11f2108/ : SUCCESS in 9m 17s\n- cross-keystone-py27 http://logs.openstack.org/44/666844/1/gate/cross-keystone-py27/b4a2617/ : SUCCESS in 11m 20s\n- cross-keystone-py35 http://logs.openstack.org/44/666844/1/gate/cross-keystone-py35/a61398f/ : SUCCESS in 18m 10s\n- cross-neutron-py27 http://logs.openstack.org/44/666844/1/gate/cross-neutron-py27/f053061/ : SUCCESS in 29m 38s\n- cross-neutron-py35 http://logs.openstack.org/44/666844/1/gate/cross-neutron-py35/c807244/ : SUCCESS in 24m 15s\n- cross-nova-py27 http://logs.openstack.org/44/666844/1/gate/cross-nova-py27/cb8de07/ : SUCCESS in 14m 18s\n- cross-nova-py35 http://logs.openstack.org/44/666844/1/gate/cross-nova-py35/36e77d3/ : SUCCESS in 22m 12s\n- cross-nova-functional http://logs.openstack.org/44/666844/1/gate/cross-nova-functional/0ee867f/ : SUCCESS in 18m 26s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/44/666844/1/gate/cross-osc-build-sphinx-docs/ba4d094/html/ : SUCCESS in 5m 21s\n- cross-swift-py27 http://logs.openstack.org/44/666844/1/gate/cross-swift-py27/dab3cb9/ : SUCCESS in 7m 11s\n- cross-swift-py35 http://logs.openstack.org/44/666844/1/gate/cross-swift-py35/06c91f3/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/44/666844/1/gate/openstack-tox-py36/b1e6616/ : SUCCESS in 3m 25s\n- requirements-integration http://logs.openstack.org/44/666844/1/gate/requirements-integration/f258a05/ : SUCCESS in 52m 49s\n- neutron-grenade-multinode http://logs.openstack.org/44/666844/1/gate/neutron-grenade-multinode/64df0ce/ : FAILURE in 19m 53s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/44/666844/1/gate/legacy-tempest-dsvm-neutron-full/c5e81e9/ : SUCCESS in 1h 27m 14s\n- neutron-grenade http://logs.openstack.org/44/666844/1/gate/neutron-grenade/f225cab/ : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f75fbcc5023bb97a81664f59e33fba5d49a0e6e2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 22:25:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1612d67f6cf73d64baefe6fba56a3470232de296","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 23:53:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/44/666844/1/check/openstack-tox-pypy/37b91fe/ : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/44/666844/1/check/openstack-tox-pep8/f5f4e37/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/44/666844/1/check/openstack-tox-py27/6c5ca15/ : SUCCESS in 3m 46s\n- openstack-tox-py35 http://logs.openstack.org/44/666844/1/check/openstack-tox-py35/2ef9f03/ : SUCCESS in 3m 49s\n- openstack-tox-validate http://logs.openstack.org/44/666844/1/check/openstack-tox-validate/bf23043/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/44/666844/1/check/requirements-tox-babel/cd9d09d/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/44/666844/1/check/requirements-tox-py27-check-uc/fce9750/ : SUCCESS in 5m 59s\n- requirements-tox-py35-check-uc http://logs.openstack.org/44/666844/1/check/requirements-tox-py35-check-uc/e2999d6/ : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc http://logs.openstack.org/44/666844/1/check/requirements-tox-py36-check-uc/da9b236/ : SUCCESS in 6m 00s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/44/666844/1/check/cross-cinder-py27/fd13211/ : SUCCESS in 7m 59s\n- cross-cinder-py35 http://logs.openstack.org/44/666844/1/check/cross-cinder-py35/ba32810/ : SUCCESS in 8m 16s\n- cross-glance-py27 http://logs.openstack.org/44/666844/1/check/cross-glance-py27/87954bb/ : SUCCESS in 4m 40s\n- cross-glance-py35 http://logs.openstack.org/44/666844/1/check/cross-glance-py35/cc1d482/ : SUCCESS in 4m 17s\n- cross-horizon-py27 http://logs.openstack.org/44/666844/1/check/cross-horizon-py27/f130787/ : SUCCESS in 10m 18s\n- cross-horizon-py35 http://logs.openstack.org/44/666844/1/check/cross-horizon-py35/89f7846/ : SUCCESS in 9m 36s\n- cross-keystone-py27 http://logs.openstack.org/44/666844/1/check/cross-keystone-py27/a4d4d06/ : SUCCESS in 12m 11s\n- cross-keystone-py35 http://logs.openstack.org/44/666844/1/check/cross-keystone-py35/36fe1ac/ : SUCCESS in 12m 20s\n- cross-neutron-py27 http://logs.openstack.org/44/666844/1/check/cross-neutron-py27/f8d60c8/ : SUCCESS in 25m 31s\n- cross-neutron-py35 http://logs.openstack.org/44/666844/1/check/cross-neutron-py35/36120ec/ : SUCCESS in 25m 14s\n- cross-nova-py27 http://logs.openstack.org/44/666844/1/check/cross-nova-py27/c6b5d8c/ : SUCCESS in 12m 32s\n- cross-nova-py35 http://logs.openstack.org/44/666844/1/check/cross-nova-py35/1c23373/ : SUCCESS in 12m 09s\n- cross-nova-functional http://logs.openstack.org/44/666844/1/check/cross-nova-functional/dedd806/ : SUCCESS in 17m 47s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/44/666844/1/check/cross-osc-build-sphinx-docs/9767d1b/html/ : SUCCESS in 5m 07s\n- cross-swift-py27 http://logs.openstack.org/44/666844/1/check/cross-swift-py27/d21d97b/ : SUCCESS in 7m 24s\n- cross-swift-py35 http://logs.openstack.org/44/666844/1/check/cross-swift-py35/df48530/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/44/666844/1/check/openstack-tox-py36/581f8f1/ : SUCCESS in 3m 25s\n- requirements-integration http://logs.openstack.org/44/666844/1/check/requirements-integration/8cdd02d/ : SUCCESS in 1h 00m 20s\n- neutron-grenade-multinode http://logs.openstack.org/44/666844/1/check/neutron-grenade-multinode/f22ddd2/ : SUCCESS in 1h 04m 10s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/44/666844/1/check/legacy-tempest-dsvm-neutron-full/635bef1/ : SUCCESS in 1h 23m 04s\n- neutron-grenade http://logs.openstack.org/44/666844/1/check/neutron-grenade/8ef9f9b/ : SUCCESS in 53m 13s","accounts_in_message":[],"_revision_number":1},{"id":"33c457e4835c241900234fe248f1d67aeeaf789b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 23:53:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b882f23ebb16da3bebd3eb0242b7c43e26aa73a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 01:28:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/44/666844/1/gate/openstack-tox-pep8/78639b7/ : SUCCESS in 8m 00s\n- openstack-tox-py27 http://logs.openstack.org/44/666844/1/gate/openstack-tox-py27/4f85000/ : SUCCESS in 8m 18s\n- openstack-tox-py35 http://logs.openstack.org/44/666844/1/gate/openstack-tox-py35/4dc5ab3/ : SUCCESS in 8m 24s\n- openstack-tox-validate http://logs.openstack.org/44/666844/1/gate/openstack-tox-validate/b5a4396/ : SUCCESS in 4m 26s\n- requirements-tox-babel http://logs.openstack.org/44/666844/1/gate/requirements-tox-babel/ec02ca1/ : SUCCESS in 4m 27s\n- requirements-tox-py27-check-uc http://logs.openstack.org/44/666844/1/gate/requirements-tox-py27-check-uc/de506cd/ : SUCCESS in 10m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/44/666844/1/gate/requirements-tox-py35-check-uc/85fbe0f/ : SUCCESS in 6m 54s\n- cross-cinder-py27 http://logs.openstack.org/44/666844/1/gate/cross-cinder-py27/13d3d19/ : SUCCESS in 11m 00s\n- cross-cinder-py35 http://logs.openstack.org/44/666844/1/gate/cross-cinder-py35/50cdef9/ : SUCCESS in 11m 23s\n- cross-glance-py27 http://logs.openstack.org/44/666844/1/gate/cross-glance-py27/1964cfc/ : SUCCESS in 5m 32s\n- cross-glance-py35 http://logs.openstack.org/44/666844/1/gate/cross-glance-py35/20f8893/ : SUCCESS in 5m 13s\n- cross-horizon-py27 http://logs.openstack.org/44/666844/1/gate/cross-horizon-py27/0aa783e/ : SUCCESS in 10m 51s\n- cross-horizon-py35 http://logs.openstack.org/44/666844/1/gate/cross-horizon-py35/7ac218d/ : SUCCESS in 9m 24s\n- cross-keystone-py27 http://logs.openstack.org/44/666844/1/gate/cross-keystone-py27/8e20ecf/ : SUCCESS in 17m 17s\n- cross-keystone-py35 http://logs.openstack.org/44/666844/1/gate/cross-keystone-py35/17428b5/ : SUCCESS in 19m 17s\n- cross-neutron-py27 http://logs.openstack.org/44/666844/1/gate/cross-neutron-py27/67170a2/ : SUCCESS in 21m 25s\n- cross-neutron-py35 http://logs.openstack.org/44/666844/1/gate/cross-neutron-py35/fe12305/ : SUCCESS in 26m 00s\n- cross-nova-py27 http://logs.openstack.org/44/666844/1/gate/cross-nova-py27/ab3e97f/ : SUCCESS in 14m 42s\n- cross-nova-py35 http://logs.openstack.org/44/666844/1/gate/cross-nova-py35/2dd2f51/ : SUCCESS in 17m 01s\n- cross-nova-functional http://logs.openstack.org/44/666844/1/gate/cross-nova-functional/04743bd/ : SUCCESS in 16m 45s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/44/666844/1/gate/cross-osc-build-sphinx-docs/0669f0d/html/ : SUCCESS in 6m 40s\n- cross-swift-py27 http://logs.openstack.org/44/666844/1/gate/cross-swift-py27/dc24f51/ : SUCCESS in 9m 51s\n- cross-swift-py35 http://logs.openstack.org/44/666844/1/gate/cross-swift-py35/56152c5/ : SUCCESS in 5m 01s\n- openstack-tox-py36 http://logs.openstack.org/44/666844/1/gate/openstack-tox-py36/4596365/ : SUCCESS in 4m 25s\n- requirements-integration http://logs.openstack.org/44/666844/1/gate/requirements-integration/743338b/ : SUCCESS in 56m 23s\n- neutron-grenade-multinode http://logs.openstack.org/44/666844/1/gate/neutron-grenade-multinode/1b846b6/ : SUCCESS in 57m 28s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/44/666844/1/gate/legacy-tempest-dsvm-neutron-full/e6ab61f/ : SUCCESS in 1h 34m 16s\n- neutron-grenade http://logs.openstack.org/44/666844/1/gate/neutron-grenade/0031d7c/ : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e90a880e2def364013946a9ae47d425cb06ad3e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 01:28:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"064f3eb30b0ce08aea3124ad1978ab5ee50219cb","revisions":{"064f3eb30b0ce08aea3124ad1978ab5ee50219cb":{"kind":"REWORK","_number":1,"created":"2019-06-21 14:05:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/666844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/666844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/666844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/666844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/666844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/666844/1"}}},"commit":{"parents":[{"commit":"14579e115c6250ca45e8afa936374630089f2338","subject":"update constraint for ovsdbapp to new release 0.12.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14579e115c6250ca45e8afa936374630089f2338"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 14:05:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 14:05:08.000000000","tz":0},"subject":"update constraint for os-net-config to new release 9.4.0","message":"update constraint for os-net-config to new release 9.4.0\n\nChange-Id: Ib0caaa2e1c26204f9fddb1027092f04064780371\nmeta:version: 9.4.0\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: I1fa3140f68ebc5534c5d86ab7ffea45f89251f44\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/064f3eb30b0ce08aea3124ad1978ab5ee50219cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/064f3eb30b0ce08aea3124ad1978ab5ee50219cb"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
