)]}'
{"id":"openstack%2Frequirements~666873","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I62c766233518b15d34962c32dbf1a86856524bbd","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I62c766233518b15d34962c32dbf1a86856524bbd","subject":"update constraint for os-apply-config to new release 8.3.2","status":"MERGED","created":"2019-06-21 15:40:40.000000000","updated":"2019-06-22 23:48:35.000000000","submitted":"2019-06-22 23:48:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666873-1561247315323-b95397b7","meta_rev_id":"67c4a0f00aed630d226ee35c84aa50a886d55bff","_number":666873,"virtual_id_number":666873,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-22 23:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-21 20:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-22 06:07:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-22 06:07:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-22 23:48:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c666657517d9aa19cd55a8ae6706d6cc71603a57","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-21 15:40:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20765323ccf198fddb2855ad89699183634d4d1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 20:45:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/73/666873/1/check/openstack-tox-pypy/dfd652d/ : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/666873/1/check/openstack-tox-pep8/c286cff/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/73/666873/1/check/openstack-tox-py27/f1b780c/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/73/666873/1/check/openstack-tox-py35/9d80998/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/73/666873/1/check/openstack-tox-validate/5c9a125/ : SUCCESS in 3m 31s\n- requirements-tox-babel http://logs.openstack.org/73/666873/1/check/requirements-tox-babel/056411b/ : SUCCESS in 3m 07s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/666873/1/check/requirements-tox-py27-check-uc/bd439c7/ : SUCCESS in 5m 49s\n- requirements-tox-py35-check-uc http://logs.openstack.org/73/666873/1/check/requirements-tox-py35-check-uc/fb873a6/ : SUCCESS in 6m 46s\n- cross-cinder-py27 http://logs.openstack.org/73/666873/1/check/cross-cinder-py27/eb79565/ : SUCCESS in 9m 05s\n- cross-cinder-py35 http://logs.openstack.org/73/666873/1/check/cross-cinder-py35/2047e0e/ : SUCCESS in 7m 56s\n- cross-glance-py27 http://logs.openstack.org/73/666873/1/check/cross-glance-py27/b105a31/ : SUCCESS in 5m 01s\n- cross-glance-py35 http://logs.openstack.org/73/666873/1/check/cross-glance-py35/02e2796/ : SUCCESS in 5m 32s\n- cross-horizon-py27 http://logs.openstack.org/73/666873/1/check/cross-horizon-py27/6d24c34/ : SUCCESS in 11m 53s\n- cross-horizon-py35 http://logs.openstack.org/73/666873/1/check/cross-horizon-py35/065f64d/ : SUCCESS in 11m 21s\n- cross-keystone-py27 http://logs.openstack.org/73/666873/1/check/cross-keystone-py27/856951e/ : SUCCESS in 11m 33s\n- cross-keystone-py35 http://logs.openstack.org/73/666873/1/check/cross-keystone-py35/8a0d9d4/ : SUCCESS in 11m 51s\n- cross-neutron-py27 http://logs.openstack.org/73/666873/1/check/cross-neutron-py27/630f5d5/ : SUCCESS in 23m 57s\n- cross-neutron-py35 http://logs.openstack.org/73/666873/1/check/cross-neutron-py35/405d3b3/ : SUCCESS in 32m 12s\n- cross-nova-py27 http://logs.openstack.org/73/666873/1/check/cross-nova-py27/714c0f7/ : SUCCESS in 16m 18s\n- cross-nova-py35 http://logs.openstack.org/73/666873/1/check/cross-nova-py35/8fb59b7/ : SUCCESS in 14m 00s\n- cross-nova-functional http://logs.openstack.org/73/666873/1/check/cross-nova-functional/151d7f5/ : SUCCESS in 20m 52s\n- cross-swift-py27 http://logs.openstack.org/73/666873/1/check/cross-swift-py27/096bf4f/ : SUCCESS in 5m 55s\n- cross-swift-py35 http://logs.openstack.org/73/666873/1/check/cross-swift-py35/337be7b/ : SUCCESS in 3m 37s\n- requirements-integration http://logs.openstack.org/73/666873/1/check/requirements-integration/993de2b/ : SUCCESS in 51m 51s\n- neutron-grenade-multinode http://logs.openstack.org/73/666873/1/check/neutron-grenade-multinode/83d3221/ : SUCCESS in 1h 09m 09s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/73/666873/1/check/legacy-tempest-dsvm-neutron-full/574bbe9/ : SUCCESS in 1h 17m 39s\n- neutron-grenade http://logs.openstack.org/73/666873/1/check/neutron-grenade/c0aa041/ : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":1},{"id":"812f9c2a18b92e5c231fe67376c5e1cb8c473d57","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-21 20:50:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7abc30df712863eb8ecd322d8c7df7edf1135772","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 06:07:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aafb2ad32b6d812ec18d50eacfe9d9576fc06333","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 06:07:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"132fcf991e20a5abc21c5ee498f977464f566dca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 10:13:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/73/666873/1/gate/openstack-tox-pep8/f4831d1/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/73/666873/1/gate/openstack-tox-py27/af8a8f1/ : SUCCESS in 6m 44s\n- openstack-tox-py35 http://logs.openstack.org/73/666873/1/gate/openstack-tox-py35/b1fbe70/ : FAILURE in 7m 12s\n- openstack-tox-validate http://logs.openstack.org/73/666873/1/gate/openstack-tox-validate/9be4736/ : SUCCESS in 3m 42s\n- requirements-tox-babel http://logs.openstack.org/73/666873/1/gate/requirements-tox-babel/a1dfc83/ : SUCCESS in 2m 47s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/666873/1/gate/requirements-tox-py27-check-uc/62b2b60/ : SUCCESS in 8m 59s\n- requirements-tox-py35-check-uc http://logs.openstack.org/73/666873/1/gate/requirements-tox-py35-check-uc/60af6f2/ : SUCCESS in 7m 58s\n- cross-cinder-py27 http://logs.openstack.org/73/666873/1/gate/cross-cinder-py27/7d5b5b5/ : SUCCESS in 8m 23s\n- cross-cinder-py35 http://logs.openstack.org/73/666873/1/gate/cross-cinder-py35/fffc016/ : SUCCESS in 15m 10s\n- cross-glance-py27 http://logs.openstack.org/73/666873/1/gate/cross-glance-py27/2b7237a/ : RETRY_LIMIT in 1m 57s\n- cross-glance-py35 http://logs.openstack.org/73/666873/1/gate/cross-glance-py35/5b39807/ : SUCCESS in 7m 14s\n- cross-horizon-py27 http://logs.openstack.org/73/666873/1/gate/cross-horizon-py27/9ff99bc/ : SUCCESS in 10m 36s\n- cross-horizon-py35 http://logs.openstack.org/73/666873/1/gate/cross-horizon-py35/93413f6/ : POST_FAILURE in 24m 16s\n- cross-keystone-py27 http://logs.openstack.org/73/666873/1/gate/cross-keystone-py27/e2f28e0/ : SUCCESS in 12m 23s\n- cross-keystone-py35 http://logs.openstack.org/73/666873/1/gate/cross-keystone-py35/48af692/ : FAILURE in 5m 31s\n- cross-neutron-py27 http://logs.openstack.org/73/666873/1/gate/cross-neutron-py27/83528da/ : SUCCESS in 29m 04s\n- cross-neutron-py35 http://logs.openstack.org/73/666873/1/gate/cross-neutron-py35/35ca2d0/ : SUCCESS in 23m 43s\n- cross-nova-py27 http://logs.openstack.org/73/666873/1/gate/cross-nova-py27/50f5247/ : SUCCESS in 13m 37s\n- cross-nova-py35 http://logs.openstack.org/73/666873/1/gate/cross-nova-py35/3fdc8fd/ : SUCCESS in 13m 02s\n- cross-nova-functional http://logs.openstack.org/73/666873/1/gate/cross-nova-functional/cd36cf3/ : SUCCESS in 21m 14s\n- cross-swift-py27 http://logs.openstack.org/73/666873/1/gate/cross-swift-py27/903f8d9/ : SUCCESS in 6m 26s\n- cross-swift-py35 http://logs.openstack.org/73/666873/1/gate/cross-swift-py35/c776b4a/ : SUCCESS in 4m 23s\n- requirements-integration http://logs.openstack.org/73/666873/1/gate/requirements-integration/ad60747/ : SUCCESS in 49m 42s\n- neutron-grenade-multinode http://logs.openstack.org/73/666873/1/gate/neutron-grenade-multinode/89ead7e/ : SUCCESS in 1h 00m 54s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/73/666873/1/gate/legacy-tempest-dsvm-neutron-full/c7b7d9c/ : SUCCESS in 1h 29m 20s\n- neutron-grenade http://logs.openstack.org/73/666873/1/gate/neutron-grenade/f6bc6bd/ : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d4aef7fcb13e03cfdab14259edbd797b353a19fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 17:14:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3324be9865287bfbbbe1ab7ea501c25a4e602d2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 19:00:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/73/666873/1/check/openstack-tox-pypy/888e223/ : POST_FAILURE in 10m 32s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/666873/1/check/openstack-tox-pep8/77b2d41/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/73/666873/1/check/openstack-tox-py27/5fcc116/ : SUCCESS in 6m 07s\n- openstack-tox-py35 http://logs.openstack.org/73/666873/1/check/openstack-tox-py35/80f5263/ : SUCCESS in 3m 06s\n- openstack-tox-validate http://logs.openstack.org/73/666873/1/check/openstack-tox-validate/06167d0/ : SUCCESS in 3m 33s\n- requirements-tox-babel http://logs.openstack.org/73/666873/1/check/requirements-tox-babel/300db87/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/666873/1/check/requirements-tox-py27-check-uc/54ee558/ : SUCCESS in 5m 17s\n- requirements-tox-py35-check-uc http://logs.openstack.org/73/666873/1/check/requirements-tox-py35-check-uc/a897a95/ : SUCCESS in 5m 31s\n- cross-cinder-py27 http://logs.openstack.org/73/666873/1/check/cross-cinder-py27/2f6ced6/ : FAILURE in 10m 09s\n- cross-cinder-py35 http://logs.openstack.org/73/666873/1/check/cross-cinder-py35/2a85ff4/ : SUCCESS in 7m 29s\n- cross-glance-py27 http://logs.openstack.org/73/666873/1/check/cross-glance-py27/1eef185/ : SUCCESS in 6m 27s\n- cross-glance-py35 http://logs.openstack.org/73/666873/1/check/cross-glance-py35/0f428e7/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/73/666873/1/check/cross-horizon-py27/42bbfe4/ : SUCCESS in 12m 14s\n- cross-horizon-py35 http://logs.openstack.org/73/666873/1/check/cross-horizon-py35/fb48a7b/ : SUCCESS in 11m 06s\n- cross-keystone-py27 http://logs.openstack.org/73/666873/1/check/cross-keystone-py27/fe587f3/ : SUCCESS in 12m 59s\n- cross-keystone-py35 http://logs.openstack.org/73/666873/1/check/cross-keystone-py35/da730e3/ : SUCCESS in 13m 25s\n- cross-neutron-py27 http://logs.openstack.org/73/666873/1/check/cross-neutron-py27/e8f4553/ : SUCCESS in 25m 35s\n- cross-neutron-py35 http://logs.openstack.org/73/666873/1/check/cross-neutron-py35/56ae15a/ : SUCCESS in 27m 33s\n- cross-nova-py27 http://logs.openstack.org/73/666873/1/check/cross-nova-py27/9830fef/ : SUCCESS in 10m 47s\n- cross-nova-py35 http://logs.openstack.org/73/666873/1/check/cross-nova-py35/2aeb530/ : SUCCESS in 18m 26s\n- cross-nova-functional http://logs.openstack.org/73/666873/1/check/cross-nova-functional/6896f6b/ : SUCCESS in 17m 50s\n- cross-swift-py27 http://logs.openstack.org/73/666873/1/check/cross-swift-py27/bdcbdce/ : SUCCESS in 5m 57s\n- cross-swift-py35 http://logs.openstack.org/73/666873/1/check/cross-swift-py35/15e3248/ : SUCCESS in 4m 26s\n- requirements-integration http://logs.openstack.org/73/666873/1/check/requirements-integration/0232201/ : SUCCESS in 41m 40s\n- neutron-grenade-multinode http://logs.openstack.org/73/666873/1/check/neutron-grenade-multinode/19bd194/ : SUCCESS in 55m 19s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/73/666873/1/check/legacy-tempest-dsvm-neutron-full/f3c4a1a/ : SUCCESS in 1h 21m 05s\n- neutron-grenade http://logs.openstack.org/73/666873/1/check/neutron-grenade/b64a806/ : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1696b8afea920dd705ff282e9678b858bfc4256e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 19:14:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"061be51466568e5b8904664a7f677f0c85186a45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 20:40:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/73/666873/1/check/openstack-tox-pypy/fc53a64/ : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/666873/1/check/openstack-tox-pep8/3677ee7/ : SUCCESS in 2m 49s\n- openstack-tox-py27 http://logs.openstack.org/73/666873/1/check/openstack-tox-py27/4341865/ : SUCCESS in 3m 16s\n- openstack-tox-py35 http://logs.openstack.org/73/666873/1/check/openstack-tox-py35/a99073c/ : SUCCESS in 4m 19s\n- openstack-tox-validate http://logs.openstack.org/73/666873/1/check/openstack-tox-validate/397be86/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/73/666873/1/check/requirements-tox-babel/ddd0a10/ : SUCCESS in 2m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/666873/1/check/requirements-tox-py27-check-uc/fe3e9dd/ : SUCCESS in 7m 20s\n- requirements-tox-py35-check-uc http://logs.openstack.org/73/666873/1/check/requirements-tox-py35-check-uc/9b92f8a/ : SUCCESS in 7m 31s\n- cross-cinder-py27 http://logs.openstack.org/73/666873/1/check/cross-cinder-py27/11aa3c1/ : SUCCESS in 9m 20s\n- cross-cinder-py35 http://logs.openstack.org/73/666873/1/check/cross-cinder-py35/8857bff/ : SUCCESS in 8m 57s\n- cross-glance-py27 http://logs.openstack.org/73/666873/1/check/cross-glance-py27/2484c24/ : SUCCESS in 5m 05s\n- cross-glance-py35 http://logs.openstack.org/73/666873/1/check/cross-glance-py35/f84bd22/ : SUCCESS in 5m 08s\n- cross-horizon-py27 http://logs.openstack.org/73/666873/1/check/cross-horizon-py27/fde6a86/ : SUCCESS in 11m 30s\n- cross-horizon-py35 http://logs.openstack.org/73/666873/1/check/cross-horizon-py35/fd4b8ee/ : SUCCESS in 10m 51s\n- cross-keystone-py27 http://logs.openstack.org/73/666873/1/check/cross-keystone-py27/6375a08/ : SUCCESS in 12m 04s\n- cross-keystone-py35 http://logs.openstack.org/73/666873/1/check/cross-keystone-py35/367891d/ : SUCCESS in 11m 43s\n- cross-neutron-py27 http://logs.openstack.org/73/666873/1/check/cross-neutron-py27/f2438a8/ : SUCCESS in 24m 05s\n- cross-neutron-py35 http://logs.openstack.org/73/666873/1/check/cross-neutron-py35/b9d525b/ : SUCCESS in 23m 35s\n- cross-nova-py27 http://logs.openstack.org/73/666873/1/check/cross-nova-py27/0b38038/ : SUCCESS in 12m 22s\n- cross-nova-py35 http://logs.openstack.org/73/666873/1/check/cross-nova-py35/f6f7f92/ : SUCCESS in 12m 13s\n- cross-nova-functional http://logs.openstack.org/73/666873/1/check/cross-nova-functional/9002c4c/ : SUCCESS in 21m 05s\n- cross-swift-py27 http://logs.openstack.org/73/666873/1/check/cross-swift-py27/27ddd95/ : SUCCESS in 5m 30s\n- cross-swift-py35 http://logs.openstack.org/73/666873/1/check/cross-swift-py35/f26957f/ : SUCCESS in 3m 18s\n- requirements-integration http://logs.openstack.org/73/666873/1/check/requirements-integration/29b177d/ : SUCCESS in 48m 57s\n- neutron-grenade-multinode http://logs.openstack.org/73/666873/1/check/neutron-grenade-multinode/ce292d6/ : SUCCESS in 1h 08m 01s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/73/666873/1/check/legacy-tempest-dsvm-neutron-full/9d25b7e/ : SUCCESS in 1h 20m 30s\n- neutron-grenade http://logs.openstack.org/73/666873/1/check/neutron-grenade/3e61375/ : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f54aad211b7572e3eab241530ac1bdef0be41f6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 20:40:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae79fb453da2a622d6dcec505da3c859034b01fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 23:48:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6e3741d1b0f847d1de0a47f100c2303dd18c8cdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 23:48:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/73/666873/1/gate/openstack-tox-pep8/5cde061/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/73/666873/1/gate/openstack-tox-py27/334e700/ : SUCCESS in 3m 55s\n- openstack-tox-py35 http://logs.openstack.org/73/666873/1/gate/openstack-tox-py35/df4f697/ : SUCCESS in 4m 32s\n- openstack-tox-validate http://logs.openstack.org/73/666873/1/gate/openstack-tox-validate/d56b614/ : SUCCESS in 3m 22s\n- requirements-tox-babel http://logs.openstack.org/73/666873/1/gate/requirements-tox-babel/61311d1/ : SUCCESS in 4m 11s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/666873/1/gate/requirements-tox-py27-check-uc/529fb48/ : SUCCESS in 9m 14s\n- requirements-tox-py35-check-uc http://logs.openstack.org/73/666873/1/gate/requirements-tox-py35-check-uc/e787686/ : SUCCESS in 6m 26s\n- cross-cinder-py27 http://logs.openstack.org/73/666873/1/gate/cross-cinder-py27/749152d/ : SUCCESS in 10m 04s\n- cross-cinder-py35 http://logs.openstack.org/73/666873/1/gate/cross-cinder-py35/21ec7d6/ : SUCCESS in 9m 25s\n- cross-glance-py27 http://logs.openstack.org/73/666873/1/gate/cross-glance-py27/836320c/ : SUCCESS in 5m 13s\n- cross-glance-py35 http://logs.openstack.org/73/666873/1/gate/cross-glance-py35/da4f0e1/ : SUCCESS in 6m 08s\n- cross-horizon-py27 http://logs.openstack.org/73/666873/1/gate/cross-horizon-py27/8158371/ : SUCCESS in 10m 03s\n- cross-horizon-py35 http://logs.openstack.org/73/666873/1/gate/cross-horizon-py35/211a0ca/ : SUCCESS in 9m 37s\n- cross-keystone-py27 http://logs.openstack.org/73/666873/1/gate/cross-keystone-py27/9c65849/ : SUCCESS in 14m 10s\n- cross-keystone-py35 http://logs.openstack.org/73/666873/1/gate/cross-keystone-py35/05a019a/ : SUCCESS in 12m 17s\n- cross-neutron-py27 http://logs.openstack.org/73/666873/1/gate/cross-neutron-py27/773bd22/ : SUCCESS in 24m 20s\n- cross-neutron-py35 http://logs.openstack.org/73/666873/1/gate/cross-neutron-py35/254e8d2/ : SUCCESS in 21m 51s\n- cross-nova-py27 http://logs.openstack.org/73/666873/1/gate/cross-nova-py27/ee21a81/ : SUCCESS in 12m 26s\n- cross-nova-py35 http://logs.openstack.org/73/666873/1/gate/cross-nova-py35/495aae5/ : SUCCESS in 13m 09s\n- cross-nova-functional http://logs.openstack.org/73/666873/1/gate/cross-nova-functional/4a87a65/ : SUCCESS in 20m 27s\n- cross-swift-py27 http://logs.openstack.org/73/666873/1/gate/cross-swift-py27/9797b1d/ : SUCCESS in 5m 59s\n- cross-swift-py35 http://logs.openstack.org/73/666873/1/gate/cross-swift-py35/4978f27/ : SUCCESS in 5m 07s\n- requirements-integration http://logs.openstack.org/73/666873/1/gate/requirements-integration/cda9436/ : SUCCESS in 44m 35s\n- neutron-grenade-multinode http://logs.openstack.org/73/666873/1/gate/neutron-grenade-multinode/1f375d4/ : SUCCESS in 1h 00m 59s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/73/666873/1/gate/legacy-tempest-dsvm-neutron-full/787413d/ : SUCCESS in 1h 14m 33s\n- neutron-grenade http://logs.openstack.org/73/666873/1/gate/neutron-grenade/dfb572c/ : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73da9e00c177ee2474a743122c3c052dcc4512ed","revisions":{"73da9e00c177ee2474a743122c3c052dcc4512ed":{"kind":"REWORK","_number":1,"created":"2019-06-21 15:40:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/666873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/666873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/666873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/666873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/666873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/666873/1"}}},"commit":{"parents":[{"commit":"a0f0b32b8ab4a34e05c4a48e635405116fc19222","subject":"Merge \"update constraint for python-octaviaclient to new release 1.4.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0f0b32b8ab4a34e05c4a48e635405116fc19222"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 15:40:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 15:40:35.000000000","tz":0},"subject":"update constraint for os-apply-config to new release 8.3.2","message":"update constraint for os-apply-config to new release 8.3.2\n\nChange-Id: I62c766233518b15d34962c32dbf1a86856524bbd\nmeta:version: 8.3.2\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: Ib9bfed2ec66d97fe56ed045ad702df06b79996c8\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73da9e00c177ee2474a743122c3c052dcc4512ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/73da9e00c177ee2474a743122c3c052dcc4512ed"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
