)]}'
{"id":"openstack%2Frequirements~666875","triplet_id":"openstack%2Frequirements~stable%2Fqueens~Ibae52c62b83d999f3cd46e43304d246d3c4674e5","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"Ibae52c62b83d999f3cd46e43304d246d3c4674e5","subject":"update constraint for paunch to new release 2.5.2","status":"MERGED","created":"2019-06-21 15:41:21.000000000","updated":"2019-06-22 10:27:04.000000000","submitted":"2019-06-22 10:27:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666875-1561199224066-bfd37fb7","meta_rev_id":"9568ab061117b4f02ab272077cd6e1a88a0101ab","_number":666875,"virtual_id_number":666875,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-22 10:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-21 21:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-22 06:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-22 06:07:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-22 10:27:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"516a38386da0c24c007e73ad098e8991cd8b3923","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-21 15:41:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ab277d78e87944f6ed9a630fe4793e62d8bd500","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 21:04:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/75/666875/1/check/openstack-tox-pypy/f454020/ : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/75/666875/1/check/openstack-tox-pep8/3b6edec/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/75/666875/1/check/openstack-tox-py27/c624498/ : SUCCESS in 2m 53s\n- openstack-tox-py35 http://logs.openstack.org/75/666875/1/check/openstack-tox-py35/7557c47/ : SUCCESS in 3m 24s\n- openstack-tox-validate http://logs.openstack.org/75/666875/1/check/openstack-tox-validate/222f566/ : SUCCESS in 2m 52s\n- requirements-tox-babel http://logs.openstack.org/75/666875/1/check/requirements-tox-babel/aad72b0/ : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/75/666875/1/check/requirements-tox-py27-check-uc/1d8aa09/ : SUCCESS in 5m 37s\n- requirements-tox-py35-check-uc http://logs.openstack.org/75/666875/1/check/requirements-tox-py35-check-uc/20eaf23/ : SUCCESS in 6m 16s\n- cross-cinder-py27 http://logs.openstack.org/75/666875/1/check/cross-cinder-py27/081d4cb/ : SUCCESS in 9m 49s\n- cross-cinder-py35 http://logs.openstack.org/75/666875/1/check/cross-cinder-py35/fa5b136/ : SUCCESS in 8m 34s\n- cross-glance-py27 http://logs.openstack.org/75/666875/1/check/cross-glance-py27/92ac042/ : SUCCESS in 5m 23s\n- cross-glance-py35 http://logs.openstack.org/75/666875/1/check/cross-glance-py35/6de5587/ : SUCCESS in 5m 11s\n- cross-horizon-py27 http://logs.openstack.org/75/666875/1/check/cross-horizon-py27/f1b8af5/ : SUCCESS in 12m 40s\n- cross-horizon-py35 http://logs.openstack.org/75/666875/1/check/cross-horizon-py35/29831e7/ : SUCCESS in 10m 41s\n- cross-keystone-py27 http://logs.openstack.org/75/666875/1/check/cross-keystone-py27/8a2215b/ : SUCCESS in 30m 28s\n- cross-keystone-py35 http://logs.openstack.org/75/666875/1/check/cross-keystone-py35/2876be0/ : SUCCESS in 17m 29s\n- cross-neutron-py27 http://logs.openstack.org/75/666875/1/check/cross-neutron-py27/a587c7f/ : SUCCESS in 24m 13s\n- cross-neutron-py35 http://logs.openstack.org/75/666875/1/check/cross-neutron-py35/5879552/ : SUCCESS in 27m 27s\n- cross-nova-py27 http://logs.openstack.org/75/666875/1/check/cross-nova-py27/1f164c4/ : SUCCESS in 12m 47s\n- cross-nova-py35 http://logs.openstack.org/75/666875/1/check/cross-nova-py35/e96282d/ : SUCCESS in 17m 52s\n- cross-nova-functional http://logs.openstack.org/75/666875/1/check/cross-nova-functional/04259bd/ : SUCCESS in 28m 06s\n- cross-swift-py27 http://logs.openstack.org/75/666875/1/check/cross-swift-py27/9fbb23f/ : SUCCESS in 5m 55s\n- cross-swift-py35 http://logs.openstack.org/75/666875/1/check/cross-swift-py35/ce1ad71/ : SUCCESS in 3m 41s\n- requirements-integration http://logs.openstack.org/75/666875/1/check/requirements-integration/1658e1e/ : SUCCESS in 55m 11s\n- neutron-grenade-multinode http://logs.openstack.org/75/666875/1/check/neutron-grenade-multinode/b2d3d5c/ : SUCCESS in 59m 36s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/75/666875/1/check/legacy-tempest-dsvm-neutron-full/ffdf61f/ : SUCCESS in 1h 21m 28s\n- neutron-grenade http://logs.openstack.org/75/666875/1/check/neutron-grenade/29757b7/ : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":1},{"id":"107e32baba722b1e5c1c4aef8c89289b2d60bdea","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-21 21:12:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"49977cc3a24a1052a1e367902825ebdb6226fb04","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 06:07:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1a48a91b981d2d4c5947d129ea4eb3ec638a0f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 06:07:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c43425718e05d25772d2989e0f22b1f7e95a2237","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 10:27:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/666875/1/gate/openstack-tox-pep8/e3ceff3/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/75/666875/1/gate/openstack-tox-py27/4241d86/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/75/666875/1/gate/openstack-tox-py35/aba5104/ : SUCCESS in 3m 33s\n- openstack-tox-validate http://logs.openstack.org/75/666875/1/gate/openstack-tox-validate/e768e1a/ : SUCCESS in 3m 34s\n- requirements-tox-babel http://logs.openstack.org/75/666875/1/gate/requirements-tox-babel/28dd42d/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/75/666875/1/gate/requirements-tox-py27-check-uc/6e0d86b/ : SUCCESS in 5m 42s\n- requirements-tox-py35-check-uc http://logs.openstack.org/75/666875/1/gate/requirements-tox-py35-check-uc/3df236a/ : SUCCESS in 5m 57s\n- cross-cinder-py27 http://logs.openstack.org/75/666875/1/gate/cross-cinder-py27/4f0ecdc/ : SUCCESS in 8m 51s\n- cross-cinder-py35 http://logs.openstack.org/75/666875/1/gate/cross-cinder-py35/b6c895d/ : SUCCESS in 8m 08s\n- cross-glance-py27 http://logs.openstack.org/75/666875/1/gate/cross-glance-py27/0ee9356/ : SUCCESS in 5m 28s\n- cross-glance-py35 http://logs.openstack.org/75/666875/1/gate/cross-glance-py35/c56732a/ : SUCCESS in 4m 54s\n- cross-horizon-py27 http://logs.openstack.org/75/666875/1/gate/cross-horizon-py27/d54b962/ : SUCCESS in 13m 33s\n- cross-horizon-py35 http://logs.openstack.org/75/666875/1/gate/cross-horizon-py35/0ca58c8/ : SUCCESS in 13m 10s\n- cross-keystone-py27 http://logs.openstack.org/75/666875/1/gate/cross-keystone-py27/94c071a/ : SUCCESS in 12m 11s\n- cross-keystone-py35 http://logs.openstack.org/75/666875/1/gate/cross-keystone-py35/10a0d2d/ : SUCCESS in 18m 11s\n- cross-neutron-py27 http://logs.openstack.org/75/666875/1/gate/cross-neutron-py27/6fc200b/ : SUCCESS in 28m 01s\n- cross-neutron-py35 http://logs.openstack.org/75/666875/1/gate/cross-neutron-py35/daf3fa4/ : SUCCESS in 29m 36s\n- cross-nova-py27 http://logs.openstack.org/75/666875/1/gate/cross-nova-py27/ed80cd1/ : SUCCESS in 17m 15s\n- cross-nova-py35 http://logs.openstack.org/75/666875/1/gate/cross-nova-py35/8efec58/ : SUCCESS in 18m 01s\n- cross-nova-functional http://logs.openstack.org/75/666875/1/gate/cross-nova-functional/c2856a0/ : SUCCESS in 21m 48s\n- cross-swift-py27 http://logs.openstack.org/75/666875/1/gate/cross-swift-py27/866435f/ : SUCCESS in 5m 59s\n- cross-swift-py35 http://logs.openstack.org/75/666875/1/gate/cross-swift-py35/d260b1d/ : SUCCESS in 4m 30s\n- requirements-integration http://logs.openstack.org/75/666875/1/gate/requirements-integration/abb01d1/ : SUCCESS in 49m 42s\n- neutron-grenade-multinode http://logs.openstack.org/75/666875/1/gate/neutron-grenade-multinode/7af9f6f/ : SUCCESS in 1h 01m 37s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/75/666875/1/gate/legacy-tempest-dsvm-neutron-full/a93a3bd/ : SUCCESS in 1h 29m 30s\n- neutron-grenade http://logs.openstack.org/75/666875/1/gate/neutron-grenade/4bcdab3/ : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4269ad53283cae744c35a5075c9c778f9a309562","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 10:27:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd53e35546650b118f9c59735a720687f2d0c068","revisions":{"dd53e35546650b118f9c59735a720687f2d0c068":{"kind":"REWORK","_number":1,"created":"2019-06-21 15:41:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/666875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/666875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/666875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/666875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/666875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/666875/1"}}},"commit":{"parents":[{"commit":"a0f0b32b8ab4a34e05c4a48e635405116fc19222","subject":"Merge \"update constraint for python-octaviaclient to new release 1.4.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0f0b32b8ab4a34e05c4a48e635405116fc19222"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 15:41:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 15:41:19.000000000","tz":0},"subject":"update constraint for paunch to new release 2.5.2","message":"update constraint for paunch to new release 2.5.2\n\nChange-Id: Ibae52c62b83d999f3cd46e43304d246d3c4674e5\nmeta:version: 2.5.2\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: Ib9bfed2ec66d97fe56ed045ad702df06b79996c8\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd53e35546650b118f9c59735a720687f2d0c068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd53e35546650b118f9c59735a720687f2d0c068"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
