)]}'
{"id":"openstack%2Frequirements~666877","triplet_id":"openstack%2Frequirements~stable%2Fqueens~Ibcf26d8a99ba65c99acf4de78b9daef0522742be","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"Ibcf26d8a99ba65c99acf4de78b9daef0522742be","subject":"update constraint for os-net-config to new release 8.5.0","status":"MERGED","created":"2019-06-21 15:42:47.000000000","updated":"2019-06-23 01:28:45.000000000","submitted":"2019-06-23 01:28:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666877-1561253325451-1a2baadc","meta_rev_id":"c2c25ed15090fa987e0e86444a6839013d7b9cea","_number":666877,"virtual_id_number":666877,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-23 01:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-22 01:01:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-22 06:07:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-22 06:07:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-23 01:28:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6803f9da59b9d19dc860f5671b47e8b1101ffc42","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-21 15:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df45f5403b0ff3471904469a4988d14db4cf5eb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 21:05:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/77/666877/1/check/openstack-tox-pypy/ebbea16/ : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/77/666877/1/check/openstack-tox-pep8/c750003/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/77/666877/1/check/openstack-tox-py27/563844b/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/77/666877/1/check/openstack-tox-py35/03281a5/ : SUCCESS in 3m 58s\n- openstack-tox-validate http://logs.openstack.org/77/666877/1/check/openstack-tox-validate/9ed28ba/ : SUCCESS in 3m 00s\n- requirements-tox-babel http://logs.openstack.org/77/666877/1/check/requirements-tox-babel/14db710/ : SUCCESS in 4m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py27-check-uc/a9d8533/ : SUCCESS in 7m 42s\n- requirements-tox-py35-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py35-check-uc/451b935/ : SUCCESS in 8m 54s\n- cross-cinder-py27 http://logs.openstack.org/77/666877/1/check/cross-cinder-py27/1ca88a2/ : SUCCESS in 8m 10s\n- cross-cinder-py35 http://logs.openstack.org/77/666877/1/check/cross-cinder-py35/ac55766/ : SUCCESS in 11m 38s\n- cross-glance-py27 http://logs.openstack.org/77/666877/1/check/cross-glance-py27/0cb57ba/ : SUCCESS in 4m 30s\n- cross-glance-py35 http://logs.openstack.org/77/666877/1/check/cross-glance-py35/3c1940c/ : SUCCESS in 4m 21s\n- cross-horizon-py27 http://logs.openstack.org/77/666877/1/check/cross-horizon-py27/3460c9c/ : SUCCESS in 12m 01s\n- cross-horizon-py35 http://logs.openstack.org/77/666877/1/check/cross-horizon-py35/1c51dfc/ : SUCCESS in 11m 50s\n- cross-keystone-py27 http://logs.openstack.org/77/666877/1/check/cross-keystone-py27/964b840/ : SUCCESS in 13m 54s\n- cross-keystone-py35 http://logs.openstack.org/77/666877/1/check/cross-keystone-py35/9c4d696/ : SUCCESS in 18m 20s\n- cross-neutron-py27 http://logs.openstack.org/77/666877/1/check/cross-neutron-py27/1bf066d/ : SUCCESS in 28m 00s\n- cross-neutron-py35 http://logs.openstack.org/77/666877/1/check/cross-neutron-py35/193cd88/ : SUCCESS in 28m 15s\n- cross-nova-py27 http://logs.openstack.org/77/666877/1/check/cross-nova-py27/c91358b/ : SUCCESS in 13m 24s\n- cross-nova-py35 http://logs.openstack.org/77/666877/1/check/cross-nova-py35/18c936f/ : SUCCESS in 14m 24s\n- cross-nova-functional http://logs.openstack.org/77/666877/1/check/cross-nova-functional/e581d0f/ : SUCCESS in 20m 13s\n- cross-swift-py27 http://logs.openstack.org/77/666877/1/check/cross-swift-py27/f5b7c02/ : FAILURE in 8m 38s\n- cross-swift-py35 http://logs.openstack.org/77/666877/1/check/cross-swift-py35/cc0f96e/ : SUCCESS in 4m 10s\n- requirements-integration http://logs.openstack.org/77/666877/1/check/requirements-integration/3bc10c1/ : SUCCESS in 44m 32s\n- neutron-grenade-multinode http://logs.openstack.org/77/666877/1/check/neutron-grenade-multinode/e3264bc/ : SUCCESS in 1h 05m 05s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/77/666877/1/check/legacy-tempest-dsvm-neutron-full/df4a19c/ : SUCCESS in 1h 21m 32s\n- neutron-grenade http://logs.openstack.org/77/666877/1/check/neutron-grenade/ea742cf/ : SUCCESS in 49m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b28b3f1e907721502283ca9c92f3fed42ea59553","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-21 21:13:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ce9dd522f4274142c70fab1fc020b47fce10bbe6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-21 23:13:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"61af77b9e8ab74b88814733065ea4ceeb8a2c06a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 00:57:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/77/666877/1/check/openstack-tox-pypy/c0f4065/ : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/77/666877/1/check/openstack-tox-pep8/791801e/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/77/666877/1/check/openstack-tox-py27/ef5d38d/ : SUCCESS in 4m 31s\n- openstack-tox-py35 http://logs.openstack.org/77/666877/1/check/openstack-tox-py35/7cac761/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/77/666877/1/check/openstack-tox-validate/b63a0fb/ : SUCCESS in 3m 30s\n- requirements-tox-babel http://logs.openstack.org/77/666877/1/check/requirements-tox-babel/e3b335f/ : SUCCESS in 3m 01s\n- requirements-tox-py27-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py27-check-uc/ef94b88/ : SUCCESS in 6m 00s\n- requirements-tox-py35-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py35-check-uc/625c95c/ : SUCCESS in 6m 05s\n- cross-cinder-py27 http://logs.openstack.org/77/666877/1/check/cross-cinder-py27/adea032/ : SUCCESS in 9m 19s\n- cross-cinder-py35 http://logs.openstack.org/77/666877/1/check/cross-cinder-py35/9633855/ : SUCCESS in 8m 38s\n- cross-glance-py27 http://logs.openstack.org/77/666877/1/check/cross-glance-py27/0d775a9/ : SUCCESS in 5m 04s\n- cross-glance-py35 http://logs.openstack.org/77/666877/1/check/cross-glance-py35/e82df71/ : SUCCESS in 6m 14s\n- cross-horizon-py27 http://logs.openstack.org/77/666877/1/check/cross-horizon-py27/bae686e/ : SUCCESS in 12m 46s\n- cross-horizon-py35 http://logs.openstack.org/77/666877/1/check/cross-horizon-py35/605187e/ : SUCCESS in 10m 48s\n- cross-keystone-py27 http://logs.openstack.org/77/666877/1/check/cross-keystone-py27/62135a0/ : SUCCESS in 12m 34s\n- cross-keystone-py35 http://logs.openstack.org/77/666877/1/check/cross-keystone-py35/9653d6e/ : SUCCESS in 11m 50s\n- cross-neutron-py27 http://logs.openstack.org/77/666877/1/check/cross-neutron-py27/15cd2b5/ : SUCCESS in 24m 31s\n- cross-neutron-py35 http://logs.openstack.org/77/666877/1/check/cross-neutron-py35/abf5a26/ : SUCCESS in 27m 35s\n- cross-nova-py27 http://logs.openstack.org/77/666877/1/check/cross-nova-py27/cd9c042/ : SUCCESS in 15m 02s\n- cross-nova-py35 http://logs.openstack.org/77/666877/1/check/cross-nova-py35/605d36d/ : SUCCESS in 15m 00s\n- cross-nova-functional http://logs.openstack.org/77/666877/1/check/cross-nova-functional/f3054e3/ : SUCCESS in 21m 35s\n- cross-swift-py27 http://logs.openstack.org/77/666877/1/check/cross-swift-py27/8003e51/ : SUCCESS in 21m 16s\n- cross-swift-py35 http://logs.openstack.org/77/666877/1/check/cross-swift-py35/f970e51/ : SUCCESS in 4m 07s\n- requirements-integration http://logs.openstack.org/77/666877/1/check/requirements-integration/0e2160f/ : SUCCESS in 51m 26s\n- neutron-grenade-multinode http://logs.openstack.org/77/666877/1/check/neutron-grenade-multinode/aa043e8/ : SUCCESS in 1h 03m 06s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/77/666877/1/check/legacy-tempest-dsvm-neutron-full/b19ddb4/ : SUCCESS in 1h 31m 23s\n- neutron-grenade http://logs.openstack.org/77/666877/1/check/neutron-grenade/926641a/ : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6e8fd7017db0c25f7040f1ccbfbd82e671db2672","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 01:01:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2a16316cc73eff41a062a4eaae4c2e0f17cd2f1a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 06:07:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"710cc514f900ae072214947bff410c09ff492dd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 06:08:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8932c3c8b553860010b62ecca527b2a9fb59d227","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 10:29:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/77/666877/1/gate/openstack-tox-pep8/029d4be/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/77/666877/1/gate/openstack-tox-py27/d146a60/ : SUCCESS in 5m 48s\n- openstack-tox-py35 http://logs.openstack.org/77/666877/1/gate/openstack-tox-py35/ed532ea/ : SUCCESS in 4m 54s\n- openstack-tox-validate http://logs.openstack.org/77/666877/1/gate/openstack-tox-validate/25f6948/ : SUCCESS in 4m 30s\n- requirements-tox-babel http://logs.openstack.org/77/666877/1/gate/requirements-tox-babel/36f9588/ : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc http://logs.openstack.org/77/666877/1/gate/requirements-tox-py27-check-uc/bc5e6c7/ : SUCCESS in 6m 19s\n- requirements-tox-py35-check-uc http://logs.openstack.org/77/666877/1/gate/requirements-tox-py35-check-uc/f16b4ce/ : SUCCESS in 7m 00s\n- cross-cinder-py27 http://logs.openstack.org/77/666877/1/gate/cross-cinder-py27/da09b69/ : SUCCESS in 8m 44s\n- cross-cinder-py35 http://logs.openstack.org/77/666877/1/gate/cross-cinder-py35/5654f9a/ : SUCCESS in 11m 25s\n- cross-glance-py27 http://logs.openstack.org/77/666877/1/gate/cross-glance-py27/cb47e4a/ : SUCCESS in 6m 26s\n- cross-glance-py35 http://logs.openstack.org/77/666877/1/gate/cross-glance-py35/5064f40/ : SUCCESS in 5m 27s\n- cross-horizon-py27 http://logs.openstack.org/77/666877/1/gate/cross-horizon-py27/b649d32/ : SUCCESS in 11m 03s\n- cross-horizon-py35 http://logs.openstack.org/77/666877/1/gate/cross-horizon-py35/a7945ee/ : SUCCESS in 12m 32s\n- cross-keystone-py27 http://logs.openstack.org/77/666877/1/gate/cross-keystone-py27/c17575f/ : SUCCESS in 13m 44s\n- cross-keystone-py35 http://logs.openstack.org/77/666877/1/gate/cross-keystone-py35/db6db5c/ : SUCCESS in 13m 50s\n- cross-neutron-py27 http://logs.openstack.org/77/666877/1/gate/cross-neutron-py27/a4b7eb5/ : SUCCESS in 22m 04s\n- cross-neutron-py35 http://logs.openstack.org/77/666877/1/gate/cross-neutron-py35/25869de/ : SUCCESS in 28m 07s\n- cross-nova-py27 http://logs.openstack.org/77/666877/1/gate/cross-nova-py27/8c6e182/ : SUCCESS in 14m 29s\n- cross-nova-py35 http://logs.openstack.org/77/666877/1/gate/cross-nova-py35/d318ae4/ : SUCCESS in 17m 19s\n- cross-nova-functional http://logs.openstack.org/77/666877/1/gate/cross-nova-functional/44da3be/ : SUCCESS in 17m 19s\n- cross-swift-py27 http://logs.openstack.org/77/666877/1/gate/cross-swift-py27/176203b/ : SUCCESS in 7m 36s\n- cross-swift-py35 http://logs.openstack.org/77/666877/1/gate/cross-swift-py35/7a1cc45/ : RETRY_LIMIT in 1m 34s\n- requirements-integration http://logs.openstack.org/77/666877/1/gate/requirements-integration/7ae173e/ : SUCCESS in 48m 21s\n- neutron-grenade-multinode http://logs.openstack.org/77/666877/1/gate/neutron-grenade-multinode/54e8e2b/ : FAILURE in 48m 08s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/77/666877/1/gate/legacy-tempest-dsvm-neutron-full/14c7297/ : SUCCESS in 1h 31m 53s\n- neutron-grenade http://logs.openstack.org/77/666877/1/gate/neutron-grenade/7830c6e/ : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":1},{"id":"35ea49ac96640cb55cd581ebf1b19c515414f61d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 17:14:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2a9436a3434fa8023a507b4caaeb0713f2cf9d1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 18:40:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/77/666877/1/check/openstack-tox-pypy/5933225/ : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/77/666877/1/check/openstack-tox-pep8/a5988d3/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/77/666877/1/check/openstack-tox-py27/984a789/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/77/666877/1/check/openstack-tox-py35/f503c17/ : SUCCESS in 4m 36s\n- openstack-tox-validate http://logs.openstack.org/77/666877/1/check/openstack-tox-validate/794aaad/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/77/666877/1/check/requirements-tox-babel/498cd19/ : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py27-check-uc/d6bed3c/ : SUCCESS in 6m 16s\n- requirements-tox-py35-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py35-check-uc/5b6c0db/ : SUCCESS in 6m 32s\n- cross-cinder-py27 http://logs.openstack.org/77/666877/1/check/cross-cinder-py27/8403952/ : SUCCESS in 8m 55s\n- cross-cinder-py35 http://logs.openstack.org/77/666877/1/check/cross-cinder-py35/a3dc9be/ : SUCCESS in 11m 43s\n- cross-glance-py27 http://logs.openstack.org/77/666877/1/check/cross-glance-py27/373bd04/ : SUCCESS in 4m 56s\n- cross-glance-py35 http://logs.openstack.org/77/666877/1/check/cross-glance-py35/f34f546/ : SUCCESS in 5m 34s\n- cross-horizon-py27 http://logs.openstack.org/77/666877/1/check/cross-horizon-py27/dcf2ae9/ : SUCCESS in 10m 28s\n- cross-horizon-py35 http://logs.openstack.org/77/666877/1/check/cross-horizon-py35/220f53c/ : SUCCESS in 9m 35s\n- cross-keystone-py27 http://logs.openstack.org/77/666877/1/check/cross-keystone-py27/77c37d2/ : SUCCESS in 15m 28s\n- cross-keystone-py35 http://logs.openstack.org/77/666877/1/check/cross-keystone-py35/0b8f269/ : SUCCESS in 23m 58s\n- cross-neutron-py27 http://logs.openstack.org/77/666877/1/check/cross-neutron-py27/eeab1ff/ : SUCCESS in 19m 21s\n- cross-neutron-py35 http://logs.openstack.org/77/666877/1/check/cross-neutron-py35/65cb531/ : SUCCESS in 38m 57s\n- cross-nova-py27 http://logs.openstack.org/77/666877/1/check/cross-nova-py27/7296912/ : SUCCESS in 13m 48s\n- cross-nova-py35 http://logs.openstack.org/77/666877/1/check/cross-nova-py35/b641aa9/ : SUCCESS in 17m 11s\n- cross-nova-functional http://logs.openstack.org/77/666877/1/check/cross-nova-functional/f177340/ : SUCCESS in 21m 35s\n- cross-swift-py27 http://logs.openstack.org/77/666877/1/check/cross-swift-py27/84c94c1/ : SUCCESS in 6m 25s\n- cross-swift-py35 http://logs.openstack.org/77/666877/1/check/cross-swift-py35/8f31afe/ : SUCCESS in 3m 38s\n- requirements-integration http://logs.openstack.org/77/666877/1/check/requirements-integration/56cda59/ : SUCCESS in 54m 36s\n- neutron-grenade-multinode http://logs.openstack.org/77/666877/1/check/neutron-grenade-multinode/2f39177/ : SUCCESS in 1h 05m 36s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/77/666877/1/check/legacy-tempest-dsvm-neutron-full/fb61943/ : SUCCESS in 1h 19m 38s\n- neutron-grenade http://logs.openstack.org/77/666877/1/check/neutron-grenade/3e409c8/ : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"30796fc74a93dcd9b15a007656453b733e4ff0aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 18:40:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b8179db3869af2362f25ba814018bcb793f3c32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 20:07:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/77/666877/1/gate/openstack-tox-pep8/ef90958/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/77/666877/1/gate/openstack-tox-py27/087c4da/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/77/666877/1/gate/openstack-tox-py35/a869b49/ : SUCCESS in 4m 45s\n- openstack-tox-validate http://logs.openstack.org/77/666877/1/gate/openstack-tox-validate/18461ca/ : SUCCESS in 4m 26s\n- requirements-tox-babel http://logs.openstack.org/77/666877/1/gate/requirements-tox-babel/59cb888/ : SUCCESS in 4m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/77/666877/1/gate/requirements-tox-py27-check-uc/e302595/ : SUCCESS in 7m 07s\n- requirements-tox-py35-check-uc http://logs.openstack.org/77/666877/1/gate/requirements-tox-py35-check-uc/1afce3e/ : SUCCESS in 5m 52s\n- cross-cinder-py27 http://logs.openstack.org/77/666877/1/gate/cross-cinder-py27/c5df771/ : SUCCESS in 11m 45s\n- cross-cinder-py35 http://logs.openstack.org/77/666877/1/gate/cross-cinder-py35/c46d0f5/ : SUCCESS in 10m 26s\n- cross-glance-py27 http://logs.openstack.org/77/666877/1/gate/cross-glance-py27/a5627b0/ : SUCCESS in 5m 24s\n- cross-glance-py35 http://logs.openstack.org/77/666877/1/gate/cross-glance-py35/3294a1f/ : SUCCESS in 6m 01s\n- cross-horizon-py27 http://logs.openstack.org/77/666877/1/gate/cross-horizon-py27/d7a5ee9/ : SUCCESS in 9m 57s\n- cross-horizon-py35 http://logs.openstack.org/77/666877/1/gate/cross-horizon-py35/add64fe/ : SUCCESS in 11m 36s\n- cross-keystone-py27 http://logs.openstack.org/77/666877/1/gate/cross-keystone-py27/efd33dd/ : RETRY_LIMIT in 2m 06s\n- cross-keystone-py35 http://logs.openstack.org/77/666877/1/gate/cross-keystone-py35/3555dd7/ : SUCCESS in 19m 22s\n- cross-neutron-py27 http://logs.openstack.org/77/666877/1/gate/cross-neutron-py27/f51e4ce/ : SUCCESS in 25m 57s\n- cross-neutron-py35 http://logs.openstack.org/77/666877/1/gate/cross-neutron-py35/e6d65b1/ : SUCCESS in 23m 12s\n- cross-nova-py27 http://logs.openstack.org/77/666877/1/gate/cross-nova-py27/88ef297/ : SUCCESS in 16m 04s\n- cross-nova-py35 http://logs.openstack.org/77/666877/1/gate/cross-nova-py35/81c8f95/ : SUCCESS in 15m 00s\n- cross-nova-functional http://logs.openstack.org/77/666877/1/gate/cross-nova-functional/9d4d394/ : SUCCESS in 17m 38s\n- cross-swift-py27 http://logs.openstack.org/77/666877/1/gate/cross-swift-py27/f8c72e6/ : SUCCESS in 8m 37s\n- cross-swift-py35 http://logs.openstack.org/77/666877/1/gate/cross-swift-py35/c6827a8/ : SUCCESS in 3m 43s\n- requirements-integration http://logs.openstack.org/77/666877/1/gate/requirements-integration/f2ce097/ : SUCCESS in 50m 09s\n- neutron-grenade-multinode http://logs.openstack.org/77/666877/1/gate/neutron-grenade-multinode/af99c70/ : SUCCESS in 1h 02m 02s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/77/666877/1/gate/legacy-tempest-dsvm-neutron-full/439021e/ : SUCCESS in 1h 17m 40s\n- neutron-grenade http://logs.openstack.org/77/666877/1/gate/neutron-grenade/d489454/ : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":1},{"id":"45dfa5e5bf366472c6b13807416b1659beed9bf2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 20:34:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8115604ee1bd689a51ccf3c196af330adadefdae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 22:06:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/77/666877/1/check/openstack-tox-pypy/a8fafb2/ : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/77/666877/1/check/openstack-tox-pep8/e8fd8a6/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/77/666877/1/check/openstack-tox-py27/0bbec9c/ : SUCCESS in 6m 51s\n- openstack-tox-py35 http://logs.openstack.org/77/666877/1/check/openstack-tox-py35/08ae896/ : SUCCESS in 6m 35s\n- openstack-tox-validate http://logs.openstack.org/77/666877/1/check/openstack-tox-validate/78648a2/ : SUCCESS in 3m 21s\n- requirements-tox-babel http://logs.openstack.org/77/666877/1/check/requirements-tox-babel/8f5bca4/ : SUCCESS in 4m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py27-check-uc/259357f/ : SUCCESS in 8m 56s\n- requirements-tox-py35-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py35-check-uc/1851d85/ : SUCCESS in 7m 00s\n- cross-cinder-py27 http://logs.openstack.org/77/666877/1/check/cross-cinder-py27/d07d5a2/ : SUCCESS in 11m 21s\n- cross-cinder-py35 http://logs.openstack.org/77/666877/1/check/cross-cinder-py35/b0994ac/ : SUCCESS in 11m 54s\n- cross-glance-py27 http://logs.openstack.org/77/666877/1/check/cross-glance-py27/2eb72f8/ : SUCCESS in 4m 56s\n- cross-glance-py35 http://logs.openstack.org/77/666877/1/check/cross-glance-py35/e241ead/ : SUCCESS in 7m 17s\n- cross-horizon-py27 http://logs.openstack.org/77/666877/1/check/cross-horizon-py27/81e3706/ : SUCCESS in 12m 15s\n- cross-horizon-py35 http://logs.openstack.org/77/666877/1/check/cross-horizon-py35/e07506f/ : SUCCESS in 10m 40s\n- cross-keystone-py27 http://logs.openstack.org/77/666877/1/check/cross-keystone-py27/392ef00/ : SUCCESS in 18m 01s\n- cross-keystone-py35 http://logs.openstack.org/77/666877/1/check/cross-keystone-py35/094a2aa/ : SUCCESS in 12m 04s\n- cross-neutron-py27 http://logs.openstack.org/77/666877/1/check/cross-neutron-py27/1d4cd38/ : SUCCESS in 32m 03s\n- cross-neutron-py35 http://logs.openstack.org/77/666877/1/check/cross-neutron-py35/ce5d634/ : SUCCESS in 20m 52s\n- cross-nova-py27 http://logs.openstack.org/77/666877/1/check/cross-nova-py27/41386c8/ : SUCCESS in 14m 35s\n- cross-nova-py35 http://logs.openstack.org/77/666877/1/check/cross-nova-py35/ba18f35/ : SUCCESS in 15m 58s\n- cross-nova-functional http://logs.openstack.org/77/666877/1/check/cross-nova-functional/2e1b958/ : SUCCESS in 20m 32s\n- cross-swift-py27 http://logs.openstack.org/77/666877/1/check/cross-swift-py27/59e1520/ : SUCCESS in 6m 36s\n- cross-swift-py35 http://logs.openstack.org/77/666877/1/check/cross-swift-py35/4d61bd0/ : SUCCESS in 4m 45s\n- requirements-integration http://logs.openstack.org/77/666877/1/check/requirements-integration/c00428a/ : SUCCESS in 54m 21s\n- neutron-grenade-multinode http://logs.openstack.org/77/666877/1/check/neutron-grenade-multinode/9d63bb0/ : FAILURE in 59m 01s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/77/666877/1/check/legacy-tempest-dsvm-neutron-full/9114591/ : SUCCESS in 1h 23m 38s\n- neutron-grenade http://logs.openstack.org/77/666877/1/check/neutron-grenade/121a5d5/ : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fe5a862820543886281e940661f673d4b2e47ca4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 22:26:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"96ea5e38c0c46f1974b2416e5c6af46733b884e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 00:03:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/77/666877/1/check/openstack-tox-pypy/3b15de9/ : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/77/666877/1/check/openstack-tox-pep8/b4597ae/ : SUCCESS in 2m 59s\n- openstack-tox-py27 http://logs.openstack.org/77/666877/1/check/openstack-tox-py27/6352fb6/ : SUCCESS in 7m 19s\n- openstack-tox-py35 http://logs.openstack.org/77/666877/1/check/openstack-tox-py35/f53db37/ : SUCCESS in 3m 27s\n- openstack-tox-validate http://logs.openstack.org/77/666877/1/check/openstack-tox-validate/5a15de5/ : SUCCESS in 3m 17s\n- requirements-tox-babel http://logs.openstack.org/77/666877/1/check/requirements-tox-babel/1d01664/ : SUCCESS in 2m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py27-check-uc/376d2a3/ : SUCCESS in 5m 59s\n- requirements-tox-py35-check-uc http://logs.openstack.org/77/666877/1/check/requirements-tox-py35-check-uc/5699c96/ : SUCCESS in 6m 19s\n- cross-cinder-py27 http://logs.openstack.org/77/666877/1/check/cross-cinder-py27/911e40f/ : SUCCESS in 8m 37s\n- cross-cinder-py35 http://logs.openstack.org/77/666877/1/check/cross-cinder-py35/b0aeee1/ : SUCCESS in 9m 41s\n- cross-glance-py27 http://logs.openstack.org/77/666877/1/check/cross-glance-py27/bace634/ : SUCCESS in 6m 16s\n- cross-glance-py35 http://logs.openstack.org/77/666877/1/check/cross-glance-py35/64637fb/ : SUCCESS in 5m 02s\n- cross-horizon-py27 http://logs.openstack.org/77/666877/1/check/cross-horizon-py27/18df37e/ : SUCCESS in 11m 28s\n- cross-horizon-py35 http://logs.openstack.org/77/666877/1/check/cross-horizon-py35/f15faae/ : SUCCESS in 10m 44s\n- cross-keystone-py27 http://logs.openstack.org/77/666877/1/check/cross-keystone-py27/8936382/ : SUCCESS in 30m 14s\n- cross-keystone-py35 http://logs.openstack.org/77/666877/1/check/cross-keystone-py35/2256e0f/ : SUCCESS in 12m 11s\n- cross-neutron-py27 http://logs.openstack.org/77/666877/1/check/cross-neutron-py27/e25ba2f/ : SUCCESS in 24m 03s\n- cross-neutron-py35 http://logs.openstack.org/77/666877/1/check/cross-neutron-py35/47ec595/ : SUCCESS in 34m 53s\n- cross-nova-py27 http://logs.openstack.org/77/666877/1/check/cross-nova-py27/463ff6d/ : SUCCESS in 13m 49s\n- cross-nova-py35 http://logs.openstack.org/77/666877/1/check/cross-nova-py35/db48a8d/ : SUCCESS in 13m 54s\n- cross-nova-functional http://logs.openstack.org/77/666877/1/check/cross-nova-functional/3fce7f3/ : SUCCESS in 20m 56s\n- cross-swift-py27 http://logs.openstack.org/77/666877/1/check/cross-swift-py27/bd00838/ : SUCCESS in 5m 53s\n- cross-swift-py35 http://logs.openstack.org/77/666877/1/check/cross-swift-py35/4e765a3/ : SUCCESS in 4m 58s\n- requirements-integration http://logs.openstack.org/77/666877/1/check/requirements-integration/f5c94a9/ : SUCCESS in 51m 21s\n- neutron-grenade-multinode http://logs.openstack.org/77/666877/1/check/neutron-grenade-multinode/4a08241/ : SUCCESS in 1h 10m 57s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/77/666877/1/check/legacy-tempest-dsvm-neutron-full/4c4408b/ : SUCCESS in 1h 29m 29s\n- neutron-grenade http://logs.openstack.org/77/666877/1/check/neutron-grenade/d44bc2c/ : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4d65d9bd4480a27f5360ce2de9946331f03be92b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 00:03:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c39d4d2c23bde45d10379f3554939b995e092dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 01:28:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6276233cd8bde4dc132c479d80288abe3d9e52c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-23 01:28:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/77/666877/1/gate/openstack-tox-pep8/49d6a30/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/77/666877/1/gate/openstack-tox-py27/bba0871/ : SUCCESS in 3m 36s\n- openstack-tox-py35 http://logs.openstack.org/77/666877/1/gate/openstack-tox-py35/5c0f0e3/ : SUCCESS in 3m 41s\n- openstack-tox-validate http://logs.openstack.org/77/666877/1/gate/openstack-tox-validate/5f2ce6d/ : SUCCESS in 2m 45s\n- requirements-tox-babel http://logs.openstack.org/77/666877/1/gate/requirements-tox-babel/18b3836/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/77/666877/1/gate/requirements-tox-py27-check-uc/b5b334f/ : SUCCESS in 5m 49s\n- requirements-tox-py35-check-uc http://logs.openstack.org/77/666877/1/gate/requirements-tox-py35-check-uc/69cc9b4/ : SUCCESS in 6m 12s\n- cross-cinder-py27 http://logs.openstack.org/77/666877/1/gate/cross-cinder-py27/fa84888/ : SUCCESS in 9m 38s\n- cross-cinder-py35 http://logs.openstack.org/77/666877/1/gate/cross-cinder-py35/fa9fd2a/ : SUCCESS in 7m 56s\n- cross-glance-py27 http://logs.openstack.org/77/666877/1/gate/cross-glance-py27/57e2af4/ : SUCCESS in 4m 40s\n- cross-glance-py35 http://logs.openstack.org/77/666877/1/gate/cross-glance-py35/95a8333/ : SUCCESS in 4m 56s\n- cross-horizon-py27 http://logs.openstack.org/77/666877/1/gate/cross-horizon-py27/16539de/ : SUCCESS in 9m 16s\n- cross-horizon-py35 http://logs.openstack.org/77/666877/1/gate/cross-horizon-py35/d47b0c9/ : SUCCESS in 11m 01s\n- cross-keystone-py27 http://logs.openstack.org/77/666877/1/gate/cross-keystone-py27/a49ba4e/ : SUCCESS in 12m 53s\n- cross-keystone-py35 http://logs.openstack.org/77/666877/1/gate/cross-keystone-py35/8c43983/ : SUCCESS in 12m 09s\n- cross-neutron-py27 http://logs.openstack.org/77/666877/1/gate/cross-neutron-py27/95fbe08/ : SUCCESS in 23m 48s\n- cross-neutron-py35 http://logs.openstack.org/77/666877/1/gate/cross-neutron-py35/b49f917/ : SUCCESS in 23m 00s\n- cross-nova-py27 http://logs.openstack.org/77/666877/1/gate/cross-nova-py27/9aa6138/ : SUCCESS in 15m 26s\n- cross-nova-py35 http://logs.openstack.org/77/666877/1/gate/cross-nova-py35/069a95a/ : SUCCESS in 12m 26s\n- cross-nova-functional http://logs.openstack.org/77/666877/1/gate/cross-nova-functional/89da46f/ : SUCCESS in 18m 16s\n- cross-swift-py27 http://logs.openstack.org/77/666877/1/gate/cross-swift-py27/d154cdf/ : SUCCESS in 5m 56s\n- cross-swift-py35 http://logs.openstack.org/77/666877/1/gate/cross-swift-py35/b259073/ : SUCCESS in 3m 34s\n- requirements-integration http://logs.openstack.org/77/666877/1/gate/requirements-integration/853e653/ : SUCCESS in 50m 42s\n- neutron-grenade-multinode http://logs.openstack.org/77/666877/1/gate/neutron-grenade-multinode/7375d68/ : SUCCESS in 1h 01m 31s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/77/666877/1/gate/legacy-tempest-dsvm-neutron-full/ac75708/ : SUCCESS in 1h 20m 11s\n- neutron-grenade http://logs.openstack.org/77/666877/1/gate/neutron-grenade/33d4bd5/ : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6c33d6e4f213f891a36b5c8bd65f49d32af7934","revisions":{"b6c33d6e4f213f891a36b5c8bd65f49d32af7934":{"kind":"REWORK","_number":1,"created":"2019-06-21 15:42:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/666877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/666877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/666877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/666877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/666877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/666877/1"}}},"commit":{"parents":[{"commit":"a0f0b32b8ab4a34e05c4a48e635405116fc19222","subject":"Merge \"update constraint for python-octaviaclient to new release 1.4.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0f0b32b8ab4a34e05c4a48e635405116fc19222"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 15:42:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 15:42:45.000000000","tz":0},"subject":"update constraint for os-net-config to new release 8.5.0","message":"update constraint for os-net-config to new release 8.5.0\n\nChange-Id: Ibcf26d8a99ba65c99acf4de78b9daef0522742be\nmeta:version: 8.5.0\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: Ib9bfed2ec66d97fe56ed045ad702df06b79996c8\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6c33d6e4f213f891a36b5c8bd65f49d32af7934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6c33d6e4f213f891a36b5c8bd65f49d32af7934"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
