)]}'
{"id":"openstack%2Frequirements~666878","triplet_id":"openstack%2Frequirements~stable%2Fqueens~I2df12f5d24b6485f9542299d43e23b5a49f3fbb6","project":"openstack/requirements","branch":"stable/queens","topic":"new-release","hashtags":[],"change_id":"I2df12f5d24b6485f9542299d43e23b5a49f3fbb6","subject":"update constraint for tripleo-common to new release 8.7.0","status":"MERGED","created":"2019-06-21 15:44:27.000000000","updated":"2019-06-22 21:26:56.000000000","submitted":"2019-06-22 21:26:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666878-1561238816541-230b38e8","meta_rev_id":"8385800d1bfc9ee48c66c2ffe6a4566a1061f293","_number":666878,"virtual_id_number":666878,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-22 21:26:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-22 01:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-22 06:07:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-22 06:07:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-22 21:26:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee1bc31ecda5e16cfc8a70139e039acbe0fce4b5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-21 15:44:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32e9d552b82a0050eb7a5f13154e742be1a1c4cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 21:13:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/78/666878/1/check/openstack-tox-pypy/25e6334/ : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/666878/1/check/openstack-tox-pep8/78dcbf0/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/78/666878/1/check/openstack-tox-py27/c680477/ : SUCCESS in 3m 11s\n- openstack-tox-py35 http://logs.openstack.org/78/666878/1/check/openstack-tox-py35/a7341a2/ : SUCCESS in 3m 33s\n- openstack-tox-validate http://logs.openstack.org/78/666878/1/check/openstack-tox-validate/59b4d59/ : SUCCESS in 3m 16s\n- requirements-tox-babel http://logs.openstack.org/78/666878/1/check/requirements-tox-babel/c0bc9dd/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/666878/1/check/requirements-tox-py27-check-uc/d4a5209/ : SUCCESS in 5m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/78/666878/1/check/requirements-tox-py35-check-uc/f3c48ac/ : SUCCESS in 6m 48s\n- cross-cinder-py27 http://logs.openstack.org/78/666878/1/check/cross-cinder-py27/40d6796/ : SUCCESS in 9m 39s\n- cross-cinder-py35 http://logs.openstack.org/78/666878/1/check/cross-cinder-py35/0b296c1/ : SUCCESS in 8m 20s\n- cross-glance-py27 http://logs.openstack.org/78/666878/1/check/cross-glance-py27/15807fd/ : SUCCESS in 5m 02s\n- cross-glance-py35 http://logs.openstack.org/78/666878/1/check/cross-glance-py35/8856cbc/ : SUCCESS in 5m 12s\n- cross-horizon-py27 http://logs.openstack.org/78/666878/1/check/cross-horizon-py27/40949fe/ : SUCCESS in 12m 29s\n- cross-horizon-py35 http://logs.openstack.org/78/666878/1/check/cross-horizon-py35/6cde1cb/ : SUCCESS in 10m 25s\n- cross-keystone-py27 http://logs.openstack.org/78/666878/1/check/cross-keystone-py27/af39114/ : SUCCESS in 14m 03s\n- cross-keystone-py35 http://logs.openstack.org/78/666878/1/check/cross-keystone-py35/f35c2d4/ : SUCCESS in 13m 14s\n- cross-neutron-py27 http://logs.openstack.org/78/666878/1/check/cross-neutron-py27/2a46caa/ : SUCCESS in 27m 30s\n- cross-neutron-py35 http://logs.openstack.org/78/666878/1/check/cross-neutron-py35/2d9a3dd/ : SUCCESS in 29m 11s\n- cross-nova-py27 http://logs.openstack.org/78/666878/1/check/cross-nova-py27/b626ecc/ : SUCCESS in 12m 21s\n- cross-nova-py35 http://logs.openstack.org/78/666878/1/check/cross-nova-py35/7a88920/ : SUCCESS in 18m 22s\n- cross-nova-functional http://logs.openstack.org/78/666878/1/check/cross-nova-functional/ebea4b3/ : SUCCESS in 27m 19s\n- cross-swift-py27 http://logs.openstack.org/78/666878/1/check/cross-swift-py27/d3a71c5/ : SUCCESS in 7m 41s\n- cross-swift-py35 http://logs.openstack.org/78/666878/1/check/cross-swift-py35/8d828e9/ : SUCCESS in 3m 40s\n- requirements-integration http://logs.openstack.org/78/666878/1/check/requirements-integration/1da5170/ : SUCCESS in 49m 33s\n- neutron-grenade-multinode http://logs.openstack.org/78/666878/1/check/neutron-grenade-multinode/be5a5de/ : FAILURE in 7m 04s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/78/666878/1/check/legacy-tempest-dsvm-neutron-full/65f82db/ : SUCCESS in 1h 26m 32s\n- neutron-grenade http://logs.openstack.org/78/666878/1/check/neutron-grenade/5e0c18d/ : FAILURE in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"aba90af97447e8f98b9b123b2ca15e63e25584b6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-21 21:15:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bb11ab2e70ced29d393e54898fa5a5bf760551e6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-21 23:13:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"53d5f1bd898b0d1eb50fdf37981b11e01355ae7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 01:06:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/78/666878/1/check/openstack-tox-pypy/92ad8eb/ : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/666878/1/check/openstack-tox-pep8/e36ea0a/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/78/666878/1/check/openstack-tox-py27/c1dfe40/ : SUCCESS in 3m 19s\n- openstack-tox-py35 http://logs.openstack.org/78/666878/1/check/openstack-tox-py35/96166a9/ : SUCCESS in 4m 23s\n- openstack-tox-validate http://logs.openstack.org/78/666878/1/check/openstack-tox-validate/2e02dea/ : SUCCESS in 3m 26s\n- requirements-tox-babel http://logs.openstack.org/78/666878/1/check/requirements-tox-babel/0d5dada/ : SUCCESS in 3m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/666878/1/check/requirements-tox-py27-check-uc/692955a/ : SUCCESS in 7m 28s\n- requirements-tox-py35-check-uc http://logs.openstack.org/78/666878/1/check/requirements-tox-py35-check-uc/8e5a888/ : SUCCESS in 6m 53s\n- cross-cinder-py27 http://logs.openstack.org/78/666878/1/check/cross-cinder-py27/550bb1b/ : SUCCESS in 8m 21s\n- cross-cinder-py35 http://logs.openstack.org/78/666878/1/check/cross-cinder-py35/51ac659/ : SUCCESS in 9m 03s\n- cross-glance-py27 http://logs.openstack.org/78/666878/1/check/cross-glance-py27/b372362/ : SUCCESS in 11m 04s\n- cross-glance-py35 http://logs.openstack.org/78/666878/1/check/cross-glance-py35/bd75288/ : SUCCESS in 11m 08s\n- cross-horizon-py27 http://logs.openstack.org/78/666878/1/check/cross-horizon-py27/045e1d7/ : SUCCESS in 15m 07s\n- cross-horizon-py35 http://logs.openstack.org/78/666878/1/check/cross-horizon-py35/e3e697f/ : SUCCESS in 11m 30s\n- cross-keystone-py27 http://logs.openstack.org/78/666878/1/check/cross-keystone-py27/9151fbe/ : SUCCESS in 12m 41s\n- cross-keystone-py35 http://logs.openstack.org/78/666878/1/check/cross-keystone-py35/be66c2c/ : SUCCESS in 12m 41s\n- cross-neutron-py27 http://logs.openstack.org/78/666878/1/check/cross-neutron-py27/b1a5ff9/ : SUCCESS in 24m 08s\n- cross-neutron-py35 http://logs.openstack.org/78/666878/1/check/cross-neutron-py35/cd76370/ : SUCCESS in 25m 17s\n- cross-nova-py27 http://logs.openstack.org/78/666878/1/check/cross-nova-py27/996481c/ : SUCCESS in 12m 54s\n- cross-nova-py35 http://logs.openstack.org/78/666878/1/check/cross-nova-py35/d2dd99f/ : SUCCESS in 12m 44s\n- cross-nova-functional http://logs.openstack.org/78/666878/1/check/cross-nova-functional/41e01b4/ : SUCCESS in 18m 53s\n- cross-swift-py27 http://logs.openstack.org/78/666878/1/check/cross-swift-py27/0053764/ : SUCCESS in 6m 37s\n- cross-swift-py35 http://logs.openstack.org/78/666878/1/check/cross-swift-py35/b59f56f/ : SUCCESS in 3m 54s\n- requirements-integration http://logs.openstack.org/78/666878/1/check/requirements-integration/47f316e/ : SUCCESS in 51m 26s\n- neutron-grenade-multinode http://logs.openstack.org/78/666878/1/check/neutron-grenade-multinode/7c59de9/ : SUCCESS in 1h 03m 12s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/78/666878/1/check/legacy-tempest-dsvm-neutron-full/b6f0353/ : SUCCESS in 1h 24m 13s\n- neutron-grenade http://logs.openstack.org/78/666878/1/check/neutron-grenade/f4c2535/ : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":1},{"id":"39b90f454c810202a67722188cf36d81c84dc30a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 01:07:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4071cc835012f673af9cd11ce533ecbcb13c3db9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 06:07:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd5c0d60d4030d9724de4111cff707ed5d46e2e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 06:08:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74a4670aa9f1716f0bd4c1c5cc66a5bf9c541d8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 10:43:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/78/666878/1/gate/openstack-tox-pep8/ae9b8b7/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/78/666878/1/gate/openstack-tox-py27/a9a9505/ : SUCCESS in 3m 02s\n- openstack-tox-py35 http://logs.openstack.org/78/666878/1/gate/openstack-tox-py35/7c0cf44/ : SUCCESS in 3m 29s\n- openstack-tox-validate http://logs.openstack.org/78/666878/1/gate/openstack-tox-validate/90c70fc/ : SUCCESS in 3m 31s\n- requirements-tox-babel http://logs.openstack.org/78/666878/1/gate/requirements-tox-babel/94309ff/ : SUCCESS in 8m 31s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/666878/1/gate/requirements-tox-py27-check-uc/9c1ad7e/ : SUCCESS in 5m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/78/666878/1/gate/requirements-tox-py35-check-uc/eea47b2/ : SUCCESS in 5m 51s\n- cross-cinder-py27 http://logs.openstack.org/78/666878/1/gate/cross-cinder-py27/bce3a37/ : SUCCESS in 7m 36s\n- cross-cinder-py35 http://logs.openstack.org/78/666878/1/gate/cross-cinder-py35/d825065/ : SUCCESS in 9m 03s\n- cross-glance-py27 http://logs.openstack.org/78/666878/1/gate/cross-glance-py27/d44b9f0/ : SUCCESS in 5m 58s\n- cross-glance-py35 http://logs.openstack.org/78/666878/1/gate/cross-glance-py35/690baf2/ : SUCCESS in 5m 18s\n- cross-horizon-py27 http://logs.openstack.org/78/666878/1/gate/cross-horizon-py27/16b9f48/ : SUCCESS in 11m 31s\n- cross-horizon-py35 http://logs.openstack.org/78/666878/1/gate/cross-horizon-py35/266f92d/ : SUCCESS in 11m 45s\n- cross-keystone-py27 http://logs.openstack.org/78/666878/1/gate/cross-keystone-py27/c8400c2/ : SUCCESS in 19m 35s\n- cross-keystone-py35 http://logs.openstack.org/78/666878/1/gate/cross-keystone-py35/f348a72/ : SUCCESS in 17m 03s\n- cross-neutron-py27 http://logs.openstack.org/78/666878/1/gate/cross-neutron-py27/f8ab23d/ : SUCCESS in 24m 11s\n- cross-neutron-py35 http://logs.openstack.org/78/666878/1/gate/cross-neutron-py35/6354974/ : SUCCESS in 26m 01s\n- cross-nova-py27 http://logs.openstack.org/78/666878/1/gate/cross-nova-py27/9530a86/ : SUCCESS in 12m 11s\n- cross-nova-py35 http://logs.openstack.org/78/666878/1/gate/cross-nova-py35/a8a3103/ : SUCCESS in 13m 31s\n- cross-nova-functional http://logs.openstack.org/78/666878/1/gate/cross-nova-functional/7670730/ : SUCCESS in 20m 26s\n- cross-swift-py27 http://logs.openstack.org/78/666878/1/gate/cross-swift-py27/3cce62d/ : SUCCESS in 5m 55s\n- cross-swift-py35 http://logs.openstack.org/78/666878/1/gate/cross-swift-py35/2a03379/ : SUCCESS in 3m 53s\n- requirements-integration http://logs.openstack.org/78/666878/1/gate/requirements-integration/3fcc9c2/ : FAILURE in 32m 49s\n- neutron-grenade-multinode http://logs.openstack.org/78/666878/1/gate/neutron-grenade-multinode/e884ba2/ : SUCCESS in 59m 15s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/78/666878/1/gate/legacy-tempest-dsvm-neutron-full/beca79c/ : SUCCESS in 1h 27m 12s\n- neutron-grenade http://logs.openstack.org/78/666878/1/gate/neutron-grenade/b3ee123/ : SUCCESS in 57m 04s","accounts_in_message":[],"_revision_number":1},{"id":"05d06ecd263ebae61dd329a24ed2e66f7e64d283","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 17:14:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1a33a8c95b9dbb4c7dee7a615e6874a46674bbc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-22 18:59:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ce2f7e658b6dbae498f1fa7fac458dc8eb8341d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 19:25:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/78/666878/1/check/openstack-tox-pypy/8717688/ : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/666878/1/check/openstack-tox-pep8/ab98352/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/78/666878/1/check/openstack-tox-py27/94db1a4/ : SUCCESS in 3m 18s\n- openstack-tox-py35 http://logs.openstack.org/78/666878/1/check/openstack-tox-py35/99128f1/ : SUCCESS in 3m 21s\n- openstack-tox-validate http://logs.openstack.org/78/666878/1/check/openstack-tox-validate/d4ce3bf/ : SUCCESS in 2m 46s\n- requirements-tox-babel http://logs.openstack.org/78/666878/1/check/requirements-tox-babel/031f757/ : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/666878/1/check/requirements-tox-py27-check-uc/6de86da/ : SUCCESS in 5m 31s\n- requirements-tox-py35-check-uc http://logs.openstack.org/78/666878/1/check/requirements-tox-py35-check-uc/760ff49/ : SUCCESS in 6m 29s\n- cross-cinder-py27 http://logs.openstack.org/78/666878/1/check/cross-cinder-py27/51b0d24/ : SUCCESS in 9m 16s\n- cross-cinder-py35 http://logs.openstack.org/78/666878/1/check/cross-cinder-py35/30e3a73/ : SUCCESS in 8m 09s\n- cross-glance-py27 http://logs.openstack.org/78/666878/1/check/cross-glance-py27/eb26e7a/ : SUCCESS in 4m 29s\n- cross-glance-py35 http://logs.openstack.org/78/666878/1/check/cross-glance-py35/9498ba2/ : SUCCESS in 4m 52s\n- cross-horizon-py27 http://logs.openstack.org/78/666878/1/check/cross-horizon-py27/aa73f2b/ : SUCCESS in 11m 23s\n- cross-horizon-py35 http://logs.openstack.org/78/666878/1/check/cross-horizon-py35/447cd7c/ : SUCCESS in 12m 35s\n- cross-keystone-py27 http://logs.openstack.org/78/666878/1/check/cross-keystone-py27/0fb53e5/ : SUCCESS in 19m 30s\n- cross-keystone-py35 http://logs.openstack.org/78/666878/1/check/cross-keystone-py35/c080719/ : SUCCESS in 29m 39s\n- cross-neutron-py27 http://logs.openstack.org/78/666878/1/check/cross-neutron-py27/1fb2331/ : SUCCESS in 22m 43s\n- cross-neutron-py35 http://logs.openstack.org/78/666878/1/check/cross-neutron-py35/89248d7/ : SUCCESS in 39m 29s\n- cross-nova-py27 http://logs.openstack.org/78/666878/1/check/cross-nova-py27/ebb718a/ : SUCCESS in 18m 10s\n- cross-nova-py35 http://logs.openstack.org/78/666878/1/check/cross-nova-py35/d3eb7ea/ : SUCCESS in 19m 55s\n- cross-nova-functional http://logs.openstack.org/78/666878/1/check/cross-nova-functional/8498ca2/ : SUCCESS in 27m 04s\n- cross-swift-py27 http://logs.openstack.org/78/666878/1/check/cross-swift-py27/41e06ea/ : SUCCESS in 5m 39s\n- cross-swift-py35 http://logs.openstack.org/78/666878/1/check/cross-swift-py35/b5f59e3/ : SUCCESS in 3m 40s\n- requirements-integration http://logs.openstack.org/78/666878/1/check/requirements-integration/031177b/ : SUCCESS in 48m 31s\n- neutron-grenade-multinode http://logs.openstack.org/78/666878/1/check/neutron-grenade-multinode/e8b3e61/ : SUCCESS in 57m 12s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/78/666878/1/check/legacy-tempest-dsvm-neutron-full/c287525/ : SUCCESS in 1h 37m 25s\n- neutron-grenade http://logs.openstack.org/78/666878/1/check/neutron-grenade/dc341d6/ : SUCCESS in 47m 47s","accounts_in_message":[],"_revision_number":1},{"id":"99fe8ae372f31e40b0bd652589f23a1ba7f7444f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 19:25:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"945feabc165db81868192c06aba25d4a537d0bf5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 21:26:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f805163039e685a35e41f5338c4955b42e601e34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-22 21:26:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/78/666878/1/gate/openstack-tox-pep8/7ef956f/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/78/666878/1/gate/openstack-tox-py27/e28f009/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/78/666878/1/gate/openstack-tox-py35/6888ee7/ : SUCCESS in 3m 06s\n- openstack-tox-validate http://logs.openstack.org/78/666878/1/gate/openstack-tox-validate/cba7df8/ : SUCCESS in 2m 43s\n- requirements-tox-babel http://logs.openstack.org/78/666878/1/gate/requirements-tox-babel/7f12e7d/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/666878/1/gate/requirements-tox-py27-check-uc/06da1e5/ : SUCCESS in 5m 46s\n- requirements-tox-py35-check-uc http://logs.openstack.org/78/666878/1/gate/requirements-tox-py35-check-uc/392d959/ : SUCCESS in 5m 41s\n- cross-cinder-py27 http://logs.openstack.org/78/666878/1/gate/cross-cinder-py27/ae00d70/ : SUCCESS in 8m 08s\n- cross-cinder-py35 http://logs.openstack.org/78/666878/1/gate/cross-cinder-py35/1b9685f/ : SUCCESS in 7m 51s\n- cross-glance-py27 http://logs.openstack.org/78/666878/1/gate/cross-glance-py27/34266ec/ : SUCCESS in 4m 53s\n- cross-glance-py35 http://logs.openstack.org/78/666878/1/gate/cross-glance-py35/d4d5106/ : SUCCESS in 4m 36s\n- cross-horizon-py27 http://logs.openstack.org/78/666878/1/gate/cross-horizon-py27/e1b0c1c/ : SUCCESS in 12m 19s\n- cross-horizon-py35 http://logs.openstack.org/78/666878/1/gate/cross-horizon-py35/048e3ea/ : SUCCESS in 10m 51s\n- cross-keystone-py27 http://logs.openstack.org/78/666878/1/gate/cross-keystone-py27/997b288/ : SUCCESS in 12m 26s\n- cross-keystone-py35 http://logs.openstack.org/78/666878/1/gate/cross-keystone-py35/8a01825/ : SUCCESS in 11m 29s\n- cross-neutron-py27 http://logs.openstack.org/78/666878/1/gate/cross-neutron-py27/b57942f/ : SUCCESS in 24m 56s\n- cross-neutron-py35 http://logs.openstack.org/78/666878/1/gate/cross-neutron-py35/6d7e5f2/ : SUCCESS in 29m 47s\n- cross-nova-py27 http://logs.openstack.org/78/666878/1/gate/cross-nova-py27/dc3cbac/ : SUCCESS in 11m 46s\n- cross-nova-py35 http://logs.openstack.org/78/666878/1/gate/cross-nova-py35/14a65b7/ : SUCCESS in 12m 58s\n- cross-nova-functional http://logs.openstack.org/78/666878/1/gate/cross-nova-functional/b01af39/ : SUCCESS in 21m 17s\n- cross-swift-py27 http://logs.openstack.org/78/666878/1/gate/cross-swift-py27/a85006b/ : SUCCESS in 5m 28s\n- cross-swift-py35 http://logs.openstack.org/78/666878/1/gate/cross-swift-py35/84ed4d7/ : SUCCESS in 3m 33s\n- requirements-integration http://logs.openstack.org/78/666878/1/gate/requirements-integration/5373943/ : SUCCESS in 48m 05s\n- neutron-grenade-multinode http://logs.openstack.org/78/666878/1/gate/neutron-grenade-multinode/0cf51aa/ : SUCCESS in 1h 00m 08s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/78/666878/1/gate/legacy-tempest-dsvm-neutron-full/33204a7/ : SUCCESS in 1h 20m 09s\n- neutron-grenade http://logs.openstack.org/78/666878/1/gate/neutron-grenade/8721a82/ : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be20da1f876b0a474b469d66da95ea50a4f78a0a","revisions":{"be20da1f876b0a474b469d66da95ea50a4f78a0a":{"kind":"REWORK","_number":1,"created":"2019-06-21 15:44:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/666878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/666878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/666878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/666878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/666878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/666878/1"}}},"commit":{"parents":[{"commit":"a0f0b32b8ab4a34e05c4a48e635405116fc19222","subject":"Merge \"update constraint for python-octaviaclient to new release 1.4.1\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0f0b32b8ab4a34e05c4a48e635405116fc19222"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 15:44:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-21 15:44:25.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 8.7.0","message":"update constraint for tripleo-common to new release 8.7.0\n\nChange-Id: I2df12f5d24b6485f9542299d43e23b5a49f3fbb6\nmeta:version: 8.7.0\nmeta:diff-start: -\nmeta:series: queens\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Commit: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Change-Id: Ib9bfed2ec66d97fe56ed045ad702df06b79996c8\nmeta:release:Code-Review+1: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be20da1f876b0a474b469d66da95ea50a4f78a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be20da1f876b0a474b469d66da95ea50a4f78a0a"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
