)]}'
{"id":"openstack%2Frequirements~667790","triplet_id":"openstack%2Frequirements~master~I1b570061f2d23889a2bbcab4dfa2725b99cea1dd","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I1b570061f2d23889a2bbcab4dfa2725b99cea1dd","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-06-27 06:20:55.000000000","updated":"2019-06-28 02:38:38.000000000","submitted":"2019-06-28 02:38:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667790-1561689518380-ca750131","meta_rev_id":"1ae060ec5002b2aba19868ea1e722422c81338be","_number":667790,"virtual_id_number":667790,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-06-28 02:38:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-06-27 14:35:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-06-27 14:35:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 14:35:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-06-28 02:38:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aff545a7782bc5c234c3ebdd796f609f86d9a619","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-27 06:20:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d07f84fd56aef6d7507de9a701f5985bcfbc5332","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-27 06:30:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9c9cfcf921e24a889d1953f5200da6113f48be9f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-27 06:31:00.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"9024eff230251d68019e0879327124d9cba6037c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:26:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/90/667790/2/check/openstack-tox-docs/cb5322c/html/ : SUCCESS in 4m 30s\n- openstack-tox-pypy http://logs.openstack.org/90/667790/2/check/openstack-tox-pypy/c7e9e6f/ : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/90/667790/2/check/openstack-tox-pep8/cc023f7/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/90/667790/2/check/openstack-tox-py27/845da61/ : SUCCESS in 2m 48s\n- openstack-tox-py36 http://logs.openstack.org/90/667790/2/check/openstack-tox-py36/0560a9e/ : SUCCESS in 3m 02s\n- openstack-tox-validate http://logs.openstack.org/90/667790/2/check/openstack-tox-validate/3f1e585/ : SUCCESS in 2m 52s\n- requirements-tox-babel http://logs.openstack.org/90/667790/2/check/requirements-tox-babel/b034801/ : SUCCESS in 2m 42s\n- requirements-tox-py27-check-uc http://logs.openstack.org/90/667790/2/check/requirements-tox-py27-check-uc/91d31d2/ : SUCCESS in 7m 02s\n- requirements-tox-py36-check-uc http://logs.openstack.org/90/667790/2/check/requirements-tox-py36-check-uc/3df36ab/ : SUCCESS in 14m 28s\n- cross-cinder-py27 http://logs.openstack.org/90/667790/2/check/cross-cinder-py27/89ee06e/ : SUCCESS in 8m 31s\n- cross-cinder-py36 http://logs.openstack.org/90/667790/2/check/cross-cinder-py36/a48330c/ : RETRY_LIMIT in 2m 33s\n- cross-glance-py27 http://logs.openstack.org/90/667790/2/check/cross-glance-py27/0ee36f5/ : RETRY_LIMIT in 2m 27s\n- cross-glance-py36 http://logs.openstack.org/90/667790/2/check/cross-glance-py36/578e9b3/ : SUCCESS in 9m 06s\n- cross-horizon-py27 http://logs.openstack.org/90/667790/2/check/cross-horizon-py27/f3bd22a/ : SUCCESS in 11m 36s\n- cross-horizon-py36 http://logs.openstack.org/90/667790/2/check/cross-horizon-py36/6084419/ : SUCCESS in 9m 45s\n- cross-horizon-npm http://logs.openstack.org/90/667790/2/check/cross-horizon-npm/d1d8fce/ : SUCCESS in 6m 02s\n- cross-keystone-py27 http://logs.openstack.org/90/667790/2/check/cross-keystone-py27/b6afedf/ : SUCCESS in 23m 54s\n- cross-keystone-py36 http://logs.openstack.org/90/667790/2/check/cross-keystone-py36/eb79b0d/ : SUCCESS in 20m 49s\n- cross-neutron-py27 http://logs.openstack.org/90/667790/2/check/cross-neutron-py27/8c95bc5/ : SUCCESS in 20m 30s\n- cross-neutron-py36 http://logs.openstack.org/90/667790/2/check/cross-neutron-py36/737f389/ : SUCCESS in 23m 36s\n- cross-nova-py27 http://logs.openstack.org/90/667790/2/check/cross-nova-py27/656594f/ : SUCCESS in 12m 31s\n- cross-nova-py36 http://logs.openstack.org/90/667790/2/check/cross-nova-py36/8a70500/ : SUCCESS in 11m 29s\n- cross-nova-functional http://logs.openstack.org/90/667790/2/check/cross-nova-functional/863babc/ : SUCCESS in 18m 07s\n- cross-osc-tox-docs http://logs.openstack.org/90/667790/2/check/cross-osc-tox-docs/479f736/html/ : SUCCESS in 3m 41s\n- cross-swift-py27 http://logs.openstack.org/90/667790/2/check/cross-swift-py27/9cbf04e/ : SUCCESS in 7m 44s\n- cross-swift-py36 http://logs.openstack.org/90/667790/2/check/cross-swift-py36/44b5a54/ : SUCCESS in 8m 25s\n- requirements-integration http://logs.openstack.org/90/667790/2/check/requirements-integration/b627da4/ : SUCCESS in 54m 08s\n- neutron-grenade-multinode http://logs.openstack.org/90/667790/2/check/neutron-grenade-multinode/d322445/ : SUCCESS in 1h 04m 18s\n- tempest-full http://logs.openstack.org/90/667790/2/check/tempest-full/b793b4c/ : SUCCESS in 1h 41m 19s\n- neutron-grenade http://logs.openstack.org/90/667790/2/check/neutron-grenade/3dd0a1f/ : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":2},{"id":"57023a49468ffa819e52933d6c072020b886b454","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-06-27 14:35:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"14a279e56a82272d55811886b2b6125c2bcebb6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 23:55:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/667790/2/check/openstack-tox-docs/a16db89/html/ : SUCCESS in 2m 28s\n- openstack-tox-pypy http://logs.openstack.org/90/667790/2/check/openstack-tox-pypy/17d269b/ : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/90/667790/2/check/openstack-tox-pep8/26e5a08/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/90/667790/2/check/openstack-tox-py27/50112ae/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/90/667790/2/check/openstack-tox-py36/b784901/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/90/667790/2/check/openstack-tox-validate/8e51cc1/ : SUCCESS in 4m 27s\n- requirements-tox-babel http://logs.openstack.org/90/667790/2/check/requirements-tox-babel/199d858/ : SUCCESS in 4m 15s\n- requirements-tox-py27-check-uc http://logs.openstack.org/90/667790/2/check/requirements-tox-py27-check-uc/0dc1005/ : SUCCESS in 8m 23s\n- requirements-tox-py36-check-uc http://logs.openstack.org/90/667790/2/check/requirements-tox-py36-check-uc/6be6dec/ : SUCCESS in 9m 54s\n- cross-cinder-py27 http://logs.openstack.org/90/667790/2/check/cross-cinder-py27/9968605/ : SUCCESS in 9m 08s\n- cross-cinder-py36 http://logs.openstack.org/90/667790/2/check/cross-cinder-py36/30f4b9c/ : SUCCESS in 7m 42s\n- cross-glance-py27 http://logs.openstack.org/90/667790/2/check/cross-glance-py27/ae23e3c/ : SUCCESS in 5m 27s\n- cross-glance-py36 http://logs.openstack.org/90/667790/2/check/cross-glance-py36/5a24439/ : SUCCESS in 4m 26s\n- cross-horizon-py27 http://logs.openstack.org/90/667790/2/check/cross-horizon-py27/2855623/ : SUCCESS in 10m 48s\n- cross-horizon-py36 http://logs.openstack.org/90/667790/2/check/cross-horizon-py36/87acd2f/ : SUCCESS in 8m 37s\n- cross-horizon-npm http://logs.openstack.org/90/667790/2/check/cross-horizon-npm/49a5ec3/ : SUCCESS in 5m 56s\n- cross-keystone-py27 http://logs.openstack.org/90/667790/2/check/cross-keystone-py27/5a56752/ : SUCCESS in 26m 08s\n- cross-keystone-py36 http://logs.openstack.org/90/667790/2/check/cross-keystone-py36/1892572/ : SUCCESS in 30m 57s\n- cross-neutron-py27 http://logs.openstack.org/90/667790/2/check/cross-neutron-py27/fcf855a/ : SUCCESS in 24m 35s\n- cross-neutron-py36 http://logs.openstack.org/90/667790/2/check/cross-neutron-py36/c7fcb5d/ : SUCCESS in 23m 55s\n- cross-nova-py27 http://logs.openstack.org/90/667790/2/check/cross-nova-py27/fdf08f5/ : SUCCESS in 12m 22s\n- cross-nova-py36 http://logs.openstack.org/90/667790/2/check/cross-nova-py36/b781951/ : SUCCESS in 12m 22s\n- cross-nova-functional http://logs.openstack.org/90/667790/2/check/cross-nova-functional/da2682a/ : SUCCESS in 20m 14s\n- cross-osc-tox-docs http://logs.openstack.org/90/667790/2/check/cross-osc-tox-docs/19b7fb9/html/ : SUCCESS in 3m 58s\n- cross-swift-py27 http://logs.openstack.org/90/667790/2/check/cross-swift-py27/bd30cef/ : SUCCESS in 7m 25s\n- cross-swift-py36 http://logs.openstack.org/90/667790/2/check/cross-swift-py36/361acce/ : SUCCESS in 8m 05s\n- requirements-integration http://logs.openstack.org/90/667790/2/check/requirements-integration/b25a911/ : SUCCESS in 54m 36s\n- neutron-grenade-multinode http://logs.openstack.org/90/667790/2/check/neutron-grenade-multinode/cf0ee64/ : SUCCESS in 1h 05m 16s\n- tempest-full http://logs.openstack.org/90/667790/2/check/tempest-full/f66adc8/ : SUCCESS in 2h 04m 30s\n- neutron-grenade http://logs.openstack.org/90/667790/2/check/neutron-grenade/dcdeac3/ : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":2},{"id":"daec12bbbb65012cd1f3b1dba0661ebe452b9424","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 23:56:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"638b1a04940841f59959c87f50a25e3aa29bdab2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 02:38:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/90/667790/2/gate/openstack-tox-docs/e3171fc/html/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/90/667790/2/gate/openstack-tox-pep8/b89173e/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/90/667790/2/gate/openstack-tox-py27/35d8b7b/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/90/667790/2/gate/openstack-tox-py36/4ce97e7/ : SUCCESS in 4m 08s\n- openstack-tox-validate http://logs.openstack.org/90/667790/2/gate/openstack-tox-validate/a0106b5/ : SUCCESS in 3m 19s\n- requirements-tox-babel http://logs.openstack.org/90/667790/2/gate/requirements-tox-babel/e618765/ : SUCCESS in 3m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/90/667790/2/gate/requirements-tox-py27-check-uc/b67a8ab/ : SUCCESS in 7m 27s\n- requirements-tox-py36-check-uc http://logs.openstack.org/90/667790/2/gate/requirements-tox-py36-check-uc/5275f1c/ : SUCCESS in 6m 38s\n- cross-cinder-py27 http://logs.openstack.org/90/667790/2/gate/cross-cinder-py27/08bb897/ : SUCCESS in 8m 19s\n- cross-cinder-py36 http://logs.openstack.org/90/667790/2/gate/cross-cinder-py36/bbdcf2c/ : SUCCESS in 8m 43s\n- cross-glance-py27 http://logs.openstack.org/90/667790/2/gate/cross-glance-py27/bfb60b1/ : SUCCESS in 5m 30s\n- cross-glance-py36 http://logs.openstack.org/90/667790/2/gate/cross-glance-py36/539d67c/ : SUCCESS in 4m 33s\n- cross-horizon-py27 http://logs.openstack.org/90/667790/2/gate/cross-horizon-py27/3dc43d4/ : SUCCESS in 11m 11s\n- cross-horizon-py36 http://logs.openstack.org/90/667790/2/gate/cross-horizon-py36/6fa640e/ : SUCCESS in 9m 59s\n- cross-horizon-npm http://logs.openstack.org/90/667790/2/gate/cross-horizon-npm/7d6c3b2/ : SUCCESS in 6m 13s\n- cross-keystone-py27 http://logs.openstack.org/90/667790/2/gate/cross-keystone-py27/6047311/ : SUCCESS in 23m 51s\n- cross-keystone-py36 http://logs.openstack.org/90/667790/2/gate/cross-keystone-py36/a5a8abe/ : SUCCESS in 23m 34s\n- cross-neutron-py27 http://logs.openstack.org/90/667790/2/gate/cross-neutron-py27/1420c19/ : SUCCESS in 22m 34s\n- cross-neutron-py36 http://logs.openstack.org/90/667790/2/gate/cross-neutron-py36/571e2bb/ : SUCCESS in 24m 41s\n- cross-nova-py27 http://logs.openstack.org/90/667790/2/gate/cross-nova-py27/7cf2f1b/ : SUCCESS in 11m 45s\n- cross-nova-py36 http://logs.openstack.org/90/667790/2/gate/cross-nova-py36/f7ee154/ : SUCCESS in 10m 34s\n- cross-nova-functional http://logs.openstack.org/90/667790/2/gate/cross-nova-functional/eeaa7af/ : SUCCESS in 20m 13s\n- cross-osc-tox-docs http://logs.openstack.org/90/667790/2/gate/cross-osc-tox-docs/5bc370d/html/ : SUCCESS in 4m 28s\n- cross-swift-py27 http://logs.openstack.org/90/667790/2/gate/cross-swift-py27/b479ecc/ : SUCCESS in 8m 31s\n- cross-swift-py36 http://logs.openstack.org/90/667790/2/gate/cross-swift-py36/b1cfcf9/ : SUCCESS in 9m 29s\n- requirements-integration http://logs.openstack.org/90/667790/2/gate/requirements-integration/a02881d/ : SUCCESS in 1h 01m 05s\n- neutron-grenade-multinode http://logs.openstack.org/90/667790/2/gate/neutron-grenade-multinode/915637c/ : SUCCESS in 1h 11m 27s\n- tempest-full http://logs.openstack.org/90/667790/2/gate/tempest-full/5682301/ : SUCCESS in 1h 49m 03s\n- neutron-grenade http://logs.openstack.org/90/667790/2/gate/neutron-grenade/b9b23d6/ : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":2},{"id":"1ae060ec5002b2aba19868ea1e722422c81338be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 02:38:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c468983657ce53328a22c5beb550ea992d4b96dc","revisions":{"7cc3c6d0ed85046b4f2ac4f3da6851b3365a2a84":{"kind":"REWORK","_number":1,"created":"2019-06-27 06:20:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/667790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/667790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/667790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/667790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/667790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/667790/1"}}},"commit":{"parents":[{"commit":"23f7ac020b1c48c4f52b9a85dd1f94c137412afc","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23f7ac020b1c48c4f52b9a85dd1f94c137412afc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-27 06:20:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-27 06:20:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1b570061f2d23889a2bbcab4dfa2725b99cea1dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7cc3c6d0ed85046b4f2ac4f3da6851b3365a2a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7cc3c6d0ed85046b4f2ac4f3da6851b3365a2a84"}]},"branch":"refs/heads/master"},"c468983657ce53328a22c5beb550ea992d4b96dc":{"kind":"REWORK","_number":2,"created":"2019-06-27 06:30:54.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/90/667790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/667790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/667790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/667790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/667790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/667790/2"}}},"commit":{"parents":[{"commit":"23f7ac020b1c48c4f52b9a85dd1f94c137412afc","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/23f7ac020b1c48c4f52b9a85dd1f94c137412afc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-27 06:20:52.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-06-27 06:30:42.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1b570061f2d23889a2bbcab4dfa2725b99cea1dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c468983657ce53328a22c5beb550ea992d4b96dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c468983657ce53328a22c5beb550ea992d4b96dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
