)]}'
{"id":"openstack%2Frequirements~668928","triplet_id":"openstack%2Frequirements~master~I5b82b6d734cbf37d15b61d6b83fcb1e53d1eb2b9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5b82b6d734cbf37d15b61d6b83fcb1e53d1eb2b9","subject":"update constraint for neutron-lib to new release 1.28.0","status":"MERGED","created":"2019-07-03 14:47:43.000000000","updated":"2019-07-04 01:56:02.000000000","submitted":"2019-07-04 01:56:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668928-1562205362733-795c03e7","meta_rev_id":"b3afc59696d111ec132b336caa221a471c404b17","_number":668928,"virtual_id_number":668928,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-07-04 01:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-03 23:39:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-07-03 18:56:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-03 23:39:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 18:56:07.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-07-03 23:39:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-04 01:56:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd501119334ff16e7fe320d90e7bcc81d9419eb5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-03 14:47:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b42f96e757c0b7964ca2f8979a7453535cb84dfb","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-03 18:56:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a4ae2553ca411da2b0332d2559bbe63cbbbd11c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:18:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/668928/1/check/openstack-tox-docs/eb6d36d/html/ : SUCCESS in 2m 42s\n- openstack-tox-pypy http://logs.openstack.org/28/668928/1/check/openstack-tox-pypy/ce1159e/ : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/28/668928/1/check/openstack-tox-pep8/2be7643/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/28/668928/1/check/openstack-tox-py27/2ee8d58/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/28/668928/1/check/openstack-tox-py36/558586e/ : SUCCESS in 3m 39s\n- openstack-tox-validate http://logs.openstack.org/28/668928/1/check/openstack-tox-validate/bf1749d/ : SUCCESS in 2m 54s\n- requirements-tox-babel http://logs.openstack.org/28/668928/1/check/requirements-tox-babel/e84f425/ : SUCCESS in 3m 00s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/668928/1/check/requirements-tox-py27-check-uc/f9966a2/ : SUCCESS in 9m 36s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/668928/1/check/requirements-tox-py36-check-uc/865889c/ : SUCCESS in 9m 22s\n- cross-cinder-py27 http://logs.openstack.org/28/668928/1/check/cross-cinder-py27/576edf6/ : SUCCESS in 11m 06s\n- cross-cinder-py36 http://logs.openstack.org/28/668928/1/check/cross-cinder-py36/cb21934/ : SUCCESS in 12m 16s\n- cross-glance-py27 http://logs.openstack.org/28/668928/1/check/cross-glance-py27/b065b8a/ : SUCCESS in 4m 08s\n- cross-glance-py36 http://logs.openstack.org/28/668928/1/check/cross-glance-py36/52a1c54/ : SUCCESS in 4m 55s\n- cross-horizon-py27 http://logs.openstack.org/28/668928/1/check/cross-horizon-py27/bc5be23/ : SUCCESS in 10m 56s\n- cross-horizon-py36 http://logs.openstack.org/28/668928/1/check/cross-horizon-py36/b4280bf/ : SUCCESS in 9m 08s\n- cross-horizon-npm http://logs.openstack.org/28/668928/1/check/cross-horizon-npm/7a54fe8/ : SUCCESS in 7m 17s\n- cross-keystone-py27 http://logs.openstack.org/28/668928/1/check/cross-keystone-py27/dc66e54/ : SUCCESS in 24m 55s\n- cross-keystone-py36 http://logs.openstack.org/28/668928/1/check/cross-keystone-py36/eec932b/ : SUCCESS in 31m 09s\n- cross-neutron-py27 http://logs.openstack.org/28/668928/1/check/cross-neutron-py27/2e30fa8/ : SUCCESS in 29m 31s\n- cross-neutron-py36 http://logs.openstack.org/28/668928/1/check/cross-neutron-py36/bd2c40c/ : SUCCESS in 24m 09s\n- cross-nova-py27 http://logs.openstack.org/28/668928/1/check/cross-nova-py27/460cf28/ : SUCCESS in 14m 35s\n- cross-nova-py36 http://logs.openstack.org/28/668928/1/check/cross-nova-py36/24ebb26/ : SUCCESS in 11m 14s\n- cross-nova-functional http://logs.openstack.org/28/668928/1/check/cross-nova-functional/bcba49a/ : SUCCESS in 20m 42s\n- cross-osc-tox-docs http://logs.openstack.org/28/668928/1/check/cross-osc-tox-docs/e006c17/html/ : SUCCESS in 5m 15s\n- cross-swift-py27 http://logs.openstack.org/28/668928/1/check/cross-swift-py27/e3cd753/ : SUCCESS in 8m 35s\n- cross-swift-py36 http://logs.openstack.org/28/668928/1/check/cross-swift-py36/72dda4c/ : SUCCESS in 7m 41s\n- requirements-integration http://logs.openstack.org/28/668928/1/check/requirements-integration/6e704d5/ : SUCCESS in 51m 53s\n- neutron-grenade-multinode http://logs.openstack.org/28/668928/1/check/neutron-grenade-multinode/e84b094/ : SUCCESS in 1h 18m 19s\n- tempest-full http://logs.openstack.org/28/668928/1/check/tempest-full/8701123/ : SUCCESS in 1h 42m 10s\n- neutron-grenade http://logs.openstack.org/28/668928/1/check/neutron-grenade/ee0fd6b/ : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"05c8cdc1a173a1024a5d1b55cca3c0a7fca339ec","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-03 23:39:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b05f2bf583787cbfbae0a531ed0d484a0e2c5764","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 23:40:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ad4a22e51391d6eb769ca0d7cb88788d5fa42f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 01:56:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/668928/1/gate/openstack-tox-docs/dba0cd5/html/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/28/668928/1/gate/openstack-tox-pep8/4b0b0ee/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/28/668928/1/gate/openstack-tox-py27/4d2843b/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/28/668928/1/gate/openstack-tox-py36/67b1738/ : SUCCESS in 3m 54s\n- openstack-tox-validate http://logs.openstack.org/28/668928/1/gate/openstack-tox-validate/5b56239/ : SUCCESS in 3m 16s\n- requirements-tox-babel http://logs.openstack.org/28/668928/1/gate/requirements-tox-babel/a04d3d8/ : SUCCESS in 3m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/28/668928/1/gate/requirements-tox-py27-check-uc/d048f32/ : SUCCESS in 6m 48s\n- requirements-tox-py36-check-uc http://logs.openstack.org/28/668928/1/gate/requirements-tox-py36-check-uc/207a80d/ : SUCCESS in 6m 51s\n- cross-cinder-py27 http://logs.openstack.org/28/668928/1/gate/cross-cinder-py27/334c316/ : SUCCESS in 8m 56s\n- cross-cinder-py36 http://logs.openstack.org/28/668928/1/gate/cross-cinder-py36/2f4988f/ : SUCCESS in 7m 35s\n- cross-glance-py27 http://logs.openstack.org/28/668928/1/gate/cross-glance-py27/413598d/ : SUCCESS in 4m 33s\n- cross-glance-py36 http://logs.openstack.org/28/668928/1/gate/cross-glance-py36/5197fa2/ : SUCCESS in 4m 18s\n- cross-horizon-py27 http://logs.openstack.org/28/668928/1/gate/cross-horizon-py27/46bb00d/ : SUCCESS in 10m 14s\n- cross-horizon-py36 http://logs.openstack.org/28/668928/1/gate/cross-horizon-py36/2e05310/ : SUCCESS in 8m 43s\n- cross-horizon-npm http://logs.openstack.org/28/668928/1/gate/cross-horizon-npm/7d50eed/ : SUCCESS in 6m 31s\n- cross-keystone-py27 http://logs.openstack.org/28/668928/1/gate/cross-keystone-py27/36ba86f/ : SUCCESS in 41m 12s\n- cross-keystone-py36 http://logs.openstack.org/28/668928/1/gate/cross-keystone-py36/5c63150/ : SUCCESS in 41m 20s\n- cross-neutron-py27 http://logs.openstack.org/28/668928/1/gate/cross-neutron-py27/fd43e88/ : SUCCESS in 35m 12s\n- cross-neutron-py36 http://logs.openstack.org/28/668928/1/gate/cross-neutron-py36/8e80963/ : SUCCESS in 18m 47s\n- cross-nova-py27 http://logs.openstack.org/28/668928/1/gate/cross-nova-py27/fd24076/ : SUCCESS in 12m 32s\n- cross-nova-py36 http://logs.openstack.org/28/668928/1/gate/cross-nova-py36/1c3842c/ : SUCCESS in 12m 43s\n- cross-nova-functional http://logs.openstack.org/28/668928/1/gate/cross-nova-functional/c1fa773/ : SUCCESS in 16m 48s\n- cross-osc-tox-docs http://logs.openstack.org/28/668928/1/gate/cross-osc-tox-docs/4d0b283/html/ : SUCCESS in 3m 56s\n- cross-swift-py27 http://logs.openstack.org/28/668928/1/gate/cross-swift-py27/2571477/ : SUCCESS in 8m 38s\n- cross-swift-py36 http://logs.openstack.org/28/668928/1/gate/cross-swift-py36/4aefe8f/ : SUCCESS in 8m 59s\n- requirements-integration http://logs.openstack.org/28/668928/1/gate/requirements-integration/be100cb/ : SUCCESS in 50m 43s\n- neutron-grenade-multinode http://logs.openstack.org/28/668928/1/gate/neutron-grenade-multinode/e1b51fe/ : SUCCESS in 1h 05m 17s\n- tempest-full http://logs.openstack.org/28/668928/1/gate/tempest-full/0aa4b33/ : SUCCESS in 1h 33m 44s\n- neutron-grenade http://logs.openstack.org/28/668928/1/gate/neutron-grenade/2e60ba7/ : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a7b6a4b8948170e3cb8023e64bcd3f82c3209692","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 01:56:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"526227ae9ebee78cf655de44368482f8aabfae2e","revisions":{"526227ae9ebee78cf655de44368482f8aabfae2e":{"kind":"REWORK","_number":1,"created":"2019-07-03 14:47:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/668928/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/668928/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/668928/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/668928/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/668928/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/668928/1"}}},"commit":{"parents":[{"commit":"d4c1f4dd82ddffa0a0eebeb30e37220a7d4c2044","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4c1f4dd82ddffa0a0eebeb30e37220a7d4c2044"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-03 14:47:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-03 14:47:41.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 1.28.0","message":"update constraint for neutron-lib to new release 1.28.0\n\nChange-Id: I5b82b6d734cbf37d15b61d6b83fcb1e53d1eb2b9\nmeta:version: 1.28.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Jens Harbott \u003cj.harbott@x-ion.de\u003e\nmeta:release:Commit: Jens Harbott \u003cj.harbott@x-ion.de\u003e\nmeta:release:Change-Id: If9bcfc80cb75382b70ddc5937ed39a176ab8956f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/526227ae9ebee78cf655de44368482f8aabfae2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/526227ae9ebee78cf655de44368482f8aabfae2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
