)]}'
{"id":"openstack%2Frequirements~669133","triplet_id":"openstack%2Frequirements~master~I496de2f6eae3401c674ae8530d02ba10c5d3ea75","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I496de2f6eae3401c674ae8530d02ba10c5d3ea75","subject":"update constraint for pbr to new release 5.4.0","status":"MERGED","created":"2019-07-04 10:25:22.000000000","updated":"2019-07-05 06:42:29.000000000","submitted":"2019-07-05 06:42:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669133-1562308949698-5004bf0a","meta_rev_id":"15db78934615c5933656425fa1a765a26742ad2b","_number":669133,"virtual_id_number":669133,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-05 06:42:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-05 03:08:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-05 03:08:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 03:08:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-05 06:42:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"21eb78dfa63ef57bb7efe435e0ccf73e45e15bf8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-04 10:25:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5874ebe97b4159524a0e971e4d04c1a4b0daa8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 18:29:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/669133/1/check/openstack-tox-docs/b8c0cf9/html/ : SUCCESS in 3m 14s\n- openstack-tox-pypy http://logs.openstack.org/33/669133/1/check/openstack-tox-pypy/1426966/ : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/33/669133/1/check/openstack-tox-pep8/c22f5b3/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/33/669133/1/check/openstack-tox-py27/dedf1ea/ : SUCCESS in 3m 43s\n- openstack-tox-py36 http://logs.openstack.org/33/669133/1/check/openstack-tox-py36/a7d9fdf/ : SUCCESS in 4m 57s\n- openstack-tox-validate http://logs.openstack.org/33/669133/1/check/openstack-tox-validate/7ac839d/ : SUCCESS in 4m 41s\n- requirements-tox-babel http://logs.openstack.org/33/669133/1/check/requirements-tox-babel/cea9d05/ : SUCCESS in 3m 34s\n- requirements-tox-py27-check-uc http://logs.openstack.org/33/669133/1/check/requirements-tox-py27-check-uc/05328b0/ : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc http://logs.openstack.org/33/669133/1/check/requirements-tox-py36-check-uc/f9f03b8/ : SUCCESS in 6m 05s\n- cross-cinder-py27 http://logs.openstack.org/33/669133/1/check/cross-cinder-py27/73d3d56/ : SUCCESS in 8m 20s\n- cross-cinder-py36 http://logs.openstack.org/33/669133/1/check/cross-cinder-py36/05d74f6/ : SUCCESS in 7m 51s\n- cross-glance-py27 http://logs.openstack.org/33/669133/1/check/cross-glance-py27/e734d48/ : SUCCESS in 5m 37s\n- cross-glance-py36 http://logs.openstack.org/33/669133/1/check/cross-glance-py36/39c54a5/ : SUCCESS in 5m 37s\n- cross-horizon-py27 http://logs.openstack.org/33/669133/1/check/cross-horizon-py27/e1e6b0e/ : SUCCESS in 9m 31s\n- cross-horizon-py36 http://logs.openstack.org/33/669133/1/check/cross-horizon-py36/a7d8584/ : SUCCESS in 7m 52s\n- cross-horizon-npm http://logs.openstack.org/33/669133/1/check/cross-horizon-npm/2fe600a/ : SUCCESS in 6m 17s\n- cross-keystone-py27 http://logs.openstack.org/33/669133/1/check/cross-keystone-py27/1585cbf/ : SUCCESS in 27m 49s\n- cross-keystone-py36 http://logs.openstack.org/33/669133/1/check/cross-keystone-py36/1ec20e9/ : SUCCESS in 27m 22s\n- cross-neutron-py27 http://logs.openstack.org/33/669133/1/check/cross-neutron-py27/fa4387d/ : SUCCESS in 24m 13s\n- cross-neutron-py36 http://logs.openstack.org/33/669133/1/check/cross-neutron-py36/a9e7387/ : SUCCESS in 23m 27s\n- cross-nova-py27 http://logs.openstack.org/33/669133/1/check/cross-nova-py27/ee9715d/ : SUCCESS in 14m 36s\n- cross-nova-py36 http://logs.openstack.org/33/669133/1/check/cross-nova-py36/4c4c461/ : SUCCESS in 16m 22s\n- cross-nova-functional http://logs.openstack.org/33/669133/1/check/cross-nova-functional/f4cd6da/ : SUCCESS in 18m 08s\n- cross-osc-tox-docs http://logs.openstack.org/33/669133/1/check/cross-osc-tox-docs/aee4789/html/ : SUCCESS in 5m 06s\n- cross-swift-py27 http://logs.openstack.org/33/669133/1/check/cross-swift-py27/93c1108/ : SUCCESS in 8m 05s\n- cross-swift-py36 http://logs.openstack.org/33/669133/1/check/cross-swift-py36/8a88917/ : SUCCESS in 8m 20s\n- requirements-integration http://logs.openstack.org/33/669133/1/check/requirements-integration/1b47464/ : SUCCESS in 59m 16s\n- neutron-grenade-multinode http://logs.openstack.org/33/669133/1/check/neutron-grenade-multinode/052c558/ : SUCCESS in 1h 13m 43s\n- tempest-full http://logs.openstack.org/33/669133/1/check/tempest-full/d54dd7e/ : SUCCESS in 1h 54m 24s\n- neutron-grenade http://logs.openstack.org/33/669133/1/check/neutron-grenade/881780c/ : SUCCESS in 1h 18m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4e75e2c36089b075a0ed1dd6c047ec3e6f8b3139","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-05 03:08:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19c752abf45119f9bf412f9ebc29cc3cf1f6ce72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 03:08:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b26ff0fc26de5ad991f3b856a39aacdea0b307bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:42:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"90b7de04cd155ebac756af1b87ca5b6144d82996","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:42:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/669133/1/gate/openstack-tox-docs/180e7b2/html/ : SUCCESS in 3m 54s\n- openstack-tox-pep8 http://logs.openstack.org/33/669133/1/gate/openstack-tox-pep8/7bcf0d8/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/33/669133/1/gate/openstack-tox-py27/6f61369/ : SUCCESS in 3m 17s\n- openstack-tox-py36 http://logs.openstack.org/33/669133/1/gate/openstack-tox-py36/6c7bfbe/ : SUCCESS in 3m 21s\n- openstack-tox-validate http://logs.openstack.org/33/669133/1/gate/openstack-tox-validate/1cf4752/ : SUCCESS in 4m 00s\n- requirements-tox-babel http://logs.openstack.org/33/669133/1/gate/requirements-tox-babel/97677b6/ : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/33/669133/1/gate/requirements-tox-py27-check-uc/2185405/ : SUCCESS in 10m 33s\n- requirements-tox-py36-check-uc http://logs.openstack.org/33/669133/1/gate/requirements-tox-py36-check-uc/5d5a5ce/ : SUCCESS in 5m 57s\n- cross-cinder-py27 http://logs.openstack.org/33/669133/1/gate/cross-cinder-py27/28b4e97/ : SUCCESS in 9m 30s\n- cross-cinder-py36 http://logs.openstack.org/33/669133/1/gate/cross-cinder-py36/ca2fd31/ : SUCCESS in 8m 02s\n- cross-glance-py27 http://logs.openstack.org/33/669133/1/gate/cross-glance-py27/c2686ad/ : SUCCESS in 6m 32s\n- cross-glance-py36 http://logs.openstack.org/33/669133/1/gate/cross-glance-py36/4952268/ : SUCCESS in 3m 54s\n- cross-horizon-py27 http://logs.openstack.org/33/669133/1/gate/cross-horizon-py27/0ed480d/ : SUCCESS in 10m 15s\n- cross-horizon-py36 http://logs.openstack.org/33/669133/1/gate/cross-horizon-py36/b61e5c4/ : SUCCESS in 7m 37s\n- cross-horizon-npm http://logs.openstack.org/33/669133/1/gate/cross-horizon-npm/99730c8/ : SUCCESS in 7m 13s\n- cross-keystone-py27 http://logs.openstack.org/33/669133/1/gate/cross-keystone-py27/6cee0c3/ : SUCCESS in 23m 59s\n- cross-keystone-py36 http://logs.openstack.org/33/669133/1/gate/cross-keystone-py36/1f5c785/ : SUCCESS in 26m 33s\n- cross-neutron-py27 http://logs.openstack.org/33/669133/1/gate/cross-neutron-py27/4b07c24/ : SUCCESS in 21m 38s\n- cross-neutron-py36 http://logs.openstack.org/33/669133/1/gate/cross-neutron-py36/1e6a5ec/ : SUCCESS in 20m 24s\n- cross-nova-py27 http://logs.openstack.org/33/669133/1/gate/cross-nova-py27/eb8e458/ : SUCCESS in 12m 55s\n- cross-nova-py36 http://logs.openstack.org/33/669133/1/gate/cross-nova-py36/72862c0/ : SUCCESS in 12m 34s\n- cross-nova-functional http://logs.openstack.org/33/669133/1/gate/cross-nova-functional/c45d537/ : SUCCESS in 17m 53s\n- cross-osc-tox-docs http://logs.openstack.org/33/669133/1/gate/cross-osc-tox-docs/4f8a82b/html/ : SUCCESS in 3m 53s\n- cross-swift-py27 http://logs.openstack.org/33/669133/1/gate/cross-swift-py27/b45a3a9/ : SUCCESS in 7m 57s\n- cross-swift-py36 http://logs.openstack.org/33/669133/1/gate/cross-swift-py36/fde4453/ : SUCCESS in 7m 35s\n- requirements-integration http://logs.openstack.org/33/669133/1/gate/requirements-integration/cfdd1c3/ : SUCCESS in 50m 59s\n- neutron-grenade-multinode http://logs.openstack.org/33/669133/1/gate/neutron-grenade-multinode/b97f40e/ : SUCCESS in 1h 06m 37s\n- tempest-full http://logs.openstack.org/33/669133/1/gate/tempest-full/21effb1/ : SUCCESS in 1h 44m 11s\n- neutron-grenade http://logs.openstack.org/33/669133/1/gate/neutron-grenade/998cca3/ : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c3f0bf16cb6f4ce6dbd17af037c31f509b72809","revisions":{"1c3f0bf16cb6f4ce6dbd17af037c31f509b72809":{"kind":"REWORK","_number":1,"created":"2019-07-04 10:25:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/669133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/669133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/669133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/669133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/669133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/669133/1"}}},"commit":{"parents":[{"commit":"526227ae9ebee78cf655de44368482f8aabfae2e","subject":"update constraint for neutron-lib to new release 1.28.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/526227ae9ebee78cf655de44368482f8aabfae2e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-04 10:25:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-04 10:25:17.000000000","tz":0},"subject":"update constraint for pbr to new release 5.4.0","message":"update constraint for pbr to new release 5.4.0\n\nChange-Id: I496de2f6eae3401c674ae8530d02ba10c5d3ea75\nmeta:version: 5.4.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I216cfb0c14af6f88b3a9a2210ad8b03c088440fe\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c3f0bf16cb6f4ce6dbd17af037c31f509b72809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c3f0bf16cb6f4ce6dbd17af037c31f509b72809"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
