)]}'
{"id":"openstack%2Frequirements~669134","triplet_id":"openstack%2Frequirements~master~I6dd55d56d31d7d12cb9c2225747df55a8272e834","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6dd55d56d31d7d12cb9c2225747df55a8272e834","subject":"update constraint for python-qinlingclient to new release 2.2.0","status":"MERGED","created":"2019-07-04 10:25:56.000000000","updated":"2019-07-05 06:42:28.000000000","submitted":"2019-07-05 06:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669134-1562308948185-eaa21e4c","meta_rev_id":"3948c67f4bcc118bcdc449a1209e5992b092185b","_number":669134,"virtual_id_number":669134,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-05 06:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-05 03:08:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-05 03:08:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 03:08:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-05 06:42:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"402eafd530652b5368c189bc386410adcc55e8c3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-04 10:25:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7faa6b6489658805f3caf556db2f08a15e4ad339","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 18:37:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/34/669134/1/check/openstack-tox-docs/5a9f815/html/ : SUCCESS in 2m 58s\n- openstack-tox-pypy http://logs.openstack.org/34/669134/1/check/openstack-tox-pypy/c753745/ : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/34/669134/1/check/openstack-tox-pep8/cc928b7/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/34/669134/1/check/openstack-tox-py27/4e43b67/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/34/669134/1/check/openstack-tox-py36/b3772a4/ : SUCCESS in 4m 23s\n- openstack-tox-validate http://logs.openstack.org/34/669134/1/check/openstack-tox-validate/0f41823/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/34/669134/1/check/requirements-tox-babel/8613d74/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/34/669134/1/check/requirements-tox-py27-check-uc/4205629/ : SUCCESS in 7m 56s\n- requirements-tox-py36-check-uc http://logs.openstack.org/34/669134/1/check/requirements-tox-py36-check-uc/e42c575/ : SUCCESS in 8m 29s\n- cross-cinder-py27 http://logs.openstack.org/34/669134/1/check/cross-cinder-py27/7597ec6/ : SUCCESS in 8m 02s\n- cross-cinder-py36 http://logs.openstack.org/34/669134/1/check/cross-cinder-py36/e6c28f8/ : SUCCESS in 7m 52s\n- cross-glance-py27 http://logs.openstack.org/34/669134/1/check/cross-glance-py27/eeaeb21/ : SUCCESS in 4m 44s\n- cross-glance-py36 http://logs.openstack.org/34/669134/1/check/cross-glance-py36/30cc4c8/ : SUCCESS in 4m 07s\n- cross-horizon-py27 http://logs.openstack.org/34/669134/1/check/cross-horizon-py27/50be843/ : SUCCESS in 9m 56s\n- cross-horizon-py36 http://logs.openstack.org/34/669134/1/check/cross-horizon-py36/19a0af4/ : SUCCESS in 9m 30s\n- cross-horizon-npm http://logs.openstack.org/34/669134/1/check/cross-horizon-npm/1814836/ : SUCCESS in 6m 08s\n- cross-keystone-py27 http://logs.openstack.org/34/669134/1/check/cross-keystone-py27/903e223/ : SUCCESS in 28m 10s\n- cross-keystone-py36 http://logs.openstack.org/34/669134/1/check/cross-keystone-py36/b5ed45e/ : SUCCESS in 25m 27s\n- cross-neutron-py27 http://logs.openstack.org/34/669134/1/check/cross-neutron-py27/5672e36/ : SUCCESS in 26m 00s\n- cross-neutron-py36 http://logs.openstack.org/34/669134/1/check/cross-neutron-py36/9665baf/ : SUCCESS in 19m 23s\n- cross-nova-py27 http://logs.openstack.org/34/669134/1/check/cross-nova-py27/8396dfa/ : SUCCESS in 12m 07s\n- cross-nova-py36 http://logs.openstack.org/34/669134/1/check/cross-nova-py36/0aec509/ : SUCCESS in 12m 52s\n- cross-nova-functional http://logs.openstack.org/34/669134/1/check/cross-nova-functional/e790f48/ : SUCCESS in 18m 19s\n- cross-osc-tox-docs http://logs.openstack.org/34/669134/1/check/cross-osc-tox-docs/45a3af2/html/ : SUCCESS in 4m 29s\n- cross-swift-py27 http://logs.openstack.org/34/669134/1/check/cross-swift-py27/8519b42/ : SUCCESS in 7m 25s\n- cross-swift-py36 http://logs.openstack.org/34/669134/1/check/cross-swift-py36/110a3ac/ : SUCCESS in 7m 40s\n- requirements-integration http://logs.openstack.org/34/669134/1/check/requirements-integration/0f1aa8e/ : SUCCESS in 56m 41s\n- neutron-grenade-multinode http://logs.openstack.org/34/669134/1/check/neutron-grenade-multinode/b484607/ : SUCCESS in 1h 13m 53s\n- tempest-full http://logs.openstack.org/34/669134/1/check/tempest-full/3bad26c/ : SUCCESS in 1h 56m 40s\n- neutron-grenade http://logs.openstack.org/34/669134/1/check/neutron-grenade/8a64d55/ : SUCCESS in 55m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9517cf9e3ef19306ae04ae5e2b504cd497b671fc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-05 03:08:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff8e65bd62610541b58e5da5d961fbf084c91cdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 03:08:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec1c2a19bf87909fd3e02b012cff49b18cd07764","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:42:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/34/669134/1/gate/openstack-tox-docs/b57162e/html/ : SUCCESS in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/34/669134/1/gate/openstack-tox-pep8/6aa75db/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/34/669134/1/gate/openstack-tox-py27/5fb8270/ : SUCCESS in 2m 54s\n- openstack-tox-py36 http://logs.openstack.org/34/669134/1/gate/openstack-tox-py36/6313811/ : SUCCESS in 3m 18s\n- openstack-tox-validate http://logs.openstack.org/34/669134/1/gate/openstack-tox-validate/43f9dfa/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/34/669134/1/gate/requirements-tox-babel/8c40bbf/ : SUCCESS in 3m 26s\n- requirements-tox-py27-check-uc http://logs.openstack.org/34/669134/1/gate/requirements-tox-py27-check-uc/ca3a070/ : SUCCESS in 6m 55s\n- requirements-tox-py36-check-uc http://logs.openstack.org/34/669134/1/gate/requirements-tox-py36-check-uc/5442f59/ : SUCCESS in 6m 16s\n- cross-cinder-py27 http://logs.openstack.org/34/669134/1/gate/cross-cinder-py27/bc95486/ : SUCCESS in 10m 29s\n- cross-cinder-py36 http://logs.openstack.org/34/669134/1/gate/cross-cinder-py36/4e20bb7/ : SUCCESS in 9m 53s\n- cross-glance-py27 http://logs.openstack.org/34/669134/1/gate/cross-glance-py27/27e34c9/ : SUCCESS in 3m 58s\n- cross-glance-py36 http://logs.openstack.org/34/669134/1/gate/cross-glance-py36/2a86d2c/ : SUCCESS in 4m 33s\n- cross-horizon-py27 http://logs.openstack.org/34/669134/1/gate/cross-horizon-py27/1ce4fe7/ : SUCCESS in 11m 58s\n- cross-horizon-py36 http://logs.openstack.org/34/669134/1/gate/cross-horizon-py36/2c02b48/ : SUCCESS in 9m 29s\n- cross-horizon-npm http://logs.openstack.org/34/669134/1/gate/cross-horizon-npm/2ab70f8/ : SUCCESS in 6m 18s\n- cross-keystone-py27 http://logs.openstack.org/34/669134/1/gate/cross-keystone-py27/bba7a91/ : SUCCESS in 25m 50s\n- cross-keystone-py36 http://logs.openstack.org/34/669134/1/gate/cross-keystone-py36/77cc80b/ : SUCCESS in 29m 38s\n- cross-neutron-py27 http://logs.openstack.org/34/669134/1/gate/cross-neutron-py27/2bf660a/ : SUCCESS in 25m 40s\n- cross-neutron-py36 http://logs.openstack.org/34/669134/1/gate/cross-neutron-py36/b8f7882/ : SUCCESS in 21m 50s\n- cross-nova-py27 http://logs.openstack.org/34/669134/1/gate/cross-nova-py27/0afe93a/ : SUCCESS in 13m 57s\n- cross-nova-py36 http://logs.openstack.org/34/669134/1/gate/cross-nova-py36/5a5c5ee/ : SUCCESS in 11m 28s\n- cross-nova-functional http://logs.openstack.org/34/669134/1/gate/cross-nova-functional/fd28b84/ : SUCCESS in 17m 20s\n- cross-osc-tox-docs http://logs.openstack.org/34/669134/1/gate/cross-osc-tox-docs/8292669/html/ : SUCCESS in 3m 47s\n- cross-swift-py27 http://logs.openstack.org/34/669134/1/gate/cross-swift-py27/e15e197/ : SUCCESS in 8m 00s\n- cross-swift-py36 http://logs.openstack.org/34/669134/1/gate/cross-swift-py36/a28dd59/ : SUCCESS in 7m 52s\n- requirements-integration http://logs.openstack.org/34/669134/1/gate/requirements-integration/7762cac/ : SUCCESS in 55m 07s\n- neutron-grenade-multinode http://logs.openstack.org/34/669134/1/gate/neutron-grenade-multinode/0272fb5/ : SUCCESS in 1h 12m 19s\n- tempest-full http://logs.openstack.org/34/669134/1/gate/tempest-full/ed67528/ : SUCCESS in 1h 44m 42s\n- neutron-grenade http://logs.openstack.org/34/669134/1/gate/neutron-grenade/c4f9107/ : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":1},{"id":"63a0b4fed4d10f4c0bb4e72cb1fd5e133301d397","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:42:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"537d05ef32128813951d351f34b6064a2fbb61c0","revisions":{"537d05ef32128813951d351f34b6064a2fbb61c0":{"kind":"REWORK","_number":1,"created":"2019-07-04 10:25:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/669134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/669134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/669134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/669134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/669134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/669134/1"}}},"commit":{"parents":[{"commit":"526227ae9ebee78cf655de44368482f8aabfae2e","subject":"update constraint for neutron-lib to new release 1.28.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/526227ae9ebee78cf655de44368482f8aabfae2e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-04 10:25:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-04 10:25:53.000000000","tz":0},"subject":"update constraint for python-qinlingclient to new release 2.2.0","message":"update constraint for python-qinlingclient to new release 2.2.0\n\nChange-Id: I6dd55d56d31d7d12cb9c2225747df55a8272e834\nmeta:version: 2.2.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Commit: Lingxian Kong \u003canlin.kong@gmail.com\u003e\nmeta:release:Change-Id: Ie283fc131be43a0a33efe94a61529eb1effd2278\nmeta:release:Code-Review+1: Gaëtan Trellu \u003cgaetan.trellu@incloudus.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/537d05ef32128813951d351f34b6064a2fbb61c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/537d05ef32128813951d351f34b6064a2fbb61c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
