)]}'
{"id":"openstack%2Frequirements~669135","triplet_id":"openstack%2Frequirements~master~I5260bf2aea190c556fa885dd4a0423bf429c266a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5260bf2aea190c556fa885dd4a0423bf429c266a","subject":"update constraint for oslo.config to new release 6.11.0","status":"MERGED","created":"2019-07-04 10:25:59.000000000","updated":"2019-07-05 06:42:27.000000000","submitted":"2019-07-05 06:42:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669135-1562308947072-769ffea3","meta_rev_id":"77e572115e91582bb15b6cd2413c56a6ec743aa0","_number":669135,"virtual_id_number":669135,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-05 06:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-05 03:08:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-05 03:08:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 03:08:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-05 06:42:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4f42612480ecfdc4aa159818074cfdea0c4f2d3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-04 10:25:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"865b9844969bf7281dc950c5af17382d5f2f8801","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 18:39:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/35/669135/1/check/openstack-tox-docs/e065746/html/ : SUCCESS in 3m 02s\n- openstack-tox-pypy http://logs.openstack.org/35/669135/1/check/openstack-tox-pypy/ead6baa/ : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/35/669135/1/check/openstack-tox-pep8/5387763/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/35/669135/1/check/openstack-tox-py27/869aaaa/ : SUCCESS in 3m 27s\n- openstack-tox-py36 http://logs.openstack.org/35/669135/1/check/openstack-tox-py36/906ab03/ : SUCCESS in 4m 09s\n- openstack-tox-validate http://logs.openstack.org/35/669135/1/check/openstack-tox-validate/f23e42c/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/35/669135/1/check/requirements-tox-babel/2a52f38/ : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc http://logs.openstack.org/35/669135/1/check/requirements-tox-py27-check-uc/3c8f661/ : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/35/669135/1/check/requirements-tox-py36-check-uc/dfaa1cf/ : SUCCESS in 6m 14s\n- cross-cinder-py27 http://logs.openstack.org/35/669135/1/check/cross-cinder-py27/811ce2e/ : SUCCESS in 9m 12s\n- cross-cinder-py36 http://logs.openstack.org/35/669135/1/check/cross-cinder-py36/51b5dbb/ : SUCCESS in 8m 43s\n- cross-glance-py27 http://logs.openstack.org/35/669135/1/check/cross-glance-py27/ddc9e2e/ : SUCCESS in 4m 01s\n- cross-glance-py36 http://logs.openstack.org/35/669135/1/check/cross-glance-py36/54d0572/ : SUCCESS in 4m 28s\n- cross-horizon-py27 http://logs.openstack.org/35/669135/1/check/cross-horizon-py27/58eea42/ : SUCCESS in 10m 26s\n- cross-horizon-py36 http://logs.openstack.org/35/669135/1/check/cross-horizon-py36/514ffc0/ : SUCCESS in 9m 43s\n- cross-horizon-npm http://logs.openstack.org/35/669135/1/check/cross-horizon-npm/200831c/ : SUCCESS in 5m 53s\n- cross-keystone-py27 http://logs.openstack.org/35/669135/1/check/cross-keystone-py27/94956dd/ : SUCCESS in 31m 53s\n- cross-keystone-py36 http://logs.openstack.org/35/669135/1/check/cross-keystone-py36/0163937/ : SUCCESS in 34m 52s\n- cross-neutron-py27 http://logs.openstack.org/35/669135/1/check/cross-neutron-py27/c6bf471/ : SUCCESS in 19m 46s\n- cross-neutron-py36 http://logs.openstack.org/35/669135/1/check/cross-neutron-py36/c5879be/ : SUCCESS in 24m 10s\n- cross-nova-py27 http://logs.openstack.org/35/669135/1/check/cross-nova-py27/79e5b79/ : SUCCESS in 13m 31s\n- cross-nova-py36 http://logs.openstack.org/35/669135/1/check/cross-nova-py36/dc1a26d/ : SUCCESS in 10m 22s\n- cross-nova-functional http://logs.openstack.org/35/669135/1/check/cross-nova-functional/579d9c0/ : SUCCESS in 18m 35s\n- cross-osc-tox-docs http://logs.openstack.org/35/669135/1/check/cross-osc-tox-docs/6638831/html/ : SUCCESS in 4m 40s\n- cross-swift-py27 http://logs.openstack.org/35/669135/1/check/cross-swift-py27/b91a240/ : SUCCESS in 9m 02s\n- cross-swift-py36 http://logs.openstack.org/35/669135/1/check/cross-swift-py36/859a93e/ : SUCCESS in 8m 24s\n- requirements-integration http://logs.openstack.org/35/669135/1/check/requirements-integration/f9bb6cd/ : SUCCESS in 53m 52s\n- neutron-grenade-multinode http://logs.openstack.org/35/669135/1/check/neutron-grenade-multinode/5369d24/ : SUCCESS in 1h 07m 44s\n- tempest-full http://logs.openstack.org/35/669135/1/check/tempest-full/a43e1b4/ : SUCCESS in 1h 48m 44s\n- neutron-grenade http://logs.openstack.org/35/669135/1/check/neutron-grenade/a646253/ : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":1},{"id":"979a13cf88d0b2adec778e18ac4431f4cda3a543","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-05 03:08:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e05eead11e094b58c55d17d7fcca9807e95abb79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 03:08:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"852b1879ddee5f1a9de81c413cf0a32fe44ce619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:42:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/35/669135/1/gate/openstack-tox-docs/1123a74/html/ : SUCCESS in 3m 01s\n- openstack-tox-pep8 http://logs.openstack.org/35/669135/1/gate/openstack-tox-pep8/4fff256/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/35/669135/1/gate/openstack-tox-py27/61bb722/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/35/669135/1/gate/openstack-tox-py36/4b8814d/ : SUCCESS in 4m 07s\n- openstack-tox-validate http://logs.openstack.org/35/669135/1/gate/openstack-tox-validate/3f37da5/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/35/669135/1/gate/requirements-tox-babel/6941b71/ : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/35/669135/1/gate/requirements-tox-py27-check-uc/a94729c/ : SUCCESS in 7m 34s\n- requirements-tox-py36-check-uc http://logs.openstack.org/35/669135/1/gate/requirements-tox-py36-check-uc/66a7340/ : SUCCESS in 9m 38s\n- cross-cinder-py27 http://logs.openstack.org/35/669135/1/gate/cross-cinder-py27/584d2a3/ : SUCCESS in 10m 57s\n- cross-cinder-py36 http://logs.openstack.org/35/669135/1/gate/cross-cinder-py36/07397b6/ : SUCCESS in 8m 54s\n- cross-glance-py27 http://logs.openstack.org/35/669135/1/gate/cross-glance-py27/e83374c/ : SUCCESS in 4m 27s\n- cross-glance-py36 http://logs.openstack.org/35/669135/1/gate/cross-glance-py36/e88e126/ : SUCCESS in 4m 59s\n- cross-horizon-py27 http://logs.openstack.org/35/669135/1/gate/cross-horizon-py27/db65f90/ : SUCCESS in 9m 38s\n- cross-horizon-py36 http://logs.openstack.org/35/669135/1/gate/cross-horizon-py36/3764343/ : SUCCESS in 7m 59s\n- cross-horizon-npm http://logs.openstack.org/35/669135/1/gate/cross-horizon-npm/dc7306f/ : SUCCESS in 7m 23s\n- cross-keystone-py27 http://logs.openstack.org/35/669135/1/gate/cross-keystone-py27/6eb1cdb/ : SUCCESS in 30m 21s\n- cross-keystone-py36 http://logs.openstack.org/35/669135/1/gate/cross-keystone-py36/41438de/ : SUCCESS in 30m 04s\n- cross-neutron-py27 http://logs.openstack.org/35/669135/1/gate/cross-neutron-py27/16ae58c/ : SUCCESS in 21m 41s\n- cross-neutron-py36 http://logs.openstack.org/35/669135/1/gate/cross-neutron-py36/f2fe00c/ : SUCCESS in 19m 14s\n- cross-nova-py27 http://logs.openstack.org/35/669135/1/gate/cross-nova-py27/5ace3db/ : SUCCESS in 13m 00s\n- cross-nova-py36 http://logs.openstack.org/35/669135/1/gate/cross-nova-py36/fcbd430/ : SUCCESS in 14m 57s\n- cross-nova-functional http://logs.openstack.org/35/669135/1/gate/cross-nova-functional/1517158/ : SUCCESS in 21m 24s\n- cross-osc-tox-docs http://logs.openstack.org/35/669135/1/gate/cross-osc-tox-docs/126f1c2/html/ : SUCCESS in 3m 49s\n- cross-swift-py27 http://logs.openstack.org/35/669135/1/gate/cross-swift-py27/9f26a4d/ : SUCCESS in 8m 00s\n- cross-swift-py36 http://logs.openstack.org/35/669135/1/gate/cross-swift-py36/2d8b5a5/ : SUCCESS in 9m 47s\n- requirements-integration http://logs.openstack.org/35/669135/1/gate/requirements-integration/8f71c73/ : SUCCESS in 53m 40s\n- neutron-grenade-multinode http://logs.openstack.org/35/669135/1/gate/neutron-grenade-multinode/cd2010d/ : SUCCESS in 1h 16m 32s\n- tempest-full http://logs.openstack.org/35/669135/1/gate/tempest-full/087ae8a/ : SUCCESS in 1h 44m 10s\n- neutron-grenade http://logs.openstack.org/35/669135/1/gate/neutron-grenade/c935fe6/ : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":1},{"id":"3ef1776163fbaeec679c1095d66e6ae0acc372b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:42:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97432fbad31603f3d99adbaf41d763eea10ab1e8","revisions":{"97432fbad31603f3d99adbaf41d763eea10ab1e8":{"kind":"REWORK","_number":1,"created":"2019-07-04 10:25:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/669135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/669135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/669135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/669135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/669135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/669135/1"}}},"commit":{"parents":[{"commit":"526227ae9ebee78cf655de44368482f8aabfae2e","subject":"update constraint for neutron-lib to new release 1.28.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/526227ae9ebee78cf655de44368482f8aabfae2e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-04 10:25:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-04 10:25:54.000000000","tz":0},"subject":"update constraint for oslo.config to new release 6.11.0","message":"update constraint for oslo.config to new release 6.11.0\n\nChange-Id: I5260bf2aea190c556fa885dd4a0423bf429c266a\nmeta:version: 6.11.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I216cfb0c14af6f88b3a9a2210ad8b03c088440fe\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97432fbad31603f3d99adbaf41d763eea10ab1e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/97432fbad31603f3d99adbaf41d763eea10ab1e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
