)]}'
{"id":"openstack%2Frequirements~669138","triplet_id":"openstack%2Frequirements~master~I52cf003d36ba7b57bd1e5f754def2a3fbc54e715","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I52cf003d36ba7b57bd1e5f754def2a3fbc54e715","subject":"update constraint for oslo.upgradecheck to new release 0.3.0","status":"MERGED","created":"2019-07-04 10:28:49.000000000","updated":"2019-07-05 06:42:26.000000000","submitted":"2019-07-05 06:42:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669138-1562308946075-eaa21e4c","meta_rev_id":"a43e027d4937696e51b3ccfd41a95acdae0a8b86","_number":669138,"virtual_id_number":669138,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-05 06:42:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-05 03:08:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-05 03:08:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 03:08:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-05 06:42:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bc9e1c90b919662c9e14d2f1c0499b1af632b5f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-04 10:28:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"228ce483b0b63021c3dfe6431ad767f0843f9189","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 18:55:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/669138/1/check/openstack-tox-docs/12cce0b/html/ : SUCCESS in 3m 13s\n- openstack-tox-pypy http://logs.openstack.org/38/669138/1/check/openstack-tox-pypy/d280b87/ : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/38/669138/1/check/openstack-tox-pep8/8e86316/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/38/669138/1/check/openstack-tox-py27/cd489e6/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/38/669138/1/check/openstack-tox-py36/365fe2f/ : SUCCESS in 3m 15s\n- openstack-tox-validate http://logs.openstack.org/38/669138/1/check/openstack-tox-validate/c1a759b/ : SUCCESS in 3m 03s\n- requirements-tox-babel http://logs.openstack.org/38/669138/1/check/requirements-tox-babel/a9bd984/ : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/669138/1/check/requirements-tox-py27-check-uc/96e331c/ : SUCCESS in 6m 08s\n- requirements-tox-py36-check-uc http://logs.openstack.org/38/669138/1/check/requirements-tox-py36-check-uc/5d34129/ : SUCCESS in 6m 18s\n- cross-cinder-py27 http://logs.openstack.org/38/669138/1/check/cross-cinder-py27/655e427/ : SUCCESS in 8m 16s\n- cross-cinder-py36 http://logs.openstack.org/38/669138/1/check/cross-cinder-py36/e90878c/ : SUCCESS in 7m 57s\n- cross-glance-py27 http://logs.openstack.org/38/669138/1/check/cross-glance-py27/d1c11c2/ : SUCCESS in 3m 56s\n- cross-glance-py36 http://logs.openstack.org/38/669138/1/check/cross-glance-py36/985186f/ : SUCCESS in 5m 00s\n- cross-horizon-py27 http://logs.openstack.org/38/669138/1/check/cross-horizon-py27/1921cda/ : SUCCESS in 12m 30s\n- cross-horizon-py36 http://logs.openstack.org/38/669138/1/check/cross-horizon-py36/ee6ef94/ : SUCCESS in 6m 57s\n- cross-horizon-npm http://logs.openstack.org/38/669138/1/check/cross-horizon-npm/cf85e2d/ : SUCCESS in 7m 12s\n- cross-keystone-py27 http://logs.openstack.org/38/669138/1/check/cross-keystone-py27/ef4af24/ : SUCCESS in 28m 20s\n- cross-keystone-py36 http://logs.openstack.org/38/669138/1/check/cross-keystone-py36/259abf7/ : SUCCESS in 30m 05s\n- cross-neutron-py27 http://logs.openstack.org/38/669138/1/check/cross-neutron-py27/51e9542/ : SUCCESS in 26m 35s\n- cross-neutron-py36 http://logs.openstack.org/38/669138/1/check/cross-neutron-py36/fc7b334/ : SUCCESS in 24m 34s\n- cross-nova-py27 http://logs.openstack.org/38/669138/1/check/cross-nova-py27/6b00df5/ : SUCCESS in 11m 57s\n- cross-nova-py36 http://logs.openstack.org/38/669138/1/check/cross-nova-py36/406a419/ : SUCCESS in 12m 28s\n- cross-nova-functional http://logs.openstack.org/38/669138/1/check/cross-nova-functional/5bba45c/ : SUCCESS in 18m 38s\n- cross-osc-tox-docs http://logs.openstack.org/38/669138/1/check/cross-osc-tox-docs/c074fb0/html/ : SUCCESS in 3m 44s\n- cross-swift-py27 http://logs.openstack.org/38/669138/1/check/cross-swift-py27/f90649e/ : SUCCESS in 8m 46s\n- cross-swift-py36 http://logs.openstack.org/38/669138/1/check/cross-swift-py36/93210ac/ : SUCCESS in 8m 09s\n- requirements-integration http://logs.openstack.org/38/669138/1/check/requirements-integration/c824825/ : SUCCESS in 1h 13m 40s\n- neutron-grenade-multinode http://logs.openstack.org/38/669138/1/check/neutron-grenade-multinode/eef91cf/ : SUCCESS in 1h 09m 23s\n- tempest-full http://logs.openstack.org/38/669138/1/check/tempest-full/3361565/ : SUCCESS in 1h 56m 32s\n- neutron-grenade http://logs.openstack.org/38/669138/1/check/neutron-grenade/25b0666/ : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"996347dee4343e83302646caa582b9be94c09499","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-05 03:08:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eba6bbe2fb217b44ff21c58eb3d04482708942d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 03:08:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"63a399dfb785442ee53d9d3fd199bb6a7e2eda69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:42:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/669138/1/gate/openstack-tox-docs/b5cbdc1/html/ : SUCCESS in 3m 07s\n- openstack-tox-pep8 http://logs.openstack.org/38/669138/1/gate/openstack-tox-pep8/05731df/ : SUCCESS in 3m 06s\n- openstack-tox-py27 http://logs.openstack.org/38/669138/1/gate/openstack-tox-py27/24c049c/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/38/669138/1/gate/openstack-tox-py36/f143c8e/ : SUCCESS in 4m 26s\n- openstack-tox-validate http://logs.openstack.org/38/669138/1/gate/openstack-tox-validate/4c58253/ : SUCCESS in 2m 57s\n- requirements-tox-babel http://logs.openstack.org/38/669138/1/gate/requirements-tox-babel/a1cda66/ : SUCCESS in 2m 55s\n- requirements-tox-py27-check-uc http://logs.openstack.org/38/669138/1/gate/requirements-tox-py27-check-uc/8f93b5e/ : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc http://logs.openstack.org/38/669138/1/gate/requirements-tox-py36-check-uc/29f18b1/ : SUCCESS in 11m 03s\n- cross-cinder-py27 http://logs.openstack.org/38/669138/1/gate/cross-cinder-py27/900aa84/ : SUCCESS in 8m 34s\n- cross-cinder-py36 http://logs.openstack.org/38/669138/1/gate/cross-cinder-py36/d990a08/ : SUCCESS in 7m 56s\n- cross-glance-py27 http://logs.openstack.org/38/669138/1/gate/cross-glance-py27/e4cfe18/ : SUCCESS in 4m 13s\n- cross-glance-py36 http://logs.openstack.org/38/669138/1/gate/cross-glance-py36/34cda8a/ : SUCCESS in 3m 56s\n- cross-horizon-py27 http://logs.openstack.org/38/669138/1/gate/cross-horizon-py27/fa56b50/ : SUCCESS in 10m 18s\n- cross-horizon-py36 http://logs.openstack.org/38/669138/1/gate/cross-horizon-py36/cf66f33/ : SUCCESS in 7m 46s\n- cross-horizon-npm http://logs.openstack.org/38/669138/1/gate/cross-horizon-npm/028be02/ : SUCCESS in 5m 58s\n- cross-keystone-py27 http://logs.openstack.org/38/669138/1/gate/cross-keystone-py27/00f48dc/ : SUCCESS in 22m 09s\n- cross-keystone-py36 http://logs.openstack.org/38/669138/1/gate/cross-keystone-py36/69a3d68/ : SUCCESS in 29m 30s\n- cross-neutron-py27 http://logs.openstack.org/38/669138/1/gate/cross-neutron-py27/61447f7/ : SUCCESS in 27m 17s\n- cross-neutron-py36 http://logs.openstack.org/38/669138/1/gate/cross-neutron-py36/41aed77/ : SUCCESS in 19m 16s\n- cross-nova-py27 http://logs.openstack.org/38/669138/1/gate/cross-nova-py27/0d46fbc/ : SUCCESS in 14m 21s\n- cross-nova-py36 http://logs.openstack.org/38/669138/1/gate/cross-nova-py36/a3c5e7c/ : SUCCESS in 13m 30s\n- cross-nova-functional http://logs.openstack.org/38/669138/1/gate/cross-nova-functional/538260f/ : SUCCESS in 17m 05s\n- cross-osc-tox-docs http://logs.openstack.org/38/669138/1/gate/cross-osc-tox-docs/65d2ec6/html/ : SUCCESS in 4m 32s\n- cross-swift-py27 http://logs.openstack.org/38/669138/1/gate/cross-swift-py27/2920f6c/ : SUCCESS in 8m 35s\n- cross-swift-py36 http://logs.openstack.org/38/669138/1/gate/cross-swift-py36/6c93ff2/ : SUCCESS in 7m 58s\n- requirements-integration http://logs.openstack.org/38/669138/1/gate/requirements-integration/813a6cb/ : SUCCESS in 55m 17s\n- neutron-grenade-multinode http://logs.openstack.org/38/669138/1/gate/neutron-grenade-multinode/8ed8a07/ : SUCCESS in 1h 08m 56s\n- tempest-full http://logs.openstack.org/38/669138/1/gate/tempest-full/66b101e/ : SUCCESS in 1h 49m 16s\n- neutron-grenade http://logs.openstack.org/38/669138/1/gate/neutron-grenade/7fe473f/ : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":1},{"id":"46fc8cdee390aafe9378989c2712504998a5ca6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 06:42:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af8115a90668bd305944c59d585bed051929ac48","revisions":{"af8115a90668bd305944c59d585bed051929ac48":{"kind":"REWORK","_number":1,"created":"2019-07-04 10:28:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/38/669138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/669138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/669138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/669138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/669138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/669138/1"}}},"commit":{"parents":[{"commit":"526227ae9ebee78cf655de44368482f8aabfae2e","subject":"update constraint for neutron-lib to new release 1.28.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/526227ae9ebee78cf655de44368482f8aabfae2e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-04 10:28:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-04 10:28:47.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 0.3.0","message":"update constraint for oslo.upgradecheck to new release 0.3.0\n\nChange-Id: I52cf003d36ba7b57bd1e5f754def2a3fbc54e715\nmeta:version: 0.3.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Commit: Ben Nemec \u003cbnemec@redhat.com\u003e\nmeta:release:Change-Id: I216cfb0c14af6f88b3a9a2210ad8b03c088440fe\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af8115a90668bd305944c59d585bed051929ac48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/af8115a90668bd305944c59d585bed051929ac48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
