)]}'
{"id":"openstack%2Frequirements~669350","triplet_id":"openstack%2Frequirements~master~Ib3027afefa30ab05611807f80d9b88ec76e28558","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib3027afefa30ab05611807f80d9b88ec76e28558","subject":"update constraint for oslotest to new release 3.8.0","status":"MERGED","created":"2019-07-05 12:58:27.000000000","updated":"2019-07-05 18:25:12.000000000","submitted":"2019-07-05 18:25:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669350-1562351112239-31ef7618","meta_rev_id":"585da701f127c2c657202909396e9ce5e09dcc2e","_number":669350,"virtual_id_number":669350,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-05 18:25:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-05 15:51:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-05 15:51:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 15:51:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-05 18:25:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d562695cf5372a154b01ed820a0f38b1f8d68ea8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-05 12:58:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8fe3234d155a3c492a78b84b642d1418715e2ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 15:45:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/669350/1/check/openstack-tox-docs/aacb24d/html/ : SUCCESS in 4m 21s\n- openstack-tox-pypy http://logs.openstack.org/50/669350/1/check/openstack-tox-pypy/fe30150/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/50/669350/1/check/openstack-tox-pep8/5aa1f04/ : SUCCESS in 3m 04s\n- openstack-tox-py27 http://logs.openstack.org/50/669350/1/check/openstack-tox-py27/891e98c/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/50/669350/1/check/openstack-tox-py36/a42256a/ : SUCCESS in 3m 23s\n- openstack-tox-validate http://logs.openstack.org/50/669350/1/check/openstack-tox-validate/09f5b71/ : SUCCESS in 3m 13s\n- requirements-tox-babel http://logs.openstack.org/50/669350/1/check/requirements-tox-babel/fc99636/ : SUCCESS in 3m 16s\n- requirements-tox-py27-check-uc http://logs.openstack.org/50/669350/1/check/requirements-tox-py27-check-uc/fbf684c/ : SUCCESS in 6m 12s\n- requirements-tox-py36-check-uc http://logs.openstack.org/50/669350/1/check/requirements-tox-py36-check-uc/afe1e81/ : SUCCESS in 6m 28s\n- cross-cinder-py27 http://logs.openstack.org/50/669350/1/check/cross-cinder-py27/8242d51/ : SUCCESS in 9m 38s\n- cross-cinder-py36 http://logs.openstack.org/50/669350/1/check/cross-cinder-py36/f0ff89b/ : SUCCESS in 10m 05s\n- cross-glance-py27 http://logs.openstack.org/50/669350/1/check/cross-glance-py27/c99d945/ : SUCCESS in 5m 48s\n- cross-glance-py36 http://logs.openstack.org/50/669350/1/check/cross-glance-py36/e7196cf/ : SUCCESS in 6m 30s\n- cross-horizon-py27 http://logs.openstack.org/50/669350/1/check/cross-horizon-py27/6a1f27a/ : SUCCESS in 9m 55s\n- cross-horizon-py36 http://logs.openstack.org/50/669350/1/check/cross-horizon-py36/5cb4b47/ : SUCCESS in 8m 44s\n- cross-horizon-npm http://logs.openstack.org/50/669350/1/check/cross-horizon-npm/74facf3/ : SUCCESS in 8m 01s\n- cross-keystone-py27 http://logs.openstack.org/50/669350/1/check/cross-keystone-py27/815227f/ : SUCCESS in 28m 27s\n- cross-keystone-py36 http://logs.openstack.org/50/669350/1/check/cross-keystone-py36/4448a75/ : SUCCESS in 31m 11s\n- cross-neutron-py27 http://logs.openstack.org/50/669350/1/check/cross-neutron-py27/da7e212/ : SUCCESS in 25m 23s\n- cross-neutron-py36 http://logs.openstack.org/50/669350/1/check/cross-neutron-py36/42782da/ : SUCCESS in 23m 36s\n- cross-nova-py27 http://logs.openstack.org/50/669350/1/check/cross-nova-py27/3890ead/ : SUCCESS in 11m 22s\n- cross-nova-py36 http://logs.openstack.org/50/669350/1/check/cross-nova-py36/906055d/ : SUCCESS in 12m 42s\n- cross-nova-functional http://logs.openstack.org/50/669350/1/check/cross-nova-functional/4100656/ : SUCCESS in 20m 31s\n- cross-osc-tox-docs http://logs.openstack.org/50/669350/1/check/cross-osc-tox-docs/e683152/html/ : SUCCESS in 4m 31s\n- cross-swift-py27 http://logs.openstack.org/50/669350/1/check/cross-swift-py27/ec809a5/ : SUCCESS in 7m 38s\n- cross-swift-py36 http://logs.openstack.org/50/669350/1/check/cross-swift-py36/1f4f410/ : SUCCESS in 7m 56s\n- requirements-integration http://logs.openstack.org/50/669350/1/check/requirements-integration/3265bfa/ : SUCCESS in 59m 53s\n- neutron-grenade-multinode http://logs.openstack.org/50/669350/1/check/neutron-grenade-multinode/38092aa/ : SUCCESS in 1h 14m 40s\n- tempest-full http://logs.openstack.org/50/669350/1/check/tempest-full/dd3eca9/ : SUCCESS in 1h 50m 21s\n- neutron-grenade http://logs.openstack.org/50/669350/1/check/neutron-grenade/1ac29bb/ : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":1},{"id":"132a35742993db1d14db3955f5e752fcec6d30ef","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-05 15:51:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e05cd043a6a2835061a206dd3d38db4bb958b98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 15:51:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db3a28f3a8259f4849c64668fd3c445a3a74135d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 18:25:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/50/669350/1/gate/openstack-tox-docs/e5ea3b8/html/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/50/669350/1/gate/openstack-tox-pep8/dea5cf0/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/50/669350/1/gate/openstack-tox-py27/6e83260/ : SUCCESS in 3m 39s\n- openstack-tox-py36 http://logs.openstack.org/50/669350/1/gate/openstack-tox-py36/cb759f6/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/50/669350/1/gate/openstack-tox-validate/1244771/ : SUCCESS in 3m 36s\n- requirements-tox-babel http://logs.openstack.org/50/669350/1/gate/requirements-tox-babel/257a9c7/ : SUCCESS in 2m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/50/669350/1/gate/requirements-tox-py27-check-uc/f3c37a1/ : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc http://logs.openstack.org/50/669350/1/gate/requirements-tox-py36-check-uc/9b2681c/ : SUCCESS in 6m 23s\n- cross-cinder-py27 http://logs.openstack.org/50/669350/1/gate/cross-cinder-py27/1881e5a/ : SUCCESS in 12m 31s\n- cross-cinder-py36 http://logs.openstack.org/50/669350/1/gate/cross-cinder-py36/ed324c8/ : SUCCESS in 12m 07s\n- cross-glance-py27 http://logs.openstack.org/50/669350/1/gate/cross-glance-py27/2299aa8/ : SUCCESS in 4m 22s\n- cross-glance-py36 http://logs.openstack.org/50/669350/1/gate/cross-glance-py36/04aded2/ : SUCCESS in 4m 02s\n- cross-horizon-py27 http://logs.openstack.org/50/669350/1/gate/cross-horizon-py27/35cf656/ : SUCCESS in 10m 55s\n- cross-horizon-py36 http://logs.openstack.org/50/669350/1/gate/cross-horizon-py36/0fdf92a/ : SUCCESS in 14m 04s\n- cross-horizon-npm http://logs.openstack.org/50/669350/1/gate/cross-horizon-npm/8aec846/ : SUCCESS in 11m 06s\n- cross-keystone-py27 http://logs.openstack.org/50/669350/1/gate/cross-keystone-py27/500db6d/ : SUCCESS in 27m 39s\n- cross-keystone-py36 http://logs.openstack.org/50/669350/1/gate/cross-keystone-py36/4e4d5ac/ : SUCCESS in 29m 15s\n- cross-neutron-py27 http://logs.openstack.org/50/669350/1/gate/cross-neutron-py27/8706121/ : SUCCESS in 24m 58s\n- cross-neutron-py36 http://logs.openstack.org/50/669350/1/gate/cross-neutron-py36/55bd7f5/ : SUCCESS in 22m 45s\n- cross-nova-py27 http://logs.openstack.org/50/669350/1/gate/cross-nova-py27/0ebf86a/ : SUCCESS in 14m 21s\n- cross-nova-py36 http://logs.openstack.org/50/669350/1/gate/cross-nova-py36/219a218/ : SUCCESS in 11m 24s\n- cross-nova-functional http://logs.openstack.org/50/669350/1/gate/cross-nova-functional/3f7cac5/ : SUCCESS in 23m 54s\n- cross-osc-tox-docs http://logs.openstack.org/50/669350/1/gate/cross-osc-tox-docs/e41b368/html/ : SUCCESS in 6m 18s\n- cross-swift-py27 http://logs.openstack.org/50/669350/1/gate/cross-swift-py27/65605cd/ : SUCCESS in 8m 19s\n- cross-swift-py36 http://logs.openstack.org/50/669350/1/gate/cross-swift-py36/196ef00/ : SUCCESS in 9m 01s\n- requirements-integration http://logs.openstack.org/50/669350/1/gate/requirements-integration/938cf2f/ : SUCCESS in 59m 38s\n- neutron-grenade-multinode http://logs.openstack.org/50/669350/1/gate/neutron-grenade-multinode/49d6273/ : SUCCESS in 1h 23m 08s\n- tempest-full http://logs.openstack.org/50/669350/1/gate/tempest-full/ca1a276/ : SUCCESS in 1h 43m 42s\n- neutron-grenade http://logs.openstack.org/50/669350/1/gate/neutron-grenade/d10d286/ : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":1},{"id":"114e3c81483be942fc8f00c202b147b52c083a7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 18:25:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12d99428a0101c99fdd9a0b40b4fce51fe6eb4a9","revisions":{"12d99428a0101c99fdd9a0b40b4fce51fe6eb4a9":{"kind":"REWORK","_number":1,"created":"2019-07-05 12:58:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/669350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/669350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/669350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/669350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/669350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/669350/1"}}},"commit":{"parents":[{"commit":"4a2a45ee5bdb81ef46359ab347d3b80f18c10207","subject":"Merge \"update constraint for pbr to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a2a45ee5bdb81ef46359ab347d3b80f18c10207"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-05 12:58:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-05 12:58:25.000000000","tz":0},"subject":"update constraint for oslotest to new release 3.8.0","message":"update constraint for oslotest to new release 3.8.0\n\nChange-Id: Ib3027afefa30ab05611807f80d9b88ec76e28558\nmeta:version: 3.8.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: yes\nmeta:release:Author: Thomas Bechtold \u003ctbechtold@suse.com\u003e\nmeta:release:Commit: Thomas Bechtold \u003ctbechtold@suse.com\u003e\nmeta:release:Change-Id: I7a2f5abc50361ef4690cf6eb0858dfc909eb45d8\nmeta:release:Code-Review+1: Ben Nemec \u003copenstack@nemebean.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12d99428a0101c99fdd9a0b40b4fce51fe6eb4a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/12d99428a0101c99fdd9a0b40b4fce51fe6eb4a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
