)]}'
{"id":"openstack%2Frequirements~669353","triplet_id":"openstack%2Frequirements~stable%2Fstein~I376d3714fdd8d90b94ed45c6135bce1356617966","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I376d3714fdd8d90b94ed45c6135bce1356617966","subject":"update constraint for tripleo-common to new release 10.8.0","status":"MERGED","created":"2019-07-05 13:05:03.000000000","updated":"2019-07-06 20:55:57.000000000","submitted":"2019-07-06 20:55:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669353-1562446557703-7de5f3e4","meta_rev_id":"37de5cd8f3eac71f8339587bf7d280ccea92a03e","_number":669353,"virtual_id_number":669353,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-06 20:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-05 15:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-06 18:56:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-06 18:56:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-06 20:55:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d0e54cac8a1ca8e18ade401d4f386b648640fb2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-05 13:05:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c62877b8573e8f816589c16f760440ab144b04ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 15:38:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/53/669353/1/check/openstack-tox-pypy/5d80f6d/ : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/53/669353/1/check/openstack-tox-pep8/9cf1c01/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/53/669353/1/check/openstack-tox-py27/256f282/ : SUCCESS in 3m 55s\n- openstack-tox-py35 http://logs.openstack.org/53/669353/1/check/openstack-tox-py35/a5bfc39/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/53/669353/1/check/openstack-tox-py36/ddf3398/ : SUCCESS in 4m 12s\n- openstack-tox-validate http://logs.openstack.org/53/669353/1/check/openstack-tox-validate/1df9365/ : SUCCESS in 3m 42s\n- requirements-tox-babel http://logs.openstack.org/53/669353/1/check/requirements-tox-babel/7c483ce/ : SUCCESS in 3m 27s\n- requirements-tox-py27-check-uc http://logs.openstack.org/53/669353/1/check/requirements-tox-py27-check-uc/b434fa4/ : SUCCESS in 8m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/53/669353/1/check/requirements-tox-py35-check-uc/0ff0547/ : SUCCESS in 7m 49s\n- requirements-tox-py36-check-uc http://logs.openstack.org/53/669353/1/check/requirements-tox-py36-check-uc/9cfc294/ : SUCCESS in 6m 11s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/53/669353/1/check/cross-cinder-py27/a0cab19/ : SUCCESS in 8m 50s\n- cross-cinder-py35 http://logs.openstack.org/53/669353/1/check/cross-cinder-py35/04dc5e9/ : SUCCESS in 12m 23s\n- cross-glance-py27 http://logs.openstack.org/53/669353/1/check/cross-glance-py27/b1582b6/ : SUCCESS in 5m 04s\n- cross-glance-py35 http://logs.openstack.org/53/669353/1/check/cross-glance-py35/b49b5fc/ : SUCCESS in 4m 55s\n- cross-horizon-py27 http://logs.openstack.org/53/669353/1/check/cross-horizon-py27/3449a77/ : SUCCESS in 9m 07s\n- cross-horizon-py35 http://logs.openstack.org/53/669353/1/check/cross-horizon-py35/74f1e4a/ : SUCCESS in 8m 24s\n- cross-horizon-npm http://logs.openstack.org/53/669353/1/check/cross-horizon-npm/21259e5/ : SUCCESS in 7m 11s\n- cross-keystone-py27 http://logs.openstack.org/53/669353/1/check/cross-keystone-py27/5fdcd09/ : SUCCESS in 24m 14s\n- cross-keystone-py35 http://logs.openstack.org/53/669353/1/check/cross-keystone-py35/2b238bc/ : SUCCESS in 26m 41s\n- cross-neutron-py27 http://logs.openstack.org/53/669353/1/check/cross-neutron-py27/109f2fb/ : SUCCESS in 24m 35s\n- cross-neutron-py35 http://logs.openstack.org/53/669353/1/check/cross-neutron-py35/512339b/ : SUCCESS in 26m 31s\n- cross-nova-py27 http://logs.openstack.org/53/669353/1/check/cross-nova-py27/283d6db/ : SUCCESS in 14m 54s\n- cross-nova-py35 http://logs.openstack.org/53/669353/1/check/cross-nova-py35/119471b/ : SUCCESS in 14m 06s\n- cross-nova-functional http://logs.openstack.org/53/669353/1/check/cross-nova-functional/66db988/ : SUCCESS in 19m 37s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/53/669353/1/check/cross-osc-build-sphinx-docs/b228b5d/html/ : SUCCESS in 4m 57s\n- cross-swift-py27 http://logs.openstack.org/53/669353/1/check/cross-swift-py27/9f140e0/ : SUCCESS in 7m 58s\n- cross-swift-py35 http://logs.openstack.org/53/669353/1/check/cross-swift-py35/a010ac6/ : SUCCESS in 8m 50s\n- requirements-integration http://logs.openstack.org/53/669353/1/check/requirements-integration/efc9d64/ : SUCCESS in 50m 34s\n- neutron-grenade-multinode http://logs.openstack.org/53/669353/1/check/neutron-grenade-multinode/981f5be/ : SUCCESS in 1h 07m 50s\n- tempest-full http://logs.openstack.org/53/669353/1/check/tempest-full/9486d73/ : SUCCESS in 1h 40m 53s\n- neutron-grenade http://logs.openstack.org/53/669353/1/check/neutron-grenade/92c3e0f/ : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7fc5a31e0538579ae0303dbc6f9703d59dbb0543","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-05 15:51:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9c4dacd83e9bb0fcc9038db9dfc7e04dc5abef4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-06 18:56:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9a1721b2ea54d18bbf702764697fe4c25a6dc8d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 18:57:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4798373bf52fdf4696d890c74a457e0cf9be0e24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 20:55:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/53/669353/1/gate/openstack-tox-pep8/e212eac/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/53/669353/1/gate/openstack-tox-py27/6a8b0d2/ : SUCCESS in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/53/669353/1/gate/openstack-tox-py35/7945761/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/53/669353/1/gate/openstack-tox-py36/8100b56/ : SUCCESS in 2m 59s\n- openstack-tox-validate http://logs.openstack.org/53/669353/1/gate/openstack-tox-validate/028ae51/ : SUCCESS in 2m 53s\n- requirements-tox-babel http://logs.openstack.org/53/669353/1/gate/requirements-tox-babel/c020bbf/ : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/53/669353/1/gate/requirements-tox-py27-check-uc/7f0d154/ : SUCCESS in 5m 54s\n- requirements-tox-py35-check-uc http://logs.openstack.org/53/669353/1/gate/requirements-tox-py35-check-uc/2e2b692/ : SUCCESS in 9m 27s\n- cross-cinder-py27 http://logs.openstack.org/53/669353/1/gate/cross-cinder-py27/a6dd3f6/ : SUCCESS in 8m 07s\n- cross-cinder-py35 http://logs.openstack.org/53/669353/1/gate/cross-cinder-py35/56f1b8f/ : SUCCESS in 11m 03s\n- cross-glance-py27 http://logs.openstack.org/53/669353/1/gate/cross-glance-py27/b6c4dcb/ : SUCCESS in 3m 43s\n- cross-glance-py35 http://logs.openstack.org/53/669353/1/gate/cross-glance-py35/778ba90/ : SUCCESS in 4m 08s\n- cross-horizon-py27 http://logs.openstack.org/53/669353/1/gate/cross-horizon-py27/3537add/ : SUCCESS in 10m 42s\n- cross-horizon-py35 http://logs.openstack.org/53/669353/1/gate/cross-horizon-py35/3359925/ : SUCCESS in 8m 42s\n- cross-horizon-npm http://logs.openstack.org/53/669353/1/gate/cross-horizon-npm/7756e95/ : SUCCESS in 6m 06s\n- cross-keystone-py27 http://logs.openstack.org/53/669353/1/gate/cross-keystone-py27/7f52121/ : SUCCESS in 20m 04s\n- cross-keystone-py35 http://logs.openstack.org/53/669353/1/gate/cross-keystone-py35/f95f341/ : SUCCESS in 25m 51s\n- cross-neutron-py27 http://logs.openstack.org/53/669353/1/gate/cross-neutron-py27/1464ab1/ : SUCCESS in 22m 04s\n- cross-neutron-py35 http://logs.openstack.org/53/669353/1/gate/cross-neutron-py35/a2716ba/ : SUCCESS in 23m 37s\n- cross-nova-py27 http://logs.openstack.org/53/669353/1/gate/cross-nova-py27/67924b0/ : SUCCESS in 14m 40s\n- cross-nova-py35 http://logs.openstack.org/53/669353/1/gate/cross-nova-py35/7cd1d68/ : SUCCESS in 13m 15s\n- cross-nova-functional http://logs.openstack.org/53/669353/1/gate/cross-nova-functional/b95722a/ : SUCCESS in 20m 30s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/53/669353/1/gate/cross-osc-build-sphinx-docs/348f372/html/ : SUCCESS in 4m 17s\n- cross-swift-py27 http://logs.openstack.org/53/669353/1/gate/cross-swift-py27/3b35113/ : SUCCESS in 7m 07s\n- cross-swift-py35 http://logs.openstack.org/53/669353/1/gate/cross-swift-py35/eef0d10/ : SUCCESS in 7m 46s\n- requirements-integration http://logs.openstack.org/53/669353/1/gate/requirements-integration/ff10c28/ : SUCCESS in 58m 00s\n- neutron-grenade-multinode http://logs.openstack.org/53/669353/1/gate/neutron-grenade-multinode/a8761db/ : SUCCESS in 1h 03m 31s\n- tempest-full http://logs.openstack.org/53/669353/1/gate/tempest-full/bfe7b23/ : SUCCESS in 1h 27m 02s\n- neutron-grenade http://logs.openstack.org/53/669353/1/gate/neutron-grenade/8b6f659/ : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3cc6de01898fb9268a43ff65a226dcd6b5f88fd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 20:55:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd56a6c8b57e27cfd9de342e389a5fd212405873","revisions":{"fd56a6c8b57e27cfd9de342e389a5fd212405873":{"kind":"REWORK","_number":1,"created":"2019-07-05 13:05:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/669353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/669353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/669353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/669353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/669353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/669353/1"}}},"commit":{"parents":[{"commit":"72a83df4edd56fb45d8d66c71e4df56442cc4940","subject":"update constraint for python-saharaclient to new release 2.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/72a83df4edd56fb45d8d66c71e4df56442cc4940"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-05 13:05:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-05 13:05:02.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 10.8.0","message":"update constraint for tripleo-common to new release 10.8.0\n\nChange-Id: I376d3714fdd8d90b94ed45c6135bce1356617966\nmeta:version: 10.8.0\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: yatinkarel \u003cykarel@redhat.com\u003e\nmeta:release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Change-Id: I52718eaef779f7be933f937148ebe6319af2a4a5\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd56a6c8b57e27cfd9de342e389a5fd212405873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd56a6c8b57e27cfd9de342e389a5fd212405873"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
