)]}'
{"id":"openstack%2Frequirements~669488","triplet_id":"openstack%2Frequirements~master~Iab77d920bc5c46a7d7d21c05def60ba278b56983","project":"openstack/requirements","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Iab77d920bc5c46a7d7d21c05def60ba278b56983","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 19:58:26.000000000","updated":"2019-07-06 20:34:44.000000000","submitted":"2019-07-06 20:34:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669488-1562445284005-9c046318","meta_rev_id":"34c2116c78bf207754ecd8ea156366fc074061ff","_number":669488,"virtual_id_number":669488,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-06 20:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-05 21:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-06 18:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-06 18:56:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 21:27:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-06 18:56:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-06 20:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ec63f3678ca9f6041a2cad4d4ab0594609d3583","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 19:58:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44ab2a7e8c9cb73e7a928943b2c760b4aade0c2e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-05 21:27:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29dd804e05c4472fe2a85d8a90e3872511899fd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 22:09:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/88/669488/1/check/test-release-openstack/3ae4679/ : SUCCESS in 3m 15s\n- openstack-tox-docs http://logs.openstack.org/88/669488/1/check/openstack-tox-docs/aa0cb11/html/ : SUCCESS in 3m 39s\n- openstack-tox-pypy http://logs.openstack.org/88/669488/1/check/openstack-tox-pypy/cf42492/ : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/88/669488/1/check/openstack-tox-pep8/06a0f90/ : SUCCESS in 2m 43s\n- openstack-tox-py27 http://logs.openstack.org/88/669488/1/check/openstack-tox-py27/d3fcdb9/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/88/669488/1/check/openstack-tox-py36/fdf2d98/ : SUCCESS in 3m 09s\n- openstack-tox-py37 http://logs.openstack.org/88/669488/1/check/openstack-tox-py37/662d2c5/ : SUCCESS in 3m 12s\n- openstack-tox-validate http://logs.openstack.org/88/669488/1/check/openstack-tox-validate/93ee2f0/ : SUCCESS in 3m 50s\n- requirements-tox-babel http://logs.openstack.org/88/669488/1/check/requirements-tox-babel/b9325d7/ : SUCCESS in 3m 13s\n- requirements-integration http://logs.openstack.org/88/669488/1/check/requirements-integration/d48c2ce/ : SUCCESS in 54m 13s\n- neutron-grenade-multinode http://logs.openstack.org/88/669488/1/check/neutron-grenade-multinode/5334faa/ : SUCCESS in 1h 11m 30s\n- tempest-full http://logs.openstack.org/88/669488/1/check/tempest-full/b68b136/ : SUCCESS in 1h 59m 37s\n- neutron-grenade http://logs.openstack.org/88/669488/1/check/neutron-grenade/a52ad6a/ : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":1},{"id":"bfd4c366575d8408051d8802a26709b7dbd781b8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-06 18:56:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c26934bb8f5a3d7d9c2563706b95cfbcbe95c8a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 18:56:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e111915a944c204310f566fb93a8044b51eb5af8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 20:34:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/88/669488/1/gate/test-release-openstack/715da6c/ : SUCCESS in 2m 27s\n- openstack-tox-docs http://logs.openstack.org/88/669488/1/gate/openstack-tox-docs/126b9bd/html/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/88/669488/1/gate/openstack-tox-pep8/0c3d53b/ : SUCCESS in 2m 36s\n- openstack-tox-py27 http://logs.openstack.org/88/669488/1/gate/openstack-tox-py27/0a1f163/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/88/669488/1/gate/openstack-tox-py36/b060610/ : SUCCESS in 3m 04s\n- openstack-tox-py37 http://logs.openstack.org/88/669488/1/gate/openstack-tox-py37/8d30dab/ : SUCCESS in 2m 47s\n- openstack-tox-validate http://logs.openstack.org/88/669488/1/gate/openstack-tox-validate/84e123d/ : SUCCESS in 2m 55s\n- requirements-tox-babel http://logs.openstack.org/88/669488/1/gate/requirements-tox-babel/826d0eb/ : SUCCESS in 3m 51s\n- requirements-integration http://logs.openstack.org/88/669488/1/gate/requirements-integration/2272051/ : SUCCESS in 56m 45s\n- neutron-grenade-multinode http://logs.openstack.org/88/669488/1/gate/neutron-grenade-multinode/d2a00b7/ : SUCCESS in 1h 05m 18s\n- tempest-full http://logs.openstack.org/88/669488/1/gate/tempest-full/8b2c331/ : SUCCESS in 1h 25m 18s\n- neutron-grenade http://logs.openstack.org/88/669488/1/gate/neutron-grenade/f7037cd/ : SUCCESS in 49m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0fc5b2b9aa18372ed442b8a2d73f2b6fa234d16a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 20:34:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e0eec0790b1ba5d82253c5c25db61ed174e6a87","revisions":{"7e0eec0790b1ba5d82253c5c25db61ed174e6a87":{"kind":"REWORK","_number":1,"created":"2019-07-05 19:58:26.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/88/669488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/669488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/669488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/669488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/669488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/669488/1"}}},"commit":{"parents":[{"commit":"4a2a45ee5bdb81ef46359ab347d3b80f18c10207","subject":"Merge \"update constraint for pbr to new release 5.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4a2a45ee5bdb81ef46359ab347d3b80f18c10207"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:59:23.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 19:58:26.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Iab77d920bc5c46a7d7d21c05def60ba278b56983\nStory: #2005924\nTask: #34243\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e0eec0790b1ba5d82253c5c25db61ed174e6a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e0eec0790b1ba5d82253c5c25db61ed174e6a87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
