)]}'
{"id":"openstack%2Frequirements~669798","triplet_id":"openstack%2Frequirements~master~I569d425ce4c1eb94f027cf48913858b651b77415","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I569d425ce4c1eb94f027cf48913858b651b77415","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-07-09 06:40:27.000000000","updated":"2019-07-10 01:06:51.000000000","submitted":"2019-07-10 01:06:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669798-1562720811475-0696cc2b","meta_rev_id":"c95c01da627633bbce8f6f17ad3c5624097e35ac","_number":669798,"virtual_id_number":669798,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-10 01:06:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-09 20:01:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-09 20:01:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-09 20:01:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-10 01:06:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c5c5588e8cc1a7d6d37adedc2622840ec646085","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-09 06:40:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4661c0a7b21fff95995ec67e6179af14d21064e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 09:52:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/669798/1/check/openstack-tox-docs/ea29644/html/ : SUCCESS in 3m 44s\n- openstack-tox-pypy http://logs.openstack.org/98/669798/1/check/openstack-tox-pypy/820ecc5/ : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/669798/1/check/openstack-tox-pep8/30bc01f/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/98/669798/1/check/openstack-tox-py27/7989109/ : SUCCESS in 5m 51s\n- openstack-tox-py36 http://logs.openstack.org/98/669798/1/check/openstack-tox-py36/b5bd1a4/ : SUCCESS in 3m 21s\n- openstack-tox-py37 http://logs.openstack.org/98/669798/1/check/openstack-tox-py37/8365f3f/ : SUCCESS in 4m 09s\n- openstack-tox-validate http://logs.openstack.org/98/669798/1/check/openstack-tox-validate/cace4b3/ : SUCCESS in 2m 53s\n- requirements-tox-babel http://logs.openstack.org/98/669798/1/check/requirements-tox-babel/3c294cc/ : SUCCESS in 3m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/669798/1/check/requirements-tox-py27-check-uc/0af52eb/ : FAILURE in 6m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/669798/1/check/requirements-tox-py36-check-uc/cf8c624/ : FAILURE in 6m 30s\n- cross-cinder-py27 http://logs.openstack.org/98/669798/1/check/cross-cinder-py27/20a164f/ : SUCCESS in 8m 40s\n- cross-cinder-py36 http://logs.openstack.org/98/669798/1/check/cross-cinder-py36/812677c/ : SUCCESS in 12m 29s\n- cross-glance-py27 http://logs.openstack.org/98/669798/1/check/cross-glance-py27/842a2e7/ : SUCCESS in 6m 10s\n- cross-glance-py36 http://logs.openstack.org/98/669798/1/check/cross-glance-py36/c043b5b/ : SUCCESS in 5m 55s\n- cross-horizon-py27 http://logs.openstack.org/98/669798/1/check/cross-horizon-py27/e5bf118/ : SUCCESS in 11m 27s\n- cross-horizon-py36 http://logs.openstack.org/98/669798/1/check/cross-horizon-py36/19988ab/ : SUCCESS in 9m 07s\n- cross-horizon-npm http://logs.openstack.org/98/669798/1/check/cross-horizon-npm/6245c36/ : SUCCESS in 18m 52s\n- cross-keystone-py27 http://logs.openstack.org/98/669798/1/check/cross-keystone-py27/39dae42/ : SUCCESS in 30m 46s\n- cross-keystone-py36 http://logs.openstack.org/98/669798/1/check/cross-keystone-py36/887eab5/ : SUCCESS in 27m 11s\n- cross-neutron-py27 http://logs.openstack.org/98/669798/1/check/cross-neutron-py27/fe19475/ : SUCCESS in 23m 55s\n- cross-neutron-py36 http://logs.openstack.org/98/669798/1/check/cross-neutron-py36/8999c8e/ : SUCCESS in 22m 06s\n- cross-nova-py27 http://logs.openstack.org/98/669798/1/check/cross-nova-py27/8c35ca1/ : SUCCESS in 17m 02s\n- cross-nova-py36 http://logs.openstack.org/98/669798/1/check/cross-nova-py36/f15b69b/ : SUCCESS in 11m 25s\n- cross-nova-functional http://logs.openstack.org/98/669798/1/check/cross-nova-functional/faa204b/ : SUCCESS in 25m 42s\n- cross-osc-tox-docs http://logs.openstack.org/98/669798/1/check/cross-osc-tox-docs/81580d9/html/ : SUCCESS in 5m 07s\n- cross-swift-py27 http://logs.openstack.org/98/669798/1/check/cross-swift-py27/1314d8e/ : SUCCESS in 8m 23s\n- cross-swift-py36 http://logs.openstack.org/98/669798/1/check/cross-swift-py36/0a0960c/ : SUCCESS in 8m 44s\n- requirements-integration http://logs.openstack.org/98/669798/1/check/requirements-integration/f2143ab/ : SUCCESS in 54m 10s\n- neutron-grenade-multinode http://logs.openstack.org/98/669798/1/check/neutron-grenade-multinode/f337216/ : SUCCESS in 1h 05m 57s\n- tempest-full http://logs.openstack.org/98/669798/1/check/tempest-full/3006063/ : SUCCESS in 1h 56m 22s\n- neutron-grenade http://logs.openstack.org/98/669798/1/check/neutron-grenade/1f1b5fb/ : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5a628212ea17e4ea462db5a7a4ece1577fbf84a7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-09 15:17:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e5494fedf22d00a3e4db470664b3b9da6b299a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-09 15:17:07.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"59a52000fcd1f100fee5c73f86494f402d2467b2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 19:45:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/98/669798/2/check/openstack-tox-docs/6305deb/html/ : SUCCESS in 3m 29s\n- openstack-tox-pypy http://logs.openstack.org/98/669798/2/check/openstack-tox-pypy/b34c563/ : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/669798/2/check/openstack-tox-pep8/791c5b6/ : SUCCESS in 2m 44s\n- openstack-tox-py27 http://logs.openstack.org/98/669798/2/check/openstack-tox-py27/465d80f/ : SUCCESS in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/98/669798/2/check/openstack-tox-py36/c84479e/ : SUCCESS in 3m 12s\n- openstack-tox-py37 http://logs.openstack.org/98/669798/2/check/openstack-tox-py37/db101d6/ : SUCCESS in 3m 40s\n- openstack-tox-validate http://logs.openstack.org/98/669798/2/check/openstack-tox-validate/52c7591/ : SUCCESS in 3m 13s\n- requirements-tox-babel http://logs.openstack.org/98/669798/2/check/requirements-tox-babel/e05188e/ : SUCCESS in 3m 14s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/669798/2/check/requirements-tox-py27-check-uc/4cd6a3a/ : SUCCESS in 9m 47s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/669798/2/check/requirements-tox-py36-check-uc/c7a4b90/ : SUCCESS in 9m 23s\n- cross-cinder-py27 http://logs.openstack.org/98/669798/2/check/cross-cinder-py27/a760a6a/ : SUCCESS in 9m 23s\n- cross-cinder-py36 http://logs.openstack.org/98/669798/2/check/cross-cinder-py36/4f681c8/ : SUCCESS in 8m 47s\n- cross-glance-py27 http://logs.openstack.org/98/669798/2/check/cross-glance-py27/47fcb8b/ : SUCCESS in 4m 47s\n- cross-glance-py36 http://logs.openstack.org/98/669798/2/check/cross-glance-py36/20283a5/ : SUCCESS in 5m 55s\n- cross-horizon-py27 http://logs.openstack.org/98/669798/2/check/cross-horizon-py27/bb0a32a/ : SUCCESS in 9m 17s\n- cross-horizon-py36 http://logs.openstack.org/98/669798/2/check/cross-horizon-py36/0971e78/ : SUCCESS in 9m 24s\n- cross-horizon-npm http://logs.openstack.org/98/669798/2/check/cross-horizon-npm/77a34c9/ : SUCCESS in 6m 47s\n- cross-keystone-py27 http://logs.openstack.org/98/669798/2/check/cross-keystone-py27/74a5498/ : SUCCESS in 30m 20s\n- cross-keystone-py36 http://logs.openstack.org/98/669798/2/check/cross-keystone-py36/c60ba91/ : SUCCESS in 32m 26s\n- cross-neutron-py27 http://logs.openstack.org/98/669798/2/check/cross-neutron-py27/446adc5/ : SUCCESS in 23m 08s\n- cross-neutron-py36 http://logs.openstack.org/98/669798/2/check/cross-neutron-py36/bd93483/ : FAILURE in 23m 00s\n- cross-nova-py27 http://logs.openstack.org/98/669798/2/check/cross-nova-py27/e06139c/ : SUCCESS in 17m 05s\n- cross-nova-py36 http://logs.openstack.org/98/669798/2/check/cross-nova-py36/ccc2dc1/ : SUCCESS in 15m 27s\n- cross-nova-functional http://logs.openstack.org/98/669798/2/check/cross-nova-functional/d3f807f/ : SUCCESS in 24m 40s\n- cross-osc-tox-docs http://logs.openstack.org/98/669798/2/check/cross-osc-tox-docs/09638e8/html/ : SUCCESS in 3m 44s\n- cross-swift-py27 http://logs.openstack.org/98/669798/2/check/cross-swift-py27/c296bab/ : SUCCESS in 7m 43s\n- cross-swift-py36 http://logs.openstack.org/98/669798/2/check/cross-swift-py36/4aff66d/ : SUCCESS in 7m 36s\n- requirements-integration http://logs.openstack.org/98/669798/2/check/requirements-integration/605098b/ : SUCCESS in 1h 04m 10s\n- neutron-grenade-multinode http://logs.openstack.org/98/669798/2/check/neutron-grenade-multinode/5b949fb/ : SUCCESS in 1h 17m 49s\n- tempest-full http://logs.openstack.org/98/669798/2/check/tempest-full/b45c53b/ : SUCCESS in 1h 43m 03s\n- neutron-grenade http://logs.openstack.org/98/669798/2/check/neutron-grenade/7d6bc75/ : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9d3697351c9f81ad3a444e31a10a6d18fdc320d9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-09 20:01:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"27bcaa3d9b82df387eae1e4cab75c0f4f1113b7a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 22:21:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/669798/2/check/openstack-tox-docs/48ac622/html/ : SUCCESS in 2m 56s\n- openstack-tox-pypy http://logs.openstack.org/98/669798/2/check/openstack-tox-pypy/e8498ee/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/98/669798/2/check/openstack-tox-pep8/bef1c4c/ : SUCCESS in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/98/669798/2/check/openstack-tox-py27/9023e50/ : SUCCESS in 3m 55s\n- openstack-tox-py36 http://logs.openstack.org/98/669798/2/check/openstack-tox-py36/bc993f7/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/98/669798/2/check/openstack-tox-py37/debb346/ : SUCCESS in 3m 29s\n- openstack-tox-validate http://logs.openstack.org/98/669798/2/check/openstack-tox-validate/d612fcb/ : SUCCESS in 3m 19s\n- requirements-tox-babel http://logs.openstack.org/98/669798/2/check/requirements-tox-babel/09ebc8f/ : SUCCESS in 5m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/669798/2/check/requirements-tox-py27-check-uc/3054bba/ : SUCCESS in 6m 45s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/669798/2/check/requirements-tox-py36-check-uc/dd3ce86/ : SUCCESS in 8m 56s\n- cross-cinder-py27 http://logs.openstack.org/98/669798/2/check/cross-cinder-py27/d6bcecd/ : SUCCESS in 10m 48s\n- cross-cinder-py36 http://logs.openstack.org/98/669798/2/check/cross-cinder-py36/2de59c5/ : SUCCESS in 7m 42s\n- cross-glance-py27 http://logs.openstack.org/98/669798/2/check/cross-glance-py27/d9f4781/ : SUCCESS in 4m 38s\n- cross-glance-py36 http://logs.openstack.org/98/669798/2/check/cross-glance-py36/f091a80/ : SUCCESS in 4m 16s\n- cross-horizon-py27 http://logs.openstack.org/98/669798/2/check/cross-horizon-py27/9a0de81/ : SUCCESS in 10m 11s\n- cross-horizon-py36 http://logs.openstack.org/98/669798/2/check/cross-horizon-py36/853147a/ : SUCCESS in 11m 01s\n- cross-horizon-npm http://logs.openstack.org/98/669798/2/check/cross-horizon-npm/73e754f/ : SUCCESS in 6m 06s\n- cross-keystone-py27 http://logs.openstack.org/98/669798/2/check/cross-keystone-py27/9f2a905/ : SUCCESS in 28m 41s\n- cross-keystone-py36 http://logs.openstack.org/98/669798/2/check/cross-keystone-py36/13fbca1/ : SUCCESS in 29m 40s\n- cross-neutron-py27 http://logs.openstack.org/98/669798/2/check/cross-neutron-py27/61281e0/ : SUCCESS in 27m 35s\n- cross-neutron-py36 http://logs.openstack.org/98/669798/2/check/cross-neutron-py36/1dc6d16/ : SUCCESS in 24m 07s\n- cross-nova-py27 http://logs.openstack.org/98/669798/2/check/cross-nova-py27/fbd746a/ : SUCCESS in 18m 52s\n- cross-nova-py36 http://logs.openstack.org/98/669798/2/check/cross-nova-py36/7f2ed90/ : SUCCESS in 16m 58s\n- cross-nova-functional http://logs.openstack.org/98/669798/2/check/cross-nova-functional/3430e11/ : SUCCESS in 25m 01s\n- cross-osc-tox-docs http://logs.openstack.org/98/669798/2/check/cross-osc-tox-docs/2fe6acd/html/ : SUCCESS in 3m 28s\n- cross-swift-py27 http://logs.openstack.org/98/669798/2/check/cross-swift-py27/24da70a/ : SUCCESS in 9m 48s\n- cross-swift-py36 http://logs.openstack.org/98/669798/2/check/cross-swift-py36/dd3950a/ : SUCCESS in 9m 02s\n- requirements-integration http://logs.openstack.org/98/669798/2/check/requirements-integration/6ada1f8/ : SUCCESS in 58m 26s\n- neutron-grenade-multinode http://logs.openstack.org/98/669798/2/check/neutron-grenade-multinode/f77d75f/ : SUCCESS in 1h 08m 46s\n- tempest-full http://logs.openstack.org/98/669798/2/check/tempest-full/10e9531/ : SUCCESS in 1h 54m 17s\n- neutron-grenade http://logs.openstack.org/98/669798/2/check/neutron-grenade/7ccc96c/ : SUCCESS in 1h 03m 56s","accounts_in_message":[],"_revision_number":2},{"id":"b95943a9a09c8da689c9483c4b4746e54cc28e61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 22:21:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e35df3f45b83143be04aa95daaa6bfee006fde5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 23:13:31.000000000","message":"Patch Set 2:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07091e294353cee43435b4c79f8aceec14c9e98f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 01:06:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/669798/2/gate/openstack-tox-docs/d488278/html/ : SUCCESS in 4m 43s\n- openstack-tox-pep8 http://logs.openstack.org/98/669798/2/gate/openstack-tox-pep8/da0612b/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/98/669798/2/gate/openstack-tox-py27/69b74e0/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/98/669798/2/gate/openstack-tox-py36/08203ad/ : SUCCESS in 4m 59s\n- openstack-tox-py37 http://logs.openstack.org/98/669798/2/gate/openstack-tox-py37/1a1e5ac/ : SUCCESS in 6m 05s\n- openstack-tox-validate http://logs.openstack.org/98/669798/2/gate/openstack-tox-validate/6af85df/ : SUCCESS in 4m 19s\n- requirements-tox-babel http://logs.openstack.org/98/669798/2/gate/requirements-tox-babel/4190795/ : SUCCESS in 4m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/98/669798/2/gate/requirements-tox-py27-check-uc/6cbdbfc/ : SUCCESS in 7m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/98/669798/2/gate/requirements-tox-py36-check-uc/eda864c/ : SUCCESS in 7m 31s\n- cross-cinder-py27 http://logs.openstack.org/98/669798/2/gate/cross-cinder-py27/399db67/ : SUCCESS in 10m 36s\n- cross-cinder-py36 http://logs.openstack.org/98/669798/2/gate/cross-cinder-py36/ad62c3f/ : SUCCESS in 9m 22s\n- cross-glance-py27 http://logs.openstack.org/98/669798/2/gate/cross-glance-py27/3d08aa5/ : SUCCESS in 5m 55s\n- cross-glance-py36 http://logs.openstack.org/98/669798/2/gate/cross-glance-py36/4d8ffb8/ : SUCCESS in 5m 51s\n- cross-horizon-py27 http://logs.openstack.org/98/669798/2/gate/cross-horizon-py27/423d865/ : SUCCESS in 9m 14s\n- cross-horizon-py36 http://logs.openstack.org/98/669798/2/gate/cross-horizon-py36/e4f8ebf/ : SUCCESS in 12m 21s\n- cross-horizon-npm http://logs.openstack.org/98/669798/2/gate/cross-horizon-npm/28aed3a/ : SUCCESS in 7m 50s\n- cross-keystone-py27 http://logs.openstack.org/98/669798/2/gate/cross-keystone-py27/d623b3c/ : SUCCESS in 28m 26s\n- cross-keystone-py36 http://logs.openstack.org/98/669798/2/gate/cross-keystone-py36/f38e96e/ : SUCCESS in 28m 01s\n- cross-neutron-py27 http://logs.openstack.org/98/669798/2/gate/cross-neutron-py27/6b914f9/ : SUCCESS in 22m 19s\n- cross-neutron-py36 http://logs.openstack.org/98/669798/2/gate/cross-neutron-py36/60513b7/ : SUCCESS in 27m 12s\n- cross-nova-py27 http://logs.openstack.org/98/669798/2/gate/cross-nova-py27/bcb4c0a/ : SUCCESS in 19m 28s\n- cross-nova-py36 http://logs.openstack.org/98/669798/2/gate/cross-nova-py36/5561e74/ : SUCCESS in 17m 59s\n- cross-nova-functional http://logs.openstack.org/98/669798/2/gate/cross-nova-functional/b33d588/ : SUCCESS in 24m 14s\n- cross-osc-tox-docs http://logs.openstack.org/98/669798/2/gate/cross-osc-tox-docs/049a0ad/html/ : SUCCESS in 4m 33s\n- cross-swift-py27 http://logs.openstack.org/98/669798/2/gate/cross-swift-py27/38fde8e/ : SUCCESS in 11m 07s\n- cross-swift-py36 http://logs.openstack.org/98/669798/2/gate/cross-swift-py36/74dc255/ : SUCCESS in 9m 09s\n- requirements-integration http://logs.openstack.org/98/669798/2/gate/requirements-integration/e4956af/ : SUCCESS in 1h 19m 03s\n- neutron-grenade-multinode http://logs.openstack.org/98/669798/2/gate/neutron-grenade-multinode/3200784/ : SUCCESS in 1h 12m 24s\n- tempest-full http://logs.openstack.org/98/669798/2/gate/tempest-full/ff1d359/ : SUCCESS in 1h 47m 43s\n- neutron-grenade http://logs.openstack.org/98/669798/2/gate/neutron-grenade/b8aa3f2/ : SUCCESS in 1h 10m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c95c01da627633bbce8f6f17ad3c5624097e35ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 01:06:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea3408ce2d5d422383b8b0b5924ca53eb4bc41b2","revisions":{"48f40a0264a1a749b9a9f7e6386470ab8fd01a6b":{"kind":"REWORK","_number":1,"created":"2019-07-09 06:40:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/669798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/669798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/669798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/669798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/669798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/669798/1"}}},"commit":{"parents":[{"commit":"c26e18bf7b2add5a78ee377ddfacdb76faa32c0e","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c26e18bf7b2add5a78ee377ddfacdb76faa32c0e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-09 06:40:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-09 06:40:22.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I569d425ce4c1eb94f027cf48913858b651b77415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48f40a0264a1a749b9a9f7e6386470ab8fd01a6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/48f40a0264a1a749b9a9f7e6386470ab8fd01a6b"}]},"branch":"refs/heads/master"},"ea3408ce2d5d422383b8b0b5924ca53eb4bc41b2":{"kind":"REWORK","_number":2,"created":"2019-07-09 15:17:03.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/98/669798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/669798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/669798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/669798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/669798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/669798/2"}}},"commit":{"parents":[{"commit":"c26e18bf7b2add5a78ee377ddfacdb76faa32c0e","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c26e18bf7b2add5a78ee377ddfacdb76faa32c0e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-09 06:40:22.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-09 15:16:52.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I569d425ce4c1eb94f027cf48913858b651b77415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea3408ce2d5d422383b8b0b5924ca53eb4bc41b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea3408ce2d5d422383b8b0b5924ca53eb4bc41b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
