)]}'
{"id":"openstack%2Frequirements~670258","triplet_id":"openstack%2Frequirements~master~I29259b7cfaf8ed033a41919bbdb5352bcdeff62c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I29259b7cfaf8ed033a41919bbdb5352bcdeff62c","subject":"update constraint for os-brick to new release 2.9.1","status":"MERGED","created":"2019-07-11 09:31:54.000000000","updated":"2019-07-11 23:19:34.000000000","submitted":"2019-07-11 23:19:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670258-1562887174457-a65eadfb","meta_rev_id":"c404901f60dc7a052b9c33cbef0a1ec1942b88bc","_number":670258,"virtual_id_number":670258,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-11 23:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-11 11:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-11 14:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-11 14:26:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 11:48:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-11 14:26:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-11 23:19:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"440b45757f011d6fd50897508356e498c4a13144","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-11 09:31:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"477412ce7d3b3aa2fe6177d539d30309519268f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 11:09:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/670258/1/check/openstack-tox-docs/9e6bd54/html/ : SUCCESS in 3m 26s\n- openstack-tox-pypy http://logs.openstack.org/58/670258/1/check/openstack-tox-pypy/6ee3f88/ : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/58/670258/1/check/openstack-tox-pep8/b66d5a5/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/58/670258/1/check/openstack-tox-py27/f564df4/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/58/670258/1/check/openstack-tox-py36/0939bb0/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/58/670258/1/check/openstack-tox-py37/79309f7/ : SUCCESS in 5m 03s\n- openstack-tox-validate http://logs.openstack.org/58/670258/1/check/openstack-tox-validate/c97fa8b/ : SUCCESS in 6m 18s\n- requirements-tox-babel http://logs.openstack.org/58/670258/1/check/requirements-tox-babel/89f2af0/ : SUCCESS in 3m 33s\n- requirements-tox-py27-check-uc http://logs.openstack.org/58/670258/1/check/requirements-tox-py27-check-uc/f7b8296/ : SUCCESS in 9m 53s\n- requirements-tox-py36-check-uc http://logs.openstack.org/58/670258/1/check/requirements-tox-py36-check-uc/20b445b/ : SUCCESS in 6m 36s\n- cross-cinder-py27 http://logs.openstack.org/58/670258/1/check/cross-cinder-py27/79d85e7/ : SUCCESS in 8m 26s\n- cross-cinder-py36 http://logs.openstack.org/58/670258/1/check/cross-cinder-py36/c8805bc/ : SUCCESS in 11m 28s\n- cross-glance-py27 http://logs.openstack.org/58/670258/1/check/cross-glance-py27/8cb6448/ : SUCCESS in 5m 14s\n- cross-glance-py36 http://logs.openstack.org/58/670258/1/check/cross-glance-py36/f8dc104/ : SUCCESS in 4m 35s\n- cross-horizon-py27 http://logs.openstack.org/58/670258/1/check/cross-horizon-py27/9a5f7c6/ : SUCCESS in 10m 29s\n- cross-horizon-py36 http://logs.openstack.org/58/670258/1/check/cross-horizon-py36/9fda962/ : SUCCESS in 8m 30s\n- cross-horizon-npm http://logs.openstack.org/58/670258/1/check/cross-horizon-npm/bbb554f/ : SUCCESS in 6m 39s\n- cross-keystone-py27 http://logs.openstack.org/58/670258/1/check/cross-keystone-py27/69ff29f/ : SUCCESS in 22m 19s\n- cross-keystone-py36 http://logs.openstack.org/58/670258/1/check/cross-keystone-py36/c821e06/ : SUCCESS in 33m 55s\n- cross-neutron-py27 http://logs.openstack.org/58/670258/1/check/cross-neutron-py27/f63c35e/ : SUCCESS in 23m 39s\n- cross-neutron-py36 http://logs.openstack.org/58/670258/1/check/cross-neutron-py36/e90d558/ : SUCCESS in 23m 01s\n- cross-nova-py27 http://logs.openstack.org/58/670258/1/check/cross-nova-py27/ae16a60/ : SUCCESS in 19m 47s\n- cross-nova-py36 http://logs.openstack.org/58/670258/1/check/cross-nova-py36/1c9b8aa/ : SUCCESS in 18m 22s\n- cross-nova-functional http://logs.openstack.org/58/670258/1/check/cross-nova-functional/07e6a50/ : SUCCESS in 19m 36s\n- cross-osc-tox-docs http://logs.openstack.org/58/670258/1/check/cross-osc-tox-docs/7eda14d/html/ : SUCCESS in 4m 04s\n- cross-swift-py27 http://logs.openstack.org/58/670258/1/check/cross-swift-py27/3960379/ : SUCCESS in 7m 23s\n- cross-swift-py36 http://logs.openstack.org/58/670258/1/check/cross-swift-py36/31caaab/ : SUCCESS in 7m 50s\n- requirements-integration http://logs.openstack.org/58/670258/1/check/requirements-integration/31f4641/ : SUCCESS in 56m 15s\n- neutron-grenade-multinode http://logs.openstack.org/58/670258/1/check/neutron-grenade-multinode/f66acc0/ : SUCCESS in 1h 23m 52s\n- tempest-full http://logs.openstack.org/58/670258/1/check/tempest-full/51b2cf2/ : SUCCESS in 1h 29m 36s\n- neutron-grenade http://logs.openstack.org/58/670258/1/check/neutron-grenade/98026e5/ : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":1},{"id":"59b3af1d5e16c6823e4cd563f48cf64c6bb3c00d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-11 11:48:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27348f58d6303c64c158caa9b32a9308dfe66712","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-11 14:26:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71a3848994e2eb610430bef7738658dbfc03c2a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 14:26:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37c44de04db340e969751469b3128b2673c694b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 16:04:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/58/670258/1/gate/openstack-tox-docs/4878e2e/html/ : SUCCESS in 2m 50s\n- openstack-tox-pep8 http://logs.openstack.org/58/670258/1/gate/openstack-tox-pep8/c5f05e3/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/58/670258/1/gate/openstack-tox-py27/18061d2/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/58/670258/1/gate/openstack-tox-py36/3e29296/ : SUCCESS in 3m 47s\n- openstack-tox-py37 http://logs.openstack.org/58/670258/1/gate/openstack-tox-py37/193307f/ : SUCCESS in 4m 26s\n- openstack-tox-validate http://logs.openstack.org/58/670258/1/gate/openstack-tox-validate/c2c8e84/ : SUCCESS in 4m 25s\n- requirements-tox-babel http://logs.openstack.org/58/670258/1/gate/requirements-tox-babel/6955a66/ : SUCCESS in 5m 37s\n- requirements-tox-py27-check-uc http://logs.openstack.org/58/670258/1/gate/requirements-tox-py27-check-uc/11dac5e/ : SUCCESS in 6m 39s\n- requirements-tox-py36-check-uc http://logs.openstack.org/58/670258/1/gate/requirements-tox-py36-check-uc/77964a2/ : SUCCESS in 6m 23s\n- cross-cinder-py27 http://logs.openstack.org/58/670258/1/gate/cross-cinder-py27/13575b0/ : SUCCESS in 9m 21s\n- cross-cinder-py36 http://logs.openstack.org/58/670258/1/gate/cross-cinder-py36/c2a56af/ : SUCCESS in 9m 24s\n- cross-glance-py27 http://logs.openstack.org/58/670258/1/gate/cross-glance-py27/3958e42/ : SUCCESS in 4m 41s\n- cross-glance-py36 http://logs.openstack.org/58/670258/1/gate/cross-glance-py36/33cdcda/ : SUCCESS in 4m 16s\n- cross-horizon-py27 http://logs.openstack.org/58/670258/1/gate/cross-horizon-py27/131ec74/ : SUCCESS in 10m 06s\n- cross-horizon-py36 http://logs.openstack.org/58/670258/1/gate/cross-horizon-py36/db7f549/ : SUCCESS in 9m 09s\n- cross-horizon-npm http://logs.openstack.org/58/670258/1/gate/cross-horizon-npm/8b80c02/ : SUCCESS in 6m 26s\n- cross-keystone-py27 http://logs.openstack.org/58/670258/1/gate/cross-keystone-py27/0f747e5/ : SUCCESS in 30m 12s\n- cross-keystone-py36 http://logs.openstack.org/58/670258/1/gate/cross-keystone-py36/518ce0e/ : SUCCESS in 29m 33s\n- cross-neutron-py27 http://logs.openstack.org/58/670258/1/gate/cross-neutron-py27/e204e2e/ : SUCCESS in 27m 18s\n- cross-neutron-py36 http://logs.openstack.org/58/670258/1/gate/cross-neutron-py36/aefb84b/ : SUCCESS in 22m 56s\n- cross-nova-py27 http://logs.openstack.org/58/670258/1/gate/cross-nova-py27/2b6ce2e/ : SUCCESS in 16m 03s\n- cross-nova-py36 http://logs.openstack.org/58/670258/1/gate/cross-nova-py36/560d2c6/ : SUCCESS in 16m 42s\n- cross-nova-functional http://logs.openstack.org/58/670258/1/gate/cross-nova-functional/34f71ed/ : SUCCESS in 24m 04s\n- cross-osc-tox-docs http://logs.openstack.org/58/670258/1/gate/cross-osc-tox-docs/f62d583/html/ : SUCCESS in 4m 26s\n- cross-swift-py27 http://logs.openstack.org/58/670258/1/gate/cross-swift-py27/309e167/ : SUCCESS in 8m 53s\n- cross-swift-py36 http://logs.openstack.org/58/670258/1/gate/cross-swift-py36/cdbaa0b/ : SUCCESS in 8m 33s\n- requirements-integration http://logs.openstack.org/58/670258/1/gate/requirements-integration/4050511/ : SUCCESS in 54m 05s\n- neutron-grenade-multinode http://logs.openstack.org/58/670258/1/gate/neutron-grenade-multinode/dd12433/ : SUCCESS in 1h 08m 20s\n- tempest-full http://logs.openstack.org/58/670258/1/gate/tempest-full/ba08d9e/ : FAILURE in 1h 30m 14s\n- neutron-grenade http://logs.openstack.org/58/670258/1/gate/neutron-grenade/0e30430/ : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":1},{"id":"666d532be3e7cc708061028a708c79de8111e102","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-11 16:12:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd9bb9250835d7714898c8eb67d542c8046d11c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 21:07:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/670258/1/check/openstack-tox-docs/6ff3095/html/ : SUCCESS in 4m 14s\n- openstack-tox-pypy http://logs.openstack.org/58/670258/1/check/openstack-tox-pypy/0e72821/ : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/58/670258/1/check/openstack-tox-pep8/f5f601a/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/58/670258/1/check/openstack-tox-py27/f300c35/ : SUCCESS in 3m 47s\n- openstack-tox-py36 http://logs.openstack.org/58/670258/1/check/openstack-tox-py36/bd117f2/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/58/670258/1/check/openstack-tox-py37/1aa2112/ : SUCCESS in 4m 08s\n- openstack-tox-validate http://logs.openstack.org/58/670258/1/check/openstack-tox-validate/21b83b0/ : SUCCESS in 3m 32s\n- requirements-tox-babel http://logs.openstack.org/58/670258/1/check/requirements-tox-babel/a9ce2bb/ : SUCCESS in 2m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/58/670258/1/check/requirements-tox-py27-check-uc/fa61335/ : SUCCESS in 6m 33s\n- requirements-tox-py36-check-uc http://logs.openstack.org/58/670258/1/check/requirements-tox-py36-check-uc/e234e68/ : SUCCESS in 9m 32s\n- cross-cinder-py27 http://logs.openstack.org/58/670258/1/check/cross-cinder-py27/1af817c/ : SUCCESS in 9m 31s\n- cross-cinder-py36 http://logs.openstack.org/58/670258/1/check/cross-cinder-py36/cd64cd2/ : SUCCESS in 9m 41s\n- cross-glance-py27 http://logs.openstack.org/58/670258/1/check/cross-glance-py27/b39d0be/ : SUCCESS in 7m 00s\n- cross-glance-py36 http://logs.openstack.org/58/670258/1/check/cross-glance-py36/fef5a06/ : SUCCESS in 6m 09s\n- cross-horizon-py27 http://logs.openstack.org/58/670258/1/check/cross-horizon-py27/dd1857e/ : SUCCESS in 10m 33s\n- cross-horizon-py36 http://logs.openstack.org/58/670258/1/check/cross-horizon-py36/f700643/ : SUCCESS in 8m 24s\n- cross-horizon-npm http://logs.openstack.org/58/670258/1/check/cross-horizon-npm/5eabd1a/ : SUCCESS in 8m 43s\n- cross-keystone-py27 http://logs.openstack.org/58/670258/1/check/cross-keystone-py27/45075f4/ : SUCCESS in 32m 15s\n- cross-keystone-py36 http://logs.openstack.org/58/670258/1/check/cross-keystone-py36/0596e54/ : SUCCESS in 29m 31s\n- cross-neutron-py27 http://logs.openstack.org/58/670258/1/check/cross-neutron-py27/06f7ee3/ : SUCCESS in 30m 36s\n- cross-neutron-py36 http://logs.openstack.org/58/670258/1/check/cross-neutron-py36/31d8a93/ : SUCCESS in 29m 26s\n- cross-nova-py27 http://logs.openstack.org/58/670258/1/check/cross-nova-py27/2626c46/ : SUCCESS in 17m 59s\n- cross-nova-py36 http://logs.openstack.org/58/670258/1/check/cross-nova-py36/0bc75a6/ : SUCCESS in 15m 38s\n- cross-nova-functional http://logs.openstack.org/58/670258/1/check/cross-nova-functional/29d2f7d/ : SUCCESS in 23m 54s\n- cross-osc-tox-docs http://logs.openstack.org/58/670258/1/check/cross-osc-tox-docs/64ee22c/html/ : SUCCESS in 4m 00s\n- cross-swift-py27 http://logs.openstack.org/58/670258/1/check/cross-swift-py27/9db75a0/ : SUCCESS in 8m 50s\n- cross-swift-py36 http://logs.openstack.org/58/670258/1/check/cross-swift-py36/5d41099/ : SUCCESS in 8m 17s\n- requirements-integration http://logs.openstack.org/58/670258/1/check/requirements-integration/51f1bc0/ : SUCCESS in 54m 40s\n- neutron-grenade-multinode http://logs.openstack.org/58/670258/1/check/neutron-grenade-multinode/709e7b0/ : SUCCESS in 1h 11m 38s\n- tempest-full http://logs.openstack.org/58/670258/1/check/tempest-full/d637d45/ : SUCCESS in 1h 46m 52s\n- neutron-grenade http://logs.openstack.org/58/670258/1/check/neutron-grenade/4a0d6eb/ : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":1},{"id":"353dfdf370f3ae5e1edea51466f51ef3041cb460","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 21:07:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92afe6e14a4ff15aa91fbaea918fb001505b9307","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 23:19:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/670258/1/gate/openstack-tox-docs/a053dfc/html/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/58/670258/1/gate/openstack-tox-pep8/9a8e5da/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/58/670258/1/gate/openstack-tox-py27/e033408/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/58/670258/1/gate/openstack-tox-py36/a9c04b4/ : SUCCESS in 3m 33s\n- openstack-tox-py37 http://logs.openstack.org/58/670258/1/gate/openstack-tox-py37/73bf524/ : SUCCESS in 3m 50s\n- openstack-tox-validate http://logs.openstack.org/58/670258/1/gate/openstack-tox-validate/1c7f7da/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/58/670258/1/gate/requirements-tox-babel/288c7cb/ : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc http://logs.openstack.org/58/670258/1/gate/requirements-tox-py27-check-uc/1a8961f/ : SUCCESS in 8m 24s\n- requirements-tox-py36-check-uc http://logs.openstack.org/58/670258/1/gate/requirements-tox-py36-check-uc/01e9229/ : SUCCESS in 8m 05s\n- cross-cinder-py27 http://logs.openstack.org/58/670258/1/gate/cross-cinder-py27/14e1eaf/ : SUCCESS in 9m 26s\n- cross-cinder-py36 http://logs.openstack.org/58/670258/1/gate/cross-cinder-py36/8b34a8c/ : SUCCESS in 11m 13s\n- cross-glance-py27 http://logs.openstack.org/58/670258/1/gate/cross-glance-py27/5fe07a2/ : SUCCESS in 4m 33s\n- cross-glance-py36 http://logs.openstack.org/58/670258/1/gate/cross-glance-py36/d64952e/ : SUCCESS in 6m 20s\n- cross-horizon-py27 http://logs.openstack.org/58/670258/1/gate/cross-horizon-py27/8facbb6/ : SUCCESS in 9m 35s\n- cross-horizon-py36 http://logs.openstack.org/58/670258/1/gate/cross-horizon-py36/41de420/ : SUCCESS in 8m 16s\n- cross-horizon-npm http://logs.openstack.org/58/670258/1/gate/cross-horizon-npm/16de16e/ : SUCCESS in 8m 15s\n- cross-keystone-py27 http://logs.openstack.org/58/670258/1/gate/cross-keystone-py27/5b3f3b8/ : SUCCESS in 26m 03s\n- cross-keystone-py36 http://logs.openstack.org/58/670258/1/gate/cross-keystone-py36/52a8d1a/ : SUCCESS in 30m 39s\n- cross-neutron-py27 http://logs.openstack.org/58/670258/1/gate/cross-neutron-py27/728eb7d/ : SUCCESS in 27m 30s\n- cross-neutron-py36 http://logs.openstack.org/58/670258/1/gate/cross-neutron-py36/65ff680/ : SUCCESS in 22m 36s\n- cross-nova-py27 http://logs.openstack.org/58/670258/1/gate/cross-nova-py27/6ed4c3f/ : SUCCESS in 19m 42s\n- cross-nova-py36 http://logs.openstack.org/58/670258/1/gate/cross-nova-py36/35d161b/ : SUCCESS in 15m 06s\n- cross-nova-functional http://logs.openstack.org/58/670258/1/gate/cross-nova-functional/4192bcd/ : SUCCESS in 21m 01s\n- cross-osc-tox-docs http://logs.openstack.org/58/670258/1/gate/cross-osc-tox-docs/dfec6a5/html/ : SUCCESS in 3m 48s\n- cross-swift-py27 http://logs.openstack.org/58/670258/1/gate/cross-swift-py27/c68579d/ : SUCCESS in 9m 14s\n- cross-swift-py36 http://logs.openstack.org/58/670258/1/gate/cross-swift-py36/800c17b/ : SUCCESS in 7m 22s\n- requirements-integration http://logs.openstack.org/58/670258/1/gate/requirements-integration/6a55794/ : SUCCESS in 51m 04s\n- neutron-grenade-multinode http://logs.openstack.org/58/670258/1/gate/neutron-grenade-multinode/448e2af/ : SUCCESS in 1h 07m 29s\n- tempest-full http://logs.openstack.org/58/670258/1/gate/tempest-full/93d7f46/ : SUCCESS in 2h 01m 57s\n- neutron-grenade http://logs.openstack.org/58/670258/1/gate/neutron-grenade/632ce2c/ : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":1},{"id":"53de3ea419dbb2842c0dbdbf60da15cdfb5d7d52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 23:19:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbdfc04c02bfa62d5a8b812942032710e26ff3bf","revisions":{"cbdfc04c02bfa62d5a8b812942032710e26ff3bf":{"kind":"REWORK","_number":1,"created":"2019-07-11 09:31:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/58/670258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/58/670258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/670258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/670258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/58/670258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/58/670258/1"}}},"commit":{"parents":[{"commit":"ea3408ce2d5d422383b8b0b5924ca53eb4bc41b2","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ea3408ce2d5d422383b8b0b5924ca53eb4bc41b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-11 09:31:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-11 09:31:52.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.9.1","message":"update constraint for os-brick to new release 2.9.1\n\nChange-Id: I29259b7cfaf8ed033a41919bbdb5352bcdeff62c\nmeta:version: 2.9.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: I8782bcb9e6560b423ca6319bfe67439c71a7f21a\nmeta:release:Code-Review+1: Jay Bryant \u003cjungleboyj@electronicjungle.net\u003e\nmeta:release:Code-Review+1: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbdfc04c02bfa62d5a8b812942032710e26ff3bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cbdfc04c02bfa62d5a8b812942032710e26ff3bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
