)]}'
{"id":"openstack%2Frequirements~670259","triplet_id":"openstack%2Frequirements~stable%2Fstein~I5b452942a4a02843ddf12570e9b1ebf98dd7b40a","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I5b452942a4a02843ddf12570e9b1ebf98dd7b40a","subject":"update constraint for python-cinderclient to new release 4.2.1","status":"MERGED","created":"2019-07-11 09:34:10.000000000","updated":"2019-07-11 18:03:57.000000000","submitted":"2019-07-11 18:03:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670259-1562868237147-77280bbb","meta_rev_id":"2ffb3ff73c5b34cd199d82f178c95c8f00f6910c","_number":670259,"virtual_id_number":670259,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-11 18:03:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-11 11:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-11 14:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-11 14:26:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 11:48:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-11 14:26:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-11 18:03:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e64d372209c17ab4f7cbc487b20058598f37d64","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-11 09:34:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"216111ba6c45e69d73359e49e2787292241a9214","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 11:10:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/59/670259/1/check/openstack-tox-pypy/ecf5489/ : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/59/670259/1/check/openstack-tox-pep8/6969e5e/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/59/670259/1/check/openstack-tox-py27/f186d1a/ : SUCCESS in 4m 17s\n- openstack-tox-py35 http://logs.openstack.org/59/670259/1/check/openstack-tox-py35/8435e9b/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/59/670259/1/check/openstack-tox-py36/8cb781e/ : SUCCESS in 3m 17s\n- openstack-tox-validate http://logs.openstack.org/59/670259/1/check/openstack-tox-validate/24b17ef/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/59/670259/1/check/requirements-tox-babel/7d19dde/ : SUCCESS in 4m 20s\n- requirements-tox-py27-check-uc http://logs.openstack.org/59/670259/1/check/requirements-tox-py27-check-uc/62bafa6/ : SUCCESS in 12m 59s\n- requirements-tox-py35-check-uc http://logs.openstack.org/59/670259/1/check/requirements-tox-py35-check-uc/935b258/ : SUCCESS in 7m 17s\n- requirements-tox-py36-check-uc http://logs.openstack.org/59/670259/1/check/requirements-tox-py36-check-uc/4fd26f0/ : SUCCESS in 12m 18s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/59/670259/1/check/cross-cinder-py27/143b404/ : SUCCESS in 12m 27s\n- cross-cinder-py35 http://logs.openstack.org/59/670259/1/check/cross-cinder-py35/d314295/ : SUCCESS in 8m 56s\n- cross-glance-py27 http://logs.openstack.org/59/670259/1/check/cross-glance-py27/f19d786/ : SUCCESS in 4m 41s\n- cross-glance-py35 http://logs.openstack.org/59/670259/1/check/cross-glance-py35/fe82438/ : SUCCESS in 4m 23s\n- cross-horizon-py27 http://logs.openstack.org/59/670259/1/check/cross-horizon-py27/fb17c6a/ : SUCCESS in 10m 03s\n- cross-horizon-py35 http://logs.openstack.org/59/670259/1/check/cross-horizon-py35/f6fb0b4/ : SUCCESS in 9m 15s\n- cross-horizon-npm http://logs.openstack.org/59/670259/1/check/cross-horizon-npm/3817865/ : SUCCESS in 10m 13s\n- cross-keystone-py27 http://logs.openstack.org/59/670259/1/check/cross-keystone-py27/5c4e8aa/ : SUCCESS in 23m 28s\n- cross-keystone-py35 http://logs.openstack.org/59/670259/1/check/cross-keystone-py35/3930db1/ : SUCCESS in 27m 25s\n- cross-neutron-py27 http://logs.openstack.org/59/670259/1/check/cross-neutron-py27/92a332e/ : SUCCESS in 24m 26s\n- cross-neutron-py35 http://logs.openstack.org/59/670259/1/check/cross-neutron-py35/7e5d3a4/ : SUCCESS in 27m 16s\n- cross-nova-py27 http://logs.openstack.org/59/670259/1/check/cross-nova-py27/c6a22c5/ : SUCCESS in 11m 29s\n- cross-nova-py35 http://logs.openstack.org/59/670259/1/check/cross-nova-py35/51c196b/ : SUCCESS in 12m 47s\n- cross-nova-functional http://logs.openstack.org/59/670259/1/check/cross-nova-functional/79f6608/ : SUCCESS in 17m 37s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/59/670259/1/check/cross-osc-build-sphinx-docs/a7ac9d9/html/ : SUCCESS in 4m 22s\n- cross-swift-py27 http://logs.openstack.org/59/670259/1/check/cross-swift-py27/b50af1c/ : SUCCESS in 8m 00s\n- cross-swift-py35 http://logs.openstack.org/59/670259/1/check/cross-swift-py35/d9f96b9/ : SUCCESS in 7m 25s\n- requirements-integration http://logs.openstack.org/59/670259/1/check/requirements-integration/39d41c7/ : SUCCESS in 53m 36s\n- neutron-grenade-multinode http://logs.openstack.org/59/670259/1/check/neutron-grenade-multinode/b601295/ : SUCCESS in 1h 01m 25s\n- tempest-full http://logs.openstack.org/59/670259/1/check/tempest-full/2013628/ : SUCCESS in 1h 30m 05s\n- neutron-grenade http://logs.openstack.org/59/670259/1/check/neutron-grenade/54e032c/ : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":1},{"id":"aba9e264a023d865f8cf3312c77fe56172c7b12e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-11 11:48:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1440ed226e0a8309c7fd9150d0f8892d65d81dc3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-11 14:26:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"459dec436c09ea8b806e4258ef8bb6f6eab81f1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 14:26:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b197df12d8a684567761e04a7e86c2e54b7a9b36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 18:03:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/670259/1/gate/openstack-tox-pep8/ea53d34/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/59/670259/1/gate/openstack-tox-py27/5b40ebe/ : SUCCESS in 3m 53s\n- openstack-tox-py35 http://logs.openstack.org/59/670259/1/gate/openstack-tox-py35/f94ceac/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/59/670259/1/gate/openstack-tox-py36/c6e4a50/ : SUCCESS in 3m 43s\n- openstack-tox-validate http://logs.openstack.org/59/670259/1/gate/openstack-tox-validate/5cbcc33/ : SUCCESS in 3m 38s\n- requirements-tox-babel http://logs.openstack.org/59/670259/1/gate/requirements-tox-babel/42c969f/ : SUCCESS in 4m 04s\n- requirements-tox-py27-check-uc http://logs.openstack.org/59/670259/1/gate/requirements-tox-py27-check-uc/01e2c41/ : SUCCESS in 9m 03s\n- requirements-tox-py35-check-uc http://logs.openstack.org/59/670259/1/gate/requirements-tox-py35-check-uc/bc12f0e/ : SUCCESS in 9m 58s\n- cross-cinder-py27 http://logs.openstack.org/59/670259/1/gate/cross-cinder-py27/0323178/ : SUCCESS in 9m 31s\n- cross-cinder-py35 http://logs.openstack.org/59/670259/1/gate/cross-cinder-py35/30fa088/ : SUCCESS in 10m 15s\n- cross-glance-py27 http://logs.openstack.org/59/670259/1/gate/cross-glance-py27/4e09fd6/ : SUCCESS in 4m 53s\n- cross-glance-py35 http://logs.openstack.org/59/670259/1/gate/cross-glance-py35/1e73edf/ : SUCCESS in 5m 20s\n- cross-horizon-py27 http://logs.openstack.org/59/670259/1/gate/cross-horizon-py27/727cc91/ : SUCCESS in 11m 09s\n- cross-horizon-py35 http://logs.openstack.org/59/670259/1/gate/cross-horizon-py35/4742643/ : SUCCESS in 10m 59s\n- cross-horizon-npm http://logs.openstack.org/59/670259/1/gate/cross-horizon-npm/6afb497/ : SUCCESS in 8m 29s\n- cross-keystone-py27 http://logs.openstack.org/59/670259/1/gate/cross-keystone-py27/eefa37a/ : SUCCESS in 29m 12s\n- cross-keystone-py35 http://logs.openstack.org/59/670259/1/gate/cross-keystone-py35/5301583/ : SUCCESS in 25m 39s\n- cross-neutron-py27 http://logs.openstack.org/59/670259/1/gate/cross-neutron-py27/caaea66/ : SUCCESS in 25m 44s\n- cross-neutron-py35 http://logs.openstack.org/59/670259/1/gate/cross-neutron-py35/309632b/ : SUCCESS in 25m 12s\n- cross-nova-py27 http://logs.openstack.org/59/670259/1/gate/cross-nova-py27/4cf48b7/ : SUCCESS in 13m 03s\n- cross-nova-py35 http://logs.openstack.org/59/670259/1/gate/cross-nova-py35/d50e081/ : SUCCESS in 13m 17s\n- cross-nova-functional http://logs.openstack.org/59/670259/1/gate/cross-nova-functional/6a96269/ : SUCCESS in 19m 49s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/59/670259/1/gate/cross-osc-build-sphinx-docs/1202e84/html/ : SUCCESS in 4m 51s\n- cross-swift-py27 http://logs.openstack.org/59/670259/1/gate/cross-swift-py27/9e7bc1a/ : SUCCESS in 7m 43s\n- cross-swift-py35 http://logs.openstack.org/59/670259/1/gate/cross-swift-py35/2020180/ : SUCCESS in 7m 52s\n- requirements-integration http://logs.openstack.org/59/670259/1/gate/requirements-integration/7ea596a/ : SUCCESS in 53m 22s\n- neutron-grenade-multinode http://logs.openstack.org/59/670259/1/gate/neutron-grenade-multinode/b43a949/ : SUCCESS in 1h 06m 28s\n- tempest-full http://logs.openstack.org/59/670259/1/gate/tempest-full/9dc5e55/ : SUCCESS in 1h 52m 09s\n- neutron-grenade http://logs.openstack.org/59/670259/1/gate/neutron-grenade/8da1b03/ : SUCCESS in 54m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6885b3fcd50c6e74ee8798abe8d4b6c448e0f45b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 18:03:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5f367647fa1b35b90361b061419fb00722ea543","revisions":{"b5f367647fa1b35b90361b061419fb00722ea543":{"kind":"REWORK","_number":1,"created":"2019-07-11 09:34:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/670259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/670259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/670259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/670259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/670259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/670259/1"}}},"commit":{"parents":[{"commit":"f397d14648448cf09b39749abd0aa330223d674a","subject":"Merge \"update constraint for tripleo-common to new release 10.8.0\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f397d14648448cf09b39749abd0aa330223d674a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-11 09:34:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-11 09:34:07.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 4.2.1","message":"update constraint for python-cinderclient to new release 4.2.1\n\nChange-Id: I5b452942a4a02843ddf12570e9b1ebf98dd7b40a\nmeta:version: 4.2.1\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Change-Id: Ic2de53f28ff21406d8f1831b18accca53fa48570\nmeta:release:Code-Review+1: Jay Bryant \u003cjungleboyj@electronicjungle.net\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5f367647fa1b35b90361b061419fb00722ea543"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5f367647fa1b35b90361b061419fb00722ea543"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
