)]}'
{"id":"openstack%2Frequirements~670265","triplet_id":"openstack%2Frequirements~stable%2Fstein~I2bd8bbc1b6b7c2a8dba987e70c7726542244e91d","project":"openstack/requirements","branch":"stable/stein","topic":"new-release","hashtags":[],"change_id":"I2bd8bbc1b6b7c2a8dba987e70c7726542244e91d","subject":"update constraint for os-brick to new release 2.8.2","status":"MERGED","created":"2019-07-11 09:46:10.000000000","updated":"2019-07-11 18:04:00.000000000","submitted":"2019-07-11 18:04:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670265-1562868240993-ef2d6f53","meta_rev_id":"ef9218b3578dceb6142f59eccbae3ae8c01c56b4","_number":670265,"virtual_id_number":670265,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-11 18:04:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-11 14:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-11 14:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-07-11 14:34:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 14:32:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-11 14:34:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-11 18:04:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6addf80019674c2f692135a188aa0069b6eca941","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-11 09:46:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77bf7027ec2f3b6d60d0b3bcadf2a4fd629c1396","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 11:43:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/65/670265/1/check/openstack-tox-pypy/96928c4/ : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/670265/1/check/openstack-tox-pep8/6d824fc/ : SUCCESS in 3m 14s\n- openstack-tox-py27 http://logs.openstack.org/65/670265/1/check/openstack-tox-py27/2bbc14c/ : SUCCESS in 3m 08s\n- openstack-tox-py35 http://logs.openstack.org/65/670265/1/check/openstack-tox-py35/a2557c9/ : SUCCESS in 3m 31s\n- openstack-tox-py36 http://logs.openstack.org/65/670265/1/check/openstack-tox-py36/5c4f342/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/65/670265/1/check/openstack-tox-validate/a94e29b/ : SUCCESS in 3m 11s\n- requirements-tox-babel http://logs.openstack.org/65/670265/1/check/requirements-tox-babel/bdb887a/ : SUCCESS in 2m 27s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/670265/1/check/requirements-tox-py27-check-uc/2f28ac9/ : SUCCESS in 6m 08s\n- requirements-tox-py35-check-uc http://logs.openstack.org/65/670265/1/check/requirements-tox-py35-check-uc/21ca972/ : SUCCESS in 6m 40s\n- requirements-tox-py36-check-uc http://logs.openstack.org/65/670265/1/check/requirements-tox-py36-check-uc/8509a95/ : SUCCESS in 6m 24s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/65/670265/1/check/cross-cinder-py27/accff41/ : SUCCESS in 8m 23s\n- cross-cinder-py35 http://logs.openstack.org/65/670265/1/check/cross-cinder-py35/5baf49c/ : SUCCESS in 8m 11s\n- cross-glance-py27 http://logs.openstack.org/65/670265/1/check/cross-glance-py27/cbd35c8/ : SUCCESS in 4m 09s\n- cross-glance-py35 http://logs.openstack.org/65/670265/1/check/cross-glance-py35/44e0835/ : SUCCESS in 4m 23s\n- cross-horizon-py27 http://logs.openstack.org/65/670265/1/check/cross-horizon-py27/b0a3fff/ : SUCCESS in 10m 34s\n- cross-horizon-py35 http://logs.openstack.org/65/670265/1/check/cross-horizon-py35/72a00e2/ : SUCCESS in 9m 40s\n- cross-horizon-npm http://logs.openstack.org/65/670265/1/check/cross-horizon-npm/45338cf/ : SUCCESS in 9m 33s\n- cross-keystone-py27 http://logs.openstack.org/65/670265/1/check/cross-keystone-py27/4eb29cc/ : SUCCESS in 25m 45s\n- cross-keystone-py35 http://logs.openstack.org/65/670265/1/check/cross-keystone-py35/5b294ee/ : SUCCESS in 25m 47s\n- cross-neutron-py27 http://logs.openstack.org/65/670265/1/check/cross-neutron-py27/8fda40f/ : SUCCESS in 21m 24s\n- cross-neutron-py35 http://logs.openstack.org/65/670265/1/check/cross-neutron-py35/67dba5d/ : SUCCESS in 22m 00s\n- cross-nova-py27 http://logs.openstack.org/65/670265/1/check/cross-nova-py27/c974aa1/ : SUCCESS in 13m 53s\n- cross-nova-py35 http://logs.openstack.org/65/670265/1/check/cross-nova-py35/38d97a1/ : SUCCESS in 11m 46s\n- cross-nova-functional http://logs.openstack.org/65/670265/1/check/cross-nova-functional/4d55ebb/ : SUCCESS in 21m 49s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/65/670265/1/check/cross-osc-build-sphinx-docs/20acf26/html/ : SUCCESS in 3m 41s\n- cross-swift-py27 http://logs.openstack.org/65/670265/1/check/cross-swift-py27/678f1fa/ : SUCCESS in 9m 11s\n- cross-swift-py35 http://logs.openstack.org/65/670265/1/check/cross-swift-py35/c73f774/ : SUCCESS in 8m 49s\n- requirements-integration http://logs.openstack.org/65/670265/1/check/requirements-integration/31c2c8f/ : SUCCESS in 52m 13s\n- neutron-grenade-multinode http://logs.openstack.org/65/670265/1/check/neutron-grenade-multinode/b39f0a9/ : SUCCESS in 1h 03m 16s\n- tempest-full http://logs.openstack.org/65/670265/1/check/tempest-full/f8b4e8e/ : SUCCESS in 1h 53m 11s\n- neutron-grenade http://logs.openstack.org/65/670265/1/check/neutron-grenade/cbaf621/ : FAILURE in 59m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6aa38c06fd1ec5d3ed12ca8013455b91d4762e92","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-11 11:46:45.000000000","message":"Patch Set 1:\n\nrecheck\n\ntempest.lib.exceptions.SSHTimeout: Connection to the 172.24.5.153 via SSH timed out.","accounts_in_message":[],"_revision_number":1},{"id":"e32a8ba8f507031c24fbc1c86ad9394d32259c81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 14:01:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/65/670265/1/check/openstack-tox-pypy/2bb31d4/ : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/65/670265/1/check/openstack-tox-pep8/cafe45d/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/65/670265/1/check/openstack-tox-py27/fc4787e/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/65/670265/1/check/openstack-tox-py35/ae02c7f/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/65/670265/1/check/openstack-tox-py36/18fe16c/ : SUCCESS in 5m 13s\n- openstack-tox-validate http://logs.openstack.org/65/670265/1/check/openstack-tox-validate/c464f3a/ : SUCCESS in 2m 41s\n- requirements-tox-babel http://logs.openstack.org/65/670265/1/check/requirements-tox-babel/2991a6e/ : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/670265/1/check/requirements-tox-py27-check-uc/524a2c3/ : SUCCESS in 9m 06s\n- requirements-tox-py35-check-uc http://logs.openstack.org/65/670265/1/check/requirements-tox-py35-check-uc/589c216/ : SUCCESS in 11m 26s\n- requirements-tox-py36-check-uc http://logs.openstack.org/65/670265/1/check/requirements-tox-py36-check-uc/b463798/ : SUCCESS in 9m 26s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/65/670265/1/check/cross-cinder-py27/2631827/ : SUCCESS in 10m 34s\n- cross-cinder-py35 http://logs.openstack.org/65/670265/1/check/cross-cinder-py35/cac518c/ : SUCCESS in 9m 36s\n- cross-glance-py27 http://logs.openstack.org/65/670265/1/check/cross-glance-py27/11042dd/ : SUCCESS in 4m 22s\n- cross-glance-py35 http://logs.openstack.org/65/670265/1/check/cross-glance-py35/c068283/ : SUCCESS in 5m 42s\n- cross-horizon-py27 http://logs.openstack.org/65/670265/1/check/cross-horizon-py27/5eb4c2f/ : SUCCESS in 11m 02s\n- cross-horizon-py35 http://logs.openstack.org/65/670265/1/check/cross-horizon-py35/e78771a/ : SUCCESS in 10m 19s\n- cross-horizon-npm http://logs.openstack.org/65/670265/1/check/cross-horizon-npm/6399a7a/ : SUCCESS in 6m 11s\n- cross-keystone-py27 http://logs.openstack.org/65/670265/1/check/cross-keystone-py27/2ecdd89/ : SUCCESS in 20m 18s\n- cross-keystone-py35 http://logs.openstack.org/65/670265/1/check/cross-keystone-py35/65ed9dd/ : SUCCESS in 27m 09s\n- cross-neutron-py27 http://logs.openstack.org/65/670265/1/check/cross-neutron-py27/d99605d/ : SUCCESS in 22m 01s\n- cross-neutron-py35 http://logs.openstack.org/65/670265/1/check/cross-neutron-py35/10f19eb/ : SUCCESS in 22m 57s\n- cross-nova-py27 http://logs.openstack.org/65/670265/1/check/cross-nova-py27/3abd078/ : SUCCESS in 17m 08s\n- cross-nova-py35 http://logs.openstack.org/65/670265/1/check/cross-nova-py35/7cc5bd5/ : SUCCESS in 22m 23s\n- cross-nova-functional http://logs.openstack.org/65/670265/1/check/cross-nova-functional/5ad7c19/ : SUCCESS in 23m 28s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/65/670265/1/check/cross-osc-build-sphinx-docs/ea84ac8/html/ : SUCCESS in 5m 49s\n- cross-swift-py27 http://logs.openstack.org/65/670265/1/check/cross-swift-py27/bc4119a/ : SUCCESS in 7m 42s\n- cross-swift-py35 http://logs.openstack.org/65/670265/1/check/cross-swift-py35/5bc9f09/ : SUCCESS in 7m 16s\n- requirements-integration http://logs.openstack.org/65/670265/1/check/requirements-integration/ea7c207/ : SUCCESS in 1h 03m 34s\n- neutron-grenade-multinode http://logs.openstack.org/65/670265/1/check/neutron-grenade-multinode/007fc43/ : SUCCESS in 1h 13m 59s\n- tempest-full http://logs.openstack.org/65/670265/1/check/tempest-full/b76627b/ : SUCCESS in 1h 44m 00s\n- neutron-grenade http://logs.openstack.org/65/670265/1/check/neutron-grenade/2fa5963/ : SUCCESS in 53m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1e388099e636fbcd035f9079a1792f7d1afb1919","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-11 14:32:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b2d220b28c4f2a6bbe3f7ee1cb102a6f25a44e15","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-11 14:34:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"18bcd9569beab6bd3d29c45e9b7ce6d7db0bd2d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 14:35:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8497850d12f4a71011d84c397d44c586b747a82b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 18:04:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"27ead1ec5409ef1f5191aed2abd704ce6225451b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 18:04:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/670265/1/gate/openstack-tox-pep8/7db1abe/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/65/670265/1/gate/openstack-tox-py27/795e9ae/ : SUCCESS in 3m 12s\n- openstack-tox-py35 http://logs.openstack.org/65/670265/1/gate/openstack-tox-py35/063575b/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/65/670265/1/gate/openstack-tox-py36/93e8eff/ : SUCCESS in 3m 36s\n- openstack-tox-validate http://logs.openstack.org/65/670265/1/gate/openstack-tox-validate/b57d021/ : SUCCESS in 3m 10s\n- requirements-tox-babel http://logs.openstack.org/65/670265/1/gate/requirements-tox-babel/4401b7b/ : SUCCESS in 3m 05s\n- requirements-tox-py27-check-uc http://logs.openstack.org/65/670265/1/gate/requirements-tox-py27-check-uc/3c29f55/ : SUCCESS in 6m 13s\n- requirements-tox-py35-check-uc http://logs.openstack.org/65/670265/1/gate/requirements-tox-py35-check-uc/d4d0643/ : SUCCESS in 7m 00s\n- cross-cinder-py27 http://logs.openstack.org/65/670265/1/gate/cross-cinder-py27/2e43e25/ : SUCCESS in 10m 53s\n- cross-cinder-py35 http://logs.openstack.org/65/670265/1/gate/cross-cinder-py35/34b82bd/ : SUCCESS in 9m 20s\n- cross-glance-py27 http://logs.openstack.org/65/670265/1/gate/cross-glance-py27/e37953c/ : SUCCESS in 4m 28s\n- cross-glance-py35 http://logs.openstack.org/65/670265/1/gate/cross-glance-py35/808d323/ : SUCCESS in 4m 34s\n- cross-horizon-py27 http://logs.openstack.org/65/670265/1/gate/cross-horizon-py27/292ad9b/ : SUCCESS in 9m 57s\n- cross-horizon-py35 http://logs.openstack.org/65/670265/1/gate/cross-horizon-py35/12b59d3/ : SUCCESS in 9m 23s\n- cross-horizon-npm http://logs.openstack.org/65/670265/1/gate/cross-horizon-npm/f11a7f8/ : SUCCESS in 7m 35s\n- cross-keystone-py27 http://logs.openstack.org/65/670265/1/gate/cross-keystone-py27/0947e00/ : SUCCESS in 22m 37s\n- cross-keystone-py35 http://logs.openstack.org/65/670265/1/gate/cross-keystone-py35/9439eeb/ : SUCCESS in 25m 06s\n- cross-neutron-py27 http://logs.openstack.org/65/670265/1/gate/cross-neutron-py27/ef35159/ : SUCCESS in 25m 10s\n- cross-neutron-py35 http://logs.openstack.org/65/670265/1/gate/cross-neutron-py35/48bc0c2/ : SUCCESS in 24m 26s\n- cross-nova-py27 http://logs.openstack.org/65/670265/1/gate/cross-nova-py27/3defa48/ : SUCCESS in 15m 23s\n- cross-nova-py35 http://logs.openstack.org/65/670265/1/gate/cross-nova-py35/2d04f28/ : SUCCESS in 13m 28s\n- cross-nova-functional http://logs.openstack.org/65/670265/1/gate/cross-nova-functional/dde8579/ : SUCCESS in 19m 34s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/65/670265/1/gate/cross-osc-build-sphinx-docs/97b5520/html/ : SUCCESS in 4m 23s\n- cross-swift-py27 http://logs.openstack.org/65/670265/1/gate/cross-swift-py27/88b45c1/ : SUCCESS in 8m 09s\n- cross-swift-py35 http://logs.openstack.org/65/670265/1/gate/cross-swift-py35/2fde389/ : SUCCESS in 7m 54s\n- requirements-integration http://logs.openstack.org/65/670265/1/gate/requirements-integration/17f286b/ : SUCCESS in 1h 02m 27s\n- neutron-grenade-multinode http://logs.openstack.org/65/670265/1/gate/neutron-grenade-multinode/08db08e/ : SUCCESS in 1h 08m 43s\n- tempest-full http://logs.openstack.org/65/670265/1/gate/tempest-full/3d717ee/ : SUCCESS in 1h 35m 12s\n- neutron-grenade http://logs.openstack.org/65/670265/1/gate/neutron-grenade/4727b61/ : SUCCESS in 52m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c7aa27274484f30db2b0bfea326cd92b9843224","revisions":{"1c7aa27274484f30db2b0bfea326cd92b9843224":{"kind":"REWORK","_number":1,"created":"2019-07-11 09:46:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/670265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/670265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/670265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/670265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/670265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/670265/1"}}},"commit":{"parents":[{"commit":"f397d14648448cf09b39749abd0aa330223d674a","subject":"Merge \"update constraint for tripleo-common to new release 10.8.0\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f397d14648448cf09b39749abd0aa330223d674a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-11 09:46:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-11 09:46:06.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.8.2","message":"update constraint for os-brick to new release 2.8.2\n\nChange-Id: I2bd8bbc1b6b7c2a8dba987e70c7726542244e91d\nmeta:version: 2.8.2\nmeta:diff-start: -\nmeta:series: stein\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Change-Id: I3d111a942fdce2b0b9936bb527a6d5a3927a8f57\nmeta:release:Code-Review+1: Jay Bryant \u003cjungleboyj@electronicjungle.net\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c7aa27274484f30db2b0bfea326cd92b9843224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1c7aa27274484f30db2b0bfea326cd92b9843224"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
