)]}'
{"id":"openstack%2Frequirements~670267","triplet_id":"openstack%2Frequirements~stable%2Frocky~Idd4cb44315b9a76d999f24479cb6d4ce5544a9e5","project":"openstack/requirements","branch":"stable/rocky","topic":"new-release","hashtags":[],"change_id":"Idd4cb44315b9a76d999f24479cb6d4ce5544a9e5","subject":"update constraint for os-brick to new release 2.5.8","status":"MERGED","created":"2019-07-11 09:56:11.000000000","updated":"2019-07-11 18:03:58.000000000","submitted":"2019-07-11 18:03:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670267-1562868238357-77280bbb","meta_rev_id":"3789e3c5a9e754b3cce73ccff8f188a531636105","_number":670267,"virtual_id_number":670267,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-11 18:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-07-11 13:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-11 14:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-07-11 14:26:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 13:46:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-11 14:26:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-11 18:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a926f41f1a007603109b535724e217d8b8999f1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-11 09:56:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d30fdbd085e5d8af2975b1e8612d9eb60d19f307","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 11:35:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pypy http://logs.openstack.org/67/670267/1/check/openstack-tox-pypy/2c867a3/ : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/670267/1/check/openstack-tox-pep8/2003cf5/ : SUCCESS in 3m 05s\n- openstack-tox-py27 http://logs.openstack.org/67/670267/1/check/openstack-tox-py27/7edcf3d/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/67/670267/1/check/openstack-tox-py35/33501ca/ : SUCCESS in 3m 05s\n- openstack-tox-validate http://logs.openstack.org/67/670267/1/check/openstack-tox-validate/0c5f844/ : SUCCESS in 2m 51s\n- requirements-tox-babel http://logs.openstack.org/67/670267/1/check/requirements-tox-babel/e4fdec1/ : SUCCESS in 2m 53s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/670267/1/check/requirements-tox-py27-check-uc/ca29170/ : SUCCESS in 6m 04s\n- requirements-tox-py35-check-uc http://logs.openstack.org/67/670267/1/check/requirements-tox-py35-check-uc/84ac1d0/ : SUCCESS in 6m 49s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/670267/1/check/requirements-tox-py36-check-uc/a228def/ : SUCCESS in 5m 35s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/67/670267/1/check/cross-cinder-py27/44b7464/ : SUCCESS in 8m 52s\n- cross-cinder-py35 http://logs.openstack.org/67/670267/1/check/cross-cinder-py35/a4f8fa9/ : SUCCESS in 8m 38s\n- cross-glance-py27 http://logs.openstack.org/67/670267/1/check/cross-glance-py27/dc49535/ : SUCCESS in 3m 56s\n- cross-glance-py35 http://logs.openstack.org/67/670267/1/check/cross-glance-py35/02a4008/ : SUCCESS in 4m 31s\n- cross-horizon-py27 http://logs.openstack.org/67/670267/1/check/cross-horizon-py27/a5ab497/ : SUCCESS in 9m 13s\n- cross-horizon-py35 http://logs.openstack.org/67/670267/1/check/cross-horizon-py35/4fcdc6d/ : SUCCESS in 8m 59s\n- cross-keystone-py27 http://logs.openstack.org/67/670267/1/check/cross-keystone-py27/a59c1cd/ : SUCCESS in 12m 28s\n- cross-keystone-py35 http://logs.openstack.org/67/670267/1/check/cross-keystone-py35/0b950d4/ : SUCCESS in 12m 19s\n- cross-neutron-py27 http://logs.openstack.org/67/670267/1/check/cross-neutron-py27/a09ffba/ : SUCCESS in 25m 15s\n- cross-neutron-py35 http://logs.openstack.org/67/670267/1/check/cross-neutron-py35/5a28d80/ : SUCCESS in 26m 06s\n- cross-nova-py27 http://logs.openstack.org/67/670267/1/check/cross-nova-py27/658ffe9/ : SUCCESS in 16m 30s\n- cross-nova-py35 http://logs.openstack.org/67/670267/1/check/cross-nova-py35/5bbc36f/ : FAILURE in 14m 34s\n- cross-nova-functional http://logs.openstack.org/67/670267/1/check/cross-nova-functional/5b449ca/ : SUCCESS in 18m 51s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/670267/1/check/cross-osc-build-sphinx-docs/f634145/html/ : SUCCESS in 4m 30s\n- cross-swift-py27 http://logs.openstack.org/67/670267/1/check/cross-swift-py27/4d63604/ : SUCCESS in 7m 41s\n- cross-swift-py35 http://logs.openstack.org/67/670267/1/check/cross-swift-py35/5144a3d/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/67/670267/1/check/openstack-tox-py36/a0dd186/ : SUCCESS in 3m 38s\n- requirements-integration http://logs.openstack.org/67/670267/1/check/requirements-integration/516b8cb/ : SUCCESS in 1h 12m 41s\n- neutron-grenade-multinode http://logs.openstack.org/67/670267/1/check/neutron-grenade-multinode/454752c/ : SUCCESS in 1h 02m 09s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/67/670267/1/check/legacy-tempest-dsvm-neutron-full/eb091b3/ : SUCCESS in 1h 26m 37s\n- neutron-grenade http://logs.openstack.org/67/670267/1/check/neutron-grenade/2eb0a90/ : SUCCESS in 55m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5e6d13744b96e4195e1b08a55d0f57afc2c2acf7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-11 11:48:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9569370d4030353b9f28c71433148d065c5167b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 13:46:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/67/670267/1/check/openstack-tox-pypy/370b82e/ : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/67/670267/1/check/openstack-tox-pep8/3607e49/ : SUCCESS in 2m 38s\n- openstack-tox-py27 http://logs.openstack.org/67/670267/1/check/openstack-tox-py27/00936c6/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/67/670267/1/check/openstack-tox-py35/0d89152/ : SUCCESS in 4m 10s\n- openstack-tox-validate http://logs.openstack.org/67/670267/1/check/openstack-tox-validate/859db82/ : SUCCESS in 3m 35s\n- requirements-tox-babel http://logs.openstack.org/67/670267/1/check/requirements-tox-babel/ec2a25d/ : SUCCESS in 2m 51s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/670267/1/check/requirements-tox-py27-check-uc/541b423/ : SUCCESS in 9m 35s\n- requirements-tox-py35-check-uc http://logs.openstack.org/67/670267/1/check/requirements-tox-py35-check-uc/5a45099/ : SUCCESS in 9m 15s\n- requirements-tox-py36-check-uc http://logs.openstack.org/67/670267/1/check/requirements-tox-py36-check-uc/dc85fc0/ : SUCCESS in 5m 58s (non-voting)\n- cross-cinder-py27 http://logs.openstack.org/67/670267/1/check/cross-cinder-py27/e7cf4b2/ : SUCCESS in 8m 52s\n- cross-cinder-py35 http://logs.openstack.org/67/670267/1/check/cross-cinder-py35/f4eecf2/ : SUCCESS in 10m 39s\n- cross-glance-py27 http://logs.openstack.org/67/670267/1/check/cross-glance-py27/3ae9d53/ : SUCCESS in 5m 13s\n- cross-glance-py35 http://logs.openstack.org/67/670267/1/check/cross-glance-py35/50b04e0/ : SUCCESS in 6m 03s\n- cross-horizon-py27 http://logs.openstack.org/67/670267/1/check/cross-horizon-py27/d8efd16/ : SUCCESS in 10m 09s\n- cross-horizon-py35 http://logs.openstack.org/67/670267/1/check/cross-horizon-py35/f8ae630/ : SUCCESS in 8m 42s\n- cross-keystone-py27 http://logs.openstack.org/67/670267/1/check/cross-keystone-py27/2f0e4ae/ : SUCCESS in 13m 31s\n- cross-keystone-py35 http://logs.openstack.org/67/670267/1/check/cross-keystone-py35/34ed658/ : SUCCESS in 10m 52s\n- cross-neutron-py27 http://logs.openstack.org/67/670267/1/check/cross-neutron-py27/7332d42/ : SUCCESS in 27m 37s\n- cross-neutron-py35 http://logs.openstack.org/67/670267/1/check/cross-neutron-py35/fe7f0d2/ : SUCCESS in 30m 27s\n- cross-nova-py27 http://logs.openstack.org/67/670267/1/check/cross-nova-py27/e9c63df/ : SUCCESS in 12m 56s\n- cross-nova-py35 http://logs.openstack.org/67/670267/1/check/cross-nova-py35/674b44d/ : SUCCESS in 16m 10s\n- cross-nova-functional http://logs.openstack.org/67/670267/1/check/cross-nova-functional/2acccc5/ : SUCCESS in 19m 25s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/670267/1/check/cross-osc-build-sphinx-docs/41a292a/html/ : SUCCESS in 5m 20s\n- cross-swift-py27 http://logs.openstack.org/67/670267/1/check/cross-swift-py27/d2e2f72/ : SUCCESS in 8m 07s\n- cross-swift-py35 http://logs.openstack.org/67/670267/1/check/cross-swift-py35/37e3712/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/67/670267/1/check/openstack-tox-py36/baaa260/ : SUCCESS in 5m 01s\n- requirements-integration http://logs.openstack.org/67/670267/1/check/requirements-integration/55ed043/ : SUCCESS in 48m 21s\n- neutron-grenade-multinode http://logs.openstack.org/67/670267/1/check/neutron-grenade-multinode/b68d7ee/ : SUCCESS in 1h 08m 43s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/67/670267/1/check/legacy-tempest-dsvm-neutron-full/5098291/ : SUCCESS in 1h 21m 33s\n- neutron-grenade http://logs.openstack.org/67/670267/1/check/neutron-grenade/73f3439/ : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":1},{"id":"396d2261ec363cb40b166f09c6b623df08d47de1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-11 13:46:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a69f944450f15d85eb48d9dae018d51153248d2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-11 14:26:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c3cbd64101530398067fe021f95847109024ec4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 14:27:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a26d76ecae2ac5781c9e50351b31ca002207766b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 18:03:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/67/670267/1/gate/openstack-tox-pep8/47f6091/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/67/670267/1/gate/openstack-tox-py27/37dd014/ : SUCCESS in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/67/670267/1/gate/openstack-tox-py35/34af73b/ : SUCCESS in 3m 47s\n- openstack-tox-validate http://logs.openstack.org/67/670267/1/gate/openstack-tox-validate/fb55e75/ : SUCCESS in 3m 28s\n- requirements-tox-babel http://logs.openstack.org/67/670267/1/gate/requirements-tox-babel/539a6e2/ : SUCCESS in 3m 12s\n- requirements-tox-py27-check-uc http://logs.openstack.org/67/670267/1/gate/requirements-tox-py27-check-uc/dfd2a10/ : SUCCESS in 7m 26s\n- requirements-tox-py35-check-uc http://logs.openstack.org/67/670267/1/gate/requirements-tox-py35-check-uc/1bff059/ : SUCCESS in 6m 55s\n- cross-cinder-py27 http://logs.openstack.org/67/670267/1/gate/cross-cinder-py27/0356edb/ : SUCCESS in 8m 45s\n- cross-cinder-py35 http://logs.openstack.org/67/670267/1/gate/cross-cinder-py35/73bd958/ : SUCCESS in 12m 41s\n- cross-glance-py27 http://logs.openstack.org/67/670267/1/gate/cross-glance-py27/2f99720/ : SUCCESS in 4m 41s\n- cross-glance-py35 http://logs.openstack.org/67/670267/1/gate/cross-glance-py35/a72b7e7/ : SUCCESS in 4m 29s\n- cross-horizon-py27 http://logs.openstack.org/67/670267/1/gate/cross-horizon-py27/224552d/ : SUCCESS in 11m 03s\n- cross-horizon-py35 http://logs.openstack.org/67/670267/1/gate/cross-horizon-py35/692c2ba/ : SUCCESS in 9m 13s\n- cross-keystone-py27 http://logs.openstack.org/67/670267/1/gate/cross-keystone-py27/65a3536/ : SUCCESS in 12m 06s\n- cross-keystone-py35 http://logs.openstack.org/67/670267/1/gate/cross-keystone-py35/cbbcecc/ : SUCCESS in 13m 37s\n- cross-neutron-py27 http://logs.openstack.org/67/670267/1/gate/cross-neutron-py27/8b02704/ : SUCCESS in 29m 45s\n- cross-neutron-py35 http://logs.openstack.org/67/670267/1/gate/cross-neutron-py35/056298c/ : SUCCESS in 29m 45s\n- cross-nova-py27 http://logs.openstack.org/67/670267/1/gate/cross-nova-py27/4272c78/ : SUCCESS in 14m 34s\n- cross-nova-py35 http://logs.openstack.org/67/670267/1/gate/cross-nova-py35/e1abf23/ : SUCCESS in 12m 54s\n- cross-nova-functional http://logs.openstack.org/67/670267/1/gate/cross-nova-functional/de2dcb2/ : SUCCESS in 23m 07s\n- cross-osc-build-sphinx-docs http://logs.openstack.org/67/670267/1/gate/cross-osc-build-sphinx-docs/62ac09a/html/ : SUCCESS in 5m 28s\n- cross-swift-py27 http://logs.openstack.org/67/670267/1/gate/cross-swift-py27/3c2a137/ : SUCCESS in 8m 05s\n- cross-swift-py35 http://logs.openstack.org/67/670267/1/gate/cross-swift-py35/15f2d8f/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/67/670267/1/gate/openstack-tox-py36/d06e74b/ : SUCCESS in 3m 35s\n- requirements-integration http://logs.openstack.org/67/670267/1/gate/requirements-integration/711e790/ : SUCCESS in 1h 07m 56s\n- neutron-grenade-multinode http://logs.openstack.org/67/670267/1/gate/neutron-grenade-multinode/2d679a8/ : SUCCESS in 1h 19m 03s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/67/670267/1/gate/legacy-tempest-dsvm-neutron-full/0ff14e8/ : SUCCESS in 1h 35m 04s\n- neutron-grenade http://logs.openstack.org/67/670267/1/gate/neutron-grenade/e981505/ : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":1},{"id":"71e6dff626ad361caeef56d87ef72ff7fd725bba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 18:03:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccd043fbbc1d2f9ace5e155fd1ed61e98c5efffb","revisions":{"ccd043fbbc1d2f9ace5e155fd1ed61e98c5efffb":{"kind":"REWORK","_number":1,"created":"2019-07-11 09:56:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/670267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/670267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/670267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/670267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/670267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/670267/1"}}},"commit":{"parents":[{"commit":"cae28776071c8d10391c5d7e3684c5011b83e1f2","subject":"Merge \"update constraint for tripleo-common to new release 9.6.0\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cae28776071c8d10391c5d7e3684c5011b83e1f2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-11 09:56:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-11 09:56:07.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.5.8","message":"update constraint for os-brick to new release 2.5.8\n\nChange-Id: Idd4cb44315b9a76d999f24479cb6d4ce5544a9e5\nmeta:version: 2.5.8\nmeta:diff-start: -\nmeta:series: rocky\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta:release:Commit: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Change-Id: I9f0cf3a08686f30106c5f0278f0fc764868d3a5c\nmeta:release:Code-Review+1: Jay Bryant \u003cjungleboyj@electronicjungle.net\u003e\nmeta:release:Code-Review+2: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+1: Matt Riedemann \u003cmriedem.os@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccd043fbbc1d2f9ace5e155fd1ed61e98c5efffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccd043fbbc1d2f9ace5e155fd1ed61e98c5efffb"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
