)]}'
{"id":"openstack%2Frequirements~670709","triplet_id":"openstack%2Frequirements~master~I16d198621fb63dea7daa0c69363056a4669ad14d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I16d198621fb63dea7daa0c69363056a4669ad14d","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-07-14 06:16:10.000000000","updated":"2019-07-16 23:22:03.000000000","submitted":"2019-07-16 23:22:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670709-1563319323072-6d7173bf","meta_rev_id":"8f750d767a19de5793d7958441180c9d674eb157","_number":670709,"virtual_id_number":670709,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-16 23:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-16 19:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-16 19:07:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 19:07:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-16 23:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3bfdc5a979697e3900dc48bcc9c53e060ca33f7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-14 06:16:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"329e08bbdbf99cad15756f9e7baf42352f32dca5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-14 08:03:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/09/670709/1/check/openstack-tox-docs/3ac6581/html/ : SUCCESS in 2m 59s\n- openstack-tox-pypy http://logs.openstack.org/09/670709/1/check/openstack-tox-pypy/9abdc28/ : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/09/670709/1/check/openstack-tox-pep8/4d32896/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/09/670709/1/check/openstack-tox-py27/afd57b8/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/09/670709/1/check/openstack-tox-py36/773a76b/ : SUCCESS in 5m 16s\n- openstack-tox-py37 http://logs.openstack.org/09/670709/1/check/openstack-tox-py37/8df6fa4/ : SUCCESS in 3m 09s\n- openstack-tox-validate http://logs.openstack.org/09/670709/1/check/openstack-tox-validate/33a5775/ : SUCCESS in 4m 32s\n- requirements-tox-babel http://logs.openstack.org/09/670709/1/check/requirements-tox-babel/773dc42/ : SUCCESS in 4m 17s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/670709/1/check/requirements-tox-py27-check-uc/c411ae5/ : FAILURE in 5m 57s\n- requirements-tox-py36-check-uc http://logs.openstack.org/09/670709/1/check/requirements-tox-py36-check-uc/bf14183/ : FAILURE in 6m 13s\n- cross-cinder-py27 http://logs.openstack.org/09/670709/1/check/cross-cinder-py27/bbcb6fb/ : SUCCESS in 8m 49s\n- cross-cinder-py36 http://logs.openstack.org/09/670709/1/check/cross-cinder-py36/f6498b7/ : SUCCESS in 9m 40s\n- cross-glance-py27 http://logs.openstack.org/09/670709/1/check/cross-glance-py27/6727984/ : SUCCESS in 4m 45s\n- cross-glance-py36 http://logs.openstack.org/09/670709/1/check/cross-glance-py36/a8ca5a8/ : SUCCESS in 5m 01s\n- cross-horizon-py27 http://logs.openstack.org/09/670709/1/check/cross-horizon-py27/1db0a05/ : SUCCESS in 11m 09s\n- cross-horizon-py36 http://logs.openstack.org/09/670709/1/check/cross-horizon-py36/86fc7d4/ : SUCCESS in 9m 39s\n- cross-horizon-npm http://logs.openstack.org/09/670709/1/check/cross-horizon-npm/0649e29/ : SUCCESS in 6m 25s\n- cross-keystone-py27 http://logs.openstack.org/09/670709/1/check/cross-keystone-py27/2861d06/ : SUCCESS in 27m 09s\n- cross-keystone-py36 http://logs.openstack.org/09/670709/1/check/cross-keystone-py36/2fc753a/ : SUCCESS in 26m 14s\n- cross-neutron-py27 http://logs.openstack.org/09/670709/1/check/cross-neutron-py27/e8977f7/ : SUCCESS in 22m 37s\n- cross-neutron-py36 http://logs.openstack.org/09/670709/1/check/cross-neutron-py36/15641c7/ : SUCCESS in 22m 51s\n- cross-nova-py27 http://logs.openstack.org/09/670709/1/check/cross-nova-py27/4bf0619/ : SUCCESS in 19m 41s\n- cross-nova-py36 http://logs.openstack.org/09/670709/1/check/cross-nova-py36/5e8dcbe/ : SUCCESS in 17m 32s\n- cross-nova-functional http://logs.openstack.org/09/670709/1/check/cross-nova-functional/78c21d9/ : SUCCESS in 25m 49s\n- cross-osc-tox-docs http://logs.openstack.org/09/670709/1/check/cross-osc-tox-docs/6403aa0/html/ : SUCCESS in 3m 50s\n- cross-swift-py27 http://logs.openstack.org/09/670709/1/check/cross-swift-py27/285ff9e/ : SUCCESS in 8m 45s\n- cross-swift-py36 http://logs.openstack.org/09/670709/1/check/cross-swift-py36/1367e17/ : SUCCESS in 9m 12s\n- requirements-integration http://logs.openstack.org/09/670709/1/check/requirements-integration/786fd10/ : SUCCESS in 1h 13m 41s\n- neutron-grenade-multinode http://logs.openstack.org/09/670709/1/check/neutron-grenade-multinode/17918d0/ : SUCCESS in 1h 04m 10s\n- tempest-full http://logs.openstack.org/09/670709/1/check/tempest-full/4f1e6ef/ : SUCCESS in 1h 40m 05s\n- neutron-grenade http://logs.openstack.org/09/670709/1/check/neutron-grenade/bb328ae/ : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"346a36db5ff804c1021ceacf2498879ab2be1107","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-15 06:18:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1aefd70214da9f3bfb8d48c3da3ecd890bbf92cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 08:29:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/09/670709/2/check/openstack-tox-docs/298b1e8/html/ : SUCCESS in 3m 31s\n- openstack-tox-pypy http://logs.openstack.org/09/670709/2/check/openstack-tox-pypy/7868a48/ : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/09/670709/2/check/openstack-tox-pep8/3117526/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/09/670709/2/check/openstack-tox-py27/a1cb815/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/09/670709/2/check/openstack-tox-py36/9b1ddd8/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/09/670709/2/check/openstack-tox-py37/6e3c10a/ : SUCCESS in 4m 08s\n- openstack-tox-validate http://logs.openstack.org/09/670709/2/check/openstack-tox-validate/df1206f/ : SUCCESS in 4m 21s\n- requirements-tox-babel http://logs.openstack.org/09/670709/2/check/requirements-tox-babel/7e815bb/ : SUCCESS in 3m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/670709/2/check/requirements-tox-py27-check-uc/a1aa825/ : FAILURE in 8m 15s\n- requirements-tox-py36-check-uc http://logs.openstack.org/09/670709/2/check/requirements-tox-py36-check-uc/f5de4c7/ : FAILURE in 10m 45s\n- cross-cinder-py27 http://logs.openstack.org/09/670709/2/check/cross-cinder-py27/97b1cf7/ : SUCCESS in 8m 36s\n- cross-cinder-py36 http://logs.openstack.org/09/670709/2/check/cross-cinder-py36/5d6402d/ : SUCCESS in 7m 54s\n- cross-glance-py27 http://logs.openstack.org/09/670709/2/check/cross-glance-py27/e3d42c4/ : SUCCESS in 4m 39s\n- cross-glance-py36 http://logs.openstack.org/09/670709/2/check/cross-glance-py36/e57a513/ : SUCCESS in 4m 39s\n- cross-horizon-py27 http://logs.openstack.org/09/670709/2/check/cross-horizon-py27/dd2bcff/ : SUCCESS in 10m 07s\n- cross-horizon-py36 http://logs.openstack.org/09/670709/2/check/cross-horizon-py36/e7b24de/ : SUCCESS in 8m 24s\n- cross-horizon-npm http://logs.openstack.org/09/670709/2/check/cross-horizon-npm/bdca90c/ : SUCCESS in 8m 40s\n- cross-keystone-py27 http://logs.openstack.org/09/670709/2/check/cross-keystone-py27/a2c66a4/ : SUCCESS in 27m 55s\n- cross-keystone-py36 http://logs.openstack.org/09/670709/2/check/cross-keystone-py36/8518a27/ : SUCCESS in 29m 15s\n- cross-neutron-py27 http://logs.openstack.org/09/670709/2/check/cross-neutron-py27/8a63386/ : SUCCESS in 21m 56s\n- cross-neutron-py36 http://logs.openstack.org/09/670709/2/check/cross-neutron-py36/49565a4/ : SUCCESS in 27m 45s\n- cross-nova-py27 http://logs.openstack.org/09/670709/2/check/cross-nova-py27/9b80209/ : SUCCESS in 20m 13s\n- cross-nova-py36 http://logs.openstack.org/09/670709/2/check/cross-nova-py36/9de73d9/ : SUCCESS in 15m 55s\n- cross-nova-functional http://logs.openstack.org/09/670709/2/check/cross-nova-functional/6b57333/ : SUCCESS in 21m 54s\n- cross-osc-tox-docs http://logs.openstack.org/09/670709/2/check/cross-osc-tox-docs/dc61bae/html/ : SUCCESS in 3m 54s\n- cross-swift-py27 http://logs.openstack.org/09/670709/2/check/cross-swift-py27/c4c82d1/ : SUCCESS in 8m 55s\n- cross-swift-py36 http://logs.openstack.org/09/670709/2/check/cross-swift-py36/a4a1f38/ : SUCCESS in 7m 55s\n- requirements-integration http://logs.openstack.org/09/670709/2/check/requirements-integration/961c128/ : SUCCESS in 54m 07s\n- neutron-grenade-multinode http://logs.openstack.org/09/670709/2/check/neutron-grenade-multinode/3b85c78/ : SUCCESS in 1h 04m 50s\n- tempest-full http://logs.openstack.org/09/670709/2/check/tempest-full/196c093/ : SUCCESS in 1h 40m 41s\n- neutron-grenade http://logs.openstack.org/09/670709/2/check/neutron-grenade/5720434/ : SUCCESS in 56m 21s","accounts_in_message":[],"_revision_number":2},{"id":"9648e8ae99b9c2fb906688686ffbdf3393ed2242","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-16 06:11:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48487bb53cb29a4feb48957b7d58a6d09474e0da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 09:13:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/09/670709/3/check/openstack-tox-docs/8ef15cc/html/ : SUCCESS in 3m 44s\n- openstack-tox-pypy http://logs.openstack.org/09/670709/3/check/openstack-tox-pypy/7af40f6/ : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/09/670709/3/check/openstack-tox-pep8/fbda56b/ : SUCCESS in 3m 02s\n- openstack-tox-py27 http://logs.openstack.org/09/670709/3/check/openstack-tox-py27/d75e8fc/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/09/670709/3/check/openstack-tox-py36/e9bef3a/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/09/670709/3/check/openstack-tox-py37/ef2b3d8/ : SUCCESS in 3m 06s\n- openstack-tox-validate http://logs.openstack.org/09/670709/3/check/openstack-tox-validate/0e1ffde/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/09/670709/3/check/requirements-tox-babel/e0c54ce/ : SUCCESS in 3m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/670709/3/check/requirements-tox-py27-check-uc/9e2afd2/ : FAILURE in 20m 31s\n- requirements-tox-py36-check-uc http://logs.openstack.org/09/670709/3/check/requirements-tox-py36-check-uc/1b4801e/ : FAILURE in 20m 30s\n- cross-cinder-py27 http://logs.openstack.org/09/670709/3/check/cross-cinder-py27/0aa9338/ : SUCCESS in 9m 07s\n- cross-cinder-py36 http://logs.openstack.org/09/670709/3/check/cross-cinder-py36/ea06520/ : SUCCESS in 8m 55s\n- cross-glance-py27 http://logs.openstack.org/09/670709/3/check/cross-glance-py27/983a340/ : SUCCESS in 4m 32s\n- cross-glance-py36 http://logs.openstack.org/09/670709/3/check/cross-glance-py36/1a3b77d/ : SUCCESS in 7m 30s\n- cross-horizon-py27 http://logs.openstack.org/09/670709/3/check/cross-horizon-py27/b9e2b26/ : SUCCESS in 13m 17s\n- cross-horizon-py36 http://logs.openstack.org/09/670709/3/check/cross-horizon-py36/b45238b/ : SUCCESS in 9m 42s\n- cross-horizon-npm http://logs.openstack.org/09/670709/3/check/cross-horizon-npm/b920bde/ : SUCCESS in 6m 03s\n- cross-keystone-py27 http://logs.openstack.org/09/670709/3/check/cross-keystone-py27/691c826/ : SUCCESS in 26m 27s\n- cross-keystone-py36 http://logs.openstack.org/09/670709/3/check/cross-keystone-py36/a662e05/ : SUCCESS in 29m 03s\n- cross-neutron-py27 http://logs.openstack.org/09/670709/3/check/cross-neutron-py27/762a950/ : SUCCESS in 25m 18s\n- cross-neutron-py36 http://logs.openstack.org/09/670709/3/check/cross-neutron-py36/0b2b6b9/ : SUCCESS in 21m 44s\n- cross-nova-py27 http://logs.openstack.org/09/670709/3/check/cross-nova-py27/da355af/ : SUCCESS in 18m 32s\n- cross-nova-py36 http://logs.openstack.org/09/670709/3/check/cross-nova-py36/5eb148c/ : SUCCESS in 15m 36s\n- cross-nova-functional http://logs.openstack.org/09/670709/3/check/cross-nova-functional/06f290e/ : SUCCESS in 20m 54s\n- cross-osc-tox-docs http://logs.openstack.org/09/670709/3/check/cross-osc-tox-docs/58b79d3/html/ : SUCCESS in 3m 30s\n- cross-swift-py27 http://logs.openstack.org/09/670709/3/check/cross-swift-py27/f981ac2/ : SUCCESS in 7m 28s\n- cross-swift-py36 http://logs.openstack.org/09/670709/3/check/cross-swift-py36/44676f1/ : SUCCESS in 8m 00s\n- requirements-integration http://logs.openstack.org/09/670709/3/check/requirements-integration/9e3690f/ : SUCCESS in 48m 40s\n- neutron-grenade-multinode http://logs.openstack.org/09/670709/3/check/neutron-grenade-multinode/279bc61/ : SUCCESS in 1h 12m 36s\n- tempest-full http://logs.openstack.org/09/670709/3/check/tempest-full/03d114e/ : SUCCESS in 1h 45m 02s\n- neutron-grenade http://logs.openstack.org/09/670709/3/check/neutron-grenade/0861989/ : SUCCESS in 58m 53s","accounts_in_message":[],"_revision_number":3},{"id":"497000cc10c533f913351595f58e28a17d48b5bc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-16 15:10:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"848086dc4b50e8080146c40c37ab0be259b41b13","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-16 15:23:20.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"ba62f99f579fdc6e4cf2a12bbe164a7785ce3237","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 18:55:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/09/670709/4/check/openstack-tox-docs/4cd5406/html/ : SUCCESS in 3m 17s\n- openstack-tox-pypy http://logs.openstack.org/09/670709/4/check/openstack-tox-pypy/cf86a47/ : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/09/670709/4/check/openstack-tox-pep8/3aca199/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/09/670709/4/check/openstack-tox-py27/aa33710/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/09/670709/4/check/openstack-tox-py36/1636a25/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/09/670709/4/check/openstack-tox-py37/d2a16f2/ : SUCCESS in 3m 24s\n- openstack-tox-validate http://logs.openstack.org/09/670709/4/check/openstack-tox-validate/edaf81e/ : SUCCESS in 3m 45s\n- requirements-tox-babel http://logs.openstack.org/09/670709/4/check/requirements-tox-babel/10c6974/ : SUCCESS in 3m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/670709/4/check/requirements-tox-py27-check-uc/f37cd12/ : SUCCESS in 6m 11s\n- requirements-tox-py36-check-uc http://logs.openstack.org/09/670709/4/check/requirements-tox-py36-check-uc/296f21e/ : SUCCESS in 6m 22s\n- cross-cinder-py27 http://logs.openstack.org/09/670709/4/check/cross-cinder-py27/6348f19/ : SUCCESS in 8m 21s\n- cross-cinder-py36 http://logs.openstack.org/09/670709/4/check/cross-cinder-py36/2ccb4a8/ : SUCCESS in 7m 53s\n- cross-glance-py27 http://logs.openstack.org/09/670709/4/check/cross-glance-py27/7f60696/ : SUCCESS in 5m 19s\n- cross-glance-py36 http://logs.openstack.org/09/670709/4/check/cross-glance-py36/c58d7b8/ : SUCCESS in 5m 10s\n- cross-horizon-py27 http://logs.openstack.org/09/670709/4/check/cross-horizon-py27/4eacc78/ : SUCCESS in 11m 31s\n- cross-horizon-py36 http://logs.openstack.org/09/670709/4/check/cross-horizon-py36/e3274f0/ : SUCCESS in 10m 02s\n- cross-horizon-npm http://logs.openstack.org/09/670709/4/check/cross-horizon-npm/807498e/ : SUCCESS in 6m 31s\n- cross-keystone-py27 http://logs.openstack.org/09/670709/4/check/cross-keystone-py27/bba5bf9/ : SUCCESS in 25m 01s\n- cross-keystone-py36 http://logs.openstack.org/09/670709/4/check/cross-keystone-py36/72c9fae/ : SUCCESS in 26m 51s\n- cross-neutron-py27 http://logs.openstack.org/09/670709/4/check/cross-neutron-py27/008ecff/ : SUCCESS in 20m 31s\n- cross-neutron-py36 http://logs.openstack.org/09/670709/4/check/cross-neutron-py36/4d150bd/ : SUCCESS in 24m 40s\n- cross-nova-py27 http://logs.openstack.org/09/670709/4/check/cross-nova-py27/4b2380a/ : SUCCESS in 17m 07s\n- cross-nova-py36 http://logs.openstack.org/09/670709/4/check/cross-nova-py36/d552b6a/ : SUCCESS in 17m 09s\n- cross-nova-functional http://logs.openstack.org/09/670709/4/check/cross-nova-functional/eb8de29/ : SUCCESS in 21m 02s\n- cross-osc-tox-docs http://logs.openstack.org/09/670709/4/check/cross-osc-tox-docs/82a4667/html/ : SUCCESS in 4m 39s\n- cross-swift-py27 http://logs.openstack.org/09/670709/4/check/cross-swift-py27/901522a/ : SUCCESS in 8m 36s\n- cross-swift-py36 http://logs.openstack.org/09/670709/4/check/cross-swift-py36/af1e9cd/ : SUCCESS in 8m 38s\n- requirements-integration http://logs.openstack.org/09/670709/4/check/requirements-integration/a77ba7e/ : SUCCESS in 1h 38m 46s\n- neutron-grenade-multinode http://logs.openstack.org/09/670709/4/check/neutron-grenade-multinode/e43199f/ : SUCCESS in 1h 13m 53s\n- tempest-full http://logs.openstack.org/09/670709/4/check/tempest-full/da23f4e/ : SUCCESS in 1h 40m 41s\n- neutron-grenade http://logs.openstack.org/09/670709/4/check/neutron-grenade/96242ca/ : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":4},{"id":"42c74aae31670a27f6d12a1938ef0932e8f1efce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-16 19:07:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d4c0a6d38c730a661f4dafcfedb9b67c2c39efc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 19:09:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ea147943b23a21cbdffeabc75458ac7c3b8f1bf8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:22:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/09/670709/4/gate/openstack-tox-docs/380ee69/html/ : SUCCESS in 3m 15s\n- openstack-tox-pep8 http://logs.openstack.org/09/670709/4/gate/openstack-tox-pep8/a008070/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/09/670709/4/gate/openstack-tox-py27/864bc1b/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/09/670709/4/gate/openstack-tox-py36/dde111a/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/09/670709/4/gate/openstack-tox-py37/82edaa8/ : SUCCESS in 3m 51s\n- openstack-tox-validate http://logs.openstack.org/09/670709/4/gate/openstack-tox-validate/498e580/ : SUCCESS in 3m 07s\n- requirements-tox-babel http://logs.openstack.org/09/670709/4/gate/requirements-tox-babel/4ab2250/ : SUCCESS in 3m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/09/670709/4/gate/requirements-tox-py27-check-uc/6266ef9/ : SUCCESS in 7m 05s\n- requirements-tox-py36-check-uc http://logs.openstack.org/09/670709/4/gate/requirements-tox-py36-check-uc/97fa433/ : SUCCESS in 6m 23s\n- cross-cinder-py27 http://logs.openstack.org/09/670709/4/gate/cross-cinder-py27/f877b39/ : SUCCESS in 8m 41s\n- cross-cinder-py36 http://logs.openstack.org/09/670709/4/gate/cross-cinder-py36/f8dc27f/ : SUCCESS in 7m 42s\n- cross-glance-py27 http://logs.openstack.org/09/670709/4/gate/cross-glance-py27/160f2b2/ : SUCCESS in 3m 50s\n- cross-glance-py36 http://logs.openstack.org/09/670709/4/gate/cross-glance-py36/be028d0/ : SUCCESS in 4m 52s\n- cross-horizon-py27 http://logs.openstack.org/09/670709/4/gate/cross-horizon-py27/e275b37/ : SUCCESS in 9m 35s\n- cross-horizon-py36 http://logs.openstack.org/09/670709/4/gate/cross-horizon-py36/a5b173b/ : SUCCESS in 9m 17s\n- cross-horizon-npm http://logs.openstack.org/09/670709/4/gate/cross-horizon-npm/4f34603/ : SUCCESS in 6m 42s\n- cross-keystone-py27 http://logs.openstack.org/09/670709/4/gate/cross-keystone-py27/a9b9a64/ : SUCCESS in 25m 32s\n- cross-keystone-py36 http://logs.openstack.org/09/670709/4/gate/cross-keystone-py36/d19d91d/ : SUCCESS in 27m 34s\n- cross-neutron-py27 http://logs.openstack.org/09/670709/4/gate/cross-neutron-py27/44ede0f/ : SUCCESS in 25m 55s\n- cross-neutron-py36 http://logs.openstack.org/09/670709/4/gate/cross-neutron-py36/7427e09/ : SUCCESS in 25m 50s\n- cross-nova-py27 http://logs.openstack.org/09/670709/4/gate/cross-nova-py27/56950c0/ : SUCCESS in 14m 08s\n- cross-nova-py36 http://logs.openstack.org/09/670709/4/gate/cross-nova-py36/a7b46dd/ : SUCCESS in 20m 41s\n- cross-nova-functional http://logs.openstack.org/09/670709/4/gate/cross-nova-functional/f68ac89/ : SUCCESS in 24m 38s\n- cross-osc-tox-docs http://logs.openstack.org/09/670709/4/gate/cross-osc-tox-docs/90c354f/html/ : SUCCESS in 4m 03s\n- cross-swift-py27 http://logs.openstack.org/09/670709/4/gate/cross-swift-py27/a1dee79/ : SUCCESS in 7m 46s\n- cross-swift-py36 http://logs.openstack.org/09/670709/4/gate/cross-swift-py36/41f9957/ : SUCCESS in 8m 00s\n- requirements-integration http://logs.openstack.org/09/670709/4/gate/requirements-integration/553c751/ : SUCCESS in 49m 54s\n- neutron-grenade-multinode http://logs.openstack.org/09/670709/4/gate/neutron-grenade-multinode/40ab0f7/ : SUCCESS in 1h 01m 58s\n- tempest-full http://logs.openstack.org/09/670709/4/gate/tempest-full/c840478/ : SUCCESS in 1h 32m 11s\n- neutron-grenade http://logs.openstack.org/09/670709/4/gate/neutron-grenade/f9ae84c/ : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":4},{"id":"8f750d767a19de5793d7958441180c9d674eb157","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:22:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1f24a4fae45d852f9afcdef0d36aa5b276b371dc","revisions":{"c712f163478bd01e49706f6e3bfe3b7576336a6d":{"kind":"REWORK","_number":1,"created":"2019-07-14 06:16:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/670709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/670709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/670709/1"}}},"commit":{"parents":[{"commit":"1329f56a3ca3dfaf2a1e65a5808736632ed20be0","subject":"Merge \"block pypowervm\u003d\u003d\u003d1.1.22 for capping taskflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1329f56a3ca3dfaf2a1e65a5808736632ed20be0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-14 06:16:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-14 06:16:08.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I16d198621fb63dea7daa0c69363056a4669ad14d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c712f163478bd01e49706f6e3bfe3b7576336a6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c712f163478bd01e49706f6e3bfe3b7576336a6d"}]},"branch":"refs/heads/master"},"8ebba5696e455c1fa8dfef5070d8ffee9a1f5cb6":{"kind":"REWORK","_number":2,"created":"2019-07-15 06:18:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/670709/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/670709/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/670709/2"}}},"commit":{"parents":[{"commit":"1329f56a3ca3dfaf2a1e65a5808736632ed20be0","subject":"Merge \"block pypowervm\u003d\u003d\u003d1.1.22 for capping taskflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1329f56a3ca3dfaf2a1e65a5808736632ed20be0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-15 06:18:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-15 06:18:51.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I16d198621fb63dea7daa0c69363056a4669ad14d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ebba5696e455c1fa8dfef5070d8ffee9a1f5cb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8ebba5696e455c1fa8dfef5070d8ffee9a1f5cb6"}]},"branch":"refs/heads/master"},"4d41e4fc54c2485c38624e1ee7810ba60779ea6a":{"kind":"REWORK","_number":3,"created":"2019-07-16 06:11:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/670709/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/670709/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/670709/3"}}},"commit":{"parents":[{"commit":"05d41f3b8ca2fa66e2f3b8f4849061911400c91a","subject":"Merge \"update constraint for python-swiftclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05d41f3b8ca2fa66e2f3b8f4849061911400c91a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-16 06:11:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-16 06:11:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I16d198621fb63dea7daa0c69363056a4669ad14d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d41e4fc54c2485c38624e1ee7810ba60779ea6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d41e4fc54c2485c38624e1ee7810ba60779ea6a"}]},"branch":"refs/heads/master"},"1f24a4fae45d852f9afcdef0d36aa5b276b371dc":{"kind":"REWORK","_number":4,"created":"2019-07-16 15:10:36.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/09/670709/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/670709/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/670709/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/670709/4"}}},"commit":{"parents":[{"commit":"05d41f3b8ca2fa66e2f3b8f4849061911400c91a","subject":"Merge \"update constraint for python-swiftclient to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05d41f3b8ca2fa66e2f3b8f4849061911400c91a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-16 06:11:04.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-16 15:10:21.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I16d198621fb63dea7daa0c69363056a4669ad14d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f24a4fae45d852f9afcdef0d36aa5b276b371dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1f24a4fae45d852f9afcdef0d36aa5b276b371dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
