)]}'
{"id":"openstack%2Frequirements~670855","triplet_id":"openstack%2Frequirements~master~I5610904a8cbac51b2622d72e23ba3e5efa5b1743","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5610904a8cbac51b2622d72e23ba3e5efa5b1743","subject":"update constraint for keystoneauth1 to new release 3.15.0","status":"MERGED","created":"2019-07-15 15:03:24.000000000","updated":"2019-07-15 22:49:56.000000000","submitted":"2019-07-15 22:49:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670855-1563230996091-9977374c","meta_rev_id":"01ee439af16090c1db4f4365d2a6fdd38e6b9596","_number":670855,"virtual_id_number":670855,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-07-15 22:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-15 21:03:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-07-15 21:56:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-15 21:03:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 21:03:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-15 21:56:52.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-15 22:49:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"673a8bf9d856c778011945dc657b6bd8a7a7cf8c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-15 15:03:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24d97be7b51265c8656357270e29b618b562f7ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 20:20:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/670855/1/check/openstack-tox-docs/4bf77eb/html/ : SUCCESS in 3m 38s\n- openstack-tox-pypy http://logs.openstack.org/55/670855/1/check/openstack-tox-pypy/110bec0/ : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/55/670855/1/check/openstack-tox-pep8/95d7419/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/55/670855/1/check/openstack-tox-py27/c93ff7d/ : SUCCESS in 3m 48s\n- openstack-tox-py36 http://logs.openstack.org/55/670855/1/check/openstack-tox-py36/2f973b6/ : SUCCESS in 5m 00s\n- openstack-tox-py37 http://logs.openstack.org/55/670855/1/check/openstack-tox-py37/c8f4159/ : SUCCESS in 4m 43s\n- openstack-tox-validate http://logs.openstack.org/55/670855/1/check/openstack-tox-validate/6b96927/ : SUCCESS in 2m 50s\n- requirements-tox-babel http://logs.openstack.org/55/670855/1/check/requirements-tox-babel/6d68698/ : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc http://logs.openstack.org/55/670855/1/check/requirements-tox-py27-check-uc/139e57b/ : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc http://logs.openstack.org/55/670855/1/check/requirements-tox-py36-check-uc/b54448b/ : SUCCESS in 7m 02s\n- cross-cinder-py27 http://logs.openstack.org/55/670855/1/check/cross-cinder-py27/5a41a5a/ : SUCCESS in 11m 52s\n- cross-cinder-py36 http://logs.openstack.org/55/670855/1/check/cross-cinder-py36/6cccac2/ : SUCCESS in 8m 48s\n- cross-glance-py27 http://logs.openstack.org/55/670855/1/check/cross-glance-py27/7b9a588/ : SUCCESS in 4m 19s\n- cross-glance-py36 http://logs.openstack.org/55/670855/1/check/cross-glance-py36/429af24/ : SUCCESS in 4m 03s\n- cross-horizon-py27 http://logs.openstack.org/55/670855/1/check/cross-horizon-py27/7187d3a/ : SUCCESS in 10m 21s\n- cross-horizon-py36 http://logs.openstack.org/55/670855/1/check/cross-horizon-py36/4b35e0e/ : SUCCESS in 9m 33s\n- cross-horizon-npm http://logs.openstack.org/55/670855/1/check/cross-horizon-npm/a5ce147/ : SUCCESS in 6m 46s\n- cross-keystone-py27 http://logs.openstack.org/55/670855/1/check/cross-keystone-py27/c2016fc/ : SUCCESS in 36m 31s\n- cross-keystone-py36 http://logs.openstack.org/55/670855/1/check/cross-keystone-py36/10d3db3/ : SUCCESS in 29m 05s\n- cross-neutron-py27 http://logs.openstack.org/55/670855/1/check/cross-neutron-py27/ab08038/ : SUCCESS in 34m 58s\n- cross-neutron-py36 http://logs.openstack.org/55/670855/1/check/cross-neutron-py36/ff90c36/ : SUCCESS in 34m 42s\n- cross-nova-py27 http://logs.openstack.org/55/670855/1/check/cross-nova-py27/914d6f4/ : SUCCESS in 21m 57s\n- cross-nova-py36 http://logs.openstack.org/55/670855/1/check/cross-nova-py36/8cc4d1e/ : SUCCESS in 19m 03s\n- cross-nova-functional http://logs.openstack.org/55/670855/1/check/cross-nova-functional/47362e4/ : SUCCESS in 23m 46s\n- cross-osc-tox-docs http://logs.openstack.org/55/670855/1/check/cross-osc-tox-docs/69c0e7c/html/ : SUCCESS in 4m 17s\n- cross-swift-py27 http://logs.openstack.org/55/670855/1/check/cross-swift-py27/7dd6747/ : SUCCESS in 9m 51s\n- cross-swift-py36 http://logs.openstack.org/55/670855/1/check/cross-swift-py36/4946ef9/ : SUCCESS in 7m 42s\n- requirements-integration http://logs.openstack.org/55/670855/1/check/requirements-integration/f0c2786/ : SUCCESS in 1h 10m 42s\n- neutron-grenade-multinode http://logs.openstack.org/55/670855/1/check/neutron-grenade-multinode/852b159/ : SUCCESS in 2h 05m 09s\n- tempest-full http://logs.openstack.org/55/670855/1/check/tempest-full/61b923a/ : SUCCESS in 1h 44m 48s\n- neutron-grenade http://logs.openstack.org/55/670855/1/check/neutron-grenade/46f5a4c/ : SUCCESS in 59m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9b8fd53656ab5af80ef3cde2144d342e00c28e5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-15 21:03:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8640bbdc5c1ce3420f432293ffacb0b2f67a920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 21:03:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1397b7dbb1f5e8e74004f5b9dec50668ec5bd949","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-15 21:56:52.000000000","message":"Patch Set 1: Code-Review+1\n\ngo go go","accounts_in_message":[],"_revision_number":1},{"id":"a742f48e932ba41a2642c4cb7bd7b4d3db89b488","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 22:49:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/55/670855/1/gate/openstack-tox-docs/fae4d61/html/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/55/670855/1/gate/openstack-tox-pep8/761457d/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/55/670855/1/gate/openstack-tox-py27/eaf674e/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/55/670855/1/gate/openstack-tox-py36/7dd2a7a/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/55/670855/1/gate/openstack-tox-py37/dccad33/ : SUCCESS in 4m 36s\n- openstack-tox-validate http://logs.openstack.org/55/670855/1/gate/openstack-tox-validate/90e93a1/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/55/670855/1/gate/requirements-tox-babel/0b0842c/ : SUCCESS in 3m 09s\n- requirements-tox-py27-check-uc http://logs.openstack.org/55/670855/1/gate/requirements-tox-py27-check-uc/b60d6f0/ : SUCCESS in 7m 12s\n- requirements-tox-py36-check-uc http://logs.openstack.org/55/670855/1/gate/requirements-tox-py36-check-uc/bf98943/ : SUCCESS in 7m 45s\n- cross-cinder-py27 http://logs.openstack.org/55/670855/1/gate/cross-cinder-py27/e0cb79a/ : SUCCESS in 9m 17s\n- cross-cinder-py36 http://logs.openstack.org/55/670855/1/gate/cross-cinder-py36/dbe0eb8/ : SUCCESS in 9m 55s\n- cross-glance-py27 http://logs.openstack.org/55/670855/1/gate/cross-glance-py27/09d65ca/ : SUCCESS in 4m 46s\n- cross-glance-py36 http://logs.openstack.org/55/670855/1/gate/cross-glance-py36/c410c9c/ : SUCCESS in 4m 12s\n- cross-horizon-py27 http://logs.openstack.org/55/670855/1/gate/cross-horizon-py27/94bb883/ : SUCCESS in 11m 02s\n- cross-horizon-py36 http://logs.openstack.org/55/670855/1/gate/cross-horizon-py36/285b7ae/ : SUCCESS in 8m 19s\n- cross-horizon-npm http://logs.openstack.org/55/670855/1/gate/cross-horizon-npm/632040e/ : SUCCESS in 7m 24s\n- cross-keystone-py27 http://logs.openstack.org/55/670855/1/gate/cross-keystone-py27/83acc90/ : SUCCESS in 22m 29s\n- cross-keystone-py36 http://logs.openstack.org/55/670855/1/gate/cross-keystone-py36/feb485d/ : SUCCESS in 27m 55s\n- cross-neutron-py27 http://logs.openstack.org/55/670855/1/gate/cross-neutron-py27/daab392/ : SUCCESS in 23m 55s\n- cross-neutron-py36 http://logs.openstack.org/55/670855/1/gate/cross-neutron-py36/ca3c110/ : SUCCESS in 25m 00s\n- cross-nova-py27 http://logs.openstack.org/55/670855/1/gate/cross-nova-py27/9207df9/ : SUCCESS in 19m 52s\n- cross-nova-py36 http://logs.openstack.org/55/670855/1/gate/cross-nova-py36/8b8be75/ : SUCCESS in 16m 17s\n- cross-nova-functional http://logs.openstack.org/55/670855/1/gate/cross-nova-functional/788fb60/ : SUCCESS in 24m 08s\n- cross-osc-tox-docs http://logs.openstack.org/55/670855/1/gate/cross-osc-tox-docs/4411493/html/ : SUCCESS in 3m 56s\n- cross-swift-py27 http://logs.openstack.org/55/670855/1/gate/cross-swift-py27/908f1b0/ : SUCCESS in 9m 52s\n- cross-swift-py36 http://logs.openstack.org/55/670855/1/gate/cross-swift-py36/b318775/ : SUCCESS in 9m 06s\n- requirements-integration http://logs.openstack.org/55/670855/1/gate/requirements-integration/82e2f00/ : SUCCESS in 1h 00m 12s\n- neutron-grenade-multinode http://logs.openstack.org/55/670855/1/gate/neutron-grenade-multinode/e70982b/ : SUCCESS in 1h 03m 07s\n- tempest-full http://logs.openstack.org/55/670855/1/gate/tempest-full/3b998f5/ : SUCCESS in 1h 39m 09s\n- neutron-grenade http://logs.openstack.org/55/670855/1/gate/neutron-grenade/6abb0fb/ : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f478745cd0433017e92b76996219f566d6501c4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 22:49:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99566d8324eaf308d2aec01db62206a4d2d792ca","revisions":{"99566d8324eaf308d2aec01db62206a4d2d792ca":{"kind":"REWORK","_number":1,"created":"2019-07-15 15:03:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/670855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/670855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/670855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/670855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/670855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/670855/1"}}},"commit":{"parents":[{"commit":"1329f56a3ca3dfaf2a1e65a5808736632ed20be0","subject":"Merge \"block pypowervm\u003d\u003d\u003d1.1.22 for capping taskflow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1329f56a3ca3dfaf2a1e65a5808736632ed20be0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-15 15:03:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-15 15:03:22.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 3.15.0","message":"update constraint for keystoneauth1 to new release 3.15.0\n\nChange-Id: I5610904a8cbac51b2622d72e23ba3e5efa5b1743\nmeta:version: 3.15.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Change-Id: Ie11383ac5c0dd2ee9b8a1730b7e365362e7c0084\nmeta:release:Code-Review+1: Colleen Murphy \u003ccolleen@gazlene.net\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Andriy Shevchenko \u003chuang.zhiping@99cloud.net\u003e\nmeta:release:Code-Review+1: 98k \u003c18552437190@163.com\u003e\nmeta:release:Code-Review+1: Vieri \u003c15050873171@163.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99566d8324eaf308d2aec01db62206a4d2d792ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/99566d8324eaf308d2aec01db62206a4d2d792ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
