)]}'
{"id":"openstack%2Frequirements~672178","triplet_id":"openstack%2Frequirements~master~Ie0fabdcb9ee4701ebd7bc3d03fe64483d8220224","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie0fabdcb9ee4701ebd7bc3d03fe64483d8220224","subject":"update constraint for openstacksdk to new release 0.31.2","status":"MERGED","created":"2019-07-22 21:15:43.000000000","updated":"2019-07-24 16:25:09.000000000","submitted":"2019-07-24 16:25:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672178-1563985509782-67437c28","meta_rev_id":"6756168ecd0ef696f6f752dc168f9b32f0118411","_number":672178,"virtual_id_number":672178,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-07-24 08:20:06.000000000","_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"value":2,"date":"2019-07-24 16:25:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-24 12:32:01.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-23 18:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-23 18:27:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-07-23 18:27:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 01:35:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-23 18:27:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-24 08:20:06.000000000","updated_by":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"reviewer":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"state":"REVIEWER"},{"updated":"2019-07-24 12:32:01.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-07-24 16:25:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c7045cfb28e0396e97da9a7825dc5a320a7a8e3","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-22 21:15:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36be0ffe8ed9f4e4cd5a80ad9ee05d0319f18ef5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 22:55:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/1/check/openstack-tox-docs/7b4d96e/html/ : SUCCESS in 2m 54s\n- openstack-tox-pypy http://logs.openstack.org/78/672178/1/check/openstack-tox-pypy/034d0eb/ : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/1/check/openstack-tox-pep8/bc36e7c/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/1/check/openstack-tox-py27/e66362b/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/1/check/openstack-tox-py36/66ba7d5/ : SUCCESS in 3m 33s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/1/check/openstack-tox-py37/6e7a0f1/ : SUCCESS in 4m 18s\n- openstack-tox-validate http://logs.openstack.org/78/672178/1/check/openstack-tox-validate/7b6a3b3/ : SUCCESS in 3m 33s\n- requirements-tox-babel http://logs.openstack.org/78/672178/1/check/requirements-tox-babel/b90cacd/ : SUCCESS in 3m 35s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/1/check/requirements-tox-py27-check-uc/1db9518/ : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/1/check/requirements-tox-py36-check-uc/bb13359/ : SUCCESS in 17m 14s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/1/check/cross-cinder-py27/69e8bef/ : SUCCESS in 8m 35s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/1/check/cross-cinder-py36/df3817b/ : SUCCESS in 7m 42s\n- cross-glance-py27 http://logs.openstack.org/78/672178/1/check/cross-glance-py27/4e263da/ : SUCCESS in 5m 27s\n- cross-glance-py36 http://logs.openstack.org/78/672178/1/check/cross-glance-py36/863daa3/ : SUCCESS in 5m 13s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/1/check/cross-horizon-py27/3583c38/ : SUCCESS in 11m 24s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/1/check/cross-horizon-py36/8c4af0d/ : SUCCESS in 10m 13s\n- cross-horizon-npm http://logs.openstack.org/78/672178/1/check/cross-horizon-npm/b650540/ : SUCCESS in 6m 15s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/1/check/cross-keystone-py27/1359b2e/ : SUCCESS in 24m 29s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/1/check/cross-keystone-py36/bd9445c/ : SUCCESS in 26m 01s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/1/check/cross-neutron-py27/9cea585/ : SUCCESS in 22m 53s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/1/check/cross-neutron-py36/6fad885/ : SUCCESS in 22m 16s\n- cross-nova-py27 http://logs.openstack.org/78/672178/1/check/cross-nova-py27/a2d51fd/ : SUCCESS in 17m 47s\n- cross-nova-py36 http://logs.openstack.org/78/672178/1/check/cross-nova-py36/ab30aa6/ : SUCCESS in 16m 02s\n- cross-nova-functional http://logs.openstack.org/78/672178/1/check/cross-nova-functional/5afeec0/ : SUCCESS in 26m 25s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/1/check/cross-osc-tox-docs/68cb43c/html/ : SUCCESS in 4m 03s\n- cross-swift-py27 http://logs.openstack.org/78/672178/1/check/cross-swift-py27/520d4c6/ : SUCCESS in 8m 55s\n- cross-swift-py36 http://logs.openstack.org/78/672178/1/check/cross-swift-py36/ff9fca8/ : SUCCESS in 8m 17s\n- requirements-integration http://logs.openstack.org/78/672178/1/check/requirements-integration/6d632f7/ : SUCCESS in 51m 14s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/1/check/neutron-grenade-multinode/c7534e4/ : SUCCESS in 1h 06m 10s\n- tempest-full http://logs.openstack.org/78/672178/1/check/tempest-full/ce40201/ : SUCCESS in 1h 34m 25s\n- neutron-grenade http://logs.openstack.org/78/672178/1/check/neutron-grenade/92b89b2/ : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a83a11720edfaa7e3a6e7d90ca826d7c15ed38d5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-23 01:35:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba69f35f712f7cf7a2259fc0db0f044f7c70f855","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 01:36:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eab64b811bfa9f18699f0ec86ff072c3d74c4454","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 04:36:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/1/gate/openstack-tox-docs/7985264/html/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/1/gate/openstack-tox-pep8/5f41c4d/ : SUCCESS in 4m 42s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/1/gate/openstack-tox-py27/dd01c41/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/1/gate/openstack-tox-py36/56b240c/ : SUCCESS in 4m 26s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/1/gate/openstack-tox-py37/d85c9d8/ : SUCCESS in 5m 32s\n- openstack-tox-validate http://logs.openstack.org/78/672178/1/gate/openstack-tox-validate/e7ba411/ : SUCCESS in 3m 01s\n- requirements-tox-babel http://logs.openstack.org/78/672178/1/gate/requirements-tox-babel/b377b44/ : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/1/gate/requirements-tox-py27-check-uc/5662500/ : SUCCESS in 6m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/1/gate/requirements-tox-py36-check-uc/b9898ea/ : SUCCESS in 6m 27s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/1/gate/cross-cinder-py27/c710606/ : SUCCESS in 8m 57s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/1/gate/cross-cinder-py36/bdb3764/ : SUCCESS in 8m 21s\n- cross-glance-py27 http://logs.openstack.org/78/672178/1/gate/cross-glance-py27/dcce1cc/ : SUCCESS in 4m 13s\n- cross-glance-py36 http://logs.openstack.org/78/672178/1/gate/cross-glance-py36/a091124/ : SUCCESS in 4m 26s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/1/gate/cross-horizon-py27/2f4f7ca/ : SUCCESS in 9m 30s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/1/gate/cross-horizon-py36/a3d42f4/ : SUCCESS in 9m 10s\n- cross-horizon-npm http://logs.openstack.org/78/672178/1/gate/cross-horizon-npm/05721dc/ : SUCCESS in 6m 29s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/1/gate/cross-keystone-py27/e169dd9/ : SUCCESS in 27m 29s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/1/gate/cross-keystone-py36/3ac51f1/ : SUCCESS in 25m 55s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/1/gate/cross-neutron-py27/320cba1/ : SUCCESS in 29m 03s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/1/gate/cross-neutron-py36/564ace0/ : FAILURE in 27m 44s\n- cross-nova-py27 http://logs.openstack.org/78/672178/1/gate/cross-nova-py27/0f71f95/ : SUCCESS in 16m 20s\n- cross-nova-py36 http://logs.openstack.org/78/672178/1/gate/cross-nova-py36/dcfbfc6/ : SUCCESS in 16m 35s\n- cross-nova-functional http://logs.openstack.org/78/672178/1/gate/cross-nova-functional/655f53c/ : SUCCESS in 27m 31s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/1/gate/cross-osc-tox-docs/94aeb62/html/ : SUCCESS in 3m 49s\n- cross-swift-py27 http://logs.openstack.org/78/672178/1/gate/cross-swift-py27/714fdb1/ : SUCCESS in 11m 00s\n- cross-swift-py36 http://logs.openstack.org/78/672178/1/gate/cross-swift-py36/8a59404/ : SUCCESS in 9m 32s\n- requirements-integration http://logs.openstack.org/78/672178/1/gate/requirements-integration/3b1a0fe/ : SUCCESS in 50m 57s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/1/gate/neutron-grenade-multinode/f0cece6/ : SUCCESS in 1h 06m 05s\n- tempest-full http://logs.openstack.org/78/672178/1/gate/tempest-full/0396c0f/ : SUCCESS in 1h 35m 48s\n- neutron-grenade http://logs.openstack.org/78/672178/1/gate/neutron-grenade/7745800/ : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7af3986e36379f3e6da38bf4f2b23feb3ceeb15d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-23 05:20:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76562efd64b8af5f27b8a8610adcab4cba5c35c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:12:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/1/check/openstack-tox-docs/e04922a/html/ : SUCCESS in 3m 58s\n- openstack-tox-pypy http://logs.openstack.org/78/672178/1/check/openstack-tox-pypy/a2e26f9/ : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/1/check/openstack-tox-pep8/f600284/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/1/check/openstack-tox-py27/35f8b8d/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/1/check/openstack-tox-py36/3beba70/ : SUCCESS in 3m 28s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/1/check/openstack-tox-py37/318fe73/ : SUCCESS in 4m 51s\n- openstack-tox-validate http://logs.openstack.org/78/672178/1/check/openstack-tox-validate/a31d869/ : SUCCESS in 5m 10s\n- requirements-tox-babel http://logs.openstack.org/78/672178/1/check/requirements-tox-babel/2e4b5ea/ : SUCCESS in 3m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/1/check/requirements-tox-py27-check-uc/b25a731/ : SUCCESS in 7m 18s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/1/check/requirements-tox-py36-check-uc/d3cc115/ : SUCCESS in 6m 56s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/1/check/cross-cinder-py27/2820645/ : SUCCESS in 9m 18s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/1/check/cross-cinder-py36/2396058/ : SUCCESS in 10m 39s\n- cross-glance-py27 http://logs.openstack.org/78/672178/1/check/cross-glance-py27/54c7f2b/ : SUCCESS in 6m 11s\n- cross-glance-py36 http://logs.openstack.org/78/672178/1/check/cross-glance-py36/0ed1299/ : SUCCESS in 4m 53s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/1/check/cross-horizon-py27/595187e/ : SUCCESS in 12m 21s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/1/check/cross-horizon-py36/487dfb6/ : SUCCESS in 8m 31s\n- cross-horizon-npm http://logs.openstack.org/78/672178/1/check/cross-horizon-npm/079a94b/ : SUCCESS in 9m 21s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/1/check/cross-keystone-py27/b64bf98/ : SUCCESS in 34m 01s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/1/check/cross-keystone-py36/16362df/ : SUCCESS in 28m 25s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/1/check/cross-neutron-py27/ae07703/ : SUCCESS in 23m 24s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/1/check/cross-neutron-py36/7f12336/ : SUCCESS in 25m 38s\n- cross-nova-py27 http://logs.openstack.org/78/672178/1/check/cross-nova-py27/f8718cb/ : SUCCESS in 15m 57s\n- cross-nova-py36 http://logs.openstack.org/78/672178/1/check/cross-nova-py36/b71b691/ : SUCCESS in 16m 37s\n- cross-nova-functional http://logs.openstack.org/78/672178/1/check/cross-nova-functional/16201d5/ : SUCCESS in 25m 47s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/1/check/cross-osc-tox-docs/93b0c74/html/ : SUCCESS in 3m 34s\n- cross-swift-py27 http://logs.openstack.org/78/672178/1/check/cross-swift-py27/8bab4e2/ : SUCCESS in 7m 38s\n- cross-swift-py36 http://logs.openstack.org/78/672178/1/check/cross-swift-py36/ded7a3e/ : SUCCESS in 12m 37s\n- requirements-integration http://logs.openstack.org/78/672178/1/check/requirements-integration/1bd4235/ : SUCCESS in 53m 51s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/1/check/neutron-grenade-multinode/b46cbe3/ : SUCCESS in 1h 12m 48s\n- tempest-full http://logs.openstack.org/78/672178/1/check/tempest-full/4902253/ : SUCCESS in 1h 50m 17s\n- neutron-grenade http://logs.openstack.org/78/672178/1/check/neutron-grenade/c8c3c49/ : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":1},{"id":"60d9c0caaacbb61ea0b2b9f18e887084371bcc51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:12:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1109973e4a12d9706e9e3b08f1ec203b20f791d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:00:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/1/gate/openstack-tox-docs/c8477d0/html/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/1/gate/openstack-tox-pep8/1b8ef8a/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/1/gate/openstack-tox-py27/69decb7/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/1/gate/openstack-tox-py36/cfa9355/ : SUCCESS in 3m 49s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/1/gate/openstack-tox-py37/3bfa0b1/ : SUCCESS in 4m 02s\n- openstack-tox-validate http://logs.openstack.org/78/672178/1/gate/openstack-tox-validate/836c2ff/ : SUCCESS in 3m 27s\n- requirements-tox-babel http://logs.openstack.org/78/672178/1/gate/requirements-tox-babel/8ecd91b/ : SUCCESS in 2m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/1/gate/requirements-tox-py27-check-uc/5e68f9a/ : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/1/gate/requirements-tox-py36-check-uc/67d2895/ : SUCCESS in 6m 20s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/1/gate/cross-cinder-py27/8bcccd7/ : SUCCESS in 8m 53s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/1/gate/cross-cinder-py36/7601ae0/ : SUCCESS in 7m 31s\n- cross-glance-py27 http://logs.openstack.org/78/672178/1/gate/cross-glance-py27/da35964/ : SUCCESS in 5m 19s\n- cross-glance-py36 http://logs.openstack.org/78/672178/1/gate/cross-glance-py36/4643c69/ : SUCCESS in 5m 19s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/1/gate/cross-horizon-py27/34b0ced/ : SUCCESS in 9m 40s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/1/gate/cross-horizon-py36/405d699/ : SUCCESS in 9m 36s\n- cross-horizon-npm http://logs.openstack.org/78/672178/1/gate/cross-horizon-npm/aa36167/ : SUCCESS in 7m 20s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/1/gate/cross-keystone-py27/3f69335/ : SUCCESS in 26m 09s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/1/gate/cross-keystone-py36/f949dd0/ : SUCCESS in 29m 38s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/1/gate/cross-neutron-py27/b07ad81/ : SUCCESS in 22m 20s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/1/gate/cross-neutron-py36/39080b9/ : SUCCESS in 22m 26s\n- cross-nova-py27 http://logs.openstack.org/78/672178/1/gate/cross-nova-py27/30cb99e/ : SUCCESS in 16m 24s\n- cross-nova-py36 http://logs.openstack.org/78/672178/1/gate/cross-nova-py36/068c6a7/ : SUCCESS in 19m 46s\n- cross-nova-functional http://logs.openstack.org/78/672178/1/gate/cross-nova-functional/2f00b74/ : SUCCESS in 25m 36s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/1/gate/cross-osc-tox-docs/8073107/html/ : SUCCESS in 3m 39s\n- cross-swift-py27 http://logs.openstack.org/78/672178/1/gate/cross-swift-py27/af58beb/ : SUCCESS in 9m 44s\n- cross-swift-py36 http://logs.openstack.org/78/672178/1/gate/cross-swift-py36/7bf7fe8/ : SUCCESS in 10m 37s\n- requirements-integration http://logs.openstack.org/78/672178/1/gate/requirements-integration/e38c3da/ : FAILURE in 34m 52s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/1/gate/neutron-grenade-multinode/af5f169/ : SUCCESS in 1h 16m 43s\n- tempest-full http://logs.openstack.org/78/672178/1/gate/tempest-full/ef284d5/ : SUCCESS in 1h 41m 32s\n- neutron-grenade http://logs.openstack.org/78/672178/1/gate/neutron-grenade/060cd99/ : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0a46ad94686a19de7eb67e4a57af19ea229ecd97","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-23 09:11:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b0a92914825410ec477dfa6bbfbe51216caec88c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 11:06:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/1/check/openstack-tox-docs/42b2117/html/ : SUCCESS in 3m 35s\n- openstack-tox-pypy http://logs.openstack.org/78/672178/1/check/openstack-tox-pypy/6dba0fd/ : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/1/check/openstack-tox-pep8/5dc1d88/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/1/check/openstack-tox-py27/d65658a/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/1/check/openstack-tox-py36/8135114/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/1/check/openstack-tox-py37/aa4a6e4/ : SUCCESS in 3m 26s\n- openstack-tox-validate http://logs.openstack.org/78/672178/1/check/openstack-tox-validate/dcefb8b/ : SUCCESS in 3m 09s\n- requirements-tox-babel http://logs.openstack.org/78/672178/1/check/requirements-tox-babel/5e6694e/ : SUCCESS in 3m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/1/check/requirements-tox-py27-check-uc/8e28798/ : SUCCESS in 6m 50s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/1/check/requirements-tox-py36-check-uc/bc06b0b/ : SUCCESS in 6m 37s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/1/check/cross-cinder-py27/4166ee8/ : SUCCESS in 11m 26s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/1/check/cross-cinder-py36/290240c/ : SUCCESS in 8m 45s\n- cross-glance-py27 http://logs.openstack.org/78/672178/1/check/cross-glance-py27/a03ec53/ : SUCCESS in 6m 43s\n- cross-glance-py36 http://logs.openstack.org/78/672178/1/check/cross-glance-py36/8985242/ : SUCCESS in 5m 28s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/1/check/cross-horizon-py27/9c7c316/ : SUCCESS in 10m 49s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/1/check/cross-horizon-py36/a6f29b7/ : SUCCESS in 10m 14s\n- cross-horizon-npm http://logs.openstack.org/78/672178/1/check/cross-horizon-npm/be9963b/ : SUCCESS in 6m 12s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/1/check/cross-keystone-py27/34ce29a/ : SUCCESS in 33m 28s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/1/check/cross-keystone-py36/99b5162/ : SUCCESS in 35m 35s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/1/check/cross-neutron-py27/50a3551/ : SUCCESS in 23m 43s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/1/check/cross-neutron-py36/42752bf/ : SUCCESS in 25m 16s\n- cross-nova-py27 http://logs.openstack.org/78/672178/1/check/cross-nova-py27/f19ff11/ : SUCCESS in 22m 44s\n- cross-nova-py36 http://logs.openstack.org/78/672178/1/check/cross-nova-py36/f296282/ : SUCCESS in 14m 40s\n- cross-nova-functional http://logs.openstack.org/78/672178/1/check/cross-nova-functional/08e7531/ : SUCCESS in 23m 20s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/1/check/cross-osc-tox-docs/c022f97/html/ : SUCCESS in 3m 53s\n- cross-swift-py27 http://logs.openstack.org/78/672178/1/check/cross-swift-py27/62726ca/ : SUCCESS in 10m 09s\n- cross-swift-py36 http://logs.openstack.org/78/672178/1/check/cross-swift-py36/32168c8/ : SUCCESS in 7m 49s\n- requirements-integration http://logs.openstack.org/78/672178/1/check/requirements-integration/bc5d96d/ : FAILURE in 39m 06s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/1/check/neutron-grenade-multinode/2387877/ : SUCCESS in 1h 24m 43s\n- tempest-full http://logs.openstack.org/78/672178/1/check/tempest-full/ef7dbb9/ : FAILURE in 21m 16s\n- neutron-grenade http://logs.openstack.org/78/672178/1/check/neutron-grenade/50c7f39/ : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":1},{"id":"4c824ed2e77f26882ba5062829037d135a2c9a40","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-23 14:16:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"089fe951a5b1af1e522852e1ea2674460fe47aae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 16:42:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/1/check/openstack-tox-docs/ed3ada4/html/ : SUCCESS in 4m 24s\n- openstack-tox-pypy http://logs.openstack.org/78/672178/1/check/openstack-tox-pypy/c9f90f0/ : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/1/check/openstack-tox-pep8/751b7fb/ : SUCCESS in 3m 40s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/1/check/openstack-tox-py27/9ede98b/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/1/check/openstack-tox-py36/1821321/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/1/check/openstack-tox-py37/7c39844/ : SUCCESS in 4m 00s\n- openstack-tox-validate http://logs.openstack.org/78/672178/1/check/openstack-tox-validate/a119faa/ : SUCCESS in 4m 11s\n- requirements-tox-babel http://logs.openstack.org/78/672178/1/check/requirements-tox-babel/71c9d83/ : SUCCESS in 3m 50s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/1/check/requirements-tox-py27-check-uc/6f4f64e/ : SUCCESS in 17m 19s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/1/check/requirements-tox-py36-check-uc/4a86b14/ : SUCCESS in 11m 34s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/1/check/cross-cinder-py27/94948e9/ : SUCCESS in 11m 04s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/1/check/cross-cinder-py36/4f032be/ : SUCCESS in 10m 30s\n- cross-glance-py27 http://logs.openstack.org/78/672178/1/check/cross-glance-py27/b702c54/ : SUCCESS in 5m 03s\n- cross-glance-py36 http://logs.openstack.org/78/672178/1/check/cross-glance-py36/2b6baad/ : SUCCESS in 6m 22s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/1/check/cross-horizon-py27/39b2925/ : SUCCESS in 9m 35s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/1/check/cross-horizon-py36/b74996a/ : SUCCESS in 9m 11s\n- cross-horizon-npm http://logs.openstack.org/78/672178/1/check/cross-horizon-npm/925a270/ : SUCCESS in 9m 26s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/1/check/cross-keystone-py27/6c94ee5/ : SUCCESS in 31m 56s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/1/check/cross-keystone-py36/4f1cceb/ : SUCCESS in 27m 23s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/1/check/cross-neutron-py27/bfad2f1/ : SUCCESS in 30m 52s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/1/check/cross-neutron-py36/7cf7d53/ : SUCCESS in 27m 12s\n- cross-nova-py27 http://logs.openstack.org/78/672178/1/check/cross-nova-py27/6906891/ : SUCCESS in 15m 23s\n- cross-nova-py36 http://logs.openstack.org/78/672178/1/check/cross-nova-py36/0d66bed/ : SUCCESS in 19m 33s\n- cross-nova-functional http://logs.openstack.org/78/672178/1/check/cross-nova-functional/195ca3c/ : SUCCESS in 23m 44s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/1/check/cross-osc-tox-docs/5b4c858/html/ : SUCCESS in 4m 41s\n- cross-swift-py27 http://logs.openstack.org/78/672178/1/check/cross-swift-py27/e26692d/ : SUCCESS in 11m 00s\n- cross-swift-py36 http://logs.openstack.org/78/672178/1/check/cross-swift-py36/726faea/ : SUCCESS in 8m 33s\n- requirements-integration http://logs.openstack.org/78/672178/1/check/requirements-integration/19fd3fb/ : FAILURE in 34m 59s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/1/check/neutron-grenade-multinode/061d5bd/ : SUCCESS in 1h 21m 50s\n- tempest-full http://logs.openstack.org/78/672178/1/check/tempest-full/c75cdf7/ : SUCCESS in 1h 41m 16s\n- neutron-grenade http://logs.openstack.org/78/672178/1/check/neutron-grenade/8d4e756/ : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":1},{"id":"c3f394fa7cb9ef3b8132b03a60d81a1ef618e20e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-23 18:27:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b17573bd6685c2949eae8d787e16f41d78e02b14","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-23 18:27:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1d3611a9d21ad1b74ef337220c611fbdc1baca65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 22:16:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/2/check/openstack-tox-docs/724a86d/html/ : SUCCESS in 3m 31s\n- openstack-tox-pypy http://logs.openstack.org/78/672178/2/check/openstack-tox-pypy/4d77b8a/ : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/2/check/openstack-tox-pep8/8e8ea56/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/2/check/openstack-tox-py27/257ac97/ : SUCCESS in 3m 38s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/2/check/openstack-tox-py36/f8b3163/ : SUCCESS in 5m 19s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/2/check/openstack-tox-py37/75c089c/ : SUCCESS in 5m 49s\n- openstack-tox-validate http://logs.openstack.org/78/672178/2/check/openstack-tox-validate/7255df7/ : SUCCESS in 4m 03s\n- requirements-tox-babel http://logs.openstack.org/78/672178/2/check/requirements-tox-babel/eb0509e/ : SUCCESS in 4m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py27-check-uc/aa241e6/ : SUCCESS in 16m 30s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py36-check-uc/f99dd9c/ : SUCCESS in 6m 21s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/2/check/cross-cinder-py27/581679a/ : SUCCESS in 11m 53s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/2/check/cross-cinder-py36/95333c9/ : SUCCESS in 8m 28s\n- cross-glance-py27 http://logs.openstack.org/78/672178/2/check/cross-glance-py27/c668ba0/ : SUCCESS in 4m 59s\n- cross-glance-py36 http://logs.openstack.org/78/672178/2/check/cross-glance-py36/e9d74ba/ : SUCCESS in 5m 15s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/2/check/cross-horizon-py27/46cab87/ : SUCCESS in 12m 05s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/2/check/cross-horizon-py36/ab829bf/ : SUCCESS in 10m 21s\n- cross-horizon-npm http://logs.openstack.org/78/672178/2/check/cross-horizon-npm/0077bfc/ : SUCCESS in 7m 18s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/2/check/cross-keystone-py27/46b63be/ : SUCCESS in 25m 57s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/2/check/cross-keystone-py36/4053c4c/ : SUCCESS in 26m 34s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/2/check/cross-neutron-py27/bfc9034/ : SUCCESS in 25m 07s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/2/check/cross-neutron-py36/8ea5126/ : SUCCESS in 25m 04s\n- cross-nova-py27 http://logs.openstack.org/78/672178/2/check/cross-nova-py27/d9340f1/ : SUCCESS in 17m 59s\n- cross-nova-py36 http://logs.openstack.org/78/672178/2/check/cross-nova-py36/9526723/ : SUCCESS in 17m 45s\n- cross-nova-functional http://logs.openstack.org/78/672178/2/check/cross-nova-functional/e1c6920/ : TIMED_OUT in 41m 21s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/2/check/cross-osc-tox-docs/c2571ee/html/ : SUCCESS in 3m 57s\n- cross-swift-py27 http://logs.openstack.org/78/672178/2/check/cross-swift-py27/6f8de7a/ : SUCCESS in 9m 36s\n- cross-swift-py36 http://logs.openstack.org/78/672178/2/check/cross-swift-py36/bfdb46e/ : SUCCESS in 10m 52s\n- requirements-integration http://logs.openstack.org/78/672178/2/check/requirements-integration/f8eafa3/ : FAILURE in 35m 24s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/2/check/neutron-grenade-multinode/12ffbf4/ : SUCCESS in 1h 05m 09s\n- tempest-full http://logs.openstack.org/78/672178/2/check/tempest-full/f59932e/ : SUCCESS in 1h 30m 23s\n- neutron-grenade http://logs.openstack.org/78/672178/2/check/neutron-grenade/509ec65/ : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":2},{"id":"35f0b7fe8627c94752cfc2f1dc338e40af140dd6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-23 22:35:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dae9cad076b36483e503d8439ae8d7cf43d5598d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 00:30:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/2/check/openstack-tox-docs/c81c563/html/ : SUCCESS in 3m 14s\n- openstack-tox-pypy http://logs.openstack.org/78/672178/2/check/openstack-tox-pypy/d2e5655/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/2/check/openstack-tox-pep8/7340447/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/2/check/openstack-tox-py27/b68310d/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/2/check/openstack-tox-py36/e03dbae/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/2/check/openstack-tox-py37/9c3beac/ : SUCCESS in 3m 20s\n- openstack-tox-validate http://logs.openstack.org/78/672178/2/check/openstack-tox-validate/857b2bb/ : SUCCESS in 3m 08s\n- requirements-tox-babel http://logs.openstack.org/78/672178/2/check/requirements-tox-babel/b27ed97/ : SUCCESS in 2m 49s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py27-check-uc/d7bdb9f/ : SUCCESS in 6m 01s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py36-check-uc/041140c/ : SUCCESS in 5m 38s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/2/check/cross-cinder-py27/de50395/ : SUCCESS in 9m 04s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/2/check/cross-cinder-py36/20fb1ae/ : SUCCESS in 8m 32s\n- cross-glance-py27 http://logs.openstack.org/78/672178/2/check/cross-glance-py27/8319076/ : SUCCESS in 4m 27s\n- cross-glance-py36 http://logs.openstack.org/78/672178/2/check/cross-glance-py36/dfdf482/ : SUCCESS in 4m 19s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/2/check/cross-horizon-py27/0427a11/ : SUCCESS in 9m 23s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/2/check/cross-horizon-py36/ffef212/ : SUCCESS in 9m 09s\n- cross-horizon-npm http://logs.openstack.org/78/672178/2/check/cross-horizon-npm/bf24f7c/ : SUCCESS in 5m 57s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/2/check/cross-keystone-py27/86a5698/ : SUCCESS in 21m 20s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/2/check/cross-keystone-py36/a383a4d/ : SUCCESS in 23m 00s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/2/check/cross-neutron-py27/6713fb6/ : SUCCESS in 20m 28s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/2/check/cross-neutron-py36/a7d6818/ : SUCCESS in 22m 03s\n- cross-nova-py27 http://logs.openstack.org/78/672178/2/check/cross-nova-py27/6a5e3a1/ : SUCCESS in 16m 12s\n- cross-nova-py36 http://logs.openstack.org/78/672178/2/check/cross-nova-py36/416fe51/ : SUCCESS in 16m 21s\n- cross-nova-functional http://logs.openstack.org/78/672178/2/check/cross-nova-functional/fcf0acf/ : SUCCESS in 21m 15s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/2/check/cross-osc-tox-docs/f015784/html/ : SUCCESS in 3m 19s\n- cross-swift-py27 http://logs.openstack.org/78/672178/2/check/cross-swift-py27/f988e6f/ : SUCCESS in 6m 36s\n- cross-swift-py36 http://logs.openstack.org/78/672178/2/check/cross-swift-py36/b25353e/ : SUCCESS in 6m 33s\n- requirements-integration http://logs.openstack.org/78/672178/2/check/requirements-integration/2a00f51/ : FAILURE in 31m 12s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/2/check/neutron-grenade-multinode/3b0f4e5/ : SUCCESS in 1h 04m 41s\n- tempest-full http://logs.openstack.org/78/672178/2/check/tempest-full/9ca2a02/ : SUCCESS in 1h 31m 15s\n- neutron-grenade http://logs.openstack.org/78/672178/2/check/neutron-grenade/19675ac/ : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":2},{"id":"644cabf155b120ab381dbce7d245fd50b53645be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-24 00:33:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0f5cc632e54be742cba8103069286e4841ffc14b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 01:59:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/2/check/openstack-tox-docs/d31a813/html/ : SUCCESS in 4m 43s\n- openstack-tox-pypy http://logs.openstack.org/78/672178/2/check/openstack-tox-pypy/1fef431/ : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/2/check/openstack-tox-pep8/ee29648/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/2/check/openstack-tox-py27/f3ce56a/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/2/check/openstack-tox-py36/12ccbef/ : SUCCESS in 4m 19s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/2/check/openstack-tox-py37/f5e3a5d/ : SUCCESS in 4m 38s\n- openstack-tox-validate http://logs.openstack.org/78/672178/2/check/openstack-tox-validate/145d9fb/ : SUCCESS in 3m 49s\n- requirements-tox-babel http://logs.openstack.org/78/672178/2/check/requirements-tox-babel/9ecaa41/ : SUCCESS in 2m 29s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py27-check-uc/a35118c/ : SUCCESS in 18m 30s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py36-check-uc/3def847/ : SUCCESS in 18m 23s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/2/check/cross-cinder-py27/26b3dd0/ : SUCCESS in 8m 05s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/2/check/cross-cinder-py36/0720c51/ : SUCCESS in 8m 22s\n- cross-glance-py27 http://logs.openstack.org/78/672178/2/check/cross-glance-py27/cc47805/ : SUCCESS in 4m 26s\n- cross-glance-py36 http://logs.openstack.org/78/672178/2/check/cross-glance-py36/30eb92b/ : SUCCESS in 5m 30s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/2/check/cross-horizon-py27/3345395/ : SUCCESS in 11m 09s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/2/check/cross-horizon-py36/70dadd6/ : SUCCESS in 10m 23s\n- cross-horizon-npm http://logs.openstack.org/78/672178/2/check/cross-horizon-npm/04c5b52/ : SUCCESS in 7m 24s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/2/check/cross-keystone-py27/c365e70/ : SUCCESS in 32m 24s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/2/check/cross-keystone-py36/5462899/ : SUCCESS in 27m 52s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/2/check/cross-neutron-py27/3581bdf/ : SUCCESS in 25m 37s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/2/check/cross-neutron-py36/790eb11/ : SUCCESS in 26m 07s\n- cross-nova-py27 http://logs.openstack.org/78/672178/2/check/cross-nova-py27/3dd3577/ : SUCCESS in 15m 35s\n- cross-nova-py36 http://logs.openstack.org/78/672178/2/check/cross-nova-py36/9dc9315/ : SUCCESS in 19m 45s\n- cross-nova-functional http://logs.openstack.org/78/672178/2/check/cross-nova-functional/f21bc2d/ : SUCCESS in 22m 25s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/2/check/cross-osc-tox-docs/ff7b46e/html/ : SUCCESS in 4m 17s\n- cross-swift-py27 http://logs.openstack.org/78/672178/2/check/cross-swift-py27/4467a3b/ : SUCCESS in 8m 24s\n- cross-swift-py36 http://logs.openstack.org/78/672178/2/check/cross-swift-py36/b8278b4/ : SUCCESS in 8m 13s\n- requirements-integration http://logs.openstack.org/78/672178/2/check/requirements-integration/1dbbab7/ : FAILURE in 37m 38s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/2/check/neutron-grenade-multinode/1e00641/ : SUCCESS in 1h 09m 27s\n- tempest-full http://logs.openstack.org/78/672178/2/check/tempest-full/c6b162b/ : SUCCESS in 1h 20m 53s\n- neutron-grenade http://logs.openstack.org/78/672178/2/check/neutron-grenade/86d035c/ : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":2},{"id":"54a119434cc92aee8edcc43746692882ff7712e3","author":{"_account_id":20676,"name":"daniel.pawlik","display_name":"Daniel Pawlik","email":"dpawlik@redhat.com","username":"daniel.pawlik"},"date":"2019-07-24 08:20:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9dec981059f9c93dc0f3e53c346232476c899776","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:05:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/2/check/openstack-tox-docs/07f329d/html/ : SUCCESS in 4m 40s\n- openstack-tox-pypy http://logs.openstack.org/78/672178/2/check/openstack-tox-pypy/27e88c4/ : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/2/check/openstack-tox-pep8/dedb484/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/2/check/openstack-tox-py27/cbf2c8e/ : SUCCESS in 5m 09s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/2/check/openstack-tox-py36/4052f82/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/2/check/openstack-tox-py37/fa71780/ : SUCCESS in 6m 59s\n- openstack-tox-validate http://logs.openstack.org/78/672178/2/check/openstack-tox-validate/1bcf8f5/ : SUCCESS in 4m 13s\n- requirements-tox-babel http://logs.openstack.org/78/672178/2/check/requirements-tox-babel/3ca6287/ : SUCCESS in 5m 48s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py27-check-uc/b79eba6/ : SUCCESS in 10m 39s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py36-check-uc/24aebec/ : SUCCESS in 7m 16s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/2/check/cross-cinder-py27/72d71a8/ : SUCCESS in 9m 34s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/2/check/cross-cinder-py36/ba78653/ : SUCCESS in 7m 19s\n- cross-glance-py27 http://logs.openstack.org/78/672178/2/check/cross-glance-py27/4b4ce6a/ : SUCCESS in 4m 46s\n- cross-glance-py36 http://logs.openstack.org/78/672178/2/check/cross-glance-py36/ee9de8e/ : SUCCESS in 4m 04s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/2/check/cross-horizon-py27/8068248/ : SUCCESS in 10m 14s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/2/check/cross-horizon-py36/e63c3db/ : SUCCESS in 9m 12s\n- cross-horizon-npm http://logs.openstack.org/78/672178/2/check/cross-horizon-npm/d230c2f/ : SUCCESS in 6m 05s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/2/check/cross-keystone-py27/009ecfa/ : SUCCESS in 23m 13s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/2/check/cross-keystone-py36/6a97c28/ : SUCCESS in 23m 40s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/2/check/cross-neutron-py27/4a60cdb/ : SUCCESS in 27m 36s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/2/check/cross-neutron-py36/09e85ce/ : SUCCESS in 23m 08s\n- cross-nova-py27 http://logs.openstack.org/78/672178/2/check/cross-nova-py27/2f65789/ : SUCCESS in 19m 49s\n- cross-nova-py36 http://logs.openstack.org/78/672178/2/check/cross-nova-py36/29fb45a/ : SUCCESS in 15m 00s\n- cross-nova-functional http://logs.openstack.org/78/672178/2/check/cross-nova-functional/5bb6efb/ : SUCCESS in 26m 05s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/2/check/cross-osc-tox-docs/44742ee/html/ : SUCCESS in 3m 42s\n- cross-swift-py27 http://logs.openstack.org/78/672178/2/check/cross-swift-py27/dfb2d09/ : SUCCESS in 9m 57s\n- cross-swift-py36 http://logs.openstack.org/78/672178/2/check/cross-swift-py36/6bbd61c/ : SUCCESS in 8m 31s\n- requirements-integration http://logs.openstack.org/78/672178/2/check/requirements-integration/1bf9ebf/ : FAILURE in 32m 45s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/2/check/neutron-grenade-multinode/af48bf0/ : SUCCESS in 1h 11m 22s\n- tempest-full http://logs.openstack.org/78/672178/2/check/tempest-full/21ce40a/ : SUCCESS in 1h 32m 41s\n- neutron-grenade http://logs.openstack.org/78/672178/2/check/neutron-grenade/db59907/ : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":2},{"id":"9981516523b284257b8227fd08e30274e435dee1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-24 12:32:01.000000000","message":"Patch Set 2:\n\nrecheck\n\nthe CI should be fixed now","accounts_in_message":[],"_revision_number":2},{"id":"3fe3e968de8f6f346eae498960ca283e85651c28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 14:24:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/2/check/openstack-tox-docs/d94ff5e/html/ : SUCCESS in 3m 47s\n- openstack-tox-pypy http://logs.openstack.org/78/672178/2/check/openstack-tox-pypy/ea07f64/ : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/2/check/openstack-tox-pep8/64d276a/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/2/check/openstack-tox-py27/6465907/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/2/check/openstack-tox-py36/7594b0e/ : SUCCESS in 4m 07s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/2/check/openstack-tox-py37/a50b4bf/ : SUCCESS in 5m 18s\n- openstack-tox-validate http://logs.openstack.org/78/672178/2/check/openstack-tox-validate/eb09d16/ : SUCCESS in 4m 20s\n- requirements-tox-babel http://logs.openstack.org/78/672178/2/check/requirements-tox-babel/0e6f3a0/ : SUCCESS in 3m 40s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py27-check-uc/1249c6e/ : SUCCESS in 8m 27s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/2/check/requirements-tox-py36-check-uc/40d0b69/ : SUCCESS in 6m 15s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/2/check/cross-cinder-py27/90ba80b/ : SUCCESS in 8m 15s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/2/check/cross-cinder-py36/056106c/ : SUCCESS in 7m 55s\n- cross-glance-py27 http://logs.openstack.org/78/672178/2/check/cross-glance-py27/b781926/ : SUCCESS in 6m 41s\n- cross-glance-py36 http://logs.openstack.org/78/672178/2/check/cross-glance-py36/146fd8d/ : SUCCESS in 5m 54s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/2/check/cross-horizon-py27/6756b1f/ : SUCCESS in 11m 39s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/2/check/cross-horizon-py36/42e057c/ : SUCCESS in 10m 05s\n- cross-horizon-npm http://logs.openstack.org/78/672178/2/check/cross-horizon-npm/57423ef/ : SUCCESS in 6m 20s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/2/check/cross-keystone-py27/bf9713d/ : SUCCESS in 27m 37s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/2/check/cross-keystone-py36/f0a52ab/ : SUCCESS in 26m 50s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/2/check/cross-neutron-py27/447ff94/ : SUCCESS in 24m 37s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/2/check/cross-neutron-py36/e25834e/ : SUCCESS in 22m 00s\n- cross-nova-py27 http://logs.openstack.org/78/672178/2/check/cross-nova-py27/036e4ab/ : SUCCESS in 18m 08s\n- cross-nova-py36 http://logs.openstack.org/78/672178/2/check/cross-nova-py36/107bbee/ : SUCCESS in 18m 32s\n- cross-nova-functional http://logs.openstack.org/78/672178/2/check/cross-nova-functional/bd60120/ : SUCCESS in 23m 40s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/2/check/cross-osc-tox-docs/39d1ec8/html/ : SUCCESS in 5m 43s\n- cross-swift-py27 http://logs.openstack.org/78/672178/2/check/cross-swift-py27/0675352/ : SUCCESS in 9m 25s\n- cross-swift-py36 http://logs.openstack.org/78/672178/2/check/cross-swift-py36/92a1641/ : SUCCESS in 8m 23s\n- requirements-integration http://logs.openstack.org/78/672178/2/check/requirements-integration/c6356ed/ : SUCCESS in 1h 18m 40s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/2/check/neutron-grenade-multinode/2406675/ : SUCCESS in 1h 23m 54s\n- tempest-full http://logs.openstack.org/78/672178/2/check/tempest-full/b31ada9/ : SUCCESS in 1h 43m 09s\n- neutron-grenade http://logs.openstack.org/78/672178/2/check/neutron-grenade/82bc579/ : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":2},{"id":"541fe5888fe16a16861f340b66f3b0c6e8d7f619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 14:26:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d93357de5fb519786ae612c1505761d9f564ec32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 16:25:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3e8f9d81e3c62b5bb50d1186f77e1814e95401c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 16:25:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/78/672178/2/gate/openstack-tox-docs/656df9b/html/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/78/672178/2/gate/openstack-tox-pep8/6d6c11d/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/78/672178/2/gate/openstack-tox-py27/f989ba0/ : SUCCESS in 4m 01s\n- openstack-tox-py36 http://logs.openstack.org/78/672178/2/gate/openstack-tox-py36/882019b/ : SUCCESS in 4m 47s\n- openstack-tox-py37 http://logs.openstack.org/78/672178/2/gate/openstack-tox-py37/5f9e1d9/ : SUCCESS in 6m 13s\n- openstack-tox-validate http://logs.openstack.org/78/672178/2/gate/openstack-tox-validate/dc453bb/ : SUCCESS in 4m 12s\n- requirements-tox-babel http://logs.openstack.org/78/672178/2/gate/requirements-tox-babel/262482e/ : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/78/672178/2/gate/requirements-tox-py27-check-uc/7d7de27/ : SUCCESS in 8m 32s\n- requirements-tox-py36-check-uc http://logs.openstack.org/78/672178/2/gate/requirements-tox-py36-check-uc/3030457/ : SUCCESS in 6m 47s\n- cross-cinder-py27 http://logs.openstack.org/78/672178/2/gate/cross-cinder-py27/7c3af18/ : SUCCESS in 11m 36s\n- cross-cinder-py36 http://logs.openstack.org/78/672178/2/gate/cross-cinder-py36/e08368a/ : SUCCESS in 10m 56s\n- cross-glance-py27 http://logs.openstack.org/78/672178/2/gate/cross-glance-py27/5056419/ : SUCCESS in 5m 04s\n- cross-glance-py36 http://logs.openstack.org/78/672178/2/gate/cross-glance-py36/1cd8c70/ : SUCCESS in 4m 59s\n- cross-horizon-py27 http://logs.openstack.org/78/672178/2/gate/cross-horizon-py27/389a3b3/ : SUCCESS in 10m 44s\n- cross-horizon-py36 http://logs.openstack.org/78/672178/2/gate/cross-horizon-py36/cd9e896/ : SUCCESS in 9m 11s\n- cross-horizon-npm http://logs.openstack.org/78/672178/2/gate/cross-horizon-npm/a054f68/ : SUCCESS in 9m 34s\n- cross-keystone-py27 http://logs.openstack.org/78/672178/2/gate/cross-keystone-py27/94dc4eb/ : SUCCESS in 29m 54s\n- cross-keystone-py36 http://logs.openstack.org/78/672178/2/gate/cross-keystone-py36/2f811ce/ : SUCCESS in 31m 12s\n- cross-neutron-py27 http://logs.openstack.org/78/672178/2/gate/cross-neutron-py27/6ffed0e/ : SUCCESS in 24m 56s\n- cross-neutron-py36 http://logs.openstack.org/78/672178/2/gate/cross-neutron-py36/8fdf8ae/ : SUCCESS in 24m 55s\n- cross-nova-py27 http://logs.openstack.org/78/672178/2/gate/cross-nova-py27/057fdc6/ : SUCCESS in 17m 49s\n- cross-nova-py36 http://logs.openstack.org/78/672178/2/gate/cross-nova-py36/c7eed09/ : SUCCESS in 16m 16s\n- cross-nova-functional http://logs.openstack.org/78/672178/2/gate/cross-nova-functional/2689383/ : SUCCESS in 24m 56s\n- cross-osc-tox-docs http://logs.openstack.org/78/672178/2/gate/cross-osc-tox-docs/6bff016/html/ : SUCCESS in 5m 30s\n- cross-swift-py27 http://logs.openstack.org/78/672178/2/gate/cross-swift-py27/a64e583/ : SUCCESS in 10m 22s\n- cross-swift-py36 http://logs.openstack.org/78/672178/2/gate/cross-swift-py36/f8a6f82/ : SUCCESS in 9m 50s\n- requirements-integration http://logs.openstack.org/78/672178/2/gate/requirements-integration/cd12d53/ : SUCCESS in 56m 15s\n- neutron-grenade-multinode http://logs.openstack.org/78/672178/2/gate/neutron-grenade-multinode/e905684/ : SUCCESS in 1h 16m 07s\n- tempest-full http://logs.openstack.org/78/672178/2/gate/tempest-full/f9dea89/ : SUCCESS in 1h 39m 22s\n- neutron-grenade http://logs.openstack.org/78/672178/2/gate/neutron-grenade/f3bd11d/ : SUCCESS in 1h 09m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"91a0c42938919c9e1a55e82412148f4f0e15a98c","revisions":{"8e5703f31e4ee9ff177b7920363888d78c0ce67c":{"kind":"REWORK","_number":1,"created":"2019-07-22 21:15:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/672178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/672178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/672178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/672178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/672178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/672178/1"}}},"commit":{"parents":[{"commit":"a603a63e67a7a57cb4ec12901c0d236c9422b52a","subject":"Merge \"Add tripleo-ansible to the requirements sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a603a63e67a7a57cb4ec12901c0d236c9422b52a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-22 21:15:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-22 21:15:42.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.31.2","message":"update constraint for openstacksdk to new release 0.31.2\n\nChange-Id: Ie0fabdcb9ee4701ebd7bc3d03fe64483d8220224\nmeta:version: 0.31.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: I49b1cf95a8f9611f5fdd440eb3d7415a01121b47\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e5703f31e4ee9ff177b7920363888d78c0ce67c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e5703f31e4ee9ff177b7920363888d78c0ce67c"}]},"branch":"refs/heads/master"},"91a0c42938919c9e1a55e82412148f4f0e15a98c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-23 18:27:32.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/78/672178/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/672178/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/672178/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/672178/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/672178/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/672178/2"}}},"commit":{"parents":[{"commit":"7e5222eafb718a97bb6ca7ed842c6f08c0e92670","subject":"Merge \"update constraint for keystonemiddleware to new release 7.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e5222eafb718a97bb6ca7ed842c6f08c0e92670"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-22 21:15:42.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-07-23 18:27:32.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.31.2","message":"update constraint for openstacksdk to new release 0.31.2\n\nChange-Id: Ie0fabdcb9ee4701ebd7bc3d03fe64483d8220224\nmeta:version: 0.31.2\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Commit: Monty Taylor \u003cmordred@inaugust.com\u003e\nmeta:release:Change-Id: I49b1cf95a8f9611f5fdd440eb3d7415a01121b47\nmeta:release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta:release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91a0c42938919c9e1a55e82412148f4f0e15a98c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91a0c42938919c9e1a55e82412148f4f0e15a98c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
