)]}'
{"id":"openstack%2Frequirements~672249","triplet_id":"openstack%2Frequirements~master~Ia6c54e6b852ef31773ee67e95764156ef2d8ef9f","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Ia6c54e6b852ef31773ee67e95764156ef2d8ef9f","subject":"Update constraint for monasca-common to new release 2.15.0","status":"ABANDONED","created":"2019-07-23 10:10:10.000000000","updated":"2019-07-23 12:25:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ad0bba418e65d1b8468318578ad6d98d4b2019ec","_number":672249,"virtual_id_number":672249,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":-1,"date":"2019-07-23 12:23:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"date":"2019-07-23 10:11:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-07-23 10:11:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2019-07-23 10:11:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"date":"2019-07-23 10:11:30.000000000","_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 10:11:01.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-07-23 10:11:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-23 10:11:13.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-23 10:11:30.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-07-23 12:23:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f5f0c51e40c378d510ee538916a55d20496010b","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-23 10:10:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c71b0b67e94bae9cb6596c26d8636265e4085a52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:23:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/49/672249/1/check/openstack-tox-docs/9732bc5/html/ : SUCCESS in 4m 18s\n- openstack-tox-pypy http://logs.openstack.org/49/672249/1/check/openstack-tox-pypy/6a32334/ : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/49/672249/1/check/openstack-tox-pep8/f329215/ : SUCCESS in 5m 21s\n- openstack-tox-py27 http://logs.openstack.org/49/672249/1/check/openstack-tox-py27/4b0dbc7/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/49/672249/1/check/openstack-tox-py36/9830550/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/49/672249/1/check/openstack-tox-py37/b037862/ : SUCCESS in 4m 01s\n- openstack-tox-validate http://logs.openstack.org/49/672249/1/check/openstack-tox-validate/1998635/ : SUCCESS in 2m 49s\n- requirements-tox-babel http://logs.openstack.org/49/672249/1/check/requirements-tox-babel/1c18de7/ : SUCCESS in 4m 59s\n- requirements-tox-py27-check-uc http://logs.openstack.org/49/672249/1/check/requirements-tox-py27-check-uc/28823e9/ : SUCCESS in 7m 00s\n- requirements-tox-py36-check-uc http://logs.openstack.org/49/672249/1/check/requirements-tox-py36-check-uc/8a31f65/ : SUCCESS in 8m 50s\n- cross-cinder-py27 http://logs.openstack.org/49/672249/1/check/cross-cinder-py27/c1247e0/ : SUCCESS in 9m 20s\n- cross-cinder-py36 http://logs.openstack.org/49/672249/1/check/cross-cinder-py36/8775144/ : SUCCESS in 9m 34s\n- cross-glance-py27 http://logs.openstack.org/49/672249/1/check/cross-glance-py27/3b366fc/ : SUCCESS in 5m 53s\n- cross-glance-py36 http://logs.openstack.org/49/672249/1/check/cross-glance-py36/cc0b01b/ : SUCCESS in 4m 19s\n- cross-horizon-py27 http://logs.openstack.org/49/672249/1/check/cross-horizon-py27/5054f08/ : SUCCESS in 10m 14s\n- cross-horizon-py36 http://logs.openstack.org/49/672249/1/check/cross-horizon-py36/90fdb24/ : SUCCESS in 11m 50s\n- cross-horizon-npm http://logs.openstack.org/49/672249/1/check/cross-horizon-npm/8489c1b/ : SUCCESS in 10m 37s\n- cross-keystone-py27 http://logs.openstack.org/49/672249/1/check/cross-keystone-py27/7b00f5d/ : SUCCESS in 26m 43s\n- cross-keystone-py36 http://logs.openstack.org/49/672249/1/check/cross-keystone-py36/9d00bf2/ : SUCCESS in 27m 45s\n- cross-neutron-py27 http://logs.openstack.org/49/672249/1/check/cross-neutron-py27/d1251a9/ : SUCCESS in 23m 23s\n- cross-neutron-py36 http://logs.openstack.org/49/672249/1/check/cross-neutron-py36/8645492/ : SUCCESS in 25m 55s\n- cross-nova-py27 http://logs.openstack.org/49/672249/1/check/cross-nova-py27/902f6e5/ : SUCCESS in 17m 42s\n- cross-nova-py36 http://logs.openstack.org/49/672249/1/check/cross-nova-py36/e22a46d/ : SUCCESS in 17m 25s\n- cross-nova-functional http://logs.openstack.org/49/672249/1/check/cross-nova-functional/44e783c/ : SUCCESS in 23m 34s\n- cross-osc-tox-docs http://logs.openstack.org/49/672249/1/check/cross-osc-tox-docs/513a8b6/html/ : SUCCESS in 5m 04s\n- cross-swift-py27 http://logs.openstack.org/49/672249/1/check/cross-swift-py27/503a80f/ : SUCCESS in 9m 28s\n- cross-swift-py36 http://logs.openstack.org/49/672249/1/check/cross-swift-py36/7278427/ : SUCCESS in 10m 41s\n- requirements-integration http://logs.openstack.org/49/672249/1/check/requirements-integration/3f48523/ : FAILURE in 32m 41s\n- neutron-grenade-multinode http://logs.openstack.org/49/672249/1/check/neutron-grenade-multinode/5cdfe1a/ : SUCCESS in 1h 12m 54s\n- tempest-full http://logs.openstack.org/49/672249/1/check/tempest-full/baa0f2c/ : TIMED_OUT in 2h 11m 01s\n- neutron-grenade http://logs.openstack.org/49/672249/1/check/neutron-grenade/0850544/ : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ad0bba418e65d1b8468318578ad6d98d4b2019ec","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2019-07-23 12:25:57.000000000","message":"Abandoned\n\nHandled by OpenStack Proposal Bot.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"14dadbc28b3a326a46ff883d02b2e700d2644403","revisions":{"14dadbc28b3a326a46ff883d02b2e700d2644403":{"kind":"REWORK","_number":1,"created":"2019-07-23 10:10:10.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/49/672249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/49/672249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/672249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/672249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/49/672249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/49/672249/1"}}},"commit":{"parents":[{"commit":"7e5222eafb718a97bb6ca7ed842c6f08c0e92670","subject":"Merge \"update constraint for keystonemiddleware to new release 7.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e5222eafb718a97bb6ca7ed842c6f08c0e92670"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-23 10:09:41.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2019-07-23 10:09:41.000000000","tz":120},"subject":"Update constraint for monasca-common to new release 2.15.0","message":"Update constraint for monasca-common to new release 2.15.0\n\nChange-Id: Ia6c54e6b852ef31773ee67e95764156ef2d8ef9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14dadbc28b3a326a46ff883d02b2e700d2644403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/14dadbc28b3a326a46ff883d02b2e700d2644403"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
