)]}'
{"id":"openstack%2Frequirements~672480","triplet_id":"openstack%2Frequirements~master~I0513e715c8b6330806d2af3fa2d388b30eba75b7","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0513e715c8b6330806d2af3fa2d388b30eba75b7","subject":"update constraint for openstack-doc-tools to new release 1.9.0","status":"ABANDONED","created":"2019-07-24 09:31:05.000000000","updated":"2019-07-24 19:37:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"555b27e13a44eaca947c0674f1dbaf31d87c8df8","_number":672480,"virtual_id_number":672480,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-07-24 19:30:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-24 19:37:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-24 19:37:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 19:30:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-24 19:37:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4a4be7fdf4567c5c36cc46fa176acd2c4f12df7e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-24 09:31:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc466828b27841f339320f4d941d2d1ac34b3a86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 11:17:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/80/672480/1/check/openstack-tox-docs/3c3a84a/html/ : SUCCESS in 4m 12s\n- openstack-tox-pypy http://logs.openstack.org/80/672480/1/check/openstack-tox-pypy/ea31101/ : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/672480/1/check/openstack-tox-pep8/318fdcf/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/80/672480/1/check/openstack-tox-py27/dc92415/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/80/672480/1/check/openstack-tox-py36/91cd402/ : SUCCESS in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/80/672480/1/check/openstack-tox-py37/33d1dd1/ : SUCCESS in 5m 10s\n- openstack-tox-validate http://logs.openstack.org/80/672480/1/check/openstack-tox-validate/6fa3b47/ : SUCCESS in 3m 50s\n- requirements-tox-babel http://logs.openstack.org/80/672480/1/check/requirements-tox-babel/48c2727/ : SUCCESS in 4m 24s\n- requirements-tox-py27-check-uc http://logs.openstack.org/80/672480/1/check/requirements-tox-py27-check-uc/6d86fca/ : SUCCESS in 7m 45s\n- requirements-tox-py36-check-uc http://logs.openstack.org/80/672480/1/check/requirements-tox-py36-check-uc/e51a915/ : SUCCESS in 8m 06s\n- cross-cinder-py27 http://logs.openstack.org/80/672480/1/check/cross-cinder-py27/a1568ab/ : SUCCESS in 8m 27s\n- cross-cinder-py36 http://logs.openstack.org/80/672480/1/check/cross-cinder-py36/c85874d/ : SUCCESS in 9m 04s\n- cross-glance-py27 http://logs.openstack.org/80/672480/1/check/cross-glance-py27/483bd99/ : SUCCESS in 4m 48s\n- cross-glance-py36 http://logs.openstack.org/80/672480/1/check/cross-glance-py36/5cdfd7d/ : SUCCESS in 5m 40s\n- cross-horizon-py27 http://logs.openstack.org/80/672480/1/check/cross-horizon-py27/51229de/ : SUCCESS in 11m 50s\n- cross-horizon-py36 http://logs.openstack.org/80/672480/1/check/cross-horizon-py36/936908d/ : SUCCESS in 11m 22s\n- cross-horizon-npm http://logs.openstack.org/80/672480/1/check/cross-horizon-npm/71e857b/ : SUCCESS in 8m 10s\n- cross-keystone-py27 http://logs.openstack.org/80/672480/1/check/cross-keystone-py27/9d68f6c/ : SUCCESS in 23m 38s\n- cross-keystone-py36 http://logs.openstack.org/80/672480/1/check/cross-keystone-py36/3174d41/ : SUCCESS in 28m 28s\n- cross-neutron-py27 http://logs.openstack.org/80/672480/1/check/cross-neutron-py27/c25a1db/ : SUCCESS in 23m 20s\n- cross-neutron-py36 http://logs.openstack.org/80/672480/1/check/cross-neutron-py36/c7aa16c/ : SUCCESS in 25m 29s\n- cross-nova-py27 http://logs.openstack.org/80/672480/1/check/cross-nova-py27/4a9bd8e/ : SUCCESS in 19m 53s\n- cross-nova-py36 http://logs.openstack.org/80/672480/1/check/cross-nova-py36/b716306/ : SUCCESS in 16m 42s\n- cross-nova-functional http://logs.openstack.org/80/672480/1/check/cross-nova-functional/ee7d669/ : SUCCESS in 26m 56s\n- cross-osc-tox-docs http://logs.openstack.org/80/672480/1/check/cross-osc-tox-docs/b90a3a2/html/ : SUCCESS in 4m 14s\n- cross-swift-py27 http://logs.openstack.org/80/672480/1/check/cross-swift-py27/765dba8/ : SUCCESS in 7m 40s\n- cross-swift-py36 http://logs.openstack.org/80/672480/1/check/cross-swift-py36/d18c909/ : SUCCESS in 8m 06s\n- requirements-integration http://logs.openstack.org/80/672480/1/check/requirements-integration/fe114f1/ : FAILURE in 39m 26s\n- neutron-grenade-multinode http://logs.openstack.org/80/672480/1/check/neutron-grenade-multinode/6f1d1fa/ : SUCCESS in 1h 07m 53s\n- tempest-full http://logs.openstack.org/80/672480/1/check/tempest-full/4cda804/ : SUCCESS in 1h 38m 13s\n- neutron-grenade http://logs.openstack.org/80/672480/1/check/neutron-grenade/2ec3fd7/ : SUCCESS in 51m 27s","accounts_in_message":[],"_revision_number":1},{"id":"86c023f282cf31d1747777bc5d145ae5a3264f7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-24 15:48:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"83abc2afad3bb68e4c5fe81381cdfd831925af79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 19:30:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/672480/1/check/openstack-tox-docs/34b3446/html/ : SUCCESS in 2m 53s\n- openstack-tox-pypy http://logs.openstack.org/80/672480/1/check/openstack-tox-pypy/3ee477f/ : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/80/672480/1/check/openstack-tox-pep8/cfd27a8/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/80/672480/1/check/openstack-tox-py27/51e33eb/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/80/672480/1/check/openstack-tox-py36/ace90ef/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/80/672480/1/check/openstack-tox-py37/b5441fa/ : SUCCESS in 3m 45s\n- openstack-tox-validate http://logs.openstack.org/80/672480/1/check/openstack-tox-validate/3db27c1/ : SUCCESS in 3m 38s\n- requirements-tox-babel http://logs.openstack.org/80/672480/1/check/requirements-tox-babel/f143fd6/ : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/80/672480/1/check/requirements-tox-py27-check-uc/e83a6ef/ : SUCCESS in 6m 47s\n- requirements-tox-py36-check-uc http://logs.openstack.org/80/672480/1/check/requirements-tox-py36-check-uc/c2bbe45/ : SUCCESS in 6m 51s\n- cross-cinder-py27 http://logs.openstack.org/80/672480/1/check/cross-cinder-py27/f52934b/ : SUCCESS in 9m 19s\n- cross-cinder-py36 http://logs.openstack.org/80/672480/1/check/cross-cinder-py36/baf8f19/ : SUCCESS in 8m 54s\n- cross-glance-py27 http://logs.openstack.org/80/672480/1/check/cross-glance-py27/48f7fd9/ : SUCCESS in 4m 17s\n- cross-glance-py36 http://logs.openstack.org/80/672480/1/check/cross-glance-py36/554d124/ : SUCCESS in 3m 47s\n- cross-horizon-py27 http://logs.openstack.org/80/672480/1/check/cross-horizon-py27/47580ae/ : SUCCESS in 10m 32s\n- cross-horizon-py36 http://logs.openstack.org/80/672480/1/check/cross-horizon-py36/df57d7e/ : SUCCESS in 9m 38s\n- cross-horizon-npm http://logs.openstack.org/80/672480/1/check/cross-horizon-npm/cbecc16/ : SUCCESS in 9m 19s\n- cross-keystone-py27 http://logs.openstack.org/80/672480/1/check/cross-keystone-py27/2de031a/ : SUCCESS in 27m 20s\n- cross-keystone-py36 http://logs.openstack.org/80/672480/1/check/cross-keystone-py36/1758c94/ : SUCCESS in 32m 01s\n- cross-neutron-py27 http://logs.openstack.org/80/672480/1/check/cross-neutron-py27/3fcd3db/ : SUCCESS in 24m 56s\n- cross-neutron-py36 http://logs.openstack.org/80/672480/1/check/cross-neutron-py36/e28d6b5/ : SUCCESS in 23m 56s\n- cross-nova-py27 http://logs.openstack.org/80/672480/1/check/cross-nova-py27/75f60f8/ : SUCCESS in 18m 35s\n- cross-nova-py36 http://logs.openstack.org/80/672480/1/check/cross-nova-py36/f0d1360/ : SUCCESS in 19m 51s\n- cross-nova-functional http://logs.openstack.org/80/672480/1/check/cross-nova-functional/c6ef6df/ : SUCCESS in 23m 37s\n- cross-osc-tox-docs http://logs.openstack.org/80/672480/1/check/cross-osc-tox-docs/5651d1f/html/ : SUCCESS in 3m 49s\n- cross-swift-py27 http://logs.openstack.org/80/672480/1/check/cross-swift-py27/af47ccb/ : SUCCESS in 9m 44s\n- cross-swift-py36 http://logs.openstack.org/80/672480/1/check/cross-swift-py36/3bd8d50/ : SUCCESS in 9m 17s\n- requirements-integration http://logs.openstack.org/80/672480/1/check/requirements-integration/b520fa2/ : SUCCESS in 1h 11m 32s\n- neutron-grenade-multinode http://logs.openstack.org/80/672480/1/check/neutron-grenade-multinode/6cb769e/ : SUCCESS in 1h 23m 24s\n- tempest-full http://logs.openstack.org/80/672480/1/check/tempest-full/e9d1da0/ : SUCCESS in 1h 35m 58s\n- neutron-grenade http://logs.openstack.org/80/672480/1/check/neutron-grenade/402ffb2/ : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5aba037de9ddd4f2a478575d10059765cae801a3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-24 19:37:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b9fd27314a83ade2d2ad5276b380be0db3534ee","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-24 19:37:41.000000000","message":"Abandoned\n\nuse the 1.10.0 one","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8c5c6a974225ea9af1b9f278f51dfce9e30ef8ef","revisions":{"8c5c6a974225ea9af1b9f278f51dfce9e30ef8ef":{"kind":"REWORK","_number":1,"created":"2019-07-24 09:31:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/672480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/672480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/672480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/672480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/672480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/672480/1"}}},"commit":{"parents":[{"commit":"7e5222eafb718a97bb6ca7ed842c6f08c0e92670","subject":"Merge \"update constraint for keystonemiddleware to new release 7.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e5222eafb718a97bb6ca7ed842c6f08c0e92670"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-24 09:31:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-24 09:31:02.000000000","tz":0},"subject":"update constraint for openstack-doc-tools to new release 1.9.0","message":"update constraint for openstack-doc-tools to new release 1.9.0\n\nChange-Id: I0513e715c8b6330806d2af3fa2d388b30eba75b7\nmeta:version: 1.9.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Commit: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Change-Id: I8f3348429c9874bd8823344d1aae230f5e8878e8\nmeta:release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+1: Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c5c6a974225ea9af1b9f278f51dfce9e30ef8ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8c5c6a974225ea9af1b9f278f51dfce9e30ef8ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
