)]}'
{"id":"openstack%2Frequirements~672873","triplet_id":"openstack%2Frequirements~master~I876aab704eea1c3d0c93f9fecb33d2039b49e75c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I876aab704eea1c3d0c93f9fecb33d2039b49e75c","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-07-26 06:19:59.000000000","updated":"2019-07-28 09:02:48.000000000","submitted":"2019-07-28 09:02:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672873-1564304568202-94390bb6","meta_rev_id":"ccb66aeb95dfec2c9ce3aba67938d55736c93152","_number":672873,"virtual_id_number":672873,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-28 09:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-28 05:31:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-28 05:31:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-28 05:31:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-28 09:02:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b14cb892c93d38bdb96dad0ef2864224047c3402","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-26 06:19:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b9e925516f6be5845e832fe43bef6f06c5f2497","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 09:00:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/73/672873/1/check/openstack-tox-docs/b588b85/ : POST_FAILURE in 5m 06s\n- openstack-tox-pypy http://logs.openstack.org/73/672873/1/check/openstack-tox-pypy/8741439/ : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/672873/1/check/openstack-tox-pep8/c0b92ef/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/73/672873/1/check/openstack-tox-py27/f6b1ad0/ : SUCCESS in 5m 47s\n- openstack-tox-py36 http://logs.openstack.org/73/672873/1/check/openstack-tox-py36/ae88fa1/ : SUCCESS in 7m 06s\n- openstack-tox-py37 http://logs.openstack.org/73/672873/1/check/openstack-tox-py37/7706231/ : SUCCESS in 5m 53s\n- openstack-tox-validate http://logs.openstack.org/73/672873/1/check/openstack-tox-validate/60da763/ : SUCCESS in 4m 50s\n- requirements-tox-babel http://logs.openstack.org/73/672873/1/check/requirements-tox-babel/2a9b344/ : SUCCESS in 3m 03s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/672873/1/check/requirements-tox-py27-check-uc/e486dca/ : FAILURE in 8m 02s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/672873/1/check/requirements-tox-py36-check-uc/5f03ba3/ : FAILURE in 5m 52s\n- cross-cinder-py27 http://logs.openstack.org/73/672873/1/check/cross-cinder-py27/fdc40a6/ : SUCCESS in 9m 30s\n- cross-cinder-py36 http://logs.openstack.org/73/672873/1/check/cross-cinder-py36/ee4db96/ : FAILURE in 4m 29s\n- cross-glance-py27 http://logs.openstack.org/73/672873/1/check/cross-glance-py27/2632396/ : SUCCESS in 5m 28s\n- cross-glance-py36 http://logs.openstack.org/73/672873/1/check/cross-glance-py36/a1ab498/ : FAILURE in 4m 53s\n- cross-horizon-py27 http://logs.openstack.org/73/672873/1/check/cross-horizon-py27/28c27da/ : SUCCESS in 13m 17s\n- cross-horizon-py36 http://logs.openstack.org/73/672873/1/check/cross-horizon-py36/fde553c/ : FAILURE in 3m 23s\n- cross-horizon-npm http://logs.openstack.org/73/672873/1/check/cross-horizon-npm/1264b8b/ : SUCCESS in 10m 14s\n- cross-keystone-py27 http://logs.openstack.org/73/672873/1/check/cross-keystone-py27/0a01a71/ : SUCCESS in 33m 52s\n- cross-keystone-py36 http://logs.openstack.org/73/672873/1/check/cross-keystone-py36/6f29ef3/ : SUCCESS in 34m 32s\n- cross-neutron-py27 http://logs.openstack.org/73/672873/1/check/cross-neutron-py27/2de278c/ : SUCCESS in 22m 11s\n- cross-neutron-py36 http://logs.openstack.org/73/672873/1/check/cross-neutron-py36/0d47848/ : SUCCESS in 24m 25s\n- cross-nova-py27 http://logs.openstack.org/73/672873/1/check/cross-nova-py27/4a7431a/ : SUCCESS in 16m 42s\n- cross-nova-py36 http://logs.openstack.org/73/672873/1/check/cross-nova-py36/6b3020c/ : SUCCESS in 18m 53s\n- cross-nova-functional http://logs.openstack.org/73/672873/1/check/cross-nova-functional/5125c12/ : SUCCESS in 22m 35s\n- cross-osc-tox-docs http://logs.openstack.org/73/672873/1/check/cross-osc-tox-docs/0fff285/ : POST_FAILURE in 2m 22s\n- cross-swift-py27 http://logs.openstack.org/73/672873/1/check/cross-swift-py27/9a3db81/ : SUCCESS in 9m 37s\n- cross-swift-py36 http://logs.openstack.org/73/672873/1/check/cross-swift-py36/3536c76/ : FAILURE in 4m 17s\n- requirements-integration http://logs.openstack.org/73/672873/1/check/requirements-integration/8452ffa/ : SUCCESS in 1h 11m 49s\n- neutron-grenade-multinode http://logs.openstack.org/73/672873/1/check/neutron-grenade-multinode/fdb383f/ : FAILURE in 1h 08m 13s\n- tempest-full http://logs.openstack.org/73/672873/1/check/tempest-full/3e4e550/ : SUCCESS in 1h 40m 39s\n- neutron-grenade http://logs.openstack.org/73/672873/1/check/neutron-grenade/cfea969/ : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c61491f16526b565deeb486b8aa7d7c130175c24","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-27 06:32:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97ca15f42ad42ab790ab03ee1d29c3109b4958a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 08:24:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/73/672873/2/check/openstack-tox-docs/1af6283/ : POST_FAILURE in 2m 55s\n- openstack-tox-pypy http://logs.openstack.org/73/672873/2/check/openstack-tox-pypy/064719c/ : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/672873/2/check/openstack-tox-pep8/78a0025/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/73/672873/2/check/openstack-tox-py27/3ef157f/ : SUCCESS in 2m 47s\n- openstack-tox-py36 http://logs.openstack.org/73/672873/2/check/openstack-tox-py36/7d5f17f/ : SUCCESS in 6m 06s\n- openstack-tox-py37 http://logs.openstack.org/73/672873/2/check/openstack-tox-py37/6533acb/ : SUCCESS in 4m 23s\n- openstack-tox-validate http://logs.openstack.org/73/672873/2/check/openstack-tox-validate/58230bf/ : SUCCESS in 3m 48s\n- requirements-tox-babel http://logs.openstack.org/73/672873/2/check/requirements-tox-babel/26b9791/ : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/672873/2/check/requirements-tox-py27-check-uc/eeaa3f3/ : FAILURE in 15m 56s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/672873/2/check/requirements-tox-py36-check-uc/d45affc/ : FAILURE in 10m 29s\n- cross-cinder-py27 http://logs.openstack.org/73/672873/2/check/cross-cinder-py27/0c94354/ : SUCCESS in 11m 14s\n- cross-cinder-py36 http://logs.openstack.org/73/672873/2/check/cross-cinder-py36/7fc163a/ : FAILURE in 3m 56s\n- cross-glance-py27 http://logs.openstack.org/73/672873/2/check/cross-glance-py27/53b21db/ : SUCCESS in 6m 52s\n- cross-glance-py36 http://logs.openstack.org/73/672873/2/check/cross-glance-py36/d60378a/ : FAILURE in 5m 49s\n- cross-horizon-py27 http://logs.openstack.org/73/672873/2/check/cross-horizon-py27/d8d501d/ : SUCCESS in 12m 25s\n- cross-horizon-py36 http://logs.openstack.org/73/672873/2/check/cross-horizon-py36/29fce03/ : FAILURE in 2m 35s\n- cross-horizon-npm http://logs.openstack.org/73/672873/2/check/cross-horizon-npm/3b28f64/ : SUCCESS in 6m 48s\n- cross-keystone-py27 http://logs.openstack.org/73/672873/2/check/cross-keystone-py27/eaf8300/ : SUCCESS in 24m 00s\n- cross-keystone-py36 http://logs.openstack.org/73/672873/2/check/cross-keystone-py36/d58244e/ : SUCCESS in 27m 01s\n- cross-neutron-py27 http://logs.openstack.org/73/672873/2/check/cross-neutron-py27/c2f83cb/ : SUCCESS in 29m 09s\n- cross-neutron-py36 http://logs.openstack.org/73/672873/2/check/cross-neutron-py36/727f6ea/ : SUCCESS in 22m 48s\n- cross-nova-py27 http://logs.openstack.org/73/672873/2/check/cross-nova-py27/efc5176/ : SUCCESS in 16m 34s\n- cross-nova-py36 http://logs.openstack.org/73/672873/2/check/cross-nova-py36/ef4424b/ : SUCCESS in 16m 50s\n- cross-nova-functional http://logs.openstack.org/73/672873/2/check/cross-nova-functional/7177857/ : SUCCESS in 25m 21s\n- cross-osc-tox-docs http://logs.openstack.org/73/672873/2/check/cross-osc-tox-docs/23b49e7/ : POST_FAILURE in 3m 17s\n- cross-swift-py27 http://logs.openstack.org/73/672873/2/check/cross-swift-py27/a23c05b/ : SUCCESS in 9m 29s\n- cross-swift-py36 http://logs.openstack.org/73/672873/2/check/cross-swift-py36/8719742/ : FAILURE in 5m 21s\n- requirements-integration http://logs.openstack.org/73/672873/2/check/requirements-integration/9057a7a/ : SUCCESS in 1h 13m 44s\n- neutron-grenade-multinode http://logs.openstack.org/73/672873/2/check/neutron-grenade-multinode/950ea63/ : FAILURE in 1h 03m 42s\n- tempest-full http://logs.openstack.org/73/672873/2/check/tempest-full/287639f/ : SUCCESS in 1h 41m 11s\n- neutron-grenade http://logs.openstack.org/73/672873/2/check/neutron-grenade/ed32c67/ : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d3d20914984f55d9e98e61bba2aabe7e22d0b33f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-27 17:57:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4e67bdac498adcf32cac3dc8e40daa5b7fe3604f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-27 17:57:55.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"5f8b61851e6ae92c6bb29245f04a489ff1d359ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 19:36:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/73/672873/3/check/openstack-tox-docs/7fbf519/ : POST_FAILURE in 2m 24s\n- openstack-tox-pypy http://logs.openstack.org/73/672873/3/check/openstack-tox-pypy/2cdc88a/ : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/672873/3/check/openstack-tox-pep8/4498d6c/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/73/672873/3/check/openstack-tox-py27/7ce4827/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/73/672873/3/check/openstack-tox-py36/8d29c43/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/73/672873/3/check/openstack-tox-py37/5c0a9fa/ : SUCCESS in 4m 13s\n- openstack-tox-validate http://logs.openstack.org/73/672873/3/check/openstack-tox-validate/4143545/ : SUCCESS in 3m 47s\n- requirements-tox-babel http://logs.openstack.org/73/672873/3/check/requirements-tox-babel/0a38c04/ : SUCCESS in 3m 52s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/672873/3/check/requirements-tox-py27-check-uc/9d8b8f0/ : SUCCESS in 17m 29s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/672873/3/check/requirements-tox-py36-check-uc/880ba52/ : FAILURE in 5m 49s\n- cross-cinder-py27 http://logs.openstack.org/73/672873/3/check/cross-cinder-py27/e3b72f4/ : SUCCESS in 10m 33s\n- cross-cinder-py36 http://logs.openstack.org/73/672873/3/check/cross-cinder-py36/2b26929/ : FAILURE in 4m 01s\n- cross-glance-py27 http://logs.openstack.org/73/672873/3/check/cross-glance-py27/b27c6dd/ : SUCCESS in 4m 09s\n- cross-glance-py36 http://logs.openstack.org/73/672873/3/check/cross-glance-py36/38fad01/ : FAILURE in 3m 56s\n- cross-horizon-py27 http://logs.openstack.org/73/672873/3/check/cross-horizon-py27/b948bdf/ : SUCCESS in 10m 12s\n- cross-horizon-py36 http://logs.openstack.org/73/672873/3/check/cross-horizon-py36/3f82158/ : FAILURE in 2m 26s\n- cross-horizon-npm http://logs.openstack.org/73/672873/3/check/cross-horizon-npm/7a07f16/ : SUCCESS in 7m 29s\n- cross-keystone-py27 http://logs.openstack.org/73/672873/3/check/cross-keystone-py27/1bf4a01/ : SUCCESS in 26m 55s\n- cross-keystone-py36 http://logs.openstack.org/73/672873/3/check/cross-keystone-py36/ada6188/ : SUCCESS in 28m 02s\n- cross-neutron-py27 http://logs.openstack.org/73/672873/3/check/cross-neutron-py27/5a084e2/ : SUCCESS in 21m 22s\n- cross-neutron-py36 http://logs.openstack.org/73/672873/3/check/cross-neutron-py36/c4472ef/ : SUCCESS in 26m 04s\n- cross-nova-py27 http://logs.openstack.org/73/672873/3/check/cross-nova-py27/56a9fa0/ : SUCCESS in 17m 54s\n- cross-nova-py36 http://logs.openstack.org/73/672873/3/check/cross-nova-py36/a2a8fbb/ : SUCCESS in 15m 41s\n- cross-nova-functional http://logs.openstack.org/73/672873/3/check/cross-nova-functional/23e1888/ : SUCCESS in 19m 13s\n- cross-osc-tox-docs http://logs.openstack.org/73/672873/3/check/cross-osc-tox-docs/de0b962/ : POST_FAILURE in 2m 11s\n- cross-swift-py27 http://logs.openstack.org/73/672873/3/check/cross-swift-py27/1dde071/ : SUCCESS in 13m 17s\n- cross-swift-py36 http://logs.openstack.org/73/672873/3/check/cross-swift-py36/60a94d3/ : FAILURE in 2m 25s\n- requirements-integration http://logs.openstack.org/73/672873/3/check/requirements-integration/088c6a4/ : SUCCESS in 1h 08m 34s\n- neutron-grenade-multinode http://logs.openstack.org/73/672873/3/check/neutron-grenade-multinode/96f6d67/ : FAILURE in 1h 16m 39s\n- tempest-full http://logs.openstack.org/73/672873/3/check/tempest-full/ed1c518/ : SUCCESS in 1h 32m 43s\n- neutron-grenade http://logs.openstack.org/73/672873/3/check/neutron-grenade/883a813/ : SUCCESS in 58m 23s","accounts_in_message":[],"_revision_number":3},{"id":"abd08aef066ce68e2c5acfc3be66fa4766e8d049","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-28 01:43:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"765b50d991d200387597f9f7f2533cf72688a670","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 03:40:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/73/672873/4/check/openstack-tox-docs/7485e1c/ : POST_FAILURE in 6m 26s\n- openstack-tox-pypy http://logs.openstack.org/73/672873/4/check/openstack-tox-pypy/3646fd5/ : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/672873/4/check/openstack-tox-pep8/5364eec/ : SUCCESS in 2m 31s\n- openstack-tox-py27 http://logs.openstack.org/73/672873/4/check/openstack-tox-py27/d67ef40/ : SUCCESS in 3m 46s\n- openstack-tox-py36 http://logs.openstack.org/73/672873/4/check/openstack-tox-py36/409c66a/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/73/672873/4/check/openstack-tox-py37/290cb42/ : SUCCESS in 2m 55s\n- openstack-tox-validate http://logs.openstack.org/73/672873/4/check/openstack-tox-validate/6c189cd/ : SUCCESS in 2m 50s\n- requirements-tox-babel http://logs.openstack.org/73/672873/4/check/requirements-tox-babel/b4546bf/ : SUCCESS in 5m 58s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/672873/4/check/requirements-tox-py27-check-uc/a6b364e/ : SUCCESS in 8m 04s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/672873/4/check/requirements-tox-py36-check-uc/d2c96f7/ : FAILURE in 4m 05s\n- cross-cinder-py27 http://logs.openstack.org/73/672873/4/check/cross-cinder-py27/e276784/ : SUCCESS in 10m 11s\n- cross-cinder-py36 http://logs.openstack.org/73/672873/4/check/cross-cinder-py36/864824a/ : FAILURE in 4m 12s\n- cross-glance-py27 http://logs.openstack.org/73/672873/4/check/cross-glance-py27/6c76a9a/ : SUCCESS in 6m 07s\n- cross-glance-py36 http://logs.openstack.org/73/672873/4/check/cross-glance-py36/6f00ae3/ : FAILURE in 3m 42s\n- cross-horizon-py27 http://logs.openstack.org/73/672873/4/check/cross-horizon-py27/bdd174d/ : SUCCESS in 9m 08s\n- cross-horizon-py36 http://logs.openstack.org/73/672873/4/check/cross-horizon-py36/d57a8ca/ : FAILURE in 3m 30s\n- cross-horizon-npm http://logs.openstack.org/73/672873/4/check/cross-horizon-npm/55352d5/ : SUCCESS in 11m 20s\n- cross-keystone-py27 http://logs.openstack.org/73/672873/4/check/cross-keystone-py27/40e8159/ : SUCCESS in 28m 05s\n- cross-keystone-py36 http://logs.openstack.org/73/672873/4/check/cross-keystone-py36/089a2b5/ : SUCCESS in 26m 04s\n- cross-neutron-py27 http://logs.openstack.org/73/672873/4/check/cross-neutron-py27/de12740/ : SUCCESS in 18m 37s\n- cross-neutron-py36 http://logs.openstack.org/73/672873/4/check/cross-neutron-py36/8270097/ : SUCCESS in 21m 21s\n- cross-nova-py27 http://logs.openstack.org/73/672873/4/check/cross-nova-py27/ee9b4b4/ : SUCCESS in 20m 05s\n- cross-nova-py36 http://logs.openstack.org/73/672873/4/check/cross-nova-py36/e0f95d0/ : SUCCESS in 16m 26s\n- cross-nova-functional http://logs.openstack.org/73/672873/4/check/cross-nova-functional/efcf4d3/ : SUCCESS in 20m 12s\n- cross-osc-tox-docs http://logs.openstack.org/73/672873/4/check/cross-osc-tox-docs/3239bda/ : POST_FAILURE in 2m 43s\n- cross-swift-py27 http://logs.openstack.org/73/672873/4/check/cross-swift-py27/033cce7/ : SUCCESS in 9m 14s\n- cross-swift-py36 http://logs.openstack.org/73/672873/4/check/cross-swift-py36/862794e/ : FAILURE in 2m 52s\n- requirements-integration http://logs.openstack.org/73/672873/4/check/requirements-integration/563bf0b/ : SUCCESS in 46m 40s\n- neutron-grenade-multinode http://logs.openstack.org/73/672873/4/check/neutron-grenade-multinode/068ff31/ : FAILURE in 52m 47s\n- tempest-full http://logs.openstack.org/73/672873/4/check/tempest-full/494fccc/ : SUCCESS in 1h 31m 13s\n- neutron-grenade http://logs.openstack.org/73/672873/4/check/neutron-grenade/386d364/ : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":4},{"id":"8573a2e7f90e9d1511d54ca16cb60e5eb29eb089","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-28 03:48:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a9ee22b42d45b9fa7bc590ee9de785df582babcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 05:22:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/73/672873/5/check/openstack-tox-docs/bc9ac9a/html/ : SUCCESS in 6m 34s\n- openstack-tox-pypy http://logs.openstack.org/73/672873/5/check/openstack-tox-pypy/4a09181/ : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/672873/5/check/openstack-tox-pep8/bb04974/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/73/672873/5/check/openstack-tox-py27/9ad7d46/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/73/672873/5/check/openstack-tox-py36/4b2eda3/ : SUCCESS in 3m 04s\n- openstack-tox-py37 http://logs.openstack.org/73/672873/5/check/openstack-tox-py37/b6c9999/ : SUCCESS in 3m 25s\n- openstack-tox-validate http://logs.openstack.org/73/672873/5/check/openstack-tox-validate/6edc881/ : SUCCESS in 3m 27s\n- requirements-tox-babel http://logs.openstack.org/73/672873/5/check/requirements-tox-babel/bcecf68/ : SUCCESS in 5m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/672873/5/check/requirements-tox-py27-check-uc/3f8d317/ : SUCCESS in 7m 32s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/672873/5/check/requirements-tox-py36-check-uc/92fa893/ : SUCCESS in 8m 56s\n- cross-cinder-py27 http://logs.openstack.org/73/672873/5/check/cross-cinder-py27/b1f7841/ : SUCCESS in 8m 52s\n- cross-cinder-py36 http://logs.openstack.org/73/672873/5/check/cross-cinder-py36/0d8c1b9/ : SUCCESS in 8m 10s\n- cross-glance-py27 http://logs.openstack.org/73/672873/5/check/cross-glance-py27/162b5d4/ : SUCCESS in 5m 18s\n- cross-glance-py36 http://logs.openstack.org/73/672873/5/check/cross-glance-py36/5fb884f/ : SUCCESS in 5m 35s\n- cross-horizon-py27 http://logs.openstack.org/73/672873/5/check/cross-horizon-py27/555cfbf/ : SUCCESS in 8m 33s\n- cross-horizon-py36 http://logs.openstack.org/73/672873/5/check/cross-horizon-py36/3c2961c/ : SUCCESS in 6m 52s\n- cross-horizon-npm http://logs.openstack.org/73/672873/5/check/cross-horizon-npm/1bf67ce/ : SUCCESS in 7m 29s\n- cross-keystone-py27 http://logs.openstack.org/73/672873/5/check/cross-keystone-py27/f4bc751/ : SUCCESS in 25m 16s\n- cross-keystone-py36 http://logs.openstack.org/73/672873/5/check/cross-keystone-py36/0026117/ : SUCCESS in 24m 07s\n- cross-neutron-py27 http://logs.openstack.org/73/672873/5/check/cross-neutron-py27/0054fb3/ : SUCCESS in 21m 05s\n- cross-neutron-py36 http://logs.openstack.org/73/672873/5/check/cross-neutron-py36/ba3212b/ : SUCCESS in 18m 48s\n- cross-nova-py27 http://logs.openstack.org/73/672873/5/check/cross-nova-py27/ed4f420/ : SUCCESS in 15m 02s\n- cross-nova-py36 http://logs.openstack.org/73/672873/5/check/cross-nova-py36/a188917/ : SUCCESS in 16m 25s\n- cross-nova-functional http://logs.openstack.org/73/672873/5/check/cross-nova-functional/51b8ade/ : SUCCESS in 23m 22s\n- cross-osc-tox-docs http://logs.openstack.org/73/672873/5/check/cross-osc-tox-docs/dff8c0f/html/ : SUCCESS in 4m 17s\n- cross-swift-py27 http://logs.openstack.org/73/672873/5/check/cross-swift-py27/106c137/ : SUCCESS in 7m 36s\n- cross-swift-py36 http://logs.openstack.org/73/672873/5/check/cross-swift-py36/31c73fe/ : FAILURE in 10m 02s\n- requirements-integration http://logs.openstack.org/73/672873/5/check/requirements-integration/b6f698c/ : SUCCESS in 54m 15s\n- neutron-grenade-multinode http://logs.openstack.org/73/672873/5/check/neutron-grenade-multinode/9ef1bfc/ : SUCCESS in 1h 08m 49s\n- tempest-full http://logs.openstack.org/73/672873/5/check/tempest-full/540db02/ : SUCCESS in 1h 31m 49s\n- neutron-grenade http://logs.openstack.org/73/672873/5/check/neutron-grenade/d7fc820/ : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":5},{"id":"931da0944d7d3c5a99951d56547e90690defeafe","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-28 05:31:22.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ef60699b5fbb2b8a1c16554f6bdc8a4d3ab8d913","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 07:22:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/672873/5/check/openstack-tox-docs/f144ae7/html/ : SUCCESS in 4m 43s\n- openstack-tox-pypy http://logs.openstack.org/73/672873/5/check/openstack-tox-pypy/20db9c3/ : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/672873/5/check/openstack-tox-pep8/6f2126e/ : SUCCESS in 2m 51s\n- openstack-tox-py27 http://logs.openstack.org/73/672873/5/check/openstack-tox-py27/23ea9af/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/73/672873/5/check/openstack-tox-py36/ae62e26/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/73/672873/5/check/openstack-tox-py37/f87c1eb/ : SUCCESS in 3m 32s\n- openstack-tox-validate http://logs.openstack.org/73/672873/5/check/openstack-tox-validate/0fdb43e/ : SUCCESS in 3m 25s\n- requirements-tox-babel http://logs.openstack.org/73/672873/5/check/requirements-tox-babel/eeed9ad/ : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/672873/5/check/requirements-tox-py27-check-uc/20be7f5/ : SUCCESS in 7m 27s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/672873/5/check/requirements-tox-py36-check-uc/19ab493/ : SUCCESS in 8m 00s\n- cross-cinder-py27 http://logs.openstack.org/73/672873/5/check/cross-cinder-py27/0cfa0e0/ : SUCCESS in 10m 49s\n- cross-cinder-py36 http://logs.openstack.org/73/672873/5/check/cross-cinder-py36/4f82771/ : SUCCESS in 8m 49s\n- cross-glance-py27 http://logs.openstack.org/73/672873/5/check/cross-glance-py27/fe891e8/ : SUCCESS in 4m 28s\n- cross-glance-py36 http://logs.openstack.org/73/672873/5/check/cross-glance-py36/c3a8600/ : SUCCESS in 3m 38s\n- cross-horizon-py27 http://logs.openstack.org/73/672873/5/check/cross-horizon-py27/5cb6fde/ : SUCCESS in 12m 27s\n- cross-horizon-py36 http://logs.openstack.org/73/672873/5/check/cross-horizon-py36/da1e158/ : SUCCESS in 6m 46s\n- cross-horizon-npm http://logs.openstack.org/73/672873/5/check/cross-horizon-npm/2f755a1/ : SUCCESS in 5m 22s\n- cross-keystone-py27 http://logs.openstack.org/73/672873/5/check/cross-keystone-py27/3dea1e7/ : SUCCESS in 20m 33s\n- cross-keystone-py36 http://logs.openstack.org/73/672873/5/check/cross-keystone-py36/7071916/ : SUCCESS in 32m 38s\n- cross-neutron-py27 http://logs.openstack.org/73/672873/5/check/cross-neutron-py27/dda1695/ : SUCCESS in 23m 56s\n- cross-neutron-py36 http://logs.openstack.org/73/672873/5/check/cross-neutron-py36/a63c475/ : SUCCESS in 18m 05s\n- cross-nova-py27 http://logs.openstack.org/73/672873/5/check/cross-nova-py27/e7aaada/ : SUCCESS in 16m 45s\n- cross-nova-py36 http://logs.openstack.org/73/672873/5/check/cross-nova-py36/4e3c8ee/ : SUCCESS in 14m 34s\n- cross-nova-functional http://logs.openstack.org/73/672873/5/check/cross-nova-functional/64c4c5a/ : SUCCESS in 22m 53s\n- cross-osc-tox-docs http://logs.openstack.org/73/672873/5/check/cross-osc-tox-docs/601c82e/html/ : SUCCESS in 3m 19s\n- cross-swift-py27 http://logs.openstack.org/73/672873/5/check/cross-swift-py27/7bf3729/ : SUCCESS in 7m 43s\n- cross-swift-py36 http://logs.openstack.org/73/672873/5/check/cross-swift-py36/60570e7/ : SUCCESS in 6m 56s\n- requirements-integration http://logs.openstack.org/73/672873/5/check/requirements-integration/28b8208/ : SUCCESS in 57m 35s\n- neutron-grenade-multinode http://logs.openstack.org/73/672873/5/check/neutron-grenade-multinode/75f4388/ : SUCCESS in 1h 11m 58s\n- tempest-full http://logs.openstack.org/73/672873/5/check/tempest-full/ff57975/ : SUCCESS in 1h 44m 57s\n- neutron-grenade http://logs.openstack.org/73/672873/5/check/neutron-grenade/0add7f3/ : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":5},{"id":"4e811b0e205e71463b05f8e029776680a9348b86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 07:22:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"28df2c649c66f28ee0ab5d9a5e7c942f2042d456","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 09:02:47.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/73/672873/5/gate/openstack-tox-docs/3a0d2ff/html/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/73/672873/5/gate/openstack-tox-pep8/91b0966/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/73/672873/5/gate/openstack-tox-py27/b989daf/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/73/672873/5/gate/openstack-tox-py36/69282fd/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/73/672873/5/gate/openstack-tox-py37/ec1d78e/ : SUCCESS in 4m 33s\n- openstack-tox-validate http://logs.openstack.org/73/672873/5/gate/openstack-tox-validate/e73e33e/ : SUCCESS in 3m 45s\n- requirements-tox-babel http://logs.openstack.org/73/672873/5/gate/requirements-tox-babel/54025ab/ : SUCCESS in 3m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/73/672873/5/gate/requirements-tox-py27-check-uc/a0141f9/ : SUCCESS in 8m 41s\n- requirements-tox-py36-check-uc http://logs.openstack.org/73/672873/5/gate/requirements-tox-py36-check-uc/ab41c88/ : SUCCESS in 8m 50s\n- cross-cinder-py27 http://logs.openstack.org/73/672873/5/gate/cross-cinder-py27/e0ee8e6/ : SUCCESS in 9m 41s\n- cross-cinder-py36 http://logs.openstack.org/73/672873/5/gate/cross-cinder-py36/4dfbde3/ : SUCCESS in 8m 42s\n- cross-glance-py27 http://logs.openstack.org/73/672873/5/gate/cross-glance-py27/69da317/ : SUCCESS in 5m 39s\n- cross-glance-py36 http://logs.openstack.org/73/672873/5/gate/cross-glance-py36/d1e21a0/ : SUCCESS in 5m 20s\n- cross-horizon-py27 http://logs.openstack.org/73/672873/5/gate/cross-horizon-py27/ab6c804/ : SUCCESS in 9m 17s\n- cross-horizon-py36 http://logs.openstack.org/73/672873/5/gate/cross-horizon-py36/6d62e59/ : SUCCESS in 7m 58s\n- cross-horizon-npm http://logs.openstack.org/73/672873/5/gate/cross-horizon-npm/580f7b2/ : SUCCESS in 8m 14s\n- cross-keystone-py27 http://logs.openstack.org/73/672873/5/gate/cross-keystone-py27/a7492f4/ : SUCCESS in 30m 59s\n- cross-keystone-py36 http://logs.openstack.org/73/672873/5/gate/cross-keystone-py36/0243302/ : SUCCESS in 33m 29s\n- cross-neutron-py27 http://logs.openstack.org/73/672873/5/gate/cross-neutron-py27/4396a2a/ : SUCCESS in 28m 29s\n- cross-neutron-py36 http://logs.openstack.org/73/672873/5/gate/cross-neutron-py36/28b78be/ : SUCCESS in 18m 50s\n- cross-nova-py27 http://logs.openstack.org/73/672873/5/gate/cross-nova-py27/65e5eff/ : SUCCESS in 17m 47s\n- cross-nova-py36 http://logs.openstack.org/73/672873/5/gate/cross-nova-py36/40f8dae/ : SUCCESS in 13m 01s\n- cross-nova-functional http://logs.openstack.org/73/672873/5/gate/cross-nova-functional/ae2138b/ : SUCCESS in 28m 08s\n- cross-osc-tox-docs http://logs.openstack.org/73/672873/5/gate/cross-osc-tox-docs/f532aa8/html/ : SUCCESS in 4m 09s\n- cross-swift-py27 http://logs.openstack.org/73/672873/5/gate/cross-swift-py27/91f517d/ : SUCCESS in 7m 51s\n- cross-swift-py36 http://logs.openstack.org/73/672873/5/gate/cross-swift-py36/1d27e30/ : SUCCESS in 9m 04s\n- requirements-integration http://logs.openstack.org/73/672873/5/gate/requirements-integration/e722260/ : SUCCESS in 50m 09s\n- neutron-grenade-multinode http://logs.openstack.org/73/672873/5/gate/neutron-grenade-multinode/7cf4ef7/ : SUCCESS in 1h 09m 36s\n- tempest-full http://logs.openstack.org/73/672873/5/gate/tempest-full/f7baf9d/ : SUCCESS in 1h 27m 29s\n- neutron-grenade http://logs.openstack.org/73/672873/5/gate/neutron-grenade/aa39ad7/ : SUCCESS in 55m 59s","accounts_in_message":[],"_revision_number":5},{"id":"ccb66aeb95dfec2c9ce3aba67938d55736c93152","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 09:02:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca","revisions":{"7da919965110c9c7ed304e315814937086dbe739":{"kind":"REWORK","_number":1,"created":"2019-07-26 06:19:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/672873/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/672873/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/672873/1"}}},"commit":{"parents":[{"commit":"636e93e9cb8098004b925e20b87464380536035d","subject":"update constraint for tempest to new release 21.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/636e93e9cb8098004b925e20b87464380536035d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-26 06:19:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-26 06:19:57.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I876aab704eea1c3d0c93f9fecb33d2039b49e75c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7da919965110c9c7ed304e315814937086dbe739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7da919965110c9c7ed304e315814937086dbe739"}]},"branch":"refs/heads/master"},"26d02adde526fe775cc1e858ac84ea7d8a896a18":{"kind":"REWORK","_number":2,"created":"2019-07-27 06:32:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/672873/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/672873/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/672873/2"}}},"commit":{"parents":[{"commit":"eb16ce0ff3c642dcac1940ac63b1093774327e91","subject":"Merge \"update constraint for os-net-config to new release 11.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb16ce0ff3c642dcac1940ac63b1093774327e91"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-27 06:32:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-27 06:32:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I876aab704eea1c3d0c93f9fecb33d2039b49e75c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26d02adde526fe775cc1e858ac84ea7d8a896a18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26d02adde526fe775cc1e858ac84ea7d8a896a18"}]},"branch":"refs/heads/master"},"01d38f73bbf35dafc22e0330cd61efb812a8ee5b":{"kind":"REWORK","_number":3,"created":"2019-07-27 17:57:52.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/672873/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/672873/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/672873/3"}}},"commit":{"parents":[{"commit":"eb16ce0ff3c642dcac1940ac63b1093774327e91","subject":"Merge \"update constraint for os-net-config to new release 11.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb16ce0ff3c642dcac1940ac63b1093774327e91"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-27 06:32:09.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-27 17:57:42.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I876aab704eea1c3d0c93f9fecb33d2039b49e75c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01d38f73bbf35dafc22e0330cd61efb812a8ee5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01d38f73bbf35dafc22e0330cd61efb812a8ee5b"}]},"branch":"refs/heads/master"},"e75d3cb09d10f08f2ad0b57a8b9bb3d537116891":{"kind":"REWORK","_number":4,"created":"2019-07-28 01:43:20.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/672873/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/672873/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/672873/4"}}},"commit":{"parents":[{"commit":"eb16ce0ff3c642dcac1940ac63b1093774327e91","subject":"Merge \"update constraint for os-net-config to new release 11.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb16ce0ff3c642dcac1940ac63b1093774327e91"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-27 06:32:09.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-28 01:43:08.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I876aab704eea1c3d0c93f9fecb33d2039b49e75c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e75d3cb09d10f08f2ad0b57a8b9bb3d537116891"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e75d3cb09d10f08f2ad0b57a8b9bb3d537116891"}]},"branch":"refs/heads/master"},"5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca":{"kind":"REWORK","_number":5,"created":"2019-07-28 03:48:28.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/672873/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/672873/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/672873/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/672873/5"}}},"commit":{"parents":[{"commit":"eb16ce0ff3c642dcac1940ac63b1093774327e91","subject":"Merge \"update constraint for os-net-config to new release 11.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb16ce0ff3c642dcac1940ac63b1093774327e91"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-27 06:32:09.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-28 03:48:18.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I876aab704eea1c3d0c93f9fecb33d2039b49e75c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
