)]}'
{"id":"openstack%2Frequirements~672944","triplet_id":"openstack%2Frequirements~master~I0f31b44daea5898f43de8449f91fedac4f0cea45","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0f31b44daea5898f43de8449f91fedac4f0cea45","subject":"update constraint for glance_store to new release 1.0.0","status":"MERGED","created":"2019-07-26 10:17:09.000000000","updated":"2019-07-26 16:30:35.000000000","submitted":"2019-07-26 16:30:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672944-1564158635303-c3a5ce02","meta_rev_id":"5fbe459ba4a5e012211e909df2862dfd8575aa79","_number":672944,"virtual_id_number":672944,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-26 16:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-26 14:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-26 14:35:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 14:35:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-26 16:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65c9ed41eae03916c01078f5d1202c9598669957","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-26 10:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e7d1b28da54f978e006caa1e48364784a0b7622","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 12:01:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/672944/1/check/openstack-tox-docs/c9a21e2/html/ : SUCCESS in 3m 38s\n- openstack-tox-pypy http://logs.openstack.org/44/672944/1/check/openstack-tox-pypy/f6a08da/ : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/44/672944/1/check/openstack-tox-pep8/3f5cbbc/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/44/672944/1/check/openstack-tox-py27/55eb2ed/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/44/672944/1/check/openstack-tox-py36/1043b1d/ : SUCCESS in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/44/672944/1/check/openstack-tox-py37/c708d0a/ : SUCCESS in 4m 52s\n- openstack-tox-validate http://logs.openstack.org/44/672944/1/check/openstack-tox-validate/136c9f9/ : SUCCESS in 3m 06s\n- requirements-tox-babel http://logs.openstack.org/44/672944/1/check/requirements-tox-babel/30b10dd/ : SUCCESS in 4m 22s\n- requirements-tox-py27-check-uc http://logs.openstack.org/44/672944/1/check/requirements-tox-py27-check-uc/a159759/ : SUCCESS in 13m 44s\n- requirements-tox-py36-check-uc http://logs.openstack.org/44/672944/1/check/requirements-tox-py36-check-uc/6c5fd6c/ : SUCCESS in 8m 21s\n- cross-cinder-py27 http://logs.openstack.org/44/672944/1/check/cross-cinder-py27/d5f637e/ : SUCCESS in 8m 15s\n- cross-cinder-py36 http://logs.openstack.org/44/672944/1/check/cross-cinder-py36/b05e8e2/ : SUCCESS in 9m 33s\n- cross-glance-py27 http://logs.openstack.org/44/672944/1/check/cross-glance-py27/73e0707/ : SUCCESS in 4m 21s\n- cross-glance-py36 http://logs.openstack.org/44/672944/1/check/cross-glance-py36/d885b72/ : SUCCESS in 4m 15s\n- cross-horizon-py27 http://logs.openstack.org/44/672944/1/check/cross-horizon-py27/6ffd74f/ : SUCCESS in 10m 50s\n- cross-horizon-py36 http://logs.openstack.org/44/672944/1/check/cross-horizon-py36/bddce1b/ : SUCCESS in 7m 35s\n- cross-horizon-npm http://logs.openstack.org/44/672944/1/check/cross-horizon-npm/6b2446a/ : SUCCESS in 6m 12s\n- cross-keystone-py27 http://logs.openstack.org/44/672944/1/check/cross-keystone-py27/a2e836e/ : SUCCESS in 25m 02s\n- cross-keystone-py36 http://logs.openstack.org/44/672944/1/check/cross-keystone-py36/49273bb/ : SUCCESS in 30m 53s\n- cross-neutron-py27 http://logs.openstack.org/44/672944/1/check/cross-neutron-py27/7fbe36e/ : SUCCESS in 22m 07s\n- cross-neutron-py36 http://logs.openstack.org/44/672944/1/check/cross-neutron-py36/7122489/ : SUCCESS in 19m 49s\n- cross-nova-py27 http://logs.openstack.org/44/672944/1/check/cross-nova-py27/f4dc524/ : SUCCESS in 18m 57s\n- cross-nova-py36 http://logs.openstack.org/44/672944/1/check/cross-nova-py36/4d18cba/ : SUCCESS in 16m 31s\n- cross-nova-functional http://logs.openstack.org/44/672944/1/check/cross-nova-functional/c793f9e/ : SUCCESS in 21m 12s\n- cross-osc-tox-docs http://logs.openstack.org/44/672944/1/check/cross-osc-tox-docs/5cb9567/html/ : SUCCESS in 4m 28s\n- cross-swift-py27 http://logs.openstack.org/44/672944/1/check/cross-swift-py27/b9a9abd/ : SUCCESS in 8m 04s\n- cross-swift-py36 http://logs.openstack.org/44/672944/1/check/cross-swift-py36/56fd44d/ : SUCCESS in 7m 24s\n- requirements-integration http://logs.openstack.org/44/672944/1/check/requirements-integration/d5b9033/ : SUCCESS in 53m 35s\n- neutron-grenade-multinode http://logs.openstack.org/44/672944/1/check/neutron-grenade-multinode/1931c83/ : SUCCESS in 1h 13m 36s\n- tempest-full http://logs.openstack.org/44/672944/1/check/tempest-full/a1939c3/ : SUCCESS in 1h 38m 54s\n- neutron-grenade http://logs.openstack.org/44/672944/1/check/neutron-grenade/c85e4db/ : SUCCESS in 1h 04m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a4822f60862d3d4990c5c4b950223afa7690aa45","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-26 14:35:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"75486add1f1eb351987130186ca7396999909c68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 14:35:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d643b337c06152c9fde51535deb37cc8cb5468a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 16:30:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"88ada76fc01c8303b2a761c4bbd99c226a4126d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 16:30:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/672944/1/gate/openstack-tox-docs/e209952/html/ : SUCCESS in 3m 21s\n- openstack-tox-pep8 http://logs.openstack.org/44/672944/1/gate/openstack-tox-pep8/ce372c0/ : SUCCESS in 6m 22s\n- openstack-tox-py27 http://logs.openstack.org/44/672944/1/gate/openstack-tox-py27/2e2d4ec/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/44/672944/1/gate/openstack-tox-py36/07fe667/ : SUCCESS in 7m 08s\n- openstack-tox-py37 http://logs.openstack.org/44/672944/1/gate/openstack-tox-py37/3811daa/ : SUCCESS in 8m 38s\n- openstack-tox-validate http://logs.openstack.org/44/672944/1/gate/openstack-tox-validate/762400f/ : SUCCESS in 3m 49s\n- requirements-tox-babel http://logs.openstack.org/44/672944/1/gate/requirements-tox-babel/03b2948/ : SUCCESS in 2m 44s\n- requirements-tox-py27-check-uc http://logs.openstack.org/44/672944/1/gate/requirements-tox-py27-check-uc/aab05f6/ : SUCCESS in 19m 53s\n- requirements-tox-py36-check-uc http://logs.openstack.org/44/672944/1/gate/requirements-tox-py36-check-uc/3af3bf7/ : SUCCESS in 20m 41s\n- cross-cinder-py27 http://logs.openstack.org/44/672944/1/gate/cross-cinder-py27/64022ea/ : SUCCESS in 14m 37s\n- cross-cinder-py36 http://logs.openstack.org/44/672944/1/gate/cross-cinder-py36/d9d7fc2/ : SUCCESS in 9m 08s\n- cross-glance-py27 http://logs.openstack.org/44/672944/1/gate/cross-glance-py27/89c973c/ : SUCCESS in 5m 34s\n- cross-glance-py36 http://logs.openstack.org/44/672944/1/gate/cross-glance-py36/d5ecb50/ : SUCCESS in 5m 16s\n- cross-horizon-py27 http://logs.openstack.org/44/672944/1/gate/cross-horizon-py27/07d868a/ : SUCCESS in 10m 55s\n- cross-horizon-py36 http://logs.openstack.org/44/672944/1/gate/cross-horizon-py36/974f4ca/ : SUCCESS in 11m 51s\n- cross-horizon-npm http://logs.openstack.org/44/672944/1/gate/cross-horizon-npm/6638f7e/ : SUCCESS in 13m 30s\n- cross-keystone-py27 http://logs.openstack.org/44/672944/1/gate/cross-keystone-py27/a4e71c4/ : SUCCESS in 27m 41s\n- cross-keystone-py36 http://logs.openstack.org/44/672944/1/gate/cross-keystone-py36/3837c7d/ : SUCCESS in 28m 46s\n- cross-neutron-py27 http://logs.openstack.org/44/672944/1/gate/cross-neutron-py27/59d00f7/ : SUCCESS in 23m 40s\n- cross-neutron-py36 http://logs.openstack.org/44/672944/1/gate/cross-neutron-py36/5d4eca9/ : SUCCESS in 24m 16s\n- cross-nova-py27 http://logs.openstack.org/44/672944/1/gate/cross-nova-py27/b19dbf6/ : SUCCESS in 20m 40s\n- cross-nova-py36 http://logs.openstack.org/44/672944/1/gate/cross-nova-py36/f929c11/ : SUCCESS in 23m 41s\n- cross-nova-functional http://logs.openstack.org/44/672944/1/gate/cross-nova-functional/aa2a4ee/ : SUCCESS in 22m 56s\n- cross-osc-tox-docs http://logs.openstack.org/44/672944/1/gate/cross-osc-tox-docs/c42f488/html/ : SUCCESS in 6m 21s\n- cross-swift-py27 http://logs.openstack.org/44/672944/1/gate/cross-swift-py27/e7572f1/ : SUCCESS in 8m 46s\n- cross-swift-py36 http://logs.openstack.org/44/672944/1/gate/cross-swift-py36/a1a532c/ : SUCCESS in 11m 17s\n- requirements-integration http://logs.openstack.org/44/672944/1/gate/requirements-integration/d0ebf0d/ : SUCCESS in 58m 36s\n- neutron-grenade-multinode http://logs.openstack.org/44/672944/1/gate/neutron-grenade-multinode/73fe92f/ : SUCCESS in 1h 14m 27s\n- tempest-full http://logs.openstack.org/44/672944/1/gate/tempest-full/ad122b9/ : SUCCESS in 1h 50m 20s\n- neutron-grenade http://logs.openstack.org/44/672944/1/gate/neutron-grenade/f4e40c7/ : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71f702b416779d4cb59b78f4d80a5cd69fb1fe51","revisions":{"71f702b416779d4cb59b78f4d80a5cd69fb1fe51":{"kind":"REWORK","_number":1,"created":"2019-07-26 10:17:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/672944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/672944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/672944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/672944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/672944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/672944/1"}}},"commit":{"parents":[{"commit":"636e93e9cb8098004b925e20b87464380536035d","subject":"update constraint for tempest to new release 21.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/636e93e9cb8098004b925e20b87464380536035d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-26 10:17:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-26 10:17:05.000000000","tz":0},"subject":"update constraint for glance_store to new release 1.0.0","message":"update constraint for glance_store to new release 1.0.0\n\nChange-Id: I0f31b44daea5898f43de8449f91fedac4f0cea45\nmeta:version: 1.0.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Commit: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta:release:Change-Id: Ib4134f707af2a832cfbb5a0bd9fe27bc1ed5b9df\nmeta:release:Code-Review+1: Erno Kuvaja \u003cjokke@usr.fi\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71f702b416779d4cb59b78f4d80a5cd69fb1fe51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71f702b416779d4cb59b78f4d80a5cd69fb1fe51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
