)]}'
{"id":"openstack%2Frequirements~673030","triplet_id":"openstack%2Frequirements~master~I22a9cbb40a8c713102266275c01bbb543404489c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I22a9cbb40a8c713102266275c01bbb543404489c","subject":"update constraint for os-net-config to new release 11.1.0","status":"MERGED","created":"2019-07-26 18:20:41.000000000","updated":"2019-07-27 04:56:33.000000000","submitted":"2019-07-27 04:56:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673030-1564203393831-e397d3ca","meta_rev_id":"ad276a7a37dbf78492c3dd668510dff0822d59f5","_number":673030,"virtual_id_number":673030,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-27 04:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-26 21:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-26 21:19:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 21:19:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-27 04:56:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34a94f93c7963de16e3a7064fc4e56b76c7502f1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-26 18:20:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0f6cf727e3c0868eee35069872f5f3da00953e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 19:57:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/673030/1/check/openstack-tox-docs/8b9a8f7/html/ : SUCCESS in 3m 51s\n- openstack-tox-pypy http://logs.openstack.org/30/673030/1/check/openstack-tox-pypy/30bb4a2/ : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/30/673030/1/check/openstack-tox-pep8/7b4c2fb/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/30/673030/1/check/openstack-tox-py27/08be38e/ : SUCCESS in 4m 22s\n- openstack-tox-py36 http://logs.openstack.org/30/673030/1/check/openstack-tox-py36/fca28fb/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/30/673030/1/check/openstack-tox-py37/b4e1352/ : SUCCESS in 4m 57s\n- openstack-tox-validate http://logs.openstack.org/30/673030/1/check/openstack-tox-validate/f96e67c/ : SUCCESS in 3m 58s\n- requirements-tox-babel http://logs.openstack.org/30/673030/1/check/requirements-tox-babel/08966cd/ : SUCCESS in 3m 32s\n- requirements-tox-py27-check-uc http://logs.openstack.org/30/673030/1/check/requirements-tox-py27-check-uc/e07c15f/ : SUCCESS in 16m 57s\n- requirements-tox-py36-check-uc http://logs.openstack.org/30/673030/1/check/requirements-tox-py36-check-uc/f95f78b/ : SUCCESS in 7m 56s\n- cross-cinder-py27 http://logs.openstack.org/30/673030/1/check/cross-cinder-py27/d298674/ : SUCCESS in 9m 32s\n- cross-cinder-py36 http://logs.openstack.org/30/673030/1/check/cross-cinder-py36/b19f2ca/ : SUCCESS in 8m 04s\n- cross-glance-py27 http://logs.openstack.org/30/673030/1/check/cross-glance-py27/6bfc04f/ : SUCCESS in 3m 59s\n- cross-glance-py36 http://logs.openstack.org/30/673030/1/check/cross-glance-py36/00805ef/ : SUCCESS in 5m 43s\n- cross-horizon-py27 http://logs.openstack.org/30/673030/1/check/cross-horizon-py27/ed08d84/ : SUCCESS in 11m 55s\n- cross-horizon-py36 http://logs.openstack.org/30/673030/1/check/cross-horizon-py36/0a5e91c/ : SUCCESS in 11m 26s\n- cross-horizon-npm http://logs.openstack.org/30/673030/1/check/cross-horizon-npm/cf22cfa/ : SUCCESS in 6m 12s\n- cross-keystone-py27 http://logs.openstack.org/30/673030/1/check/cross-keystone-py27/9d1ce9a/ : SUCCESS in 24m 51s\n- cross-keystone-py36 http://logs.openstack.org/30/673030/1/check/cross-keystone-py36/59516c1/ : SUCCESS in 28m 07s\n- cross-neutron-py27 http://logs.openstack.org/30/673030/1/check/cross-neutron-py27/20319b0/ : SUCCESS in 21m 57s\n- cross-neutron-py36 http://logs.openstack.org/30/673030/1/check/cross-neutron-py36/ae48f6e/ : SUCCESS in 24m 48s\n- cross-nova-py27 http://logs.openstack.org/30/673030/1/check/cross-nova-py27/b0d779e/ : SUCCESS in 17m 05s\n- cross-nova-py36 http://logs.openstack.org/30/673030/1/check/cross-nova-py36/a5c5f45/ : SUCCESS in 16m 18s\n- cross-nova-functional http://logs.openstack.org/30/673030/1/check/cross-nova-functional/2a56436/ : SUCCESS in 20m 39s\n- cross-osc-tox-docs http://logs.openstack.org/30/673030/1/check/cross-osc-tox-docs/cfa6703/html/ : SUCCESS in 3m 46s\n- cross-swift-py27 http://logs.openstack.org/30/673030/1/check/cross-swift-py27/d451ecb/ : SUCCESS in 7m 22s\n- cross-swift-py36 http://logs.openstack.org/30/673030/1/check/cross-swift-py36/3f16dec/ : SUCCESS in 8m 29s\n- requirements-integration http://logs.openstack.org/30/673030/1/check/requirements-integration/63b5e5c/ : SUCCESS in 48m 38s\n- neutron-grenade-multinode http://logs.openstack.org/30/673030/1/check/neutron-grenade-multinode/95439b8/ : SUCCESS in 1h 10m 17s\n- tempest-full http://logs.openstack.org/30/673030/1/check/tempest-full/ef99966/ : SUCCESS in 1h 31m 26s\n- neutron-grenade http://logs.openstack.org/30/673030/1/check/neutron-grenade/a73c5be/ : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":1},{"id":"877083adb8e82ad5c3adafdb397cfbfe300abfa6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-26 21:19:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6a0322681dde1020118ec8df054c288551dc3ded","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 21:20:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"894254a14193f39c6e0e042c7c6bdc2806d5d5aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 23:02:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/673030/1/check/openstack-tox-docs/0bc5a5b/html/ : SUCCESS in 4m 51s\n- openstack-tox-pypy http://logs.openstack.org/30/673030/1/check/openstack-tox-pypy/04770ca/ : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/30/673030/1/check/openstack-tox-pep8/448ede4/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/30/673030/1/check/openstack-tox-py27/0f03137/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/30/673030/1/check/openstack-tox-py36/593df08/ : SUCCESS in 5m 26s\n- openstack-tox-py37 http://logs.openstack.org/30/673030/1/check/openstack-tox-py37/4b7989c/ : SUCCESS in 6m 13s\n- openstack-tox-validate http://logs.openstack.org/30/673030/1/check/openstack-tox-validate/2622e06/ : SUCCESS in 4m 35s\n- requirements-tox-babel http://logs.openstack.org/30/673030/1/check/requirements-tox-babel/5494ee0/ : SUCCESS in 4m 34s\n- requirements-tox-py27-check-uc http://logs.openstack.org/30/673030/1/check/requirements-tox-py27-check-uc/8531c67/ : SUCCESS in 8m 54s\n- requirements-tox-py36-check-uc http://logs.openstack.org/30/673030/1/check/requirements-tox-py36-check-uc/f726eb3/ : SUCCESS in 16m 21s\n- cross-cinder-py27 http://logs.openstack.org/30/673030/1/check/cross-cinder-py27/71a8dee/ : SUCCESS in 10m 19s\n- cross-cinder-py36 http://logs.openstack.org/30/673030/1/check/cross-cinder-py36/fb8da49/ : SUCCESS in 9m 22s\n- cross-glance-py27 http://logs.openstack.org/30/673030/1/check/cross-glance-py27/4078f50/ : SUCCESS in 6m 04s\n- cross-glance-py36 http://logs.openstack.org/30/673030/1/check/cross-glance-py36/ddeb864/ : SUCCESS in 5m 51s\n- cross-horizon-py27 http://logs.openstack.org/30/673030/1/check/cross-horizon-py27/0ce92d6/ : SUCCESS in 13m 19s\n- cross-horizon-py36 http://logs.openstack.org/30/673030/1/check/cross-horizon-py36/1a35ec0/ : SUCCESS in 10m 37s\n- cross-horizon-npm http://logs.openstack.org/30/673030/1/check/cross-horizon-npm/6d7c96d/ : SUCCESS in 8m 53s\n- cross-keystone-py27 http://logs.openstack.org/30/673030/1/check/cross-keystone-py27/c51241c/ : SUCCESS in 32m 49s\n- cross-keystone-py36 http://logs.openstack.org/30/673030/1/check/cross-keystone-py36/ce3e62c/ : SUCCESS in 33m 00s\n- cross-neutron-py27 http://logs.openstack.org/30/673030/1/check/cross-neutron-py27/5386dac/ : SUCCESS in 30m 55s\n- cross-neutron-py36 http://logs.openstack.org/30/673030/1/check/cross-neutron-py36/239f411/ : SUCCESS in 24m 02s\n- cross-nova-py27 http://logs.openstack.org/30/673030/1/check/cross-nova-py27/5066677/ : SUCCESS in 25m 45s\n- cross-nova-py36 http://logs.openstack.org/30/673030/1/check/cross-nova-py36/6bb35c8/ : SUCCESS in 15m 34s\n- cross-nova-functional http://logs.openstack.org/30/673030/1/check/cross-nova-functional/c38fed4/ : SUCCESS in 23m 07s\n- cross-osc-tox-docs http://logs.openstack.org/30/673030/1/check/cross-osc-tox-docs/3798c00/html/ : SUCCESS in 5m 20s\n- cross-swift-py27 http://logs.openstack.org/30/673030/1/check/cross-swift-py27/a23c27b/ : SUCCESS in 10m 00s\n- cross-swift-py36 http://logs.openstack.org/30/673030/1/check/cross-swift-py36/cf2c3e2/ : SUCCESS in 8m 51s\n- requirements-integration http://logs.openstack.org/30/673030/1/check/requirements-integration/d001d88/ : SUCCESS in 1h 03m 38s\n- neutron-grenade-multinode http://logs.openstack.org/30/673030/1/check/neutron-grenade-multinode/fb32f87/ : SUCCESS in 1h 11m 25s\n- tempest-full http://logs.openstack.org/30/673030/1/check/tempest-full/80c1c51/ : SUCCESS in 1h 35m 55s\n- neutron-grenade http://logs.openstack.org/30/673030/1/check/neutron-grenade/9dbb7a1/ : SUCCESS in 51m 42s","accounts_in_message":[],"_revision_number":1},{"id":"cd41d2292bd7ec1fd256df4019fb7fcc939c7b4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 00:36:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/30/673030/1/gate/openstack-tox-docs/f0c45e9/html/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/30/673030/1/gate/openstack-tox-pep8/ffab486/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/30/673030/1/gate/openstack-tox-py27/be3b108/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/30/673030/1/gate/openstack-tox-py36/773128f/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/30/673030/1/gate/openstack-tox-py37/1076f11/ : SUCCESS in 5m 10s\n- openstack-tox-validate http://logs.openstack.org/30/673030/1/gate/openstack-tox-validate/04faf4f/ : SUCCESS in 4m 20s\n- requirements-tox-babel http://logs.openstack.org/30/673030/1/gate/requirements-tox-babel/bfce179/ : SUCCESS in 4m 19s\n- requirements-tox-py27-check-uc http://logs.openstack.org/30/673030/1/gate/requirements-tox-py27-check-uc/fcfc76f/ : SUCCESS in 7m 13s\n- requirements-tox-py36-check-uc http://logs.openstack.org/30/673030/1/gate/requirements-tox-py36-check-uc/e2dad73/ : SUCCESS in 7m 18s\n- cross-cinder-py27 http://logs.openstack.org/30/673030/1/gate/cross-cinder-py27/e188ff8/ : SUCCESS in 9m 02s\n- cross-cinder-py36 http://logs.openstack.org/30/673030/1/gate/cross-cinder-py36/d197987/ : SUCCESS in 9m 05s\n- cross-glance-py27 http://logs.openstack.org/30/673030/1/gate/cross-glance-py27/449250d/ : SUCCESS in 5m 41s\n- cross-glance-py36 http://logs.openstack.org/30/673030/1/gate/cross-glance-py36/22cbdbb/ : SUCCESS in 5m 27s\n- cross-horizon-py27 http://logs.openstack.org/30/673030/1/gate/cross-horizon-py27/6ace1ca/ : SUCCESS in 11m 20s\n- cross-horizon-py36 http://logs.openstack.org/30/673030/1/gate/cross-horizon-py36/c7fb89b/ : SUCCESS in 9m 31s\n- cross-horizon-npm http://logs.openstack.org/30/673030/1/gate/cross-horizon-npm/3bdec1a/ : SUCCESS in 7m 40s\n- cross-keystone-py27 http://logs.openstack.org/30/673030/1/gate/cross-keystone-py27/136e7b5/ : SUCCESS in 30m 06s\n- cross-keystone-py36 http://logs.openstack.org/30/673030/1/gate/cross-keystone-py36/de6fb64/ : SUCCESS in 39m 13s\n- cross-neutron-py27 http://logs.openstack.org/30/673030/1/gate/cross-neutron-py27/65a5af1/ : SUCCESS in 23m 38s\n- cross-neutron-py36 http://logs.openstack.org/30/673030/1/gate/cross-neutron-py36/bda786c/ : FAILURE in 24m 19s\n- cross-nova-py27 http://logs.openstack.org/30/673030/1/gate/cross-nova-py27/bc3b67e/ : SUCCESS in 17m 07s\n- cross-nova-py36 http://logs.openstack.org/30/673030/1/gate/cross-nova-py36/788c1bb/ : SUCCESS in 15m 53s\n- cross-nova-functional http://logs.openstack.org/30/673030/1/gate/cross-nova-functional/6f22dc1/ : TIMED_OUT in 41m 01s\n- cross-osc-tox-docs http://logs.openstack.org/30/673030/1/gate/cross-osc-tox-docs/4eb4390/html/ : SUCCESS in 4m 17s\n- cross-swift-py27 http://logs.openstack.org/30/673030/1/gate/cross-swift-py27/c84e404/ : SUCCESS in 7m 27s\n- cross-swift-py36 http://logs.openstack.org/30/673030/1/gate/cross-swift-py36/53372c2/ : SUCCESS in 9m 03s\n- requirements-integration http://logs.openstack.org/30/673030/1/gate/requirements-integration/731b512/ : SUCCESS in 55m 05s\n- neutron-grenade-multinode http://logs.openstack.org/30/673030/1/gate/neutron-grenade-multinode/9e3952c/ : SUCCESS in 1h 22m 20s\n- tempest-full http://logs.openstack.org/30/673030/1/gate/tempest-full/6ed9980/ : SUCCESS in 1h 40m 08s\n- neutron-grenade http://logs.openstack.org/30/673030/1/gate/neutron-grenade/25793e4/ : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3adb531d4eb4c9e4ab39baa54dddffcccae6053c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-27 00:44:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6dd27c465ff847a98aeefd0e70ae2acc31444fdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 03:01:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/673030/1/check/openstack-tox-docs/8433c79/html/ : SUCCESS in 3m 41s\n- openstack-tox-pypy http://logs.openstack.org/30/673030/1/check/openstack-tox-pypy/a147933/ : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/30/673030/1/check/openstack-tox-pep8/84ce835/ : SUCCESS in 3m 52s\n- openstack-tox-py27 http://logs.openstack.org/30/673030/1/check/openstack-tox-py27/334f97f/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/30/673030/1/check/openstack-tox-py36/65418cc/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/30/673030/1/check/openstack-tox-py37/a947276/ : SUCCESS in 4m 34s\n- openstack-tox-validate http://logs.openstack.org/30/673030/1/check/openstack-tox-validate/8f67bd9/ : SUCCESS in 4m 18s\n- requirements-tox-babel http://logs.openstack.org/30/673030/1/check/requirements-tox-babel/c7cb93e/ : SUCCESS in 2m 54s\n- requirements-tox-py27-check-uc http://logs.openstack.org/30/673030/1/check/requirements-tox-py27-check-uc/adc9e9d/ : SUCCESS in 6m 53s\n- requirements-tox-py36-check-uc http://logs.openstack.org/30/673030/1/check/requirements-tox-py36-check-uc/ab68ac0/ : SUCCESS in 6m 31s\n- cross-cinder-py27 http://logs.openstack.org/30/673030/1/check/cross-cinder-py27/18b154e/ : SUCCESS in 10m 38s\n- cross-cinder-py36 http://logs.openstack.org/30/673030/1/check/cross-cinder-py36/baaa2c1/ : SUCCESS in 7m 36s\n- cross-glance-py27 http://logs.openstack.org/30/673030/1/check/cross-glance-py27/70212b9/ : SUCCESS in 4m 24s\n- cross-glance-py36 http://logs.openstack.org/30/673030/1/check/cross-glance-py36/339e4b8/ : SUCCESS in 4m 22s\n- cross-horizon-py27 http://logs.openstack.org/30/673030/1/check/cross-horizon-py27/33bed9e/ : SUCCESS in 10m 13s\n- cross-horizon-py36 http://logs.openstack.org/30/673030/1/check/cross-horizon-py36/d5d7ad2/ : SUCCESS in 7m 16s\n- cross-horizon-npm http://logs.openstack.org/30/673030/1/check/cross-horizon-npm/d9c398b/ : SUCCESS in 6m 00s\n- cross-keystone-py27 http://logs.openstack.org/30/673030/1/check/cross-keystone-py27/a2142a6/ : SUCCESS in 33m 52s\n- cross-keystone-py36 http://logs.openstack.org/30/673030/1/check/cross-keystone-py36/9af67a0/ : SUCCESS in 29m 11s\n- cross-neutron-py27 http://logs.openstack.org/30/673030/1/check/cross-neutron-py27/b0e0034/ : SUCCESS in 29m 46s\n- cross-neutron-py36 http://logs.openstack.org/30/673030/1/check/cross-neutron-py36/8400e6f/ : SUCCESS in 29m 41s\n- cross-nova-py27 http://logs.openstack.org/30/673030/1/check/cross-nova-py27/6ad376a/ : SUCCESS in 19m 31s\n- cross-nova-py36 http://logs.openstack.org/30/673030/1/check/cross-nova-py36/23f5a4b/ : SUCCESS in 17m 54s\n- cross-nova-functional http://logs.openstack.org/30/673030/1/check/cross-nova-functional/e6e7517/ : SUCCESS in 26m 11s\n- cross-osc-tox-docs http://logs.openstack.org/30/673030/1/check/cross-osc-tox-docs/f067c5f/html/ : SUCCESS in 4m 05s\n- cross-swift-py27 http://logs.openstack.org/30/673030/1/check/cross-swift-py27/31b81a7/ : SUCCESS in 8m 20s\n- cross-swift-py36 http://logs.openstack.org/30/673030/1/check/cross-swift-py36/0cc0f6e/ : SUCCESS in 8m 30s\n- requirements-integration http://logs.openstack.org/30/673030/1/check/requirements-integration/46e2409/ : SUCCESS in 54m 34s\n- neutron-grenade-multinode http://logs.openstack.org/30/673030/1/check/neutron-grenade-multinode/7ff034c/ : SUCCESS in 1h 07m 02s\n- tempest-full http://logs.openstack.org/30/673030/1/check/tempest-full/d3b41a5/ : SUCCESS in 1h 44m 01s\n- neutron-grenade http://logs.openstack.org/30/673030/1/check/neutron-grenade/a434831/ : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2872f33bc148eb862543b5fc84d7b076c65e061d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 03:01:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ce00b63fa3f3e6df68dd784b9ac49e0f0854cba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 04:56:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/30/673030/1/gate/openstack-tox-docs/00f9e92/html/ : SUCCESS in 5m 30s\n- openstack-tox-pep8 http://logs.openstack.org/30/673030/1/gate/openstack-tox-pep8/f5af7c3/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/30/673030/1/gate/openstack-tox-py27/5c0abb2/ : SUCCESS in 3m 06s\n- openstack-tox-py36 http://logs.openstack.org/30/673030/1/gate/openstack-tox-py36/e506f54/ : SUCCESS in 3m 15s\n- openstack-tox-py37 http://logs.openstack.org/30/673030/1/gate/openstack-tox-py37/7958e7d/ : SUCCESS in 5m 18s\n- openstack-tox-validate http://logs.openstack.org/30/673030/1/gate/openstack-tox-validate/891645a/ : SUCCESS in 2m 59s\n- requirements-tox-babel http://logs.openstack.org/30/673030/1/gate/requirements-tox-babel/13494a5/ : SUCCESS in 4m 23s\n- requirements-tox-py27-check-uc http://logs.openstack.org/30/673030/1/gate/requirements-tox-py27-check-uc/f4262a4/ : SUCCESS in 6m 22s\n- requirements-tox-py36-check-uc http://logs.openstack.org/30/673030/1/gate/requirements-tox-py36-check-uc/0237a7a/ : SUCCESS in 14m 29s\n- cross-cinder-py27 http://logs.openstack.org/30/673030/1/gate/cross-cinder-py27/7d90bb0/ : SUCCESS in 12m 51s\n- cross-cinder-py36 http://logs.openstack.org/30/673030/1/gate/cross-cinder-py36/191ac43/ : SUCCESS in 7m 59s\n- cross-glance-py27 http://logs.openstack.org/30/673030/1/gate/cross-glance-py27/aa4ec1e/ : SUCCESS in 5m 44s\n- cross-glance-py36 http://logs.openstack.org/30/673030/1/gate/cross-glance-py36/4259933/ : SUCCESS in 4m 05s\n- cross-horizon-py27 http://logs.openstack.org/30/673030/1/gate/cross-horizon-py27/6b50fbe/ : SUCCESS in 12m 18s\n- cross-horizon-py36 http://logs.openstack.org/30/673030/1/gate/cross-horizon-py36/8502eb8/ : SUCCESS in 10m 26s\n- cross-horizon-npm http://logs.openstack.org/30/673030/1/gate/cross-horizon-npm/28d4325/ : SUCCESS in 7m 55s\n- cross-keystone-py27 http://logs.openstack.org/30/673030/1/gate/cross-keystone-py27/674d80e/ : SUCCESS in 27m 18s\n- cross-keystone-py36 http://logs.openstack.org/30/673030/1/gate/cross-keystone-py36/0d14c96/ : SUCCESS in 24m 26s\n- cross-neutron-py27 http://logs.openstack.org/30/673030/1/gate/cross-neutron-py27/2fa24d4/ : SUCCESS in 24m 53s\n- cross-neutron-py36 http://logs.openstack.org/30/673030/1/gate/cross-neutron-py36/7592cfd/ : SUCCESS in 17m 36s\n- cross-nova-py27 http://logs.openstack.org/30/673030/1/gate/cross-nova-py27/e4d9192/ : SUCCESS in 16m 02s\n- cross-nova-py36 http://logs.openstack.org/30/673030/1/gate/cross-nova-py36/2322753/ : SUCCESS in 16m 06s\n- cross-nova-functional http://logs.openstack.org/30/673030/1/gate/cross-nova-functional/6d26a5d/ : SUCCESS in 21m 17s\n- cross-osc-tox-docs http://logs.openstack.org/30/673030/1/gate/cross-osc-tox-docs/3e62d81/html/ : SUCCESS in 4m 30s\n- cross-swift-py27 http://logs.openstack.org/30/673030/1/gate/cross-swift-py27/fbe17ad/ : SUCCESS in 8m 20s\n- cross-swift-py36 http://logs.openstack.org/30/673030/1/gate/cross-swift-py36/b41c34e/ : SUCCESS in 10m 36s\n- requirements-integration http://logs.openstack.org/30/673030/1/gate/requirements-integration/f3d1d14/ : SUCCESS in 57m 49s\n- neutron-grenade-multinode http://logs.openstack.org/30/673030/1/gate/neutron-grenade-multinode/be46076/ : SUCCESS in 1h 05m 16s\n- tempest-full http://logs.openstack.org/30/673030/1/gate/tempest-full/169e21b/ : SUCCESS in 1h 29m 58s\n- neutron-grenade http://logs.openstack.org/30/673030/1/gate/neutron-grenade/9f34ff7/ : SUCCESS in 55m 36s","accounts_in_message":[],"_revision_number":1},{"id":"be5effeea73aef7d3eafa0a95e8f069e6138e2c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 04:56:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47c07643e3f11f1d1df0911f735cd27bbc390737","revisions":{"47c07643e3f11f1d1df0911f735cd27bbc390737":{"kind":"REWORK","_number":1,"created":"2019-07-26 18:20:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/673030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/673030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/673030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/673030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/673030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/673030/1"}}},"commit":{"parents":[{"commit":"71f702b416779d4cb59b78f4d80a5cd69fb1fe51","subject":"update constraint for glance_store to new release 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71f702b416779d4cb59b78f4d80a5cd69fb1fe51"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-26 18:20:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-26 18:20:38.000000000","tz":0},"subject":"update constraint for os-net-config to new release 11.1.0","message":"update constraint for os-net-config to new release 11.1.0\n\nChange-Id: I22a9cbb40a8c713102266275c01bbb543404489c\nmeta:version: 11.1.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Change-Id: I7d5887fdac03ea64be8c1790646431d4ed79a586\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47c07643e3f11f1d1df0911f735cd27bbc390737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47c07643e3f11f1d1df0911f735cd27bbc390737"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
