)]}'
{"id":"openstack%2Frequirements~673035","triplet_id":"openstack%2Frequirements~master~I12c13fa9eb3a369833bc593edc465ee5675bfdf6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I12c13fa9eb3a369833bc593edc465ee5675bfdf6","subject":"update constraint for paunch to new release 5.1.0","status":"MERGED","created":"2019-07-26 18:35:40.000000000","updated":"2019-07-27 02:11:53.000000000","submitted":"2019-07-27 02:11:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673035-1564193513057-d6eb9d84","meta_rev_id":"9cde9b359e69f3e7957769080b0ff82f84c61383","_number":673035,"virtual_id_number":673035,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-27 02:11:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-26 21:20:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-26 21:20:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 21:20:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-27 02:11:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed0f15315e52ad014a1c75276b7e3480545c3aaf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-26 18:35:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fabc053dd40dff62c1696114760c52750da60a5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 20:03:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/35/673035/1/check/openstack-tox-docs/75ce8f0/html/ : SUCCESS in 4m 21s\n- openstack-tox-pypy http://logs.openstack.org/35/673035/1/check/openstack-tox-pypy/cabbbbd/ : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/35/673035/1/check/openstack-tox-pep8/bd12f26/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/35/673035/1/check/openstack-tox-py27/5e34fd4/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/35/673035/1/check/openstack-tox-py36/da39ea1/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/35/673035/1/check/openstack-tox-py37/b2bd2fa/ : SUCCESS in 3m 44s\n- openstack-tox-validate http://logs.openstack.org/35/673035/1/check/openstack-tox-validate/d12c6fd/ : SUCCESS in 4m 24s\n- requirements-tox-babel http://logs.openstack.org/35/673035/1/check/requirements-tox-babel/52a0ca1/ : SUCCESS in 4m 26s\n- requirements-tox-py27-check-uc http://logs.openstack.org/35/673035/1/check/requirements-tox-py27-check-uc/ed58564/ : SUCCESS in 7m 47s\n- requirements-tox-py36-check-uc http://logs.openstack.org/35/673035/1/check/requirements-tox-py36-check-uc/ade6cb5/ : SUCCESS in 7m 02s\n- cross-cinder-py27 http://logs.openstack.org/35/673035/1/check/cross-cinder-py27/560de41/ : SUCCESS in 9m 16s\n- cross-cinder-py36 http://logs.openstack.org/35/673035/1/check/cross-cinder-py36/377ad69/ : SUCCESS in 10m 16s\n- cross-glance-py27 http://logs.openstack.org/35/673035/1/check/cross-glance-py27/16088ae/ : SUCCESS in 5m 44s\n- cross-glance-py36 http://logs.openstack.org/35/673035/1/check/cross-glance-py36/066e3f7/ : SUCCESS in 4m 55s\n- cross-horizon-py27 http://logs.openstack.org/35/673035/1/check/cross-horizon-py27/286feb6/ : SUCCESS in 12m 19s\n- cross-horizon-py36 http://logs.openstack.org/35/673035/1/check/cross-horizon-py36/7c48e6e/ : SUCCESS in 10m 22s\n- cross-horizon-npm http://logs.openstack.org/35/673035/1/check/cross-horizon-npm/026751c/ : SUCCESS in 7m 56s\n- cross-keystone-py27 http://logs.openstack.org/35/673035/1/check/cross-keystone-py27/57ae86e/ : SUCCESS in 31m 08s\n- cross-keystone-py36 http://logs.openstack.org/35/673035/1/check/cross-keystone-py36/554f51e/ : SUCCESS in 32m 45s\n- cross-neutron-py27 http://logs.openstack.org/35/673035/1/check/cross-neutron-py27/1c26b9d/ : SUCCESS in 31m 35s\n- cross-neutron-py36 http://logs.openstack.org/35/673035/1/check/cross-neutron-py36/63fec8a/ : SUCCESS in 23m 01s\n- cross-nova-py27 http://logs.openstack.org/35/673035/1/check/cross-nova-py27/146b890/ : SUCCESS in 19m 08s\n- cross-nova-py36 http://logs.openstack.org/35/673035/1/check/cross-nova-py36/3753661/ : SUCCESS in 17m 02s\n- cross-nova-functional http://logs.openstack.org/35/673035/1/check/cross-nova-functional/7ec9e15/ : SUCCESS in 25m 11s\n- cross-osc-tox-docs http://logs.openstack.org/35/673035/1/check/cross-osc-tox-docs/8550bd6/html/ : SUCCESS in 4m 09s\n- cross-swift-py27 http://logs.openstack.org/35/673035/1/check/cross-swift-py27/34b7a13/ : FAILURE in 11m 11s\n- cross-swift-py36 http://logs.openstack.org/35/673035/1/check/cross-swift-py36/10de1df/ : SUCCESS in 7m 11s\n- requirements-integration http://logs.openstack.org/35/673035/1/check/requirements-integration/863cf8e/ : SUCCESS in 58m 15s\n- neutron-grenade-multinode http://logs.openstack.org/35/673035/1/check/neutron-grenade-multinode/23dd774/ : SUCCESS in 1h 06m 49s\n- tempest-full http://logs.openstack.org/35/673035/1/check/tempest-full/e82ac8a/ : SUCCESS in 1h 22m 37s\n- neutron-grenade http://logs.openstack.org/35/673035/1/check/neutron-grenade/f77a24f/ : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f1281e674e6a85dcd66159105e0aab2915c4eb01","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-26 21:20:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e8200728bcfc5de65d7f7b7b8cdc1582332a8ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 23:05:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/35/673035/1/check/openstack-tox-docs/63570e7/html/ : SUCCESS in 4m 55s\n- openstack-tox-pypy http://logs.openstack.org/35/673035/1/check/openstack-tox-pypy/72e5500/ : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/35/673035/1/check/openstack-tox-pep8/2a4f771/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/35/673035/1/check/openstack-tox-py27/3c2c61b/ : SUCCESS in 6m 00s\n- openstack-tox-py36 http://logs.openstack.org/35/673035/1/check/openstack-tox-py36/742cb9d/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/35/673035/1/check/openstack-tox-py37/6b9df39/ : SUCCESS in 5m 42s\n- openstack-tox-validate http://logs.openstack.org/35/673035/1/check/openstack-tox-validate/da2db43/ : SUCCESS in 6m 01s\n- requirements-tox-babel http://logs.openstack.org/35/673035/1/check/requirements-tox-babel/3fa9f54/ : SUCCESS in 5m 49s\n- requirements-tox-py27-check-uc http://logs.openstack.org/35/673035/1/check/requirements-tox-py27-check-uc/3e1dc98/ : SUCCESS in 7m 56s\n- requirements-tox-py36-check-uc http://logs.openstack.org/35/673035/1/check/requirements-tox-py36-check-uc/a77fc5c/ : SUCCESS in 17m 05s\n- cross-cinder-py27 http://logs.openstack.org/35/673035/1/check/cross-cinder-py27/c9516e9/ : SUCCESS in 10m 01s\n- cross-cinder-py36 http://logs.openstack.org/35/673035/1/check/cross-cinder-py36/d0a35ad/ : SUCCESS in 9m 20s\n- cross-glance-py27 http://logs.openstack.org/35/673035/1/check/cross-glance-py27/b812570/ : SUCCESS in 7m 01s\n- cross-glance-py36 http://logs.openstack.org/35/673035/1/check/cross-glance-py36/a58226f/ : SUCCESS in 5m 08s\n- cross-horizon-py27 http://logs.openstack.org/35/673035/1/check/cross-horizon-py27/bcd7e05/ : SUCCESS in 12m 05s\n- cross-horizon-py36 http://logs.openstack.org/35/673035/1/check/cross-horizon-py36/7e4ef8e/ : SUCCESS in 11m 24s\n- cross-horizon-npm http://logs.openstack.org/35/673035/1/check/cross-horizon-npm/a81cd83/ : SUCCESS in 6m 39s\n- cross-keystone-py27 http://logs.openstack.org/35/673035/1/check/cross-keystone-py27/aa07402/ : SUCCESS in 27m 43s\n- cross-keystone-py36 http://logs.openstack.org/35/673035/1/check/cross-keystone-py36/04bbf87/ : SUCCESS in 32m 43s\n- cross-neutron-py27 http://logs.openstack.org/35/673035/1/check/cross-neutron-py27/813bfe9/ : SUCCESS in 21m 33s\n- cross-neutron-py36 http://logs.openstack.org/35/673035/1/check/cross-neutron-py36/55ce776/ : SUCCESS in 23m 04s\n- cross-nova-py27 http://logs.openstack.org/35/673035/1/check/cross-nova-py27/60366d1/ : SUCCESS in 14m 26s\n- cross-nova-py36 http://logs.openstack.org/35/673035/1/check/cross-nova-py36/8c4c10b/ : SUCCESS in 15m 52s\n- cross-nova-functional http://logs.openstack.org/35/673035/1/check/cross-nova-functional/a9d93e7/ : SUCCESS in 22m 50s\n- cross-osc-tox-docs http://logs.openstack.org/35/673035/1/check/cross-osc-tox-docs/2522f32/html/ : SUCCESS in 7m 18s\n- cross-swift-py27 http://logs.openstack.org/35/673035/1/check/cross-swift-py27/aacbb4a/ : SUCCESS in 9m 47s\n- cross-swift-py36 http://logs.openstack.org/35/673035/1/check/cross-swift-py36/a128340/ : SUCCESS in 9m 26s\n- requirements-integration http://logs.openstack.org/35/673035/1/check/requirements-integration/b27027a/ : SUCCESS in 57m 52s\n- neutron-grenade-multinode http://logs.openstack.org/35/673035/1/check/neutron-grenade-multinode/33f4a7a/ : SUCCESS in 1h 15m 21s\n- tempest-full http://logs.openstack.org/35/673035/1/check/tempest-full/6071fe9/ : SUCCESS in 1h 38m 16s\n- neutron-grenade http://logs.openstack.org/35/673035/1/check/neutron-grenade/9fdaba0/ : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0d51ddb012f4dca0642db5e22b9fb81a10285c75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 23:06:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"706a0693289a47abc788ee5cca5ca2001b0a7dc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 02:11:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/35/673035/1/gate/openstack-tox-docs/12a62a6/html/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/35/673035/1/gate/openstack-tox-pep8/2fc10ab/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/35/673035/1/gate/openstack-tox-py27/5f2dd60/ : SUCCESS in 4m 52s\n- openstack-tox-py36 http://logs.openstack.org/35/673035/1/gate/openstack-tox-py36/26842ea/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/35/673035/1/gate/openstack-tox-py37/d64e9a9/ : SUCCESS in 5m 20s\n- openstack-tox-validate http://logs.openstack.org/35/673035/1/gate/openstack-tox-validate/315542a/ : SUCCESS in 3m 23s\n- requirements-tox-babel http://logs.openstack.org/35/673035/1/gate/requirements-tox-babel/ce37719/ : SUCCESS in 3m 18s\n- requirements-tox-py27-check-uc http://logs.openstack.org/35/673035/1/gate/requirements-tox-py27-check-uc/6b76872/ : SUCCESS in 6m 44s\n- requirements-tox-py36-check-uc http://logs.openstack.org/35/673035/1/gate/requirements-tox-py36-check-uc/4edaf32/ : SUCCESS in 6m 41s\n- cross-cinder-py27 http://logs.openstack.org/35/673035/1/gate/cross-cinder-py27/9d99d4c/ : SUCCESS in 10m 38s\n- cross-cinder-py36 http://logs.openstack.org/35/673035/1/gate/cross-cinder-py36/581b912/ : SUCCESS in 9m 46s\n- cross-glance-py27 http://logs.openstack.org/35/673035/1/gate/cross-glance-py27/bfbae02/ : SUCCESS in 4m 45s\n- cross-glance-py36 http://logs.openstack.org/35/673035/1/gate/cross-glance-py36/dcd27cd/ : SUCCESS in 5m 11s\n- cross-horizon-py27 http://logs.openstack.org/35/673035/1/gate/cross-horizon-py27/c7229f5/ : SUCCESS in 10m 59s\n- cross-horizon-py36 http://logs.openstack.org/35/673035/1/gate/cross-horizon-py36/74f3b55/ : SUCCESS in 9m 19s\n- cross-horizon-npm http://logs.openstack.org/35/673035/1/gate/cross-horizon-npm/2f0cb39/ : SUCCESS in 7m 31s\n- cross-keystone-py27 http://logs.openstack.org/35/673035/1/gate/cross-keystone-py27/533994c/ : SUCCESS in 25m 32s\n- cross-keystone-py36 http://logs.openstack.org/35/673035/1/gate/cross-keystone-py36/155c8de/ : SUCCESS in 26m 49s\n- cross-neutron-py27 http://logs.openstack.org/35/673035/1/gate/cross-neutron-py27/6b8ecd7/ : SUCCESS in 28m 40s\n- cross-neutron-py36 http://logs.openstack.org/35/673035/1/gate/cross-neutron-py36/04ebb4a/ : SUCCESS in 19m 20s\n- cross-nova-py27 http://logs.openstack.org/35/673035/1/gate/cross-nova-py27/6aaf7b0/ : SUCCESS in 18m 44s\n- cross-nova-py36 http://logs.openstack.org/35/673035/1/gate/cross-nova-py36/ea79778/ : SUCCESS in 15m 44s\n- cross-nova-functional http://logs.openstack.org/35/673035/1/gate/cross-nova-functional/734746d/ : SUCCESS in 22m 52s\n- cross-osc-tox-docs http://logs.openstack.org/35/673035/1/gate/cross-osc-tox-docs/076c415/html/ : SUCCESS in 5m 17s\n- cross-swift-py27 http://logs.openstack.org/35/673035/1/gate/cross-swift-py27/a6f4847/ : SUCCESS in 10m 49s\n- cross-swift-py36 http://logs.openstack.org/35/673035/1/gate/cross-swift-py36/0d362f6/ : SUCCESS in 9m 04s\n- requirements-integration http://logs.openstack.org/35/673035/1/gate/requirements-integration/ca9aab9/ : SUCCESS in 55m 33s\n- neutron-grenade-multinode http://logs.openstack.org/35/673035/1/gate/neutron-grenade-multinode/bc5c16c/ : SUCCESS in 1h 18m 35s\n- tempest-full http://logs.openstack.org/35/673035/1/gate/tempest-full/25dc63b/ : SUCCESS in 1h 35m 55s\n- neutron-grenade http://logs.openstack.org/35/673035/1/gate/neutron-grenade/2481486/ : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":1},{"id":"41c6828b01974fc1cdc59dbe1857348c8f19ff5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 02:11:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3976ce1f32303059e7f59a1ad632bf3f8615c091","revisions":{"3976ce1f32303059e7f59a1ad632bf3f8615c091":{"kind":"REWORK","_number":1,"created":"2019-07-26 18:35:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/673035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/673035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/673035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/673035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/673035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/673035/1"}}},"commit":{"parents":[{"commit":"71f702b416779d4cb59b78f4d80a5cd69fb1fe51","subject":"update constraint for glance_store to new release 1.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/71f702b416779d4cb59b78f4d80a5cd69fb1fe51"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-26 18:35:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-26 18:35:35.000000000","tz":0},"subject":"update constraint for paunch to new release 5.1.0","message":"update constraint for paunch to new release 5.1.0\n\nChange-Id: I12c13fa9eb3a369833bc593edc465ee5675bfdf6\nmeta:version: 5.1.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Commit: Alex Schultz \u003caschultz@redhat.com\u003e\nmeta:release:Change-Id: I7d5887fdac03ea64be8c1790646431d4ed79a586\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Emilien Macchi \u003cemilien@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3976ce1f32303059e7f59a1ad632bf3f8615c091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3976ce1f32303059e7f59a1ad632bf3f8615c091"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
