)]}'
{"id":"openstack%2Frequirements~673160","triplet_id":"openstack%2Frequirements~master~I13f64539f51f255f54ecd17691107d8b592ddbe7","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I13f64539f51f255f54ecd17691107d8b592ddbe7","subject":"Updated from generate-constraints","status":"MERGED","created":"2019-07-28 06:20:32.000000000","updated":"2019-08-01 05:50:52.000000000","submitted":"2019-08-01 05:50:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673160-1564638652030-b796e31c","meta_rev_id":"1b6e96ed349a526d5e35c0e363d5f3573c63bce0","_number":673160,"virtual_id_number":673160,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-01 05:50:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-01 01:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-01 01:21:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 01:21:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-01 05:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ab015ea9debd61e4295cb00104ead1a6c8d7cf6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-28 06:20:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"359c1a4bfacfc20950bc6777dc451fcf620c2ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-28 07:58:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/60/673160/1/check/openstack-tox-docs/72d914e/ : POST_FAILURE in 2m 33s\n- openstack-tox-pypy http://logs.openstack.org/60/673160/1/check/openstack-tox-pypy/ca7e508/ : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/673160/1/check/openstack-tox-pep8/fc39c24/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/60/673160/1/check/openstack-tox-py27/9118929/ : SUCCESS in 5m 43s\n- openstack-tox-py36 http://logs.openstack.org/60/673160/1/check/openstack-tox-py36/85d98df/ : SUCCESS in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/60/673160/1/check/openstack-tox-py37/f0b6dcc/ : SUCCESS in 4m 07s\n- openstack-tox-validate http://logs.openstack.org/60/673160/1/check/openstack-tox-validate/be06208/ : SUCCESS in 3m 38s\n- requirements-tox-babel http://logs.openstack.org/60/673160/1/check/requirements-tox-babel/57b0f56/ : SUCCESS in 5m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/60/673160/1/check/requirements-tox-py27-check-uc/6d2b127/ : FAILURE in 6m 35s\n- requirements-tox-py36-check-uc http://logs.openstack.org/60/673160/1/check/requirements-tox-py36-check-uc/f69e3ab/ : FAILURE in 4m 52s\n- cross-cinder-py27 http://logs.openstack.org/60/673160/1/check/cross-cinder-py27/c7b4f45/ : SUCCESS in 8m 57s\n- cross-cinder-py36 http://logs.openstack.org/60/673160/1/check/cross-cinder-py36/76b674b/ : FAILURE in 5m 17s\n- cross-glance-py27 http://logs.openstack.org/60/673160/1/check/cross-glance-py27/57ab963/ : SUCCESS in 4m 48s\n- cross-glance-py36 http://logs.openstack.org/60/673160/1/check/cross-glance-py36/e5718f4/ : FAILURE in 4m 51s\n- cross-horizon-py27 http://logs.openstack.org/60/673160/1/check/cross-horizon-py27/90b3c74/ : SUCCESS in 12m 04s\n- cross-horizon-py36 http://logs.openstack.org/60/673160/1/check/cross-horizon-py36/a92ff84/ : FAILURE in 2m 26s\n- cross-horizon-npm http://logs.openstack.org/60/673160/1/check/cross-horizon-npm/77ca4c1/ : SUCCESS in 5m 45s\n- cross-keystone-py27 http://logs.openstack.org/60/673160/1/check/cross-keystone-py27/29bcef1/ : SUCCESS in 26m 46s\n- cross-keystone-py36 http://logs.openstack.org/60/673160/1/check/cross-keystone-py36/73ee313/ : SUCCESS in 29m 53s\n- cross-neutron-py27 http://logs.openstack.org/60/673160/1/check/cross-neutron-py27/e99e421/ : SUCCESS in 25m 34s\n- cross-neutron-py36 http://logs.openstack.org/60/673160/1/check/cross-neutron-py36/0d60465/ : SUCCESS in 22m 09s\n- cross-nova-py27 http://logs.openstack.org/60/673160/1/check/cross-nova-py27/441200c/ : SUCCESS in 16m 48s\n- cross-nova-py36 http://logs.openstack.org/60/673160/1/check/cross-nova-py36/b2d8745/ : FAILURE in 19m 30s\n- cross-nova-functional http://logs.openstack.org/60/673160/1/check/cross-nova-functional/8dcc6d9/ : SUCCESS in 23m 30s\n- cross-osc-tox-docs http://logs.openstack.org/60/673160/1/check/cross-osc-tox-docs/8696c46/ : POST_FAILURE in 2m 34s\n- cross-swift-py27 http://logs.openstack.org/60/673160/1/check/cross-swift-py27/6658753/ : SUCCESS in 7m 59s\n- cross-swift-py36 http://logs.openstack.org/60/673160/1/check/cross-swift-py36/918370c/ : FAILURE in 4m 26s\n- requirements-integration http://logs.openstack.org/60/673160/1/check/requirements-integration/ff5a233/ : SUCCESS in 56m 22s\n- neutron-grenade-multinode http://logs.openstack.org/60/673160/1/check/neutron-grenade-multinode/ed178e6/ : FAILURE in 52m 59s\n- tempest-full http://logs.openstack.org/60/673160/1/check/tempest-full/24f9311/ : FAILURE in 1h 27m 52s\n- neutron-grenade http://logs.openstack.org/60/673160/1/check/neutron-grenade/8cdfc38/ : SUCCESS in 1h 03m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c6478b7b1fd004c66d2a16c36cd0bfa66d577096","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-30 06:19:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c12ad3df2b7dd15239dc80875d9138832a69f4f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 09:27:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/60/673160/2/check/openstack-tox-docs/7ac12f5/ : POST_FAILURE in 4m 51s\n- openstack-tox-pypy http://logs.openstack.org/60/673160/2/check/openstack-tox-pypy/6d22ded/ : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/673160/2/check/openstack-tox-pep8/c21c25f/ : SUCCESS in 3m 58s\n- openstack-tox-py27 http://logs.openstack.org/60/673160/2/check/openstack-tox-py27/ab669a6/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/60/673160/2/check/openstack-tox-py36/139acd0/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/60/673160/2/check/openstack-tox-py37/b18d61d/ : SUCCESS in 4m 54s\n- openstack-tox-validate http://logs.openstack.org/60/673160/2/check/openstack-tox-validate/707df18/ : SUCCESS in 4m 18s\n- requirements-tox-babel http://logs.openstack.org/60/673160/2/check/requirements-tox-babel/323dc12/ : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/60/673160/2/check/requirements-tox-py27-check-uc/de1f951/ : FAILURE in 7m 28s\n- requirements-tox-py36-check-uc http://logs.openstack.org/60/673160/2/check/requirements-tox-py36-check-uc/5417d0d/ : FAILURE in 3m 32s\n- cross-cinder-py27 http://logs.openstack.org/60/673160/2/check/cross-cinder-py27/795775c/ : SUCCESS in 10m 33s\n- cross-cinder-py36 http://logs.openstack.org/60/673160/2/check/cross-cinder-py36/edccd8f/ : FAILURE in 4m 30s\n- cross-glance-py27 http://logs.openstack.org/60/673160/2/check/cross-glance-py27/95bde3a/ : SUCCESS in 5m 35s\n- cross-glance-py36 http://logs.openstack.org/60/673160/2/check/cross-glance-py36/cc85ff5/ : FAILURE in 4m 20s\n- cross-horizon-py27 http://logs.openstack.org/60/673160/2/check/cross-horizon-py27/ae920c3/ : SUCCESS in 11m 02s\n- cross-horizon-py36 http://logs.openstack.org/60/673160/2/check/cross-horizon-py36/5df4fb7/ : FAILURE in 3m 37s\n- cross-horizon-npm http://logs.openstack.org/60/673160/2/check/cross-horizon-npm/fccff8a/ : SUCCESS in 6m 15s\n- cross-keystone-py27 http://logs.openstack.org/60/673160/2/check/cross-keystone-py27/c302d4f/ : SUCCESS in 27m 35s\n- cross-keystone-py36 http://logs.openstack.org/60/673160/2/check/cross-keystone-py36/480281d/ : SUCCESS in 30m 39s\n- cross-neutron-py27 http://logs.openstack.org/60/673160/2/check/cross-neutron-py27/d7ade72/ : SUCCESS in 26m 24s\n- cross-neutron-py36 http://logs.openstack.org/60/673160/2/check/cross-neutron-py36/5d0b0c1/ : SUCCESS in 24m 32s\n- cross-nova-py27 http://logs.openstack.org/60/673160/2/check/cross-nova-py27/2a62ede/ : SUCCESS in 19m 20s\n- cross-nova-py36 http://logs.openstack.org/60/673160/2/check/cross-nova-py36/837e3e9/ : FAILURE in 16m 39s\n- cross-nova-functional http://logs.openstack.org/60/673160/2/check/cross-nova-functional/cbdffef/ : SUCCESS in 27m 03s\n- cross-osc-tox-docs http://logs.openstack.org/60/673160/2/check/cross-osc-tox-docs/7098223/ : POST_FAILURE in 3m 29s\n- cross-swift-py27 http://logs.openstack.org/60/673160/2/check/cross-swift-py27/86fb158/ : SUCCESS in 10m 24s\n- cross-swift-py36 http://logs.openstack.org/60/673160/2/check/cross-swift-py36/74c1474/ : FAILURE in 4m 11s\n- requirements-integration http://logs.openstack.org/60/673160/2/check/requirements-integration/369cf80/ : SUCCESS in 59m 18s\n- neutron-grenade-multinode http://logs.openstack.org/60/673160/2/check/neutron-grenade-multinode/348816a/ : FAILURE in 56m 48s\n- tempest-full http://logs.openstack.org/60/673160/2/check/tempest-full/be7f8cb/ : SUCCESS in 1h 44m 19s\n- neutron-grenade http://logs.openstack.org/60/673160/2/check/neutron-grenade/0f84f7e/ : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":2},{"id":"51937e6f0b3fc4c6ea4c47e132a34321c2843537","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-30 23:30:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"32b9af3263d24aa70212ca1f1ba96b38114cb8a7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-30 23:30:57.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":3},{"id":"699049d22042908da5c58e61a11fef0f1a8a31d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 01:54:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/60/673160/3/check/openstack-tox-docs/0f50609/html/ : SUCCESS in 2m 52s\n- openstack-tox-pypy http://logs.openstack.org/60/673160/3/check/openstack-tox-pypy/40a66a8/ : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/673160/3/check/openstack-tox-pep8/f524d66/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/60/673160/3/check/openstack-tox-py27/88b3960/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/60/673160/3/check/openstack-tox-py36/3c7959d/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/60/673160/3/check/openstack-tox-py37/86f946e/ : SUCCESS in 4m 57s\n- openstack-tox-validate http://logs.openstack.org/60/673160/3/check/openstack-tox-validate/fdfb43e/ : SUCCESS in 4m 35s\n- requirements-tox-babel http://logs.openstack.org/60/673160/3/check/requirements-tox-babel/f4144da/ : SUCCESS in 3m 21s\n- requirements-tox-py27-check-uc http://logs.openstack.org/60/673160/3/check/requirements-tox-py27-check-uc/4020d94/ : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc http://logs.openstack.org/60/673160/3/check/requirements-tox-py36-check-uc/2ada814/ : SUCCESS in 13m 15s\n- cross-cinder-py27 http://logs.openstack.org/60/673160/3/check/cross-cinder-py27/cba590a/ : SUCCESS in 9m 29s\n- cross-cinder-py36 http://logs.openstack.org/60/673160/3/check/cross-cinder-py36/ded112e/ : SUCCESS in 8m 10s\n- cross-glance-py27 http://logs.openstack.org/60/673160/3/check/cross-glance-py27/f9079d4/ : SUCCESS in 4m 06s\n- cross-glance-py36 http://logs.openstack.org/60/673160/3/check/cross-glance-py36/2dc9144/ : SUCCESS in 4m 02s\n- cross-horizon-py27 http://logs.openstack.org/60/673160/3/check/cross-horizon-py27/437772c/ : SUCCESS in 10m 00s\n- cross-horizon-py36 http://logs.openstack.org/60/673160/3/check/cross-horizon-py36/d499338/ : SUCCESS in 8m 50s\n- cross-horizon-npm http://logs.openstack.org/60/673160/3/check/cross-horizon-npm/fe53425/ : SUCCESS in 6m 41s\n- cross-keystone-py27 http://logs.openstack.org/60/673160/3/check/cross-keystone-py27/f0b6fe4/ : SUCCESS in 23m 04s\n- cross-keystone-py36 http://logs.openstack.org/60/673160/3/check/cross-keystone-py36/1d28e37/ : SUCCESS in 30m 54s\n- cross-neutron-py27 http://logs.openstack.org/60/673160/3/check/cross-neutron-py27/6ba8908/ : SUCCESS in 24m 33s\n- cross-neutron-py36 http://logs.openstack.org/60/673160/3/check/cross-neutron-py36/383a200/ : SUCCESS in 23m 21s\n- cross-nova-py27 http://logs.openstack.org/60/673160/3/check/cross-nova-py27/47f0379/ : SUCCESS in 18m 19s\n- cross-nova-py36 http://logs.openstack.org/60/673160/3/check/cross-nova-py36/5133bce/ : FAILURE in 17m 54s\n- cross-nova-functional http://logs.openstack.org/60/673160/3/check/cross-nova-functional/7e5f970/ : SUCCESS in 24m 42s\n- cross-osc-tox-docs http://logs.openstack.org/60/673160/3/check/cross-osc-tox-docs/d3b5137/html/ : SUCCESS in 4m 02s\n- cross-swift-py27 http://logs.openstack.org/60/673160/3/check/cross-swift-py27/4a853ef/ : SUCCESS in 8m 28s\n- cross-swift-py36 http://logs.openstack.org/60/673160/3/check/cross-swift-py36/b94c5d9/ : SUCCESS in 8m 38s\n- requirements-integration http://logs.openstack.org/60/673160/3/check/requirements-integration/9e7f8bf/ : SUCCESS in 54m 14s\n- neutron-grenade-multinode http://logs.openstack.org/60/673160/3/check/neutron-grenade-multinode/463c723/ : SUCCESS in 1h 10m 03s\n- tempest-full http://logs.openstack.org/60/673160/3/check/tempest-full/77eb5d0/ : SUCCESS in 1h 37m 31s\n- neutron-grenade http://logs.openstack.org/60/673160/3/check/neutron-grenade/2d43706/ : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":3},{"id":"0f8e130314f7cc5c9e02eabed3a011c9be569fd0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-31 05:06:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6a3ff612f20c39108e3cde7b5ac235f1a1e71841","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 06:59:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/60/673160/3/check/openstack-tox-docs/78c1abe/html/ : SUCCESS in 3m 27s\n- openstack-tox-pypy http://logs.openstack.org/60/673160/3/check/openstack-tox-pypy/07b7b18/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/673160/3/check/openstack-tox-pep8/a1a7f04/ : SUCCESS in 3m 26s\n- openstack-tox-py27 http://logs.openstack.org/60/673160/3/check/openstack-tox-py27/59cb600/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/60/673160/3/check/openstack-tox-py36/1eea702/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/60/673160/3/check/openstack-tox-py37/d00de43/ : SUCCESS in 3m 34s\n- openstack-tox-validate http://logs.openstack.org/60/673160/3/check/openstack-tox-validate/17899f5/ : SUCCESS in 3m 15s\n- requirements-tox-babel http://logs.openstack.org/60/673160/3/check/requirements-tox-babel/f6d5bc5/ : SUCCESS in 3m 06s\n- requirements-tox-py27-check-uc http://logs.openstack.org/60/673160/3/check/requirements-tox-py27-check-uc/34fd561/ : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc http://logs.openstack.org/60/673160/3/check/requirements-tox-py36-check-uc/6e24aa8/ : SUCCESS in 9m 42s\n- cross-cinder-py27 http://logs.openstack.org/60/673160/3/check/cross-cinder-py27/617925e/ : SUCCESS in 8m 10s\n- cross-cinder-py36 http://logs.openstack.org/60/673160/3/check/cross-cinder-py36/66f54a4/ : SUCCESS in 7m 40s\n- cross-glance-py27 http://logs.openstack.org/60/673160/3/check/cross-glance-py27/721e38a/ : SUCCESS in 4m 08s\n- cross-glance-py36 http://logs.openstack.org/60/673160/3/check/cross-glance-py36/d1e7eff/ : SUCCESS in 4m 08s\n- cross-horizon-py27 http://logs.openstack.org/60/673160/3/check/cross-horizon-py27/e2204aa/ : SUCCESS in 9m 59s\n- cross-horizon-py36 http://logs.openstack.org/60/673160/3/check/cross-horizon-py36/414f8a2/ : SUCCESS in 8m 20s\n- cross-horizon-npm http://logs.openstack.org/60/673160/3/check/cross-horizon-npm/20b0b28/ : SUCCESS in 7m 19s\n- cross-keystone-py27 http://logs.openstack.org/60/673160/3/check/cross-keystone-py27/a259501/ : SUCCESS in 29m 34s\n- cross-keystone-py36 http://logs.openstack.org/60/673160/3/check/cross-keystone-py36/498ac43/ : SUCCESS in 35m 34s\n- cross-neutron-py27 http://logs.openstack.org/60/673160/3/check/cross-neutron-py27/a8c2123/ : SUCCESS in 26m 25s\n- cross-neutron-py36 http://logs.openstack.org/60/673160/3/check/cross-neutron-py36/8e17c06/ : SUCCESS in 20m 06s\n- cross-nova-py27 http://logs.openstack.org/60/673160/3/check/cross-nova-py27/10772e2/ : SUCCESS in 19m 04s\n- cross-nova-py36 http://logs.openstack.org/60/673160/3/check/cross-nova-py36/6eee9a4/ : FAILURE in 18m 51s\n- cross-nova-functional http://logs.openstack.org/60/673160/3/check/cross-nova-functional/1715229/ : SUCCESS in 26m 57s\n- cross-osc-tox-docs http://logs.openstack.org/60/673160/3/check/cross-osc-tox-docs/14b37f8/html/ : SUCCESS in 4m 30s\n- cross-swift-py27 http://logs.openstack.org/60/673160/3/check/cross-swift-py27/29b8ce2/ : SUCCESS in 10m 27s\n- cross-swift-py36 http://logs.openstack.org/60/673160/3/check/cross-swift-py36/6e390f8/ : SUCCESS in 7m 41s\n- requirements-integration http://logs.openstack.org/60/673160/3/check/requirements-integration/f206740/ : SUCCESS in 55m 32s\n- neutron-grenade-multinode http://logs.openstack.org/60/673160/3/check/neutron-grenade-multinode/e2e5a73/ : FAILURE in 1h 01m 06s\n- tempest-full http://logs.openstack.org/60/673160/3/check/tempest-full/09a42d1/ : SUCCESS in 1h 42m 36s\n- neutron-grenade http://logs.openstack.org/60/673160/3/check/neutron-grenade/e94afac/ : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":3},{"id":"46be1aca4d4a0de4033dc9cdf16005d630c84fab","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-31 07:18:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"05a1abda97a2a01ff2435551d04159c0ee050b62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 09:47:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/60/673160/3/check/openstack-tox-docs/e3a9939/html/ : SUCCESS in 3m 26s\n- openstack-tox-pypy http://logs.openstack.org/60/673160/3/check/openstack-tox-pypy/c49636f/ : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/673160/3/check/openstack-tox-pep8/b1d5c39/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/60/673160/3/check/openstack-tox-py27/06327cc/ : SUCCESS in 3m 16s\n- openstack-tox-py36 http://logs.openstack.org/60/673160/3/check/openstack-tox-py36/20bffe6/ : SUCCESS in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/60/673160/3/check/openstack-tox-py37/1132724/ : SUCCESS in 4m 50s\n- openstack-tox-validate http://logs.openstack.org/60/673160/3/check/openstack-tox-validate/de060ad/ : SUCCESS in 3m 53s\n- requirements-tox-babel http://logs.openstack.org/60/673160/3/check/requirements-tox-babel/419eb2b/ : SUCCESS in 4m 02s\n- requirements-tox-py27-check-uc http://logs.openstack.org/60/673160/3/check/requirements-tox-py27-check-uc/87bb18c/ : SUCCESS in 10m 03s\n- requirements-tox-py36-check-uc http://logs.openstack.org/60/673160/3/check/requirements-tox-py36-check-uc/eed64f1/ : SUCCESS in 9m 48s\n- cross-cinder-py27 http://logs.openstack.org/60/673160/3/check/cross-cinder-py27/a5d7258/ : SUCCESS in 8m 11s\n- cross-cinder-py36 http://logs.openstack.org/60/673160/3/check/cross-cinder-py36/286d02d/ : FAILURE in 9m 17s\n- cross-glance-py27 http://logs.openstack.org/60/673160/3/check/cross-glance-py27/ac46f65/ : SUCCESS in 4m 16s\n- cross-glance-py36 http://logs.openstack.org/60/673160/3/check/cross-glance-py36/9713c93/ : SUCCESS in 5m 23s\n- cross-horizon-py27 http://logs.openstack.org/60/673160/3/check/cross-horizon-py27/dcf21d3/ : SUCCESS in 12m 28s\n- cross-horizon-py36 http://logs.openstack.org/60/673160/3/check/cross-horizon-py36/c4d4fab/ : SUCCESS in 9m 20s\n- cross-horizon-npm http://logs.openstack.org/60/673160/3/check/cross-horizon-npm/8daabd6/ : SUCCESS in 7m 09s\n- cross-keystone-py27 http://logs.openstack.org/60/673160/3/check/cross-keystone-py27/51478ae/ : SUCCESS in 25m 15s\n- cross-keystone-py36 http://logs.openstack.org/60/673160/3/check/cross-keystone-py36/d1d82c3/ : SUCCESS in 31m 55s\n- cross-neutron-py27 http://logs.openstack.org/60/673160/3/check/cross-neutron-py27/dac2f7a/ : SUCCESS in 29m 34s\n- cross-neutron-py36 http://logs.openstack.org/60/673160/3/check/cross-neutron-py36/203d241/ : SUCCESS in 26m 21s\n- cross-nova-py27 http://logs.openstack.org/60/673160/3/check/cross-nova-py27/0c397d5/ : SUCCESS in 16m 38s\n- cross-nova-py36 http://logs.openstack.org/60/673160/3/check/cross-nova-py36/547e78c/ : FAILURE in 16m 20s\n- cross-nova-functional http://logs.openstack.org/60/673160/3/check/cross-nova-functional/53ee1ec/ : SUCCESS in 24m 30s\n- cross-osc-tox-docs http://logs.openstack.org/60/673160/3/check/cross-osc-tox-docs/e4a7960/html/ : SUCCESS in 4m 28s\n- cross-swift-py27 http://logs.openstack.org/60/673160/3/check/cross-swift-py27/eaa885a/ : SUCCESS in 12m 02s\n- cross-swift-py36 http://logs.openstack.org/60/673160/3/check/cross-swift-py36/9e36a6e/ : SUCCESS in 12m 58s\n- requirements-integration http://logs.openstack.org/60/673160/3/check/requirements-integration/2aa73fb/ : SUCCESS in 1h 17m 52s\n- neutron-grenade-multinode http://logs.openstack.org/60/673160/3/check/neutron-grenade-multinode/457d932/ : SUCCESS in 1h 15m 12s\n- tempest-full http://logs.openstack.org/60/673160/3/check/tempest-full/4457f52/ : SUCCESS in 1h 43m 37s\n- neutron-grenade http://logs.openstack.org/60/673160/3/check/neutron-grenade/4829038/ : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":3},{"id":"4e67b13acb43f08a22483b7894badf4602be8a98","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-31 15:24:04.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ee18a70f4ad74ee3fc8532ddc20bd77c370c5030","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 20:29:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/60/673160/4/check/openstack-tox-docs/0868ec4/html/ : SUCCESS in 2m 57s\n- openstack-tox-pypy http://logs.openstack.org/60/673160/4/check/openstack-tox-pypy/a0e7f5f/ : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/60/673160/4/check/openstack-tox-pep8/5bece1e/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/60/673160/4/check/openstack-tox-py27/fea82c7/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/60/673160/4/check/openstack-tox-py36/f2a66d1/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/60/673160/4/check/openstack-tox-py37/36f25fa/ : SUCCESS in 5m 03s\n- openstack-tox-validate http://logs.openstack.org/60/673160/4/check/openstack-tox-validate/63fe103/ : SUCCESS in 3m 55s\n- requirements-tox-babel http://logs.openstack.org/60/673160/4/check/requirements-tox-babel/2ff3662/ : SUCCESS in 4m 36s\n- requirements-tox-py27-check-uc http://logs.openstack.org/60/673160/4/check/requirements-tox-py27-check-uc/4b84ea1/ : SUCCESS in 7m 44s\n- requirements-tox-py36-check-uc http://logs.openstack.org/60/673160/4/check/requirements-tox-py36-check-uc/8149b39/ : SUCCESS in 9m 16s\n- cross-cinder-py27 http://logs.openstack.org/60/673160/4/check/cross-cinder-py27/83c53bc/ : SUCCESS in 13m 02s\n- cross-cinder-py36 http://logs.openstack.org/60/673160/4/check/cross-cinder-py36/55b5a24/ : SUCCESS in 19m 14s\n- cross-glance-py27 http://logs.openstack.org/60/673160/4/check/cross-glance-py27/4543b74/ : SUCCESS in 7m 20s\n- cross-glance-py36 http://logs.openstack.org/60/673160/4/check/cross-glance-py36/142522e/ : SUCCESS in 7m 33s\n- cross-horizon-py27 http://logs.openstack.org/60/673160/4/check/cross-horizon-py27/606e36b/ : SUCCESS in 11m 53s\n- cross-horizon-py36 http://logs.openstack.org/60/673160/4/check/cross-horizon-py36/804ae35/ : SUCCESS in 10m 18s\n- cross-horizon-npm http://logs.openstack.org/60/673160/4/check/cross-horizon-npm/312bce3/ : SUCCESS in 19m 07s\n- cross-keystone-py27 http://logs.openstack.org/60/673160/4/check/cross-keystone-py27/c33314e/ : FAILURE in 5m 29s\n- cross-keystone-py36 http://logs.openstack.org/60/673160/4/check/cross-keystone-py36/51eb8ec/ : FAILURE in 5m 02s\n- cross-neutron-py27 http://logs.openstack.org/60/673160/4/check/cross-neutron-py27/13bcaa3/ : SUCCESS in 23m 28s\n- cross-neutron-py36 http://logs.openstack.org/60/673160/4/check/cross-neutron-py36/dcd7b0b/ : SUCCESS in 28m 02s\n- cross-nova-py27 http://logs.openstack.org/60/673160/4/check/cross-nova-py27/ed8ee6a/ : SUCCESS in 19m 42s\n- cross-nova-py36 http://logs.openstack.org/60/673160/4/check/cross-nova-py36/f127052/ : SUCCESS in 21m 07s\n- cross-nova-functional http://logs.openstack.org/60/673160/4/check/cross-nova-functional/0871c4f/ : SUCCESS in 25m 07s\n- cross-osc-tox-docs http://logs.openstack.org/60/673160/4/check/cross-osc-tox-docs/ea3d854/html/ : SUCCESS in 3m 40s\n- cross-swift-py27 http://logs.openstack.org/60/673160/4/check/cross-swift-py27/14c25bd/ : SUCCESS in 8m 03s\n- cross-swift-py36 http://logs.openstack.org/60/673160/4/check/cross-swift-py36/14eba25/ : SUCCESS in 9m 25s\n- requirements-integration http://logs.openstack.org/60/673160/4/check/requirements-integration/1860b48/ : SUCCESS in 1h 12m 20s\n- neutron-grenade-multinode http://logs.openstack.org/60/673160/4/check/neutron-grenade-multinode/d010a40/ : SUCCESS in 1h 25m 08s\n- tempest-full http://logs.openstack.org/60/673160/4/check/tempest-full/3634c10/ : SUCCESS in 1h 52m 44s\n- neutron-grenade http://logs.openstack.org/60/673160/4/check/neutron-grenade/5524ef2/ : SUCCESS in 1h 07m 24s","accounts_in_message":[],"_revision_number":4},{"id":"7357f3406f889a85e58b777106702f7283cccc6b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-31 22:53:13.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"405928f2b20b2c94ed598c2f0080d13ce63e7cbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:17:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/60/673160/4/check/openstack-tox-docs/ea10492/html/ : SUCCESS in 4m 09s\n- openstack-tox-pypy https://logs.opendev.org/60/673160/4/check/openstack-tox-pypy/ede9afd/ : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/60/673160/4/check/openstack-tox-pep8/5d1b34f/ : SUCCESS in 2m 58s\n- openstack-tox-py27 https://logs.opendev.org/60/673160/4/check/openstack-tox-py27/8773e53/ : SUCCESS in 4m 37s\n- openstack-tox-py36 https://logs.opendev.org/60/673160/4/check/openstack-tox-py36/a66ad21/ : SUCCESS in 4m 46s\n- openstack-tox-py37 https://logs.opendev.org/60/673160/4/check/openstack-tox-py37/e265257/ : SUCCESS in 3m 43s\n- openstack-tox-validate https://logs.opendev.org/60/673160/4/check/openstack-tox-validate/7188a2b/ : SUCCESS in 3m 05s\n- requirements-tox-babel https://logs.opendev.org/60/673160/4/check/requirements-tox-babel/6147958/ : SUCCESS in 3m 22s\n- requirements-tox-py27-check-uc https://logs.opendev.org/60/673160/4/check/requirements-tox-py27-check-uc/48ae778/ : SUCCESS in 8m 32s\n- requirements-tox-py36-check-uc https://logs.opendev.org/60/673160/4/check/requirements-tox-py36-check-uc/ad3bdbe/ : SUCCESS in 6m 58s\n- cross-cinder-py27 https://logs.opendev.org/60/673160/4/check/cross-cinder-py27/be805dc/ : SUCCESS in 8m 14s\n- cross-cinder-py36 https://logs.opendev.org/60/673160/4/check/cross-cinder-py36/0457505/ : SUCCESS in 10m 21s\n- cross-glance-py27 https://logs.opendev.org/60/673160/4/check/cross-glance-py27/74f4b2f/ : SUCCESS in 4m 48s\n- cross-glance-py36 https://logs.opendev.org/60/673160/4/check/cross-glance-py36/7fd1407/ : SUCCESS in 3m 44s\n- cross-horizon-py27 https://logs.opendev.org/60/673160/4/check/cross-horizon-py27/d7b07c9/ : SUCCESS in 10m 54s\n- cross-horizon-py36 https://logs.opendev.org/60/673160/4/check/cross-horizon-py36/0fb4004/ : SUCCESS in 9m 13s\n- cross-horizon-npm https://logs.opendev.org/60/673160/4/check/cross-horizon-npm/145e19a/ : SUCCESS in 7m 33s\n- cross-keystone-py27 https://logs.opendev.org/60/673160/4/check/cross-keystone-py27/00b4b84/ : SUCCESS in 26m 10s\n- cross-keystone-py36 https://logs.opendev.org/60/673160/4/check/cross-keystone-py36/b3e2725/ : SUCCESS in 33m 07s\n- cross-neutron-py27 https://logs.opendev.org/60/673160/4/check/cross-neutron-py27/983af5c/ : SUCCESS in 21m 52s\n- cross-neutron-py36 https://logs.opendev.org/60/673160/4/check/cross-neutron-py36/052ea33/ : SUCCESS in 25m 04s\n- cross-nova-py27 https://logs.opendev.org/60/673160/4/check/cross-nova-py27/83a4826/ : SUCCESS in 15m 38s\n- cross-nova-py36 https://logs.opendev.org/60/673160/4/check/cross-nova-py36/30707fc/ : SUCCESS in 20m 41s\n- cross-nova-functional https://logs.opendev.org/60/673160/4/check/cross-nova-functional/c320d50/ : SUCCESS in 25m 09s\n- cross-osc-tox-docs https://logs.opendev.org/60/673160/4/check/cross-osc-tox-docs/6b6ae26/html/ : SUCCESS in 5m 29s\n- cross-swift-py27 https://logs.opendev.org/60/673160/4/check/cross-swift-py27/670cca4/ : SUCCESS in 10m 19s\n- cross-swift-py36 https://logs.opendev.org/60/673160/4/check/cross-swift-py36/73867b4/ : SUCCESS in 9m 24s\n- requirements-integration https://logs.opendev.org/60/673160/4/check/requirements-integration/9f97138/ : SUCCESS in 59m 18s\n- neutron-grenade-multinode https://logs.opendev.org/60/673160/4/check/neutron-grenade-multinode/59357eb/ : SUCCESS in 1h 15m 51s\n- tempest-full https://logs.opendev.org/60/673160/4/check/tempest-full/21baa71/ : SUCCESS in 1h 38m 54s\n- neutron-grenade https://logs.opendev.org/60/673160/4/check/neutron-grenade/5f40d18/ : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":4},{"id":"8924160f64dc75ce8321983ba27ed44b4c990902","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-01 01:21:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b3cf87c559a20b2fa432fedf63e834587acac7e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:22:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a6a9b7078f6d4fb37abc1b34dd13b9bb818bad93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 05:50:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/60/673160/4/gate/openstack-tox-docs/9664882/html/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://logs.opendev.org/60/673160/4/gate/openstack-tox-pep8/af94e63/ : SUCCESS in 3m 33s\n- openstack-tox-py27 https://logs.opendev.org/60/673160/4/gate/openstack-tox-py27/71d75a6/ : SUCCESS in 4m 24s\n- openstack-tox-py36 https://logs.opendev.org/60/673160/4/gate/openstack-tox-py36/744ea26/ : SUCCESS in 3m 37s\n- openstack-tox-py37 https://logs.opendev.org/60/673160/4/gate/openstack-tox-py37/33b5d85/ : SUCCESS in 3m 07s\n- openstack-tox-validate https://logs.opendev.org/60/673160/4/gate/openstack-tox-validate/fdc971c/ : SUCCESS in 4m 39s\n- requirements-tox-babel https://logs.opendev.org/60/673160/4/gate/requirements-tox-babel/3d46b5f/ : SUCCESS in 3m 13s\n- requirements-tox-py27-check-uc https://logs.opendev.org/60/673160/4/gate/requirements-tox-py27-check-uc/4b025d5/ : SUCCESS in 6m 57s\n- requirements-tox-py36-check-uc https://logs.opendev.org/60/673160/4/gate/requirements-tox-py36-check-uc/acf10db/ : SUCCESS in 8m 08s\n- cross-cinder-py27 https://logs.opendev.org/60/673160/4/gate/cross-cinder-py27/b1f3a37/ : SUCCESS in 9m 09s\n- cross-cinder-py36 https://logs.opendev.org/60/673160/4/gate/cross-cinder-py36/8d7da6a/ : SUCCESS in 8m 45s\n- cross-glance-py27 https://logs.opendev.org/60/673160/4/gate/cross-glance-py27/26a803c/ : SUCCESS in 4m 30s\n- cross-glance-py36 https://logs.opendev.org/60/673160/4/gate/cross-glance-py36/5121f6e/ : SUCCESS in 4m 37s\n- cross-horizon-py27 https://logs.opendev.org/60/673160/4/gate/cross-horizon-py27/769b5d9/ : SUCCESS in 10m 12s\n- cross-horizon-py36 https://logs.opendev.org/60/673160/4/gate/cross-horizon-py36/51943e8/ : SUCCESS in 9m 15s\n- cross-horizon-npm https://logs.opendev.org/60/673160/4/gate/cross-horizon-npm/7a75fbc/ : SUCCESS in 6m 17s\n- cross-keystone-py27 https://logs.opendev.org/60/673160/4/gate/cross-keystone-py27/85a5e40/ : SUCCESS in 27m 25s\n- cross-keystone-py36 https://logs.opendev.org/60/673160/4/gate/cross-keystone-py36/8a72c91/ : SUCCESS in 29m 30s\n- cross-neutron-py27 https://logs.opendev.org/60/673160/4/gate/cross-neutron-py27/2f63f8f/ : SUCCESS in 20m 51s\n- cross-neutron-py36 https://logs.opendev.org/60/673160/4/gate/cross-neutron-py36/6b52a2c/ : SUCCESS in 22m 11s\n- cross-nova-py27 https://logs.opendev.org/60/673160/4/gate/cross-nova-py27/a2bbc81/ : SUCCESS in 17m 45s\n- cross-nova-py36 https://logs.opendev.org/60/673160/4/gate/cross-nova-py36/e880732/ : SUCCESS in 15m 12s\n- cross-nova-functional https://logs.opendev.org/60/673160/4/gate/cross-nova-functional/bd4da47/ : SUCCESS in 23m 05s\n- cross-osc-tox-docs https://logs.opendev.org/60/673160/4/gate/cross-osc-tox-docs/61596a9/html/ : SUCCESS in 3m 57s\n- cross-swift-py27 https://logs.opendev.org/60/673160/4/gate/cross-swift-py27/89af896/ : SUCCESS in 8m 46s\n- cross-swift-py36 https://logs.opendev.org/60/673160/4/gate/cross-swift-py36/cdee7b7/ : SUCCESS in 8m 02s\n- requirements-integration https://logs.opendev.org/60/673160/4/gate/requirements-integration/2dc68a3/ : SUCCESS in 57m 05s\n- neutron-grenade-multinode https://logs.opendev.org/60/673160/4/gate/neutron-grenade-multinode/fe06468/ : SUCCESS in 1h 06m 00s\n- tempest-full https://logs.opendev.org/60/673160/4/gate/tempest-full/1a9ad42/ : SUCCESS in 1h 56m 05s\n- neutron-grenade https://logs.opendev.org/60/673160/4/gate/neutron-grenade/2203d14/ : SUCCESS in 1h 07m 03s","accounts_in_message":[],"_revision_number":4},{"id":"1b6e96ed349a526d5e35c0e363d5f3573c63bce0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 05:50:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"883f90206d2d2afeef9007342f4a6f5e27190dda","revisions":{"c96df286f03d8203252c877f0285b3b5977bb520":{"kind":"REWORK","_number":1,"created":"2019-07-28 06:20:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/673160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/673160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/673160/1"}}},"commit":{"parents":[{"commit":"eb16ce0ff3c642dcac1940ac63b1093774327e91","subject":"Merge \"update constraint for os-net-config to new release 11.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb16ce0ff3c642dcac1940ac63b1093774327e91"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-28 06:20:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-28 06:20:30.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I13f64539f51f255f54ecd17691107d8b592ddbe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c96df286f03d8203252c877f0285b3b5977bb520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c96df286f03d8203252c877f0285b3b5977bb520"}]},"branch":"refs/heads/master"},"e74b8a49544ccd7cdacb0e9fde056cc58c44911c":{"kind":"REWORK","_number":2,"created":"2019-07-30 06:19:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/673160/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/673160/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/673160/2"}}},"commit":{"parents":[{"commit":"9219fa02ac07ba4454b79e7836e69497f5074a85","subject":"update constraint for openstackdocstheme to new release 1.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9219fa02ac07ba4454b79e7836e69497f5074a85"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 06:19:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 06:19:07.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I13f64539f51f255f54ecd17691107d8b592ddbe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e74b8a49544ccd7cdacb0e9fde056cc58c44911c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e74b8a49544ccd7cdacb0e9fde056cc58c44911c"}]},"branch":"refs/heads/master"},"6c3615c3be15284a59318fcfa34ad74795ba2b0d":{"kind":"REWORK","_number":3,"created":"2019-07-30 23:30:51.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/60/673160/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/673160/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/673160/3"}}},"commit":{"parents":[{"commit":"9219fa02ac07ba4454b79e7836e69497f5074a85","subject":"update constraint for openstackdocstheme to new release 1.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9219fa02ac07ba4454b79e7836e69497f5074a85"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 06:19:07.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-30 23:30:42.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I13f64539f51f255f54ecd17691107d8b592ddbe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c3615c3be15284a59318fcfa34ad74795ba2b0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6c3615c3be15284a59318fcfa34ad74795ba2b0d"}]},"branch":"refs/heads/master"},"883f90206d2d2afeef9007342f4a6f5e27190dda":{"kind":"REWORK","_number":4,"created":"2019-07-31 15:24:04.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/60/673160/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/673160/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/673160/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/673160/4"}}},"commit":{"parents":[{"commit":"9219fa02ac07ba4454b79e7836e69497f5074a85","subject":"update constraint for openstackdocstheme to new release 1.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9219fa02ac07ba4454b79e7836e69497f5074a85"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 06:19:07.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-31 15:23:32.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nholding back lxml for possible nova impacting issues\n\nChange-Id: I13f64539f51f255f54ecd17691107d8b592ddbe7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/883f90206d2d2afeef9007342f4a6f5e27190dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/883f90206d2d2afeef9007342f4a6f5e27190dda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
