)]}'
{"id":"openstack%2Frequirements~673305","triplet_id":"openstack%2Frequirements~master~I5af1190650048c2afec43feb08f2f5486fcfca26","project":"openstack/requirements","branch":"master","topic":"pyghmi","hashtags":[],"change_id":"I5af1190650048c2afec43feb08f2f5486fcfca26","subject":"Exclude pyghmi 1.4.0","status":"MERGED","created":"2019-07-29 14:42:29.000000000","updated":"2019-07-31 15:57:45.000000000","submitted":"2019-07-31 15:57:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673305-1564588665108-520616fe","meta_rev_id":"e46140647a14ada17a513c4f2034efe82d098449","_number":673305,"virtual_id_number":673305,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-31 15:57:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-29 19:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-31 13:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-07-31 13:37:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-29 15:50:26.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 19:28:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-31 13:37:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-31 15:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9ef085a64b7b7d581a353bf69f8a9ae73a5567f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-29 14:42:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2d1d6ee45b9a7bb06c925ef2cc1b90de246019e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-29 14:42:52.000000000","message":"Patch Set 1: Workflow-1\n\nTesting","accounts_in_message":[],"_revision_number":1},{"id":"307488ae7d2f2cc3687dd9216806504c4a052cd7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-29 14:45:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9ebeb325b39603fe2d946a004ee79b962f5e609e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-29 14:45:16.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d01863e3aafe3ba0f7a1006cf9addfa073439a3b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-29 15:50:26.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"fceeb9a73688d6016419b23543b3f4e876b14cc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 17:30:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/673305/2/check/openstack-tox-docs/ad5e610/html/ : SUCCESS in 3m 03s\n- openstack-tox-pypy http://logs.openstack.org/05/673305/2/check/openstack-tox-pypy/1eda98c/ : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/05/673305/2/check/openstack-tox-pep8/5bbf2ad/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/05/673305/2/check/openstack-tox-py27/4d987a6/ : SUCCESS in 6m 33s\n- openstack-tox-py36 http://logs.openstack.org/05/673305/2/check/openstack-tox-py36/ac87961/ : SUCCESS in 6m 01s\n- openstack-tox-py37 http://logs.openstack.org/05/673305/2/check/openstack-tox-py37/6070a36/ : SUCCESS in 4m 55s\n- openstack-tox-validate http://logs.openstack.org/05/673305/2/check/openstack-tox-validate/d15550c/ : SUCCESS in 5m 50s\n- requirements-integration http://logs.openstack.org/05/673305/2/check/requirements-integration/215b42d/ : SUCCESS in 1h 24m 33s\n- neutron-grenade-multinode http://logs.openstack.org/05/673305/2/check/neutron-grenade-multinode/42fc780/ : SUCCESS in 1h 20m 47s\n- tempest-full http://logs.openstack.org/05/673305/2/check/tempest-full/38d7e8e/ : SUCCESS in 1h 35m 27s\n- neutron-grenade http://logs.openstack.org/05/673305/2/check/neutron-grenade/293146d/ : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7eb854876aa129893c5a937e6b8b457121ac6aac","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-29 19:28:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f95f519a777f92b8eb2c4abfc7f5d592e4af3899","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-31 13:37:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"14439e07bb3af88ffd7c70b6d99313e9338d6ecf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 13:37:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1910de2a2a4ac26956b5dda06dad3c64de783e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 15:57:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/05/673305/2/gate/openstack-tox-docs/882bf34/html/ : SUCCESS in 4m 40s\n- openstack-tox-pep8 http://logs.openstack.org/05/673305/2/gate/openstack-tox-pep8/5bee68e/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/05/673305/2/gate/openstack-tox-py27/b961a8a/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/05/673305/2/gate/openstack-tox-py36/c2c3adc/ : SUCCESS in 5m 04s\n- openstack-tox-py37 http://logs.openstack.org/05/673305/2/gate/openstack-tox-py37/4ec5356/ : SUCCESS in 5m 16s\n- openstack-tox-validate http://logs.openstack.org/05/673305/2/gate/openstack-tox-validate/6b4fb4b/ : SUCCESS in 4m 43s\n- requirements-integration http://logs.openstack.org/05/673305/2/gate/requirements-integration/390c93a/ : SUCCESS in 1h 15m 07s\n- neutron-grenade-multinode http://logs.openstack.org/05/673305/2/gate/neutron-grenade-multinode/8b8f42b/ : SUCCESS in 1h 14m 45s\n- tempest-full http://logs.openstack.org/05/673305/2/gate/tempest-full/eddd90c/ : SUCCESS in 1h 53m 09s\n- neutron-grenade http://logs.openstack.org/05/673305/2/gate/neutron-grenade/3695725/ : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":2},{"id":"705aef5149de2b5dcd278477e3b722e39b538e1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 15:57:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7b60f6b3261bb6fb4e23ae217fc2c0d886e86ee","revisions":{"073db921cd790325e2b489dfa431c6f4b2866c17":{"kind":"REWORK","_number":1,"created":"2019-07-29 14:42:29.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/05/673305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/673305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/673305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/673305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/673305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/673305/1"}}},"commit":{"parents":[{"commit":"5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-07-29 14:41:20.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-07-29 14:41:20.000000000","tz":120},"subject":"Exclude pyghmi 1.4.0","message":"Exclude pyghmi 1.4.0\n\nIt seems to break virtualbmc, and thus ironic. It hasn\u0027t hit\nupper-constraints yet, but for some reason ironic jobs use pyghmi\nfrom git.\n\nChange-Id: I5af1190650048c2afec43feb08f2f5486fcfca26\nStory: #2006301\nTask: #36014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/073db921cd790325e2b489dfa431c6f4b2866c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/073db921cd790325e2b489dfa431c6f4b2866c17"}]},"branch":"refs/heads/master"},"f7b60f6b3261bb6fb4e23ae217fc2c0d886e86ee":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-29 14:45:05.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/05/673305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/673305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/673305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/673305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/673305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/673305/2"}}},"commit":{"parents":[{"commit":"5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-07-29 14:41:20.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-07-29 14:44:49.000000000","tz":120},"subject":"Exclude pyghmi 1.4.0","message":"Exclude pyghmi 1.4.0\n\nIt seems to break virtualbmc, and thus ironic. It hasn\u0027t hit\nupper-constraints yet, but for some reason ironic jobs use pyghmi\nfrom git.\n\nChange-Id: I5af1190650048c2afec43feb08f2f5486fcfca26\nStory: #2006301\nTask: #36015\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7b60f6b3261bb6fb4e23ae217fc2c0d886e86ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7b60f6b3261bb6fb4e23ae217fc2c0d886e86ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
