)]}'
{"id":"openstack%2Frequirements~673315","triplet_id":"openstack%2Frequirements~master~I1b22eaa5d4bbf19d6fcebeb17d9c31fe3277efc0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1b22eaa5d4bbf19d6fcebeb17d9c31fe3277efc0","subject":"update constraint for openstackdocstheme to new release 1.31.0","status":"MERGED","created":"2019-07-29 15:34:11.000000000","updated":"2019-07-29 21:23:34.000000000","submitted":"2019-07-29 21:23:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673315-1564435414148-622e802a","meta_rev_id":"01c4c5faf3354e1f48cb70c4db518033dcbe90ea","_number":673315,"virtual_id_number":673315,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-07-29 21:23:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-29 19:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-07-29 19:27:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 19:27:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-29 21:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4aa6c7e76329768e691e02b5fc8f25f9309a5a7b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-29 15:34:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf91a9351e0023db7d5fd34c742054a59a89e647","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 17:49:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/673315/1/check/openstack-tox-docs/ee868d4/html/ : SUCCESS in 4m 32s\n- openstack-tox-pypy http://logs.openstack.org/15/673315/1/check/openstack-tox-pypy/b589ead/ : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/15/673315/1/check/openstack-tox-pep8/40f06b9/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/15/673315/1/check/openstack-tox-py27/2816d94/ : SUCCESS in 2m 57s\n- openstack-tox-py36 http://logs.openstack.org/15/673315/1/check/openstack-tox-py36/94b4cd0/ : SUCCESS in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/15/673315/1/check/openstack-tox-py37/cfc02d9/ : SUCCESS in 6m 13s\n- openstack-tox-validate http://logs.openstack.org/15/673315/1/check/openstack-tox-validate/d6af89f/ : SUCCESS in 3m 43s\n- requirements-tox-babel http://logs.openstack.org/15/673315/1/check/requirements-tox-babel/137f60f/ : SUCCESS in 5m 10s\n- requirements-tox-py27-check-uc http://logs.openstack.org/15/673315/1/check/requirements-tox-py27-check-uc/bb098f2/ : SUCCESS in 6m 41s\n- requirements-tox-py36-check-uc http://logs.openstack.org/15/673315/1/check/requirements-tox-py36-check-uc/ef07425/ : SUCCESS in 8m 32s\n- cross-cinder-py27 http://logs.openstack.org/15/673315/1/check/cross-cinder-py27/38d4ca8/ : SUCCESS in 9m 59s\n- cross-cinder-py36 http://logs.openstack.org/15/673315/1/check/cross-cinder-py36/1884ffa/ : SUCCESS in 10m 48s\n- cross-glance-py27 http://logs.openstack.org/15/673315/1/check/cross-glance-py27/a26f657/ : SUCCESS in 6m 20s\n- cross-glance-py36 http://logs.openstack.org/15/673315/1/check/cross-glance-py36/a03d540/ : SUCCESS in 6m 31s\n- cross-horizon-py27 http://logs.openstack.org/15/673315/1/check/cross-horizon-py27/a4d9e46/ : SUCCESS in 12m 52s\n- cross-horizon-py36 http://logs.openstack.org/15/673315/1/check/cross-horizon-py36/00596ab/ : SUCCESS in 10m 06s\n- cross-horizon-npm http://logs.openstack.org/15/673315/1/check/cross-horizon-npm/2256184/ : SUCCESS in 7m 29s\n- cross-keystone-py27 http://logs.openstack.org/15/673315/1/check/cross-keystone-py27/dcb29e7/ : SUCCESS in 26m 13s\n- cross-keystone-py36 http://logs.openstack.org/15/673315/1/check/cross-keystone-py36/ee456ba/ : SUCCESS in 33m 04s\n- cross-neutron-py27 http://logs.openstack.org/15/673315/1/check/cross-neutron-py27/e3f85b0/ : SUCCESS in 23m 14s\n- cross-neutron-py36 http://logs.openstack.org/15/673315/1/check/cross-neutron-py36/a1a0785/ : SUCCESS in 17m 31s\n- cross-nova-py27 http://logs.openstack.org/15/673315/1/check/cross-nova-py27/1d33238/ : SUCCESS in 16m 17s\n- cross-nova-py36 http://logs.openstack.org/15/673315/1/check/cross-nova-py36/5f958e0/ : SUCCESS in 17m 35s\n- cross-nova-functional http://logs.openstack.org/15/673315/1/check/cross-nova-functional/cc61d2b/ : SUCCESS in 24m 00s\n- cross-osc-tox-docs http://logs.openstack.org/15/673315/1/check/cross-osc-tox-docs/9b5f63d/html/ : SUCCESS in 5m 36s\n- cross-swift-py27 http://logs.openstack.org/15/673315/1/check/cross-swift-py27/6ba62db/ : SUCCESS in 8m 52s\n- cross-swift-py36 http://logs.openstack.org/15/673315/1/check/cross-swift-py36/4dbb561/ : SUCCESS in 9m 14s\n- requirements-integration http://logs.openstack.org/15/673315/1/check/requirements-integration/c0c2102/ : SUCCESS in 1h 10m 37s\n- neutron-grenade-multinode http://logs.openstack.org/15/673315/1/check/neutron-grenade-multinode/44cdcb2/ : SUCCESS in 1h 01m 28s\n- tempest-full http://logs.openstack.org/15/673315/1/check/tempest-full/16f3951/ : SUCCESS in 1h 27m 02s\n- neutron-grenade http://logs.openstack.org/15/673315/1/check/neutron-grenade/ac9feee/ : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"234f0fca9d61b42b20bed55ca9eab5817b5fe9d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-29 19:27:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e924ebc2c8d6e5b0d911aea020a08eb599feb47b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 19:27:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fdc4e60a661b8bea63136bdb3670301efd5e429","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:23:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/15/673315/1/gate/openstack-tox-docs/1c03dec/html/ : SUCCESS in 5m 17s\n- openstack-tox-pep8 http://logs.openstack.org/15/673315/1/gate/openstack-tox-pep8/508b0f7/ : SUCCESS in 5m 51s\n- openstack-tox-py27 http://logs.openstack.org/15/673315/1/gate/openstack-tox-py27/446db30/ : SUCCESS in 4m 04s\n- openstack-tox-py36 http://logs.openstack.org/15/673315/1/gate/openstack-tox-py36/020575f/ : SUCCESS in 4m 05s\n- openstack-tox-py37 http://logs.openstack.org/15/673315/1/gate/openstack-tox-py37/57dccc9/ : SUCCESS in 4m 52s\n- openstack-tox-validate http://logs.openstack.org/15/673315/1/gate/openstack-tox-validate/2743438/ : SUCCESS in 3m 30s\n- requirements-tox-babel http://logs.openstack.org/15/673315/1/gate/requirements-tox-babel/424ca53/ : SUCCESS in 5m 30s\n- requirements-tox-py27-check-uc http://logs.openstack.org/15/673315/1/gate/requirements-tox-py27-check-uc/3d6d526/ : SUCCESS in 6m 58s\n- requirements-tox-py36-check-uc http://logs.openstack.org/15/673315/1/gate/requirements-tox-py36-check-uc/19fb11c/ : SUCCESS in 6m 34s\n- cross-cinder-py27 http://logs.openstack.org/15/673315/1/gate/cross-cinder-py27/6ee06a6/ : SUCCESS in 10m 03s\n- cross-cinder-py36 http://logs.openstack.org/15/673315/1/gate/cross-cinder-py36/ee9d1cb/ : SUCCESS in 8m 27s\n- cross-glance-py27 http://logs.openstack.org/15/673315/1/gate/cross-glance-py27/40c7539/ : SUCCESS in 7m 29s\n- cross-glance-py36 http://logs.openstack.org/15/673315/1/gate/cross-glance-py36/d7b3476/ : SUCCESS in 6m 03s\n- cross-horizon-py27 http://logs.openstack.org/15/673315/1/gate/cross-horizon-py27/ab9aff3/ : SUCCESS in 10m 54s\n- cross-horizon-py36 http://logs.openstack.org/15/673315/1/gate/cross-horizon-py36/82ed27f/ : SUCCESS in 9m 13s\n- cross-horizon-npm http://logs.openstack.org/15/673315/1/gate/cross-horizon-npm/11e82be/ : SUCCESS in 6m 50s\n- cross-keystone-py27 http://logs.openstack.org/15/673315/1/gate/cross-keystone-py27/7f51308/ : SUCCESS in 28m 14s\n- cross-keystone-py36 http://logs.openstack.org/15/673315/1/gate/cross-keystone-py36/c47a26f/ : SUCCESS in 34m 46s\n- cross-neutron-py27 http://logs.openstack.org/15/673315/1/gate/cross-neutron-py27/3f8c935/ : SUCCESS in 25m 02s\n- cross-neutron-py36 http://logs.openstack.org/15/673315/1/gate/cross-neutron-py36/4a5b619/ : SUCCESS in 27m 07s\n- cross-nova-py27 http://logs.openstack.org/15/673315/1/gate/cross-nova-py27/c3868e6/ : SUCCESS in 16m 46s\n- cross-nova-py36 http://logs.openstack.org/15/673315/1/gate/cross-nova-py36/026f3e1/ : SUCCESS in 17m 23s\n- cross-nova-functional http://logs.openstack.org/15/673315/1/gate/cross-nova-functional/a156f8e/ : SUCCESS in 25m 17s\n- cross-osc-tox-docs http://logs.openstack.org/15/673315/1/gate/cross-osc-tox-docs/baa0123/html/ : SUCCESS in 5m 02s\n- cross-swift-py27 http://logs.openstack.org/15/673315/1/gate/cross-swift-py27/7045852/ : SUCCESS in 9m 56s\n- cross-swift-py36 http://logs.openstack.org/15/673315/1/gate/cross-swift-py36/a784ade/ : SUCCESS in 9m 23s\n- requirements-integration http://logs.openstack.org/15/673315/1/gate/requirements-integration/63fbf22/ : SUCCESS in 1h 10m 51s\n- neutron-grenade-multinode http://logs.openstack.org/15/673315/1/gate/neutron-grenade-multinode/fe2d77b/ : SUCCESS in 1h 17m 54s\n- tempest-full http://logs.openstack.org/15/673315/1/gate/tempest-full/c94b407/ : SUCCESS in 1h 47m 54s\n- neutron-grenade http://logs.openstack.org/15/673315/1/gate/neutron-grenade/ae39d3c/ : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":1},{"id":"28878b7cef9a89dddddaef315c1216763e9a7855","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:23:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9219fa02ac07ba4454b79e7836e69497f5074a85","revisions":{"9219fa02ac07ba4454b79e7836e69497f5074a85":{"kind":"REWORK","_number":1,"created":"2019-07-29 15:34:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/673315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/673315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/673315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/673315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/673315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/673315/1"}}},"commit":{"parents":[{"commit":"5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5bf416e0598e7fe4ca1e0cd8198a0fb967e3ffca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-29 15:34:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-29 15:34:08.000000000","tz":0},"subject":"update constraint for openstackdocstheme to new release 1.31.0","message":"update constraint for openstackdocstheme to new release 1.31.0\n\nChange-Id: I1b22eaa5d4bbf19d6fcebeb17d9c31fe3277efc0\nmeta:version: 1.31.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: no\nmeta:first: no\nmeta:release:Author: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Commit: Andreas Jaeger \u003caj@suse.com\u003e\nmeta:release:Change-Id: Iaf0d24dacab4b66883f77846927d5052c7254295\nmeta:release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9219fa02ac07ba4454b79e7836e69497f5074a85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9219fa02ac07ba4454b79e7836e69497f5074a85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
