)]}'
{"id":"openstack%2Frequirements~673414","triplet_id":"openstack%2Frequirements~stable%2Fqueens~Ia4162512eeb584870e7184cb19f07a9b94074871","project":"openstack/requirements","branch":"stable/queens","topic":"fix_trollius","hashtags":[],"change_id":"Ia4162512eeb584870e7184cb19f07a9b94074871","subject":"Trollius 2.1 isn\u0027t on pypi anymore","status":"ABANDONED","created":"2019-07-30 07:21:24.000000000","updated":"2019-07-30 15:27:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"62786bf245a5e1e99d079c325af0b3201e2be647","_number":673414,"virtual_id_number":673414,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"date":"2019-07-30 15:22:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"date":"2019-07-30 07:27:17.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-07-30 15:21:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-07-30 14:21:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-07-30 14:23:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-30 15:15:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 14:21:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-07-30 14:23:37.000000000","updated_by":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"reviewer":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"state":"REVIEWER"},{"updated":"2019-07-30 15:15:43.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-07-30 15:21:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-07-30 15:22:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dad6de8b760255fa3d57aa9488c75bd550b53c8c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-30 07:21:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"906bcbc02e212da6a14024dedec8825e8238831d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-30 07:27:17.000000000","message":"Patch Set 1:\n\nAs said in the channel, I am doing this here, with the hope of backporting back to newton, which is impossible within other projects (EOL branches).\n\nI can also try to propose the patch into each project down to the lowest branch I can, and then propose here just for newton.","accounts_in_message":[],"_revision_number":1},{"id":"cc29168298c63be9d36efeadd5913f40aae9b557","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 09:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pypy http://logs.openstack.org/14/673414/1/check/openstack-tox-pypy/6fe9699/ : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/14/673414/1/check/openstack-tox-pep8/0e1ec3c/ : SUCCESS in 2m 36s\n- openstack-tox-py27 http://logs.openstack.org/14/673414/1/check/openstack-tox-py27/0925f20/ : SUCCESS in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/14/673414/1/check/openstack-tox-py35/d89f729/ : SUCCESS in 5m 24s\n- openstack-tox-validate http://logs.openstack.org/14/673414/1/check/openstack-tox-validate/395ab4e/ : SUCCESS in 3m 35s\n- requirements-tox-babel http://logs.openstack.org/14/673414/1/check/requirements-tox-babel/c10df3c/ : SUCCESS in 3m 11s\n- requirements-tox-py27-check-uc http://logs.openstack.org/14/673414/1/check/requirements-tox-py27-check-uc/d501356/ : SUCCESS in 28m 49s\n- requirements-tox-py35-check-uc http://logs.openstack.org/14/673414/1/check/requirements-tox-py35-check-uc/7b7f6e2/ : SUCCESS in 7m 47s\n- cross-cinder-py27 http://logs.openstack.org/14/673414/1/check/cross-cinder-py27/39b1e80/ : SUCCESS in 10m 19s\n- cross-cinder-py35 http://logs.openstack.org/14/673414/1/check/cross-cinder-py35/9f6a71a/ : SUCCESS in 11m 09s\n- cross-glance-py27 http://logs.openstack.org/14/673414/1/check/cross-glance-py27/6f288dc/ : SUCCESS in 5m 22s\n- cross-glance-py35 http://logs.openstack.org/14/673414/1/check/cross-glance-py35/d499114/ : SUCCESS in 5m 38s\n- cross-horizon-py27 http://logs.openstack.org/14/673414/1/check/cross-horizon-py27/79e536b/ : SUCCESS in 17m 51s\n- cross-horizon-py35 http://logs.openstack.org/14/673414/1/check/cross-horizon-py35/ffdc653/ : SUCCESS in 10m 22s\n- cross-keystone-py27 http://logs.openstack.org/14/673414/1/check/cross-keystone-py27/e07fd4b/ : SUCCESS in 14m 54s\n- cross-keystone-py35 http://logs.openstack.org/14/673414/1/check/cross-keystone-py35/3f1ee80/ : SUCCESS in 14m 37s\n- cross-neutron-py27 http://logs.openstack.org/14/673414/1/check/cross-neutron-py27/92cae90/ : SUCCESS in 26m 49s\n- cross-neutron-py35 http://logs.openstack.org/14/673414/1/check/cross-neutron-py35/0be8d1a/ : SUCCESS in 22m 50s\n- cross-nova-py27 http://logs.openstack.org/14/673414/1/check/cross-nova-py27/d1f959f/ : SUCCESS in 14m 58s\n- cross-nova-py35 http://logs.openstack.org/14/673414/1/check/cross-nova-py35/7cea697/ : SUCCESS in 15m 38s\n- cross-nova-functional http://logs.openstack.org/14/673414/1/check/cross-nova-functional/8c8736b/ : SUCCESS in 16m 41s\n- cross-swift-py27 http://logs.openstack.org/14/673414/1/check/cross-swift-py27/5399917/ : SUCCESS in 6m 56s\n- cross-swift-py35 http://logs.openstack.org/14/673414/1/check/cross-swift-py35/96b7248/ : SUCCESS in 5m 54s\n- requirements-integration http://logs.openstack.org/14/673414/1/check/requirements-integration/8932880/ : SUCCESS in 1h 00m 08s\n- neutron-grenade-multinode http://logs.openstack.org/14/673414/1/check/neutron-grenade-multinode/49116a3/ : SUCCESS in 1h 11m 50s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/14/673414/1/check/legacy-tempest-dsvm-neutron-full/d7425d1/ : SUCCESS in 1h 22m 37s\n- neutron-grenade http://logs.openstack.org/14/673414/1/check/neutron-grenade/560843f/ : SUCCESS in 56m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4d4f8a96f6340f084c14f42a82451317f83fd13c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-30 14:21:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9d5a8d45511f2e15a533b97b5dac6d5d9d6e83d4","author":{"_account_id":23928,"name":"Pete Birley","email":"petebirley@gmail.com","username":"portdirect"},"date":"2019-07-30 14:23:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"05e6876da0a485e77b0ca4d6fc8b191f6ea7cf44","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-07-30 15:15:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"920aa15f214654403743e6d16d280ca749155814","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-30 15:21:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dbc3c2dc953fc146579c7eeb2c4c62fb7b3f5c3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 15:22:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed919b8b298f341a8eb6a1db9de20daa57137481","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-30 15:26:55.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"889ee2366326da455b0751dd3022d682babb0b4a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-30 15:27:03.000000000","message":"Patch Set 1:\n\nAppears to be back? https://pypi.org/project/trollius/2.1/","accounts_in_message":[],"_revision_number":1},{"id":"407fd16227a8f5df2473480d4359431c381e984d","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-30 15:27:18.000000000","message":"Abandoned\n\nThis is back online. Woot!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba56e1d2c333186ff31133afb3c40fcba4e49109","revisions":{"ba56e1d2c333186ff31133afb3c40fcba4e49109":{"kind":"REWORK","_number":1,"created":"2019-07-30 07:21:24.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/14/673414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/673414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/673414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/673414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/673414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/673414/1"}}},"commit":{"parents":[{"commit":"47a88199b781b221f0968ec2ee0e7e6ccb4d7252","subject":"update constraint for os-brick to new release 2.3.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47a88199b781b221f0968ec2ee0e7e6ccb4d7252"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-30 07:18:46.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-07-30 07:21:10.000000000","tz":120},"subject":"Trollius 2.1 isn\u0027t on pypi anymore","message":"Trollius 2.1 isn\u0027t on pypi anymore\n\nOn the 2.1 branch of Trollius, only 2.1.post2 remains.\nTo continue insuring co-installability, we should bump to post2.\n\nI also added a python version matcher, as it doesn\u0027t make sense\nto have trollius installed for py3.\n\nChange-Id: Ia4162512eeb584870e7184cb19f07a9b94074871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba56e1d2c333186ff31133afb3c40fcba4e49109"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba56e1d2c333186ff31133afb3c40fcba4e49109"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
