)]}'
{"id":"openstack%2Frequirements~673848","triplet_id":"openstack%2Frequirements~master~I8a61667c529ce2f20035ab6bcc7dc31ce7edbdc3","project":"openstack/requirements","branch":"master","topic":"test-lxml-4.4.0","hashtags":[],"change_id":"I8a61667c529ce2f20035ab6bcc7dc31ce7edbdc3","subject":"test lxml 4.4.0","status":"ABANDONED","created":"2019-07-31 15:25:16.000000000","updated":"2019-08-30 19:52:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"39525a704321d8886147f71b1d9c012e4adc8206","_number":673848,"virtual_id_number":673848,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":-1,"date":"2019-08-30 17:15:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2019-08-30 12:52:59.000000000","_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"date":"2019-08-30 18:36:41.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 16:47:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-08-30 12:52:59.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-08-30 17:15:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-30 18:36:41.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"}],"messages":[{"id":"a75df0e7299c6467ea48738a90bd16da9f6013dc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-07-31 15:25:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"687089a3a0dd75f1c81d2ef3773cccd21b5769d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 20:18:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/48/673848/1/check/openstack-tox-docs/b68379c/html/ : SUCCESS in 3m 42s\n- openstack-tox-pypy http://logs.openstack.org/48/673848/1/check/openstack-tox-pypy/a651d0c/ : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/48/673848/1/check/openstack-tox-pep8/d5c5c86/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/48/673848/1/check/openstack-tox-py27/eb1bbeb/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/48/673848/1/check/openstack-tox-py36/6dd0230/ : SUCCESS in 13m 12s\n- openstack-tox-py37 http://logs.openstack.org/48/673848/1/check/openstack-tox-py37/2b1868c/ : SUCCESS in 5m 55s\n- openstack-tox-validate http://logs.openstack.org/48/673848/1/check/openstack-tox-validate/de410ab/ : SUCCESS in 3m 16s\n- requirements-tox-babel http://logs.openstack.org/48/673848/1/check/requirements-tox-babel/bfb165d/ : SUCCESS in 2m 56s\n- requirements-tox-py27-check-uc http://logs.openstack.org/48/673848/1/check/requirements-tox-py27-check-uc/0cf9abc/ : SUCCESS in 6m 59s\n- requirements-tox-py36-check-uc http://logs.openstack.org/48/673848/1/check/requirements-tox-py36-check-uc/6822881/ : SUCCESS in 6m 50s\n- cross-cinder-py27 http://logs.openstack.org/48/673848/1/check/cross-cinder-py27/cc24fc3/ : SUCCESS in 9m 42s\n- cross-cinder-py36 http://logs.openstack.org/48/673848/1/check/cross-cinder-py36/cd5777f/ : SUCCESS in 9m 10s\n- cross-glance-py27 http://logs.openstack.org/48/673848/1/check/cross-glance-py27/22c7888/ : SUCCESS in 13m 10s\n- cross-glance-py36 http://logs.openstack.org/48/673848/1/check/cross-glance-py36/ead9c46/ : SUCCESS in 5m 48s\n- cross-horizon-py27 http://logs.openstack.org/48/673848/1/check/cross-horizon-py27/585842d/ : SUCCESS in 19m 28s\n- cross-horizon-py36 http://logs.openstack.org/48/673848/1/check/cross-horizon-py36/243fac7/ : SUCCESS in 9m 59s\n- cross-horizon-npm http://logs.openstack.org/48/673848/1/check/cross-horizon-npm/3731ee5/ : SUCCESS in 13m 09s\n- cross-keystone-py27 http://logs.openstack.org/48/673848/1/check/cross-keystone-py27/88b0de1/ : SUCCESS in 27m 52s\n- cross-keystone-py36 http://logs.openstack.org/48/673848/1/check/cross-keystone-py36/fa3abc8/ : SUCCESS in 32m 03s\n- cross-neutron-py27 http://logs.openstack.org/48/673848/1/check/cross-neutron-py27/a456328/ : SUCCESS in 31m 30s\n- cross-neutron-py36 http://logs.openstack.org/48/673848/1/check/cross-neutron-py36/9a0affd/ : SUCCESS in 22m 30s\n- cross-nova-py27 http://logs.openstack.org/48/673848/1/check/cross-nova-py27/d13ddd2/ : SUCCESS in 26m 14s\n- cross-nova-py36 http://logs.openstack.org/48/673848/1/check/cross-nova-py36/b0e0e5b/ : FAILURE in 26m 12s\n- cross-nova-functional http://logs.openstack.org/48/673848/1/check/cross-nova-functional/e10f8e8/ : SUCCESS in 31m 30s\n- cross-osc-tox-docs http://logs.openstack.org/48/673848/1/check/cross-osc-tox-docs/ee87d53/html/ : SUCCESS in 5m 13s\n- cross-swift-py27 http://logs.openstack.org/48/673848/1/check/cross-swift-py27/53bea36/ : SUCCESS in 8m 12s\n- cross-swift-py36 http://logs.openstack.org/48/673848/1/check/cross-swift-py36/8e52022/ : SUCCESS in 11m 37s\n- requirements-integration http://logs.openstack.org/48/673848/1/check/requirements-integration/4a45066/ : SUCCESS in 57m 21s\n- neutron-grenade-multinode http://logs.openstack.org/48/673848/1/check/neutron-grenade-multinode/94f7652/ : SUCCESS in 1h 11m 27s\n- tempest-full http://logs.openstack.org/48/673848/1/check/tempest-full/80fe3a8/ : SUCCESS in 1h 38m 12s\n- neutron-grenade http://logs.openstack.org/48/673848/1/check/neutron-grenade/afa067e/ : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9247e9178fd79fc76354f3080484ca700c03f749","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-08-02 16:47:13.000000000","message":"Patch Set 1:\n\nLooks like an XML ordering issue in the unit tests in nova. But actual results look fine.","accounts_in_message":[],"_revision_number":1},{"id":"b8b4180d3050b3b76269e73d552056861c02c715","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-15 16:11:16.000000000","message":"Abandoned\n\n4.4.1 was released, has an ordering fix, is being tested/merged as part of the daily update","accounts_in_message":[],"_revision_number":1},{"id":"12b99b6ef90df2608102aa7296801520ac976ecb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-27 22:27:45.000000000","message":"Patch Set 1:\n\nthat did not fix it so we proably want to restore this\nhttps://review.opendev.org/#/c/678964/\nshould adress the nova issue","accounts_in_message":[],"_revision_number":1},{"id":"3489f29cff3274ba20ac57026d882545d47ca670","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-29 19:46:07.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"4adfa237cf990f1c96d6efa8c6e7c2ba69d88b7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 00:26:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e8c4066307a4fd0b40cb3c2df95df2e : SUCCESS in 3m 47s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/36c162a5c0e74dd6ac9a0fb47b7ef9ec : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51ef0a5fd6d2467b9663f6f77edd31e0 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7bc16f1a6bfe493481300e40a2245f51 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45a7000f80924bf7a1d7bd3971431a7a : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f9fdd98ced442cfa4da1dfd76170cbc : SUCCESS in 7m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66ea41d6727b416d980d7c346b3ace2a : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad6407945cc24606bf5a3fdc691e18dd : SUCCESS in 4m 00s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/3f07dc1a6d5b4b8aab33d24815d5a41f : SUCCESS in 14m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/747582e76677470bb27b877b91d1a468 : SUCCESS in 14m 22s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/b4bd680810a04aa296004b7300ee9628 : SUCCESS in 8m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8d3e81e53db94acea2ac71aa50187a2b : SUCCESS in 9m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/75d0c0dc007640abb1c6a1591df7bddd : SUCCESS in 4m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/15acb919a9084339a668cc930feb9ee7 : SUCCESS in 6m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/5e45e848772d4d94ab1419aa7a521be5 : SUCCESS in 10m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/377cfaee5e41469a8bc47c1b5c49b47f : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ad1d5b38983b4c1c8f1f82b57a4ce0fc : SUCCESS in 9m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/1a7d899cf88040f19c4ec22b02728ddd : SUCCESS in 38m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f1b20bf341764d0a99571861525652ca : SUCCESS in 28m 31s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/c03bed409c51432eb4bc0f9d708d2e1c : SUCCESS in 32m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7eba8255d77643e5a23eff727be2d2aa : SUCCESS in 27m 23s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/e4a86c6277db4dcea78d12e0bc0d9e24 : SUCCESS in 15m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/572596fc07dd4a25a9a9e67a9e254493 : FAILURE in 15m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/046ecd7180944e23b18023b559ea36bf : FAILURE in 14m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0559551196b84c9a99b33764015f62dc : SUCCESS in 22m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8352607609404723be84d2bdde9d6ff2 : SUCCESS in 6m 02s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2cc07eef2b0e4609a81a13ea7ee26f52 : FAILURE in 11m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3eb038602d534996b51033fcfa7011e2 : SUCCESS in 12m 28s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/05c6dd49d9b244dd8e62d60ee04d0215 : SUCCESS in 1h 12m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/599fd103e530480f9f3abbd7af40a81f : SUCCESS in 1h 16m 03s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a02b2d3c86654feab4c4544f8688feaf : SUCCESS in 1h 48m 35s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/af3931df3144400b8bd8e4894a4ab0a6 : SUCCESS in 1h 03m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4b4325fb6d3621727177b62848646dcaa8a056e5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-08-30 12:52:59.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c3e28e2d4c08e7b1edf3f594437b8ea06dbd7127","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 17:15:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656d4821104d417ab30b4150685d7e47 : SUCCESS in 5m 23s\n- openstack-tox-pypy https://zuul.opendev.org/t/openstack/build/2ed523919cad451baa4a457bed1ad02c : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e578f807b5cb477285b03f1e27e582ef : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bddd6a67966143079f3313a64f77cd91 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7f6ae9452194afe8af2331f46763b17 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/203ad7ad472e48b4abb5593b234b9f35 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f523a570ad24458a987729dea8d3ca40 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/009dfb5adbf144c0b8d782d2d9169810 : SUCCESS in 3m 57s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/b7540fc80be84f819cd7c5ec25ef5911 : SUCCESS in 9m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/553ca717bca44d959ef05ac7e0848c57 : SUCCESS in 9m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/474da373a68349c3977b1278f92c7401 : SUCCESS in 9m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/de614e174ec14aa893dfd48cb98a808e : SUCCESS in 9m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/07711922c7024edb9acfe73e1e6e6596 : SUCCESS in 5m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ef646afa7e1413d89eacb80f614b548 : FAILURE in 8m 05s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0100cd80994a4acfad9a0483f9c9c469 : SUCCESS in 10m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9bd12251bd294f8a87a13825e5c54773 : SUCCESS in 13m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/011f05c3c5a149778fd2fbdba8411d7e : SUCCESS in 8m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/de274c8c5c734124b68c5c8f7724dc26 : SUCCESS in 34m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/02351e28be4b4967aa51f8778b44f092 : SUCCESS in 32m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/d3be4e2d93104c50a28bb49900f15a94 : SUCCESS in 28m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cb48ccb912be442e9d50708fa9cbcf53 : SUCCESS in 22m 07s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/a0c76b3441ed409fb97411729fb7d6af : SUCCESS in 15m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f1a08c05c4014535a49ee9b8ef915d1c : SUCCESS in 14m 11s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a5bcb5a59d3241ceb63e9c196992864a : SUCCESS in 13m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/220c71052fc2477f95984511d19759b3 : SUCCESS in 20m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/60e909d28322403f9de9216efbfd0ec1 : SUCCESS in 5m 47s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6d87a7f307a04316a28f448a77f619b5 : SUCCESS in 11m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/daa5c7b51b9649ff9c6aa848412f1865 : SUCCESS in 11m 42s\n- requirements-integration https://zuul.opendev.org/t/openstack/build/a11b2ce085b3416e9383daac0a960ff2 : SUCCESS in 57m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5b5457328694c1f8b8b6937d34a0672 : SUCCESS in 1h 25m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9dd644008f9447d7a93028c2a3d13804 : SUCCESS in 1h 51m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9525501d3fbc445391615388a20caab4 : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":2},{"id":"58a2bc51f01a854a282e97a932124f55afa297e9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-30 19:52:25.000000000","message":"Abandoned\n\nusing the other review","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44b38e2f9dbc576f21c6beb3167b1e55d020fc9a","revisions":{"78f16a9599da68a11b5b2b306aecb016ac4ef6a6":{"kind":"REWORK","_number":1,"created":"2019-07-31 15:25:16.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/48/673848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/673848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/673848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/673848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/673848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/673848/1"}}},"commit":{"parents":[{"commit":"9219fa02ac07ba4454b79e7836e69497f5074a85","subject":"update constraint for openstackdocstheme to new release 1.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9219fa02ac07ba4454b79e7836e69497f5074a85"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-31 15:24:53.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-31 15:24:53.000000000","tz":-300},"subject":"test lxml 4.4.0","message":"test lxml 4.4.0\n\nChange-Id: I8a61667c529ce2f20035ab6bcc7dc31ce7edbdc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78f16a9599da68a11b5b2b306aecb016ac4ef6a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/78f16a9599da68a11b5b2b306aecb016ac4ef6a6"}]},"branch":"refs/heads/master"},"44b38e2f9dbc576f21c6beb3167b1e55d020fc9a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-30 12:52:59.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/48/673848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/673848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/673848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/673848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/673848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/673848/2"}}},"commit":{"parents":[{"commit":"9219fa02ac07ba4454b79e7836e69497f5074a85","subject":"update constraint for openstackdocstheme to new release 1.31.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9219fa02ac07ba4454b79e7836e69497f5074a85"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2019-07-31 15:24:53.000000000","tz":-300},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2019-08-30 12:52:30.000000000","tz":60},"subject":"test lxml 4.4.0","message":"test lxml 4.4.0\n\nDepends-On: https://review.opendev.org/#/c/678964/\nChange-Id: I8a61667c529ce2f20035ab6bcc7dc31ce7edbdc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44b38e2f9dbc576f21c6beb3167b1e55d020fc9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44b38e2f9dbc576f21c6beb3167b1e55d020fc9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
