)]}'
{"id":"openstack%2Frequirements~673908","triplet_id":"openstack%2Frequirements~master~I7949ac17d6777b9ba5e403c135ca74be066d4553","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7949ac17d6777b9ba5e403c135ca74be066d4553","subject":"update constraint for os-traits to new release 0.16.0","status":"MERGED","created":"2019-07-31 20:36:26.000000000","updated":"2019-08-01 05:50:53.000000000","submitted":"2019-08-01 05:50:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673908-1564638653415-b398d57e","meta_rev_id":"8b2d64b985e77b52aa832e8c6aec8bd203f7c27f","_number":673908,"virtual_id_number":673908,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":2,"date":"2019-08-01 05:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-01 01:22:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2019-07-31 20:39:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-01 01:22:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 20:39:05.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-08-01 01:22:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-01 05:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0cc69b148740ce9a69fa5dafcd3be2f8587f51c6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-31 20:36:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a54b23871190f651b8f677283b3004414365e20","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-31 20:39:05.000000000","message":"Patch Set 1: Code-Review+1\n\ndo it","accounts_in_message":[],"_revision_number":1},{"id":"40f623e899acb784755b1481a1479c43ea1a6672","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 23:43:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/673908/1/check/openstack-tox-docs/3638da0/html/ : SUCCESS in 5m 02s\n- openstack-tox-pypy http://logs.openstack.org/08/673908/1/check/openstack-tox-pypy/72c623b/ : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/08/673908/1/check/openstack-tox-pep8/35fea63/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/08/673908/1/check/openstack-tox-py27/cfbd48d/ : SUCCESS in 3m 58s\n- openstack-tox-py36 http://logs.openstack.org/08/673908/1/check/openstack-tox-py36/6da0462/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/08/673908/1/check/openstack-tox-py37/aa7314b/ : SUCCESS in 5m 24s\n- openstack-tox-validate http://logs.openstack.org/08/673908/1/check/openstack-tox-validate/a847241/ : SUCCESS in 3m 38s\n- requirements-tox-babel http://logs.openstack.org/08/673908/1/check/requirements-tox-babel/82427f6/ : SUCCESS in 3m 53s\n- requirements-tox-py27-check-uc http://logs.openstack.org/08/673908/1/check/requirements-tox-py27-check-uc/ac9afa4/ : SUCCESS in 11m 52s\n- requirements-tox-py36-check-uc http://logs.openstack.org/08/673908/1/check/requirements-tox-py36-check-uc/b12a42c/ : SUCCESS in 8m 09s\n- cross-cinder-py27 http://logs.openstack.org/08/673908/1/check/cross-cinder-py27/041b63b/ : SUCCESS in 9m 50s\n- cross-cinder-py36 http://logs.openstack.org/08/673908/1/check/cross-cinder-py36/b9914c3/ : SUCCESS in 9m 15s\n- cross-glance-py27 http://logs.openstack.org/08/673908/1/check/cross-glance-py27/403dd9f/ : SUCCESS in 4m 55s\n- cross-glance-py36 http://logs.openstack.org/08/673908/1/check/cross-glance-py36/86bd827/ : SUCCESS in 5m 26s\n- cross-horizon-py27 http://logs.openstack.org/08/673908/1/check/cross-horizon-py27/6acbcf3/ : SUCCESS in 10m 23s\n- cross-horizon-py36 http://logs.openstack.org/08/673908/1/check/cross-horizon-py36/2f0b5d5/ : SUCCESS in 10m 34s\n- cross-horizon-npm http://logs.openstack.org/08/673908/1/check/cross-horizon-npm/f841300/ : SUCCESS in 7m 41s\n- cross-keystone-py27 http://logs.openstack.org/08/673908/1/check/cross-keystone-py27/7321cfe/ : SUCCESS in 27m 49s\n- cross-keystone-py36 http://logs.openstack.org/08/673908/1/check/cross-keystone-py36/f4e7e07/ : SUCCESS in 31m 17s\n- cross-neutron-py27 http://logs.openstack.org/08/673908/1/check/cross-neutron-py27/f699035/ : SUCCESS in 24m 27s\n- cross-neutron-py36 http://logs.openstack.org/08/673908/1/check/cross-neutron-py36/43cd6b1/ : SUCCESS in 26m 09s\n- cross-nova-py27 http://logs.openstack.org/08/673908/1/check/cross-nova-py27/63ee293/ : SUCCESS in 21m 03s\n- cross-nova-py36 http://logs.openstack.org/08/673908/1/check/cross-nova-py36/61aab50/ : SUCCESS in 19m 19s\n- cross-nova-functional http://logs.openstack.org/08/673908/1/check/cross-nova-functional/6a6db9d/ : SUCCESS in 24m 55s\n- cross-osc-tox-docs http://logs.openstack.org/08/673908/1/check/cross-osc-tox-docs/e684b6a/html/ : SUCCESS in 3m 54s\n- cross-swift-py27 http://logs.openstack.org/08/673908/1/check/cross-swift-py27/e0fe36f/ : SUCCESS in 9m 09s\n- cross-swift-py36 http://logs.openstack.org/08/673908/1/check/cross-swift-py36/a5e1cc0/ : SUCCESS in 8m 43s\n- requirements-integration http://logs.openstack.org/08/673908/1/check/requirements-integration/5e202bc/ : SUCCESS in 1h 01m 40s\n- neutron-grenade-multinode http://logs.openstack.org/08/673908/1/check/neutron-grenade-multinode/f6faad7/ : SUCCESS in 1h 12m 45s\n- tempest-full http://logs.openstack.org/08/673908/1/check/tempest-full/d5d15c9/ : SUCCESS in 2h 02m 29s\n- neutron-grenade http://logs.openstack.org/08/673908/1/check/neutron-grenade/18ea44a/ : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":1},{"id":"42b224153da8d160a6d3d00e186adfbf2ba6a4a8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-01 01:22:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c808560cc24197167aece3b6466503b7ce752414","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 01:22:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"27554876e545029cc237e0ac193715dd0f408e10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 05:50:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/08/673908/1/gate/openstack-tox-docs/10dc4aa/html/ : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://logs.opendev.org/08/673908/1/gate/openstack-tox-pep8/4e8ee03/ : SUCCESS in 4m 16s\n- openstack-tox-py27 https://logs.opendev.org/08/673908/1/gate/openstack-tox-py27/d2b29dc/ : SUCCESS in 3m 23s\n- openstack-tox-py36 https://logs.opendev.org/08/673908/1/gate/openstack-tox-py36/31eb34c/ : SUCCESS in 3m 32s\n- openstack-tox-py37 https://logs.opendev.org/08/673908/1/gate/openstack-tox-py37/62d2021/ : SUCCESS in 3m 57s\n- openstack-tox-validate https://logs.opendev.org/08/673908/1/gate/openstack-tox-validate/3fa78c3/ : SUCCESS in 2m 36s\n- requirements-tox-babel https://logs.opendev.org/08/673908/1/gate/requirements-tox-babel/223cd26/ : SUCCESS in 3m 45s\n- requirements-tox-py27-check-uc https://logs.opendev.org/08/673908/1/gate/requirements-tox-py27-check-uc/b31850d/ : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://logs.opendev.org/08/673908/1/gate/requirements-tox-py36-check-uc/1e5a309/ : SUCCESS in 7m 50s\n- cross-cinder-py27 https://logs.opendev.org/08/673908/1/gate/cross-cinder-py27/2e8ab2b/ : SUCCESS in 11m 22s\n- cross-cinder-py36 https://logs.opendev.org/08/673908/1/gate/cross-cinder-py36/e34fbc8/ : SUCCESS in 8m 13s\n- cross-glance-py27 https://logs.opendev.org/08/673908/1/gate/cross-glance-py27/3c81bba/ : SUCCESS in 3m 54s\n- cross-glance-py36 https://logs.opendev.org/08/673908/1/gate/cross-glance-py36/f181918/ : SUCCESS in 4m 25s\n- cross-horizon-py27 https://logs.opendev.org/08/673908/1/gate/cross-horizon-py27/2b56053/ : SUCCESS in 10m 51s\n- cross-horizon-py36 https://logs.opendev.org/08/673908/1/gate/cross-horizon-py36/c5cdd40/ : SUCCESS in 8m 42s\n- cross-horizon-npm https://logs.opendev.org/08/673908/1/gate/cross-horizon-npm/202713c/ : SUCCESS in 6m 06s\n- cross-keystone-py27 https://logs.opendev.org/08/673908/1/gate/cross-keystone-py27/3b0f270/ : SUCCESS in 29m 25s\n- cross-keystone-py36 https://logs.opendev.org/08/673908/1/gate/cross-keystone-py36/7832551/ : SUCCESS in 29m 07s\n- cross-neutron-py27 https://logs.opendev.org/08/673908/1/gate/cross-neutron-py27/973f156/ : SUCCESS in 24m 20s\n- cross-neutron-py36 https://logs.opendev.org/08/673908/1/gate/cross-neutron-py36/7457c7e/ : SUCCESS in 21m 23s\n- cross-nova-py27 https://logs.opendev.org/08/673908/1/gate/cross-nova-py27/b93a876/ : SUCCESS in 19m 27s\n- cross-nova-py36 https://logs.opendev.org/08/673908/1/gate/cross-nova-py36/8d499c5/ : SUCCESS in 16m 52s\n- cross-nova-functional https://logs.opendev.org/08/673908/1/gate/cross-nova-functional/832f4ec/ : SUCCESS in 21m 33s\n- cross-osc-tox-docs https://logs.opendev.org/08/673908/1/gate/cross-osc-tox-docs/635e6de/html/ : SUCCESS in 3m 33s\n- cross-swift-py27 https://logs.opendev.org/08/673908/1/gate/cross-swift-py27/0e023b4/ : SUCCESS in 8m 08s\n- cross-swift-py36 https://logs.opendev.org/08/673908/1/gate/cross-swift-py36/fdd4c6f/ : SUCCESS in 7m 31s\n- requirements-integration https://logs.opendev.org/08/673908/1/gate/requirements-integration/5a4af24/ : SUCCESS in 56m 52s\n- neutron-grenade-multinode https://logs.opendev.org/08/673908/1/gate/neutron-grenade-multinode/fa66b54/ : SUCCESS in 1h 07m 49s\n- tempest-full https://logs.opendev.org/08/673908/1/gate/tempest-full/9130473/ : SUCCESS in 1h 35m 19s\n- neutron-grenade https://logs.opendev.org/08/673908/1/gate/neutron-grenade/50a72d8/ : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6d470a48237ec4e8b469bf442ec891e6d836cc95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 05:50:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c5dc68616e8974295e0fd7a2137b97819b2532c","revisions":{"3c5dc68616e8974295e0fd7a2137b97819b2532c":{"kind":"REWORK","_number":1,"created":"2019-07-31 20:36:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/673908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/673908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/673908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/673908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/673908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/673908/1"}}},"commit":{"parents":[{"commit":"b616150a321bc6cae9a57a58ba184af007fd31c0","subject":"Merge \"Exclude pyghmi 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b616150a321bc6cae9a57a58ba184af007fd31c0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-31 20:36:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-31 20:36:24.000000000","tz":0},"subject":"update constraint for os-traits to new release 0.16.0","message":"update constraint for os-traits to new release 0.16.0\n\nChange-Id: I7949ac17d6777b9ba5e403c135ca74be066d4553\nmeta:version: 0.16.0\nmeta:diff-start: -\nmeta:series: independent\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Commit: Eric Fried \u003copenstack@fried.cc\u003e\nmeta:release:Change-Id: Ia421ed500fbc15bf0088a8436ddeb5d8d1196256\nmeta:release:Code-Review+1: Chris Dent \u003ccdent@anticdent.org\u003e\nmeta:release:Code-Review+1: ya.wang \u003cwang.ya@99cloud.net\u003e\nmeta:release:Code-Review+2: Kendall Nelson \u003ckennelson11@gmail.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c5dc68616e8974295e0fd7a2137b97819b2532c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c5dc68616e8974295e0fd7a2137b97819b2532c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
