)]}'
{"id":"openstack%2Frequirements~674350","triplet_id":"openstack%2Frequirements~master~Ia13e555b07bf9cf677db5a6f5ae105fdec15b506","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia13e555b07bf9cf677db5a6f5ae105fdec15b506","subject":"update constraint for octavia-lib to new release 1.3.0","status":"MERGED","created":"2019-08-02 16:44:35.000000000","updated":"2019-08-03 05:33:23.000000000","submitted":"2019-08-03 05:33:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674350-1564810403699-e5490da1","meta_rev_id":"251d0d92edd6ee63834b7d171653eb31d9dd727f","_number":674350,"virtual_id_number":674350,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-03 05:33:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-02 18:55:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-02 18:55:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 18:55:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-03 05:33:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12ca544f3a7792528ae8b72c0d31b0a68082ff65","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-02 16:44:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8be640105386327c15c169d778f0108a2b6ed933","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 18:35:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/50/674350/1/check/openstack-tox-docs/f8a90e5/html/ : SUCCESS in 3m 08s\n- openstack-tox-pypy https://logs.opendev.org/50/674350/1/check/openstack-tox-pypy/022f64a/ : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/50/674350/1/check/openstack-tox-pep8/9620f57/ : SUCCESS in 5m 16s\n- openstack-tox-py27 https://logs.opendev.org/50/674350/1/check/openstack-tox-py27/4482339/ : SUCCESS in 4m 40s\n- openstack-tox-py36 https://logs.opendev.org/50/674350/1/check/openstack-tox-py36/2d77ee2/ : SUCCESS in 3m 51s\n- openstack-tox-py37 https://logs.opendev.org/50/674350/1/check/openstack-tox-py37/f4010ce/ : SUCCESS in 4m 07s\n- openstack-tox-validate https://logs.opendev.org/50/674350/1/check/openstack-tox-validate/e3de4f4/ : SUCCESS in 2m 50s\n- requirements-tox-babel https://logs.opendev.org/50/674350/1/check/requirements-tox-babel/a8a0bf3/ : SUCCESS in 3m 55s\n- requirements-tox-py27-check-uc https://logs.opendev.org/50/674350/1/check/requirements-tox-py27-check-uc/cc11cfe/ : SUCCESS in 7m 01s\n- requirements-tox-py36-check-uc https://logs.opendev.org/50/674350/1/check/requirements-tox-py36-check-uc/6816dca/ : SUCCESS in 6m 56s\n- cross-cinder-py27 https://logs.opendev.org/50/674350/1/check/cross-cinder-py27/d7c4811/ : SUCCESS in 8m 42s\n- cross-cinder-py36 https://logs.opendev.org/50/674350/1/check/cross-cinder-py36/f1469ac/ : SUCCESS in 9m 24s\n- cross-glance-py27 https://logs.opendev.org/50/674350/1/check/cross-glance-py27/fb6ed64/ : SUCCESS in 4m 35s\n- cross-glance-py36 https://logs.opendev.org/50/674350/1/check/cross-glance-py36/7077581/ : SUCCESS in 5m 13s\n- cross-horizon-py27 https://logs.opendev.org/50/674350/1/check/cross-horizon-py27/81279bd/ : SUCCESS in 11m 48s\n- cross-horizon-py36 https://logs.opendev.org/50/674350/1/check/cross-horizon-py36/8bcc03a/ : SUCCESS in 10m 23s\n- cross-horizon-npm https://logs.opendev.org/50/674350/1/check/cross-horizon-npm/5549001/ : SUCCESS in 9m 05s\n- cross-keystone-py27 https://logs.opendev.org/50/674350/1/check/cross-keystone-py27/9044595/ : SUCCESS in 25m 46s\n- cross-keystone-py36 https://logs.opendev.org/50/674350/1/check/cross-keystone-py36/83ff6cc/ : SUCCESS in 33m 32s\n- cross-neutron-py27 https://logs.opendev.org/50/674350/1/check/cross-neutron-py27/aaa3336/ : SUCCESS in 28m 56s\n- cross-neutron-py36 https://logs.opendev.org/50/674350/1/check/cross-neutron-py36/5c225ba/ : SUCCESS in 23m 22s\n- cross-nova-py27 https://logs.opendev.org/50/674350/1/check/cross-nova-py27/d89b346/ : SUCCESS in 17m 59s\n- cross-nova-py36 https://logs.opendev.org/50/674350/1/check/cross-nova-py36/a8fe151/ : SUCCESS in 19m 23s\n- cross-nova-functional https://logs.opendev.org/50/674350/1/check/cross-nova-functional/c59fe64/ : SUCCESS in 20m 38s\n- cross-osc-tox-docs https://logs.opendev.org/50/674350/1/check/cross-osc-tox-docs/6b486f7/html/ : SUCCESS in 5m 00s\n- cross-swift-py27 https://logs.opendev.org/50/674350/1/check/cross-swift-py27/cb3eb2c/ : SUCCESS in 7m 56s\n- cross-swift-py36 https://logs.opendev.org/50/674350/1/check/cross-swift-py36/6b36e8e/ : SUCCESS in 8m 25s\n- requirements-integration https://logs.opendev.org/50/674350/1/check/requirements-integration/782424b/ : SUCCESS in 55m 42s\n- neutron-grenade-multinode https://logs.opendev.org/50/674350/1/check/neutron-grenade-multinode/98dac23/ : SUCCESS in 1h 11m 06s\n- tempest-full https://logs.opendev.org/50/674350/1/check/tempest-full/5d3a234/ : SUCCESS in 1h 41m 33s\n- neutron-grenade https://logs.opendev.org/50/674350/1/check/neutron-grenade/b4782c3/ : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5898297ef5a603dd0e07b011e946e2c813124a79","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-02 18:55:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1bb16b297437191245af050f8ba1c563e039fc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 18:55:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f9b1af89fc204b62005d9626c2de29a600e6461","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 20:34:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/50/674350/1/gate/openstack-tox-docs/5e2b393/html/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://logs.opendev.org/50/674350/1/gate/openstack-tox-pep8/ba7d886/ : SUCCESS in 3m 38s\n- openstack-tox-py27 https://logs.opendev.org/50/674350/1/gate/openstack-tox-py27/58bf8db/ : SUCCESS in 5m 27s\n- openstack-tox-py36 https://logs.opendev.org/50/674350/1/gate/openstack-tox-py36/8942643/ : SUCCESS in 3m 56s\n- openstack-tox-py37 https://logs.opendev.org/50/674350/1/gate/openstack-tox-py37/a685cac/ : SUCCESS in 5m 31s\n- openstack-tox-validate https://logs.opendev.org/50/674350/1/gate/openstack-tox-validate/a1283a0/ : SUCCESS in 3m 47s\n- requirements-tox-babel https://logs.opendev.org/50/674350/1/gate/requirements-tox-babel/42450a1/ : SUCCESS in 4m 24s\n- requirements-tox-py27-check-uc https://logs.opendev.org/50/674350/1/gate/requirements-tox-py27-check-uc/9130bd7/ : SUCCESS in 10m 23s\n- requirements-tox-py36-check-uc https://logs.opendev.org/50/674350/1/gate/requirements-tox-py36-check-uc/ae5540b/ : SUCCESS in 7m 32s\n- cross-cinder-py27 https://logs.opendev.org/50/674350/1/gate/cross-cinder-py27/da1f844/ : SUCCESS in 9m 42s\n- cross-cinder-py36 https://logs.opendev.org/50/674350/1/gate/cross-cinder-py36/d99ed03/ : SUCCESS in 10m 09s\n- cross-glance-py27 https://logs.opendev.org/50/674350/1/gate/cross-glance-py27/8b6b6c8/ : SUCCESS in 4m 20s\n- cross-glance-py36 https://logs.opendev.org/50/674350/1/gate/cross-glance-py36/ab61322/ : SUCCESS in 5m 38s\n- cross-horizon-py27 https://logs.opendev.org/50/674350/1/gate/cross-horizon-py27/f4af760/ : SUCCESS in 10m 01s\n- cross-horizon-py36 https://logs.opendev.org/50/674350/1/gate/cross-horizon-py36/0eec842/ : SUCCESS in 9m 27s\n- cross-horizon-npm https://logs.opendev.org/50/674350/1/gate/cross-horizon-npm/76c7223/ : SUCCESS in 8m 40s\n- cross-keystone-py27 https://logs.opendev.org/50/674350/1/gate/cross-keystone-py27/a9b63d6/ : TIMED_OUT in 41m 20s\n- cross-keystone-py36 https://logs.opendev.org/50/674350/1/gate/cross-keystone-py36/6feaff4/ : TIMED_OUT in 41m 04s\n- cross-neutron-py27 https://logs.opendev.org/50/674350/1/gate/cross-neutron-py27/5b27dc1/ : SUCCESS in 23m 13s\n- cross-neutron-py36 https://logs.opendev.org/50/674350/1/gate/cross-neutron-py36/4fcc103/ : SUCCESS in 26m 47s\n- cross-nova-py27 https://logs.opendev.org/50/674350/1/gate/cross-nova-py27/e15e855/ : SUCCESS in 17m 09s\n- cross-nova-py36 https://logs.opendev.org/50/674350/1/gate/cross-nova-py36/eb78b08/ : SUCCESS in 17m 47s\n- cross-nova-functional https://logs.opendev.org/50/674350/1/gate/cross-nova-functional/e3e266c/ : SUCCESS in 22m 56s\n- cross-osc-tox-docs https://logs.opendev.org/50/674350/1/gate/cross-osc-tox-docs/8493db2/html/ : SUCCESS in 4m 51s\n- cross-swift-py27 https://logs.opendev.org/50/674350/1/gate/cross-swift-py27/4c32e03/ : SUCCESS in 11m 01s\n- cross-swift-py36 https://logs.opendev.org/50/674350/1/gate/cross-swift-py36/0aa1b28/ : SUCCESS in 10m 47s\n- requirements-integration https://logs.opendev.org/50/674350/1/gate/requirements-integration/695368c/ : SUCCESS in 1h 00m 02s\n- neutron-grenade-multinode https://logs.opendev.org/50/674350/1/gate/neutron-grenade-multinode/bd9a24a/ : SUCCESS in 1h 07m 36s\n- tempest-full https://logs.opendev.org/50/674350/1/gate/tempest-full/0851329/ : SUCCESS in 1h 37m 04s\n- neutron-grenade https://logs.opendev.org/50/674350/1/gate/neutron-grenade/b995119/ : FAILURE in 1h 00m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9aa606114561be033c0f409e3f5565fbf3a990dd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-02 22:01:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dc3bb45e863bd3780b5d5e878fde278b76ada3ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 23:42:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/50/674350/1/check/openstack-tox-docs/eec70b4/html/ : SUCCESS in 3m 12s\n- openstack-tox-pypy https://logs.opendev.org/50/674350/1/check/openstack-tox-pypy/927fb4f/ : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/50/674350/1/check/openstack-tox-pep8/dd774b6/ : SUCCESS in 2m 59s\n- openstack-tox-py27 https://logs.opendev.org/50/674350/1/check/openstack-tox-py27/6702311/ : SUCCESS in 3m 45s\n- openstack-tox-py36 https://logs.opendev.org/50/674350/1/check/openstack-tox-py36/fe1b5c4/ : SUCCESS in 3m 12s\n- openstack-tox-py37 https://logs.opendev.org/50/674350/1/check/openstack-tox-py37/e74a44f/ : SUCCESS in 3m 28s\n- openstack-tox-validate https://logs.opendev.org/50/674350/1/check/openstack-tox-validate/f15a00c/ : SUCCESS in 3m 13s\n- requirements-tox-babel https://logs.opendev.org/50/674350/1/check/requirements-tox-babel/ac0fe28/ : SUCCESS in 3m 02s\n- requirements-tox-py27-check-uc https://logs.opendev.org/50/674350/1/check/requirements-tox-py27-check-uc/ed90ef1/ : SUCCESS in 6m 33s\n- requirements-tox-py36-check-uc https://logs.opendev.org/50/674350/1/check/requirements-tox-py36-check-uc/ab8d69e/ : SUCCESS in 15m 24s\n- cross-cinder-py27 https://logs.opendev.org/50/674350/1/check/cross-cinder-py27/573e453/ : SUCCESS in 8m 49s\n- cross-cinder-py36 https://logs.opendev.org/50/674350/1/check/cross-cinder-py36/83e5a59/ : SUCCESS in 7m 29s\n- cross-glance-py27 https://logs.opendev.org/50/674350/1/check/cross-glance-py27/7355751/ : SUCCESS in 4m 36s\n- cross-glance-py36 https://logs.opendev.org/50/674350/1/check/cross-glance-py36/739c4ad/ : SUCCESS in 4m 12s\n- cross-horizon-py27 https://logs.opendev.org/50/674350/1/check/cross-horizon-py27/b05754e/ : SUCCESS in 9m 59s\n- cross-horizon-py36 https://logs.opendev.org/50/674350/1/check/cross-horizon-py36/8ab3b08/ : SUCCESS in 11m 29s\n- cross-horizon-npm https://logs.opendev.org/50/674350/1/check/cross-horizon-npm/c58c8e4/ : SUCCESS in 8m 56s\n- cross-keystone-py27 https://logs.opendev.org/50/674350/1/check/cross-keystone-py27/4ddee8c/ : SUCCESS in 31m 48s\n- cross-keystone-py36 https://logs.opendev.org/50/674350/1/check/cross-keystone-py36/2dc8fd0/ : SUCCESS in 29m 57s\n- cross-neutron-py27 https://logs.opendev.org/50/674350/1/check/cross-neutron-py27/438363b/ : SUCCESS in 23m 39s\n- cross-neutron-py36 https://logs.opendev.org/50/674350/1/check/cross-neutron-py36/9ddd3d4/ : SUCCESS in 23m 59s\n- cross-nova-py27 https://logs.opendev.org/50/674350/1/check/cross-nova-py27/9254f53/ : SUCCESS in 17m 29s\n- cross-nova-py36 https://logs.opendev.org/50/674350/1/check/cross-nova-py36/1f9cf9d/ : SUCCESS in 13m 59s\n- cross-nova-functional https://logs.opendev.org/50/674350/1/check/cross-nova-functional/8ec44bb/ : SUCCESS in 20m 49s\n- cross-osc-tox-docs https://logs.opendev.org/50/674350/1/check/cross-osc-tox-docs/aafaf65/html/ : SUCCESS in 3m 39s\n- cross-swift-py27 https://logs.opendev.org/50/674350/1/check/cross-swift-py27/bc3451f/ : SUCCESS in 8m 25s\n- cross-swift-py36 https://logs.opendev.org/50/674350/1/check/cross-swift-py36/08fd972/ : SUCCESS in 7m 50s\n- requirements-integration https://logs.opendev.org/50/674350/1/check/requirements-integration/ae0b310/ : SUCCESS in 59m 50s\n- neutron-grenade-multinode https://logs.opendev.org/50/674350/1/check/neutron-grenade-multinode/6fc82ca/ : SUCCESS in 1h 11m 10s\n- tempest-full https://logs.opendev.org/50/674350/1/check/tempest-full/e60a0c3/ : SUCCESS in 1h 39m 32s\n- neutron-grenade https://logs.opendev.org/50/674350/1/check/neutron-grenade/955bf89/ : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e6d5c6073e457d585eb8c602358c25b594f9dc85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 23:43:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d066262afa407c70befa20fb336be2b36a990cae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 01:53:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/50/674350/1/gate/openstack-tox-docs/7ffddfb/html/ : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://logs.opendev.org/50/674350/1/gate/openstack-tox-pep8/74c5a3c/ : SUCCESS in 3m 13s\n- openstack-tox-py27 https://logs.opendev.org/50/674350/1/gate/openstack-tox-py27/10455c6/ : SUCCESS in 2m 53s\n- openstack-tox-py36 https://logs.opendev.org/50/674350/1/gate/openstack-tox-py36/fb54ff5/ : SUCCESS in 3m 10s\n- openstack-tox-py37 https://logs.opendev.org/50/674350/1/gate/openstack-tox-py37/0526088/ : SUCCESS in 5m 11s\n- openstack-tox-validate https://logs.opendev.org/50/674350/1/gate/openstack-tox-validate/47c3673/ : SUCCESS in 2m 47s\n- requirements-tox-babel https://logs.opendev.org/50/674350/1/gate/requirements-tox-babel/6e3ae97/ : SUCCESS in 2m 59s\n- requirements-tox-py27-check-uc https://logs.opendev.org/50/674350/1/gate/requirements-tox-py27-check-uc/e8cf7f6/ : SUCCESS in 6m 12s\n- requirements-tox-py36-check-uc https://logs.opendev.org/50/674350/1/gate/requirements-tox-py36-check-uc/de2e6bf/ : SUCCESS in 5m 53s\n- cross-cinder-py27 https://logs.opendev.org/50/674350/1/gate/cross-cinder-py27/6bdb9fe/ : SUCCESS in 12m 18s\n- cross-cinder-py36 https://logs.opendev.org/50/674350/1/gate/cross-cinder-py36/d8c0a58/ : SUCCESS in 8m 11s\n- cross-glance-py27 https://logs.opendev.org/50/674350/1/gate/cross-glance-py27/2315f76/ : SUCCESS in 4m 45s\n- cross-glance-py36 https://logs.opendev.org/50/674350/1/gate/cross-glance-py36/6267e08/ : SUCCESS in 5m 44s\n- cross-horizon-py27 https://logs.opendev.org/50/674350/1/gate/cross-horizon-py27/5984539/ : SUCCESS in 9m 53s\n- cross-horizon-py36 https://logs.opendev.org/50/674350/1/gate/cross-horizon-py36/c12a7ff/ : SUCCESS in 8m 33s\n- cross-horizon-npm https://logs.opendev.org/50/674350/1/gate/cross-horizon-npm/636ecb5/ : SUCCESS in 5m 35s\n- cross-keystone-py27 https://logs.opendev.org/50/674350/1/gate/cross-keystone-py27/ce609e6/ : TIMED_OUT in 40m 52s\n- cross-keystone-py36 https://logs.opendev.org/50/674350/1/gate/cross-keystone-py36/6bc0a85/ : TIMED_OUT in 40m 54s\n- cross-neutron-py27 https://logs.opendev.org/50/674350/1/gate/cross-neutron-py27/1559388/ : SUCCESS in 20m 52s\n- cross-neutron-py36 https://logs.opendev.org/50/674350/1/gate/cross-neutron-py36/ba55b5b/ : SUCCESS in 23m 11s\n- cross-nova-py27 https://logs.opendev.org/50/674350/1/gate/cross-nova-py27/c46a15e/ : SUCCESS in 28m 55s\n- cross-nova-py36 https://logs.opendev.org/50/674350/1/gate/cross-nova-py36/4353502/ : SUCCESS in 16m 45s\n- cross-nova-functional https://logs.opendev.org/50/674350/1/gate/cross-nova-functional/50ef082/ : SUCCESS in 21m 03s\n- cross-osc-tox-docs https://logs.opendev.org/50/674350/1/gate/cross-osc-tox-docs/8752fb1/html/ : SUCCESS in 3m 58s\n- cross-swift-py27 https://logs.opendev.org/50/674350/1/gate/cross-swift-py27/58f964c/ : SUCCESS in 7m 37s\n- cross-swift-py36 https://logs.opendev.org/50/674350/1/gate/cross-swift-py36/837fb8e/ : SUCCESS in 7m 12s\n- requirements-integration https://logs.opendev.org/50/674350/1/gate/requirements-integration/880b731/ : SUCCESS in 53m 20s\n- neutron-grenade-multinode https://logs.opendev.org/50/674350/1/gate/neutron-grenade-multinode/a6d2706/ : SUCCESS in 1h 08m 48s\n- tempest-full https://logs.opendev.org/50/674350/1/gate/tempest-full/8232676/ : TIMED_OUT in 2h 08m 58s\n- neutron-grenade https://logs.opendev.org/50/674350/1/gate/neutron-grenade/40c598a/ : FAILURE in 53m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0a3cadc1f1263e796aa65b836df3d1883ab9f4bc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-03 02:27:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fc665418cf737d663d1625ab0408bf290b591768","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 04:01:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/50/674350/1/check/openstack-tox-docs/7c1051f/html/ : SUCCESS in 2m 28s\n- openstack-tox-pypy https://logs.opendev.org/50/674350/1/check/openstack-tox-pypy/2bf40af/ : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/50/674350/1/check/openstack-tox-pep8/478838d/ : SUCCESS in 3m 47s\n- openstack-tox-py27 https://logs.opendev.org/50/674350/1/check/openstack-tox-py27/6bdc8cb/ : SUCCESS in 3m 51s\n- openstack-tox-py36 https://logs.opendev.org/50/674350/1/check/openstack-tox-py36/941b454/ : SUCCESS in 3m 27s\n- openstack-tox-py37 https://logs.opendev.org/50/674350/1/check/openstack-tox-py37/520ca74/ : SUCCESS in 3m 08s\n- openstack-tox-validate https://logs.opendev.org/50/674350/1/check/openstack-tox-validate/ce2be7a/ : SUCCESS in 3m 26s\n- requirements-tox-babel https://logs.opendev.org/50/674350/1/check/requirements-tox-babel/2dcb915/ : SUCCESS in 2m 29s\n- requirements-tox-py27-check-uc https://logs.opendev.org/50/674350/1/check/requirements-tox-py27-check-uc/2898db0/ : SUCCESS in 6m 27s\n- requirements-tox-py36-check-uc https://logs.opendev.org/50/674350/1/check/requirements-tox-py36-check-uc/31faffa/ : SUCCESS in 6m 14s\n- cross-cinder-py27 https://logs.opendev.org/50/674350/1/check/cross-cinder-py27/7b16c02/ : SUCCESS in 9m 38s\n- cross-cinder-py36 https://logs.opendev.org/50/674350/1/check/cross-cinder-py36/5b65f46/ : SUCCESS in 10m 18s\n- cross-glance-py27 https://logs.opendev.org/50/674350/1/check/cross-glance-py27/dda0419/ : SUCCESS in 3m 55s\n- cross-glance-py36 https://logs.opendev.org/50/674350/1/check/cross-glance-py36/6dc831f/ : SUCCESS in 4m 10s\n- cross-horizon-py27 https://logs.opendev.org/50/674350/1/check/cross-horizon-py27/78d4150/ : SUCCESS in 8m 57s\n- cross-horizon-py36 https://logs.opendev.org/50/674350/1/check/cross-horizon-py36/ee33332/ : SUCCESS in 7m 28s\n- cross-horizon-npm https://logs.opendev.org/50/674350/1/check/cross-horizon-npm/feb5317/ : SUCCESS in 5m 30s\n- cross-keystone-py27 https://logs.opendev.org/50/674350/1/check/cross-keystone-py27/125819b/ : SUCCESS in 38m 15s\n- cross-keystone-py36 https://logs.opendev.org/50/674350/1/check/cross-keystone-py36/83974e1/ : SUCCESS in 34m 26s\n- cross-neutron-py27 https://logs.opendev.org/50/674350/1/check/cross-neutron-py27/98435d5/ : SUCCESS in 22m 35s\n- cross-neutron-py36 https://logs.opendev.org/50/674350/1/check/cross-neutron-py36/e2066c0/ : SUCCESS in 23m 08s\n- cross-nova-py27 https://logs.opendev.org/50/674350/1/check/cross-nova-py27/3585419/ : SUCCESS in 14m 39s\n- cross-nova-py36 https://logs.opendev.org/50/674350/1/check/cross-nova-py36/058c411/ : SUCCESS in 16m 43s\n- cross-nova-functional https://logs.opendev.org/50/674350/1/check/cross-nova-functional/bba0b83/ : SUCCESS in 24m 45s\n- cross-osc-tox-docs https://logs.opendev.org/50/674350/1/check/cross-osc-tox-docs/012de6f/html/ : SUCCESS in 3m 42s\n- cross-swift-py27 https://logs.opendev.org/50/674350/1/check/cross-swift-py27/db9db69/ : SUCCESS in 9m 26s\n- cross-swift-py36 https://logs.opendev.org/50/674350/1/check/cross-swift-py36/9fad731/ : SUCCESS in 7m 22s\n- requirements-integration https://logs.opendev.org/50/674350/1/check/requirements-integration/095e380/ : SUCCESS in 47m 14s\n- neutron-grenade-multinode https://logs.opendev.org/50/674350/1/check/neutron-grenade-multinode/021876b/ : SUCCESS in 1h 08m 43s\n- tempest-full https://logs.opendev.org/50/674350/1/check/tempest-full/c6c6826/ : SUCCESS in 1h 31m 04s\n- neutron-grenade https://logs.opendev.org/50/674350/1/check/neutron-grenade/b363346/ : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3d784e8260d48c0a3dddd915389290c0ab66b414","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 04:01:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac9b262b4293c029d56e9ed7e370222ffa073709","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 05:33:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/50/674350/1/gate/openstack-tox-docs/2490c2e/html/ : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://logs.opendev.org/50/674350/1/gate/openstack-tox-pep8/bba451d/ : SUCCESS in 4m 49s\n- openstack-tox-py27 https://logs.opendev.org/50/674350/1/gate/openstack-tox-py27/ac6035a/ : SUCCESS in 5m 06s\n- openstack-tox-py36 https://logs.opendev.org/50/674350/1/gate/openstack-tox-py36/91cab58/ : SUCCESS in 3m 17s\n- openstack-tox-py37 https://logs.opendev.org/50/674350/1/gate/openstack-tox-py37/cd5c34d/ : SUCCESS in 5m 21s\n- openstack-tox-validate https://logs.opendev.org/50/674350/1/gate/openstack-tox-validate/249f716/ : SUCCESS in 3m 00s\n- requirements-tox-babel https://logs.opendev.org/50/674350/1/gate/requirements-tox-babel/092c519/ : SUCCESS in 4m 47s\n- requirements-tox-py27-check-uc https://logs.opendev.org/50/674350/1/gate/requirements-tox-py27-check-uc/e6aa84f/ : SUCCESS in 6m 14s\n- requirements-tox-py36-check-uc https://logs.opendev.org/50/674350/1/gate/requirements-tox-py36-check-uc/f69c04f/ : SUCCESS in 7m 40s\n- cross-cinder-py27 https://logs.opendev.org/50/674350/1/gate/cross-cinder-py27/3bf9716/ : SUCCESS in 7m 55s\n- cross-cinder-py36 https://logs.opendev.org/50/674350/1/gate/cross-cinder-py36/4f3e260/ : SUCCESS in 7m 12s\n- cross-glance-py27 https://logs.opendev.org/50/674350/1/gate/cross-glance-py27/74b3b9b/ : SUCCESS in 4m 08s\n- cross-glance-py36 https://logs.opendev.org/50/674350/1/gate/cross-glance-py36/025ef14/ : SUCCESS in 3m 58s\n- cross-horizon-py27 https://logs.opendev.org/50/674350/1/gate/cross-horizon-py27/292d1c5/ : SUCCESS in 9m 10s\n- cross-horizon-py36 https://logs.opendev.org/50/674350/1/gate/cross-horizon-py36/d8c6776/ : SUCCESS in 9m 36s\n- cross-horizon-npm https://logs.opendev.org/50/674350/1/gate/cross-horizon-npm/a1f4c2b/ : SUCCESS in 5m 50s\n- cross-keystone-py27 https://logs.opendev.org/50/674350/1/gate/cross-keystone-py27/dde512e/ : SUCCESS in 27m 29s\n- cross-keystone-py36 https://logs.opendev.org/50/674350/1/gate/cross-keystone-py36/943f68b/ : SUCCESS in 26m 34s\n- cross-neutron-py27 https://logs.opendev.org/50/674350/1/gate/cross-neutron-py27/85e5b2b/ : SUCCESS in 22m 11s\n- cross-neutron-py36 https://logs.opendev.org/50/674350/1/gate/cross-neutron-py36/9d5ef80/ : SUCCESS in 16m 52s\n- cross-nova-py27 https://logs.opendev.org/50/674350/1/gate/cross-nova-py27/fad8eaa/ : SUCCESS in 16m 27s\n- cross-nova-py36 https://logs.opendev.org/50/674350/1/gate/cross-nova-py36/354165c/ : SUCCESS in 16m 07s\n- cross-nova-functional https://logs.opendev.org/50/674350/1/gate/cross-nova-functional/0cf24d1/ : SUCCESS in 19m 32s\n- cross-osc-tox-docs https://logs.opendev.org/50/674350/1/gate/cross-osc-tox-docs/892aa15/html/ : SUCCESS in 3m 27s\n- cross-swift-py27 https://logs.opendev.org/50/674350/1/gate/cross-swift-py27/2c96af0/ : SUCCESS in 11m 17s\n- cross-swift-py36 https://logs.opendev.org/50/674350/1/gate/cross-swift-py36/0b633f1/ : SUCCESS in 11m 38s\n- requirements-integration https://logs.opendev.org/50/674350/1/gate/requirements-integration/bc6b647/ : SUCCESS in 46m 14s\n- neutron-grenade-multinode https://logs.opendev.org/50/674350/1/gate/neutron-grenade-multinode/183a344/ : SUCCESS in 59m 02s\n- tempest-full https://logs.opendev.org/50/674350/1/gate/tempest-full/8be5153/ : SUCCESS in 1h 28m 59s\n- neutron-grenade https://logs.opendev.org/50/674350/1/gate/neutron-grenade/9a25019/ : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":1},{"id":"47d7d017f67347a803a02e7154951ffe482cdf98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 05:33:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1de770f40830febe5b8ead9d0958a222f2348c28","revisions":{"1de770f40830febe5b8ead9d0958a222f2348c28":{"kind":"REWORK","_number":1,"created":"2019-08-02 16:44:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/674350/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/50/674350/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/674350/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/674350/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/50/674350/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/50/674350/1"}}},"commit":{"parents":[{"commit":"e021ce6144933d0443eb6342ebe8accd9066daa9","subject":"update constraint for openstacksdk to new release 0.32.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e021ce6144933d0443eb6342ebe8accd9066daa9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-02 16:44:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-02 16:44:34.000000000","tz":0},"subject":"update constraint for octavia-lib to new release 1.3.0","message":"update constraint for octavia-lib to new release 1.3.0\n\nChange-Id: Ia13e555b07bf9cf677db5a6f5ae105fdec15b506\nmeta:version: 1.3.0\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta:release:Change-Id: I8310a6379a2caa156ed225d1e07be68536c56863\nmeta:release:Code-Review+1: Adam Harwell \u003cflux.adam@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Code-Review+1: Carlos Goncalves \u003ccgoncalves@redhat.com\u003e\nmeta:release:Code-Review+1: Maciej Józefczyk \u003cmjozefcz@redhat.com\u003e\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1de770f40830febe5b8ead9d0958a222f2348c28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1de770f40830febe5b8ead9d0958a222f2348c28"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
