)]}'
{"id":"openstack%2Frequirements~674560","triplet_id":"openstack%2Frequirements~master~I8f7e79c80db1d51a9705e808ec80288c440945ab","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8f7e79c80db1d51a9705e808ec80288c440945ab","subject":"update constraint for oslo.middleware to new release 3.38.1","status":"MERGED","created":"2019-08-05 09:38:59.000000000","updated":"2019-08-05 15:31:44.000000000","submitted":"2019-08-05 15:31:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674560-1565019104937-efd57ca1","meta_rev_id":"45514411931965d79e6307876bcf7b090ae41add","_number":674560,"virtual_id_number":674560,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2019-08-05 15:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2019-08-05 13:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2019-08-05 13:35:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 13:35:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2019-08-05 15:31:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03b38a013172a4199c80119617f34f8619353517","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-05 09:38:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"033e2ed28c3a62ef599141efedea107ccf4b4b04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 11:12:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/60/674560/1/check/openstack-tox-docs/5b5c1d8/html/ : SUCCESS in 4m 29s\n- openstack-tox-pypy https://logs.opendev.org/60/674560/1/check/openstack-tox-pypy/81b0e00/ : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/60/674560/1/check/openstack-tox-pep8/460cd4e/ : SUCCESS in 4m 44s\n- openstack-tox-py27 https://logs.opendev.org/60/674560/1/check/openstack-tox-py27/d0f52ce/ : SUCCESS in 4m 29s\n- openstack-tox-py36 https://logs.opendev.org/60/674560/1/check/openstack-tox-py36/82d84f0/ : SUCCESS in 4m 46s\n- openstack-tox-py37 https://logs.opendev.org/60/674560/1/check/openstack-tox-py37/f8fec7c/ : SUCCESS in 5m 05s\n- openstack-tox-validate https://logs.opendev.org/60/674560/1/check/openstack-tox-validate/ed8c8d2/ : SUCCESS in 4m 23s\n- requirements-tox-babel https://logs.opendev.org/60/674560/1/check/requirements-tox-babel/06fb557/ : SUCCESS in 4m 30s\n- requirements-tox-py27-check-uc https://logs.opendev.org/60/674560/1/check/requirements-tox-py27-check-uc/5048821/ : SUCCESS in 17m 41s\n- requirements-tox-py36-check-uc https://logs.opendev.org/60/674560/1/check/requirements-tox-py36-check-uc/a7697cf/ : SUCCESS in 15m 44s\n- cross-cinder-py27 https://logs.opendev.org/60/674560/1/check/cross-cinder-py27/f80eca5/ : SUCCESS in 12m 09s\n- cross-cinder-py36 https://logs.opendev.org/60/674560/1/check/cross-cinder-py36/b896186/ : SUCCESS in 12m 15s\n- cross-glance-py27 https://logs.opendev.org/60/674560/1/check/cross-glance-py27/690c461/ : SUCCESS in 6m 15s\n- cross-glance-py36 https://logs.opendev.org/60/674560/1/check/cross-glance-py36/6145f57/ : SUCCESS in 8m 24s\n- cross-horizon-py27 https://logs.opendev.org/60/674560/1/check/cross-horizon-py27/14b0a1c/ : SUCCESS in 10m 23s\n- cross-horizon-py36 https://logs.opendev.org/60/674560/1/check/cross-horizon-py36/db34113/ : SUCCESS in 9m 10s\n- cross-horizon-npm https://logs.opendev.org/60/674560/1/check/cross-horizon-npm/d624da0/ : SUCCESS in 6m 34s\n- cross-keystone-py27 https://logs.opendev.org/60/674560/1/check/cross-keystone-py27/ccd4e84/ : SUCCESS in 25m 38s\n- cross-keystone-py36 https://logs.opendev.org/60/674560/1/check/cross-keystone-py36/f23ad0e/ : SUCCESS in 29m 02s\n- cross-neutron-py27 https://logs.opendev.org/60/674560/1/check/cross-neutron-py27/3f172dc/ : SUCCESS in 25m 31s\n- cross-neutron-py36 https://logs.opendev.org/60/674560/1/check/cross-neutron-py36/6c3e382/ : SUCCESS in 20m 56s\n- cross-nova-py27 https://logs.opendev.org/60/674560/1/check/cross-nova-py27/eae0f6b/ : SUCCESS in 19m 01s\n- cross-nova-py36 https://logs.opendev.org/60/674560/1/check/cross-nova-py36/7364329/ : SUCCESS in 17m 17s\n- cross-nova-functional https://logs.opendev.org/60/674560/1/check/cross-nova-functional/e674d12/ : SUCCESS in 23m 09s\n- cross-osc-tox-docs https://logs.opendev.org/60/674560/1/check/cross-osc-tox-docs/869735c/html/ : SUCCESS in 5m 39s\n- cross-swift-py27 https://logs.opendev.org/60/674560/1/check/cross-swift-py27/ad7d3da/ : SUCCESS in 17m 30s\n- cross-swift-py36 https://logs.opendev.org/60/674560/1/check/cross-swift-py36/50c95ef/ : SUCCESS in 9m 32s\n- requirements-integration https://logs.opendev.org/60/674560/1/check/requirements-integration/79b2ec5/ : SUCCESS in 58m 54s\n- neutron-grenade-multinode https://logs.opendev.org/60/674560/1/check/neutron-grenade-multinode/9968d9e/ : SUCCESS in 1h 05m 59s\n- tempest-full https://logs.opendev.org/60/674560/1/check/tempest-full/3eccf02/ : SUCCESS in 1h 26m 19s\n- neutron-grenade https://logs.opendev.org/60/674560/1/check/neutron-grenade/21aac90/ : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e965e640cc6251dc5f0df1ca780e191d311f5529","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2019-08-05 13:35:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0aec43f5aea8832779c56ce369e76d5b2ab1f6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:36:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e58706ff057ef05ae8c1ae4d790d23fb84fc4e3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:31:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f61646d029570544ab213329007b3d8f31c42404","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:31:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/60/674560/1/gate/openstack-tox-docs/eba5f0f/html/ : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://logs.opendev.org/60/674560/1/gate/openstack-tox-pep8/a1bba5f/ : SUCCESS in 4m 58s\n- openstack-tox-py27 https://logs.opendev.org/60/674560/1/gate/openstack-tox-py27/5d25e80/ : SUCCESS in 3m 37s\n- openstack-tox-py36 https://logs.opendev.org/60/674560/1/gate/openstack-tox-py36/85f69be/ : SUCCESS in 6m 02s\n- openstack-tox-py37 https://logs.opendev.org/60/674560/1/gate/openstack-tox-py37/e56b005/ : SUCCESS in 5m 35s\n- openstack-tox-validate https://logs.opendev.org/60/674560/1/gate/openstack-tox-validate/3eb7600/ : SUCCESS in 4m 45s\n- requirements-tox-babel https://logs.opendev.org/60/674560/1/gate/requirements-tox-babel/5406c44/ : SUCCESS in 4m 32s\n- requirements-tox-py27-check-uc https://logs.opendev.org/60/674560/1/gate/requirements-tox-py27-check-uc/7eb354f/ : SUCCESS in 6m 51s\n- requirements-tox-py36-check-uc https://logs.opendev.org/60/674560/1/gate/requirements-tox-py36-check-uc/b028b37/ : SUCCESS in 9m 44s\n- cross-cinder-py27 https://logs.opendev.org/60/674560/1/gate/cross-cinder-py27/e0bbcdf/ : SUCCESS in 9m 04s\n- cross-cinder-py36 https://logs.opendev.org/60/674560/1/gate/cross-cinder-py36/3c5a9ad/ : SUCCESS in 10m 06s\n- cross-glance-py27 https://logs.opendev.org/60/674560/1/gate/cross-glance-py27/ad0cb26/ : SUCCESS in 5m 37s\n- cross-glance-py36 https://logs.opendev.org/60/674560/1/gate/cross-glance-py36/19b6771/ : SUCCESS in 6m 12s\n- cross-horizon-py27 https://logs.opendev.org/60/674560/1/gate/cross-horizon-py27/ceeb0a9/ : SUCCESS in 13m 47s\n- cross-horizon-py36 https://logs.opendev.org/60/674560/1/gate/cross-horizon-py36/1e8094b/ : SUCCESS in 11m 09s\n- cross-horizon-npm https://logs.opendev.org/60/674560/1/gate/cross-horizon-npm/a66291e/ : SUCCESS in 9m 23s\n- cross-keystone-py27 https://logs.opendev.org/60/674560/1/gate/cross-keystone-py27/ff2dae0/ : SUCCESS in 24m 29s\n- cross-keystone-py36 https://logs.opendev.org/60/674560/1/gate/cross-keystone-py36/5df4690/ : SUCCESS in 31m 41s\n- cross-neutron-py27 https://logs.opendev.org/60/674560/1/gate/cross-neutron-py27/699665a/ : SUCCESS in 24m 21s\n- cross-neutron-py36 https://logs.opendev.org/60/674560/1/gate/cross-neutron-py36/f96ebb7/ : SUCCESS in 25m 57s\n- cross-nova-py27 https://logs.opendev.org/60/674560/1/gate/cross-nova-py27/ae5431d/ : SUCCESS in 20m 08s\n- cross-nova-py36 https://logs.opendev.org/60/674560/1/gate/cross-nova-py36/ed51033/ : SUCCESS in 20m 52s\n- cross-nova-functional https://logs.opendev.org/60/674560/1/gate/cross-nova-functional/ba9aef7/ : SUCCESS in 24m 12s\n- cross-osc-tox-docs https://logs.opendev.org/60/674560/1/gate/cross-osc-tox-docs/86ac3c6/html/ : SUCCESS in 4m 54s\n- cross-swift-py27 https://logs.opendev.org/60/674560/1/gate/cross-swift-py27/d305927/ : SUCCESS in 8m 45s\n- cross-swift-py36 https://logs.opendev.org/60/674560/1/gate/cross-swift-py36/faf7993/ : SUCCESS in 9m 54s\n- requirements-integration https://logs.opendev.org/60/674560/1/gate/requirements-integration/e4d7056/ : SUCCESS in 58m 51s\n- neutron-grenade-multinode https://logs.opendev.org/60/674560/1/gate/neutron-grenade-multinode/a23c9ca/ : SUCCESS in 1h 09m 32s\n- tempest-full https://logs.opendev.org/60/674560/1/gate/tempest-full/4dbd047/ : SUCCESS in 1h 49m 41s\n- neutron-grenade https://logs.opendev.org/60/674560/1/gate/neutron-grenade/1fc8e17/ : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb8fd94ced1b866ed6f2171a1c54d02444703c5e","revisions":{"eb8fd94ced1b866ed6f2171a1c54d02444703c5e":{"kind":"REWORK","_number":1,"created":"2019-08-05 09:38:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/674560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/674560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/674560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/674560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/674560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/674560/1"}}},"commit":{"parents":[{"commit":"6edef0462327546e751824782163eeeab5d89508","subject":"update constraint for octavia-lib to new release 1.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6edef0462327546e751824782163eeeab5d89508"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-05 09:38:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-05 09:38:57.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 3.38.1","message":"update constraint for oslo.middleware to new release 3.38.1\n\nChange-Id: I8f7e79c80db1d51a9705e808ec80288c440945ab\nmeta:version: 3.38.1\nmeta:diff-start: -\nmeta:series: train\nmeta:release-type: release\nmeta:pypi: yes\nmeta:first: no\nmeta:release:Author: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Commit: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Change-Id: If6af2137272cc18bf2d01f26c0bab65b803ae281\nmeta:release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta:release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta:release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb8fd94ced1b866ed6f2171a1c54d02444703c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eb8fd94ced1b866ed6f2171a1c54d02444703c5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
